)]}'
{"id":"openstack%2Fkeystonemiddleware~853122","triplet_id":"openstack%2Fkeystonemiddleware~master~I45917b3ca0cd6383cb421582ec91ea65793465d4","project":"openstack/keystonemiddleware","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-08-17 00:30:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I45917b3ca0cd6383cb421582ec91ea65793465d4","subject":"remove unicode prefix from code","status":"ABANDONED","created":"2022-08-16 04:35:14.000000000","updated":"2022-08-17 00:30:05.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bf3aad1a4b057c69f992645e428806f3c8a0a2a5","_number":853122,"virtual_id_number":853122,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-16 06:02:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-16 06:02:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0522789a51cdd72e68039fbd834c478e2f9ec95f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-16 04:35:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4b46cd089486e45ab94fa2e5a1ea955e42dc2ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-16 04:37:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c45810c7ca2237cf11add0a16632734ba3e6b014","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-16 04:42:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"80c4be3e2dabca91a515d182c3a7485143db379d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 06:02:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c6b85ad879c4b71b4381fe84891f64d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55f06e0a9a7e428f927d9c4dba9c8fda : FAILURE in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d30bb5b0a945ab80a1d4920fe4f12d : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84aff0fa730a4dd783e6802057019934 : FAILURE in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7a8b3b231084093bfd96e58af4bec8c : FAILURE in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/946abb81e6854b00b6ca7bb47ae4a800 : FAILURE in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/145ced7fc9e84375a30f6cc8443a0725 : SUCCESS in 5m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4b7188188a6408d840b41c9f7c8d561 : SUCCESS in 1h 17m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edf461a29b6841538ffed2fc81b8503d : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"5999bb76f2a360f18f778482dcbda512c8c9c7fd","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-16 10:09:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ccfd4ba195e02585a7c1ce44f1812e300b18e9e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 11:51:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8564af040e947c490ee78b9ba67f380\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d1f833e38014fc48145814359751555 : FAILURE in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ebda25545044a30a665f46becdadd7a : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0389a75459d14f8b8f5575d09107fc09 : FAILURE in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ea769e0bc0b40d3a2527acf2ca8802d : FAILURE in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99223f0c746f41abacc551c585158654 : FAILURE in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c035664c1fd84991a38c019c3519e5a7 : SUCCESS in 4m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/561ec71f861b40b993fd3e9584b8560f : SUCCESS in 1h 40m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e50469cc167d4a54a5be9679cba24700 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":3},{"id":"bf3aad1a4b057c69f992645e428806f3c8a0a2a5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-17 00:30:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"94c5124c453174b6884adbd9af7aa4a7d0de54f0","revisions":{"dffd905c34ead6c488a88497c77189c986401fbb":{"kind":"REWORK","_number":1,"created":"2022-08-16 04:35:14.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/22/853122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/22/853122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/1"}}},"commit":{"parents":[{"commit":"2bda844bb219df355d74b5c5b21f86244921a1c2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2bda844bb219df355d74b5c5b21f86244921a1c2"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-16 04:34:39.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-16 04:34:45.000000000","tz":480},"subject":"remove unicode prefix from code","message":"remove unicode prefix from code\n\nChange-Id: I45917b3ca0cd6383cb421582ec91ea65793465d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/dffd905c34ead6c488a88497c77189c986401fbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/dffd905c34ead6c488a88497c77189c986401fbb"}]},"branch":"refs/heads/master"},"2a93bc3ff720df5e63ad01e26fa40c6a9e8f6f6f":{"kind":"REWORK","_number":2,"created":"2022-08-16 04:37:12.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/22/853122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/22/853122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/2"}}},"commit":{"parents":[{"commit":"2bda844bb219df355d74b5c5b21f86244921a1c2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2bda844bb219df355d74b5c5b21f86244921a1c2"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-16 04:34:39.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-16 04:36:45.000000000","tz":480},"subject":"remove unicode prefix from code","message":"remove unicode prefix from code\n\nChange-Id: I45917b3ca0cd6383cb421582ec91ea65793465d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2a93bc3ff720df5e63ad01e26fa40c6a9e8f6f6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2a93bc3ff720df5e63ad01e26fa40c6a9e8f6f6f"}]},"branch":"refs/heads/master"},"94c5124c453174b6884adbd9af7aa4a7d0de54f0":{"kind":"REWORK","_number":3,"created":"2022-08-16 04:42:22.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/22/853122/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/22/853122/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/22/853122/3"}}},"commit":{"parents":[{"commit":"2bda844bb219df355d74b5c5b21f86244921a1c2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2bda844bb219df355d74b5c5b21f86244921a1c2"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-16 04:34:39.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-16 04:41:53.000000000","tz":480},"subject":"remove unicode prefix from code","message":"remove unicode prefix from code\n\nChange-Id: I45917b3ca0cd6383cb421582ec91ea65793465d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/94c5124c453174b6884adbd9af7aa4a7d0de54f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/94c5124c453174b6884adbd9af7aa4a7d0de54f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
