)]}'
{"id":"openstack%2Fkeystonemiddleware~865325","triplet_id":"openstack%2Fkeystonemiddleware~master~I0107815b9423ea3661cef0a20a8689c6eba454ce","project":"openstack/keystonemiddleware","branch":"master","attention_set":{"30126":{"account":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"last_update":"2022-11-28 08:11:47.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0107815b9423ea3661cef0a20a8689c6eba454ce","subject":"Switch memcache_pool_socket_timeout to a float","status":"NEW","created":"2022-11-22 19:08:32.000000000","updated":"2022-11-28 08:11:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5b48748283b3c2b89e527d991952e2d40caa54ed","_number":865325,"virtual_id_number":865325,"owner":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-22 20:58:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2022-11-23 09:41:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-22 20:58:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-23 09:41:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"02d786e10f5cff4425a5434c59db38e326aca828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2022-11-22 19:08:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"181dac2b61846dfcf7b8185d159b9ee3d30a81bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 20:58:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a2d6ff241b54d76aa2c2f1f9cdeaaf2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09c9cf993bf0458795fa64db3d1da3ec : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a5010480df245e2951c768bcffcb227 : FAILURE in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c4cffaf4e314de986252da19163fa80 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6dc393a4d474d4dbe61d1a53c05c4c1 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/207c2f96d1c34d8cb5fc3b76921b8f30 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4235413e1b4740dd932a54de61c4a932 : SUCCESS in 4m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cade0e5fc32647d8af1997634c44b154 : SUCCESS in 1h 44m 22s","accounts_in_message":[],"_revision_number":1},{"id":"aec0fce947dd3e08d971f1d2874029b7de4b3660","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-23 09:41:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"790db1583d942db6d40e9d782743928243bab4f6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-23 09:43:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"365a6fb91b55c65b323df5569c338b096cfe5f9e","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2022-11-28 06:35:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b48748283b3c2b89e527d991952e2d40caa54ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 08:11:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/377c0069d4e24989b4370a5c53f9fc10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba92033cd79a4646b4f85d942f00b073 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5adbd48728a44d0ca9ea06c846a3fb65 : FAILURE in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ede4e8f589404855a1b27d004986722f : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6aadf6d631474d80a0d96591a0061b91 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/105a387595544d9096fa7606bd11fcd8 : SUCCESS in 2m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94289d4fcb054595af02643b3e82b0da : SUCCESS in 4m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/199ca81c152c4a5382540336e031deb1 : SUCCESS in 1h 31m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bca0311f94ae2f04440006efd20b84af7d440939","revisions":{"bca0311f94ae2f04440006efd20b84af7d440939":{"kind":"REWORK","_number":1,"created":"2022-11-22 19:08:32.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/25/865325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/25/865325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/25/865325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/25/865325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/25/865325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/25/865325/1"}}},"commit":{"parents":[{"commit":"aeeb64d349524433dda05cf12994ea4f8676c098","subject":"Merge \"Bump tox minversion to 3.18.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/aeeb64d349524433dda05cf12994ea4f8676c098"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2022-11-22 19:02:57.000000000","tz":60},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2022-11-22 19:08:27.000000000","tz":60},"subject":"Switch memcache_pool_socket_timeout to a float","message":"Switch memcache_pool_socket_timeout to a float\n\nIt\u0027s not unusual to set 0.25 as a socket_timeout value, and this was\nnot possible. This can have a dramatic impact for performance when one\nmemcached server is down, especially when keystone is involved.\n\nSee https://review.opendev.org/c/openstack/oslo.cache/+/519353/\nfor a similar change in oslo (and related lp#1731921).\n\nChange-Id: I0107815b9423ea3661cef0a20a8689c6eba454ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/bca0311f94ae2f04440006efd20b84af7d440939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/bca0311f94ae2f04440006efd20b84af7d440939"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
