)]}'
{"id":"openstack%2Fkeystonemiddleware~868284","triplet_id":"openstack%2Fkeystonemiddleware~stable%2Fzed~Ice8e34440a5fe1baa370646ed70b5e085c4af70e","project":"openstack/keystonemiddleware","branch":"stable/zed","topic":"lp1987355-stable/zed","attention_set":{},"removed_from_attention_set":{"34275":{"account":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"last_update":"2023-01-27 16:11:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice8e34440a5fe1baa370646ed70b5e085c4af70e","subject":"Remove cache invalidation when using expired token","status":"MERGED","created":"2022-12-21 16:56:26.000000000","updated":"2023-01-27 16:12:45.000000000","submitted":"2023-01-27 16:11:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868284-lp1987355-stable/zed","cherry_pick_of_change":860481,"cherry_pick_of_patch_set":3,"meta_rev_id":"970dfae89d2da02f9a85de55b551ee76aa5689d8","_number":868284,"virtual_id_number":868284,"owner":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-27 16:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-01-27 14:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2023-01-27 14:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-21 18:38:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 14:56:21.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"98ba2d38f7f351aca210432d269b8bc87828b327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"date":"2022-12-21 16:56:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"01c96c98b91edacb7c50cb612572a44413423500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-21 18:38:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e17780f877544ee4b7935d72a892c399\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d627ae8e78454a53b58e966afc007fb1 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78217daa17c44f0c832e41c66972bdfc : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ee837ec702d4feba553a302a9d145e2 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05558828ea184cefba7a766607928c77 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b9379a2a0784be99224de6b32e4fafa : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e978f321144692949782ea30eb3e2b : SUCCESS in 4m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e8c6a1472af4bd9afb9f238fa6efbf0 : SUCCESS in 1h 36m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6e132ed35c63e6962fda71c552b882973ca4768f","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-01-27 14:56:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8de7696a86bc55a06447b3b089e45d9a2209854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 14:56:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ae8d8f6521a169a3860d31fd83aaa4af04d0019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 16:11:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c8ed7b1083f43b89ca63b0389f0256b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75199e076f7745a9afb6fc8e9a9b986d : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8756463affb643a3bd0203dbd9730558 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d7af28739404b84bfb49ade1d89f118 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3051aea49644634a91ac99a5cd6023f : SUCCESS in 5m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fdcd7cf28eb472d9662930a5165d781 : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7157096760371af83fe4c39c5ba2bbe86d9e29af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 16:11:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"970dfae89d2da02f9a85de55b551ee76aa5689d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 16:12:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa7b56634ece44fa86e7f4cd8673749e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6aff5ead652433f83ee665923737320 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f602fa718d12e4ab15ab134256fa835db2a1b7ee","revisions":{"f602fa718d12e4ab15ab134256fa835db2a1b7ee":{"kind":"REWORK","_number":1,"created":"2022-12-21 16:56:26.000000000","uploader":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"ref":"refs/changes/84/868284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/84/868284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/84/868284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/84/868284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/84/868284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/84/868284/1"}}},"commit":{"parents":[{"commit":"46778cd00035fb2d0ac335be80e592af26a00f5e","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/46778cd00035fb2d0ac335be80e592af26a00f5e"}]}],"author":{"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","date":"2022-10-05 17:41:06.000000000","tz":-180},"committer":{"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","date":"2022-12-21 16:56:26.000000000","tz":0},"subject":"Remove cache invalidation when using expired token","message":"Remove cache invalidation when using expired token\n\nThis can create a race condition for long running services that reuse\ntheir token (eg. Kubernetes Cinder CSI plugin) in this case for\nexample:\n\n1 [user] Asks nova to attach a volume to a server\n2 ...the user\u0027s token expires\n3 [user] Asks cinder if the volume has been attached\n4 [nova] Asks cinder to attach the volume\n\nIn step 3 the token is marked as invalid in the cache and step 4 fails\neven if allow_expired is true\n\nCloses-Bug: #1987355\nChange-Id: Ice8e34440a5fe1baa370646ed70b5e085c4af70e\n(cherry picked from commit e05466c5f439cd05482f109e6eb97a50ba156698)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f602fa718d12e4ab15ab134256fa835db2a1b7ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f602fa718d12e4ab15ab134256fa835db2a1b7ee"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[]}
