)]}'
{"id":"openstack%2Fkeystonemiddleware~868734","triplet_id":"openstack%2Fkeystonemiddleware~master~I529c5b0c89933395b126e86651ef09368dd7e6b4","project":"openstack/keystonemiddleware","branch":"master","topic":"bp/enhance-oauth2-interoperability","attention_set":{},"removed_from_attention_set":{"33920":{"account":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"last_update":"2023-09-13 18:32:57.000000000","reason":"Change was submitted"},"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2023-09-13 18:32:57.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-09-13 18:32:57.000000000","reason":"Change was submitted"},"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2023-09-13 18:32:57.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-09-08 14:04:45.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-09-13 18:32:57.000000000","reason":"Change was submitted"},"34634":{"account":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"last_update":"2023-09-13 18:32:57.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-09-08 14:05:08.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"33455":{"account":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"last_update":"2023-09-13 18:32:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I529c5b0c89933395b126e86651ef09368dd7e6b4","subject":"External OAuth2.0 Authorization Server Support","status":"MERGED","created":"2022-12-28 03:40:13.000000000","updated":"2023-09-13 18:34:02.000000000","submitted":"2023-09-13 18:32:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"868734-bp/enhance-oauth2-interoperability","meta_rev_id":"ce64586ec43b1ef4597b42ceb1d0d8f26867399f","_number":868734,"virtual_id_number":868734,"owner":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-13 18:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2023-09-08 14:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2023-09-08 14:05:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-09-08 14:05:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"CC":[{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-28 06:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-03 14:56:39.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"CC"},{"updated":"2023-01-03 14:56:39.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"state":"CC"},{"updated":"2023-01-03 14:56:39.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"state":"CC"},{"updated":"2023-01-03 14:56:39.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"state":"REVIEWER"},{"updated":"2023-01-03 14:56:39.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2023-01-03 14:56:39.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2023-01-03 14:56:39.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2023-01-03 14:56:39.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-01-03 14:56:39.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2023-01-24 04:16:46.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":33292,"name":"Masaki UENO","email":"masaki.ueno.up@hco.ntt.co.jp","username":"masaki-ueno"},"state":"CC"},{"updated":"2023-07-18 05:23:53.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"545e45347adce4259f7429d2857de3e9f3829a6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2022-12-28 03:40:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87c185950c1f82875a06190935d8faba8bd68b0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2022-12-28 03:42:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a39b9a091fb0bcbaebc340d16d765c6a1bc9113c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2022-12-28 04:26:28.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dbace20da9c06e2204ec2218ecf5b2503121c9e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2022-12-28 04:34:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"706cfd7f743982e3ea50d32677c2d4130e7b64e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 06:02:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c0988cf3da14f1ea3f1b53918c2375c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c8f1e6cf7c3147d0897152aee0ac9a77 : SUCCESS in 1m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/458663927cb14715a214dc37708d14c3 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/338247abfdd14878b918a199476dcc1a : FAILURE in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d794e33cdb644f93ba130cf7285f8656 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/511764a723cd4f3c83349daaa2d96cac : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f362fe13c76a48a8a59543c5c853bddb : SUCCESS in 2m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3aac588a6c46cdaf79605db79aa57e : FAILURE in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aeb8484d76bd48b8b2bb39355123aa16 : SUCCESS in 1m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bcbc004f6c44842910ee8e96543e377 : SUCCESS in 1h 25m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8352115397dd44f582f3d0d387e4dc6e : SUCCESS in 3m 42s\n\nWarning:\n  Comments left for invalid file keystonemiddleware/oauth2_mtls_token.py","accounts_in_message":[],"_revision_number":4},{"id":"41799280a9a51b22072601a13ef744086e455904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2022-12-28 06:42:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c94ef470d5fc9fcfc46aa9c9c3b9319e80c10a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-28 08:06:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46ae4a3da7a04067a99e2ba1406f9612\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a7964cdfd06642d68cae5fa5092c579e : SUCCESS in 2m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8e22012c325482fb4c9328f41f2ccdc : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be8da36f36a442399f0f63150adf0f30 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/243099c84dbb4b0eb82e8980848ce60b : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92b9e239d84b43cea8f624acdd2a2901 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b803aeab778c4ee981563547c03f0ce7 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ff4137cc674c1ea4bd4351c7637421 : FAILURE in 3m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/967479c0b3a34de891909463482e3200 : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6d83b53bd2a4eeb9fa534b31fe22173 : SUCCESS in 1h 22m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/119d6ac250b0463da5d5bdf82dc0f329 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":5},{"id":"6d9ff6c7e2d1f3aed9de544049d31889d35c7fc0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-01-03 12:29:20.000000000","message":"Topic bp/external-authentication-server-oauth2-grant-support removed","accounts_in_message":[],"_revision_number":5},{"id":"aaaa48655e027f65032948e793a39ef15ecd0955","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-01-03 12:59:13.000000000","message":"Topic set to bp/support-oauth2-mtls","accounts_in_message":[],"_revision_number":5},{"id":"b1662aaff7e7792489be3133203bfb4f12830bc9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-01-03 12:59:37.000000000","message":"Topic bp/support-oauth2-mtls removed","accounts_in_message":[],"_revision_number":5},{"id":"a530fe8766cf81cf7896ad599d9f9d9cbea96205","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-01-03 12:59:41.000000000","message":"Topic set to bp/enhance-oauth2-interoperability","accounts_in_message":[],"_revision_number":5},{"id":"f60628b17cf739da2ac5b6f6548b8936e490d8d5","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-01-03 14:56:39.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e645c194b86ecec4234cdd1fc67618a75d5ea5a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-01-20 00:48:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9497129e8161020fc7668ff093c377595063defa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-20 02:04:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1acacce159b142898ad8a96c37007c5f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7368a4723d314db0b04152dd46a11006 : SUCCESS in 1m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/567f31dfcb22472989ce953d9530d29a : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f80bd333362040b8b74084970742cabb : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faad10a10ea5417db260fdff47f807f6 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab6b568d2df6488fbbc5ea084d576c07 : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/039019c324e746919d3aa29cbe32678a : SUCCESS in 2m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd361dab64454f6d9ff91a23eaf04d7c : FAILURE in 6m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b9d2848ab7c64d758489d57bebc108d7 : SUCCESS in 4m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d524c632c8254d5bbca6c09d20f833de : SUCCESS in 1h 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cd9581ec0164ebebb42811e1f875b19 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":6},{"id":"673aaeb6cbc36c02ecf2a22c795a2d57d8655781","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-01-24 04:16:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"28cc038916ec3120347b0724f4cd747e38656db1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-02-07 05:52:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6fc2073ee7e64fd47f5930975dd8e07b9309fea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 07:22:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3541943f69e247158d377e3727ddb8a7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc8488d830d64c718bb4a695a9c63662 : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84ebcbca84054d26872c375c85810afe : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e39dd370a834c188ea705db28053d06 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc2ba5bc2d0c47a19bd403da45148f50 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ac1b0cae8f148508e3f90343a769f52 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d20ea5885924340adf954ef693ed322 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e45e72af3e4e2bb754a4b3f2b3c232 : FAILURE in 3m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e702546cb5e54bbba54c89bf25505af7 : SUCCESS in 1m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3e8d62d23b441a596bb6af4f3baefb0 : SUCCESS in 1h 27m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a26fe113a294678a5f1c1e1b4a3b51a : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":7},{"id":"597682c95141584c40e1291a29e597457a839d6c","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-02-09 01:42:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c2e6378dbc0b33ead7aa24dda6b1bbbd2df62c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-02-22 04:51:41.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"33f03d82726094548dd7e2611f3d2aa5a046a66e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-03-03 00:43:47.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"127221010bb15d2ba09f2443f3aa02a700da1068","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-03-03 06:49:47.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"de6240dd58e92accc4ea3c7fb1db4039617115d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-03-03 07:53:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d819856483c30c56871bd1586ad63a8803f2b1d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 09:20:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f270da9aaeb0486e9cd23b031ffa21cb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/40f844ce01c746e388193a41345ad6b6 : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53f2c2c5fefb45b5b9de4d9528b67c1f : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ecc75749024345aaa503a92d7adc9f : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be8d17e2231a4117a9e5074d83dedbf7 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c75758e6d164cdaacc83730121ac0c1 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65513d46b13847198c55c777f25a6901 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c61aa9bd2774863b71d0597d3f65d7d : SUCCESS in 4m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ac5327bd5a9347c3bf595ce1fd0eb078 : SUCCESS in 2m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a26d35c2315449c9552e340ba692515 : SUCCESS in 1h 21m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c121c167cdf443ad96851c499e57693a : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":11},{"id":"c44d05992ca5431ebfa51c8a2df9e033dd4a2ef3","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-03-16 06:08:29.000000000","message":"Patch Set 11: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c4f4a69191df6f0b99fce98263230011f2d5a8be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-03-16 08:39:42.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f7cd9254693d6a2a35ced6ea5257c13377f5ccfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 10:10:33.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70dcd3d44cac475ea6e5943b43505ea3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aecea8daa01940fdaddfa2ce9eb1aca7 : SUCCESS in 1m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/161bbc3e4e954041bdb78653c72ad942 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20637cdf91124cf89836fdd6674e92b3 : FAILURE in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b647b9dc2f19467ca9867ff9d29a4b94 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/033e3cfeaef34e5ea9210a4d05d8de93 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5023465a0ff6493d83326c226746d39c : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d83170d44dfe4cb49f5559179b83b931 : SUCCESS in 5m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a5633dbdaf574f2681a3c6c1903505cf : SUCCESS in 1m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26b155744a4e49eb814cf9178516254e : SUCCESS in 1h 25m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d93460f1c28647f3bece4c7bc93c25ea : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":12},{"id":"9c878045a78ed066374286d9c0eb0e8f27318328","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-03-17 03:10:02.000000000","message":"Patch Set 12:\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6cf221b5a47c66fd897c79400e04a8b1d1abeadf","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-03-17 03:48:16.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"5c98a1f4d2d1b44ade77e5c0b8b687d79b908188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-03-20 00:59:41.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"3692553b1d33029e87ac0ec22c0f5be08e13a7d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-03-20 01:30:17.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"3c3d3a0c0027b0ddb4cf08b100e110bd0fb31508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 02:41:29.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/794124d19b0d4f7aa85f13aedc3e8784\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ef6ceb5229f6423f8d2caba227844402 : SUCCESS in 1m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2390af124ea14badbeaf7ccf6de885a8 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828102d54c70485fb76750e3db0c82b2 : FAILURE in 2m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/334fa0dbb9f64f68a5b14aea0685ed1f : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96fc65cd41f146499ec7d5c8e26ed8fa : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c534b5a6264d442391a8105ee4129a25 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b65a243d4845178a876ea1e1b042e7 : SUCCESS in 3m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ada4585db2db4730ac87ae1b7524f1b7 : SUCCESS in 2m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36b0a9f109044347bc916a1b78a0b966 : SUCCESS in 1h 09m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f776f42f93e3474281f5474027ff3579 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":14},{"id":"3b65c9264323872308e12284fa78e94cbf4c1af1","author":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"date":"2023-03-20 08:12:22.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d751044772ac70aea0ed669f9ec0f5b5b1dbde8f","author":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"date":"2023-04-06 00:03:41.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f76bfdb82fe31df29eb04c301c40aa927e1637ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 01:29:52.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1fccf0902ab47cb86f684c885a7c950\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a36e31e0a8294b41adc19e3126f028b4 : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a82e7e277084142b1f93e4243848474 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e26678b1bfcc44d386eb7392abceb6f2 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/806b4cac56f44e139dd7c86e2ee2bdfc : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d7371d91ed3482cab1e8c9f4aa812ee : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9380191f57b4774912de42ace8086fe : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c96aedc8735466d8738cb79d9593c24 : SUCCESS in 6m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5388c6755e414981956004e837f31374 : SUCCESS in 2m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/696a6345992a4ae4bcfcc72ce4adada2 : SUCCESS in 1h 23m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/590bbaf1a2c649f6a55c8579efd8842d : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":14},{"id":"d882c48ed553132faca0497f8ca5331777b6912d","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-06-14 18:42:39.000000000","message":"Patch Set 14: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":14},{"id":"2949d4fc8c0d0697b2cf648abe63c282f936755b","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2023-06-20 15:49:55.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"47c6619fa9671cf01d72b6b31f93cb1fc3f94569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"date":"2023-08-28 07:50:22.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"31bbce091a1267139b78b001f9ab6027a1a89691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 09:27:54.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0151d0049f624dd3b3ae6bf366fa3b2c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f6c81d514394b6f982ff1cac25a4835 : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e479ef466bdf4321bb80f59ac439f5b9 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d4914d517d45c9b8f4a11ddd0684f0 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/046b24c2d718400f829eb4837c49aa0e : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf656191eab641edb93db5205c8baa43 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ceea975d91d04c4b8521f69f4bd1761e : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8abf0b9924e44525af041817f94c4767 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76cfb4f182e4028b0ff75d22e51467b : SUCCESS in 5m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/365b18054e574a318a3445310af24b7f : SUCCESS in 1m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c9e0c31f4284ffdbda083872cfdfefc : SUCCESS in 1h 32m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/800c8d8aeb1c4113a5efead63bb5aee1 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":15},{"id":"b6dc3a95c33caef4f0dc7362c1313dd5d90e4854","author":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"date":"2023-08-28 09:40:07.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"c31e7790bd83d8ec0ede2228f8fa52efe2d93869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"date":"2023-08-30 13:25:28.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"7fce82b8c99091adee384b65bc88787c85a266c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 15:04:40.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c92167026585402ea89df17d3167f313\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/39f48e9db4754231ae1b119431e4b7d2 : SUCCESS in 2m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e02517122e54fd8891240089124df08 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e24be02e2ac64b9cb78b79c167fe907e : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a4983fa082d45daa026b4af2d0234a7 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9f30309f06e4448a515678b436206ee : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfc5cc1825654b98b67d1aa32b8b62dd : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c827f3f0e29445abb0c91ec13d5b572 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5733ac78612e44c88a1b1f7e15dd31c1 : SUCCESS in 5m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b2205980b93d4c91ae7c581d1666e5a2 : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea8c2bfb26c641808ba0dc626465f192 : SUCCESS in 1h 33m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ac9f934e34541029004032a86f7c791 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":16},{"id":"9a301d3c4c92dfb6368a74eadb6779f51b1c01ca","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-09-08 14:04:45.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"c976b8ac5a2884769e4467e88fbc060050a330cd","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-09-08 14:05:08.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"a62331948738a4cbfd044a4bd03017faab8c217c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 14:05:39.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"ecf1e06cc11afb548e6c5476466a1066a078c395","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 15:28:46.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32a1aa73a97141b49e170181902abf44\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6014e11e4b344a6a564b609d847e212 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649111d5023f4a80b95b8272d1b284f3 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b3fd2a2eb65465f83bba76736f176af : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec76f96a283749cd912d780096a5217b : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cec14b257b534bfb8f61cb8d2cbc59d1 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89c2f7756a84618ae539cc9a526185e : SUCCESS in 5m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4be633673cbc4e4b81caf3e78dd46825 : SUCCESS in 2m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3178e9ef27f243e28c5d01820e981fd7 : FAILURE in 1h 16m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa3ec7b74c1a43019f4f1f7a4bfea8fa : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":16},{"id":"6e8b4f893462f0e8b24c57f15769f7739ca97e68","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-09-13 15:08:57.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"1e75743a6cb40a32155b2b114d0aff6bdd5e30dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 16:42:30.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce10e13183f047e0ae76d0531efce091\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5e1378791b247eea6f1fcf56f5e19de : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8278d84706fa483587cac555abe1a0ee : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7947cb605b42fa842e56d0408179c1 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b78dd2ea282b46dfbe42f2377683d684 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6760f92e669a48cda6db1cde191a5908 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/330d4123466849a0ab001d9e225a03ab : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5741441f7baf4f09b21ef6249ddd4e14 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/945fb76579004db5817a3f04d707d2d1 : SUCCESS in 5m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ae793d5a3244b848ad3a69a68917863 : SUCCESS in 1m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41ed1169e94742c4a708291e81c20d89 : SUCCESS in 1h 31m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10e7b0dec7a344e48605b98265422740 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":16},{"id":"d6a3175729d22aa922c7a772d15a3002bd69589c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 16:43:02.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"9b455df6240fe59a2ee7b6c5637e2e08b3fabba2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 18:32:57.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e02697d15634c448dd6d3ccf2481efd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a68ac621ae4f4fb09dc6e181b7405abe : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c6abb1396274baf81182ff16165ba2e : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5607586fc49415eb0b244173e002015 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c11a4f068dd24977bfb49e5e5999745c : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b091e34162842f1bc562d3ba99137f2 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098adfeddcd042abb2b4323076188a8f : SUCCESS in 5m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7be67ccc2e664119b1e5eec1ac58abd4 : SUCCESS in 1m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9ee1b8fe2a964dabbd4a55d475788be2 : SUCCESS in 1h 42m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2eede43915845a6b77abd887c21e3d4 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":16},{"id":"15e7f038c75816b50ca36eb9ca7340c177b0ab09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 18:32:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"ce64586ec43b1ef4597b42ceb1d0d8f26867399f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 18:34:02.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27f4316ea42246488a40ec934b0fcab1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba82b024c5144a119acc5c9d11409626 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfddee6176594adda16292fe641b5be9 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"de15a610e160defb367b224258498727384d10a8","revisions":{"6c0a4b57b351f7f62bab70c44efb3d6f6eec122d":{"kind":"REWORK","_number":1,"created":"2022-12-28 03:40:13.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/1"}}},"commit":{"parents":[{"commit":"5b23a88d2832f8c39991e834ee30c415ea885803","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5b23a88d2832f8c39991e834ee30c415ea885803"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:39:57.000000000","tz":540},"subject":"xExternal Authentication Server OAuth2.0 Grant Support","message":"xExternal Authentication Server OAuth2.0 Grant Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained from\nan External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint external-authentication-server-oauth2-grant-support\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6c0a4b57b351f7f62bab70c44efb3d6f6eec122d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6c0a4b57b351f7f62bab70c44efb3d6f6eec122d"}]},"branch":"refs/heads/master"},"f14431d308f11781e2aa57924449c258fa8b4ae2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-28 03:42:03.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/2"}}},"commit":{"parents":[{"commit":"5b23a88d2832f8c39991e834ee30c415ea885803","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5b23a88d2832f8c39991e834ee30c415ea885803"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:41:51.000000000","tz":540},"subject":"xExternal Authentication Server OAuth2.0 Grant Support","message":"xExternal Authentication Server OAuth2.0 Grant Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint external-authentication-server-oauth2-grant-support\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f14431d308f11781e2aa57924449c258fa8b4ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f14431d308f11781e2aa57924449c258fa8b4ae2"}]},"branch":"refs/heads/master"},"f79b2aa53d48b533e4a2309dd920887477d84c37":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-12-28 04:26:28.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/3"}}},"commit":{"parents":[{"commit":"5b23a88d2832f8c39991e834ee30c415ea885803","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5b23a88d2832f8c39991e834ee30c415ea885803"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 04:25:18.000000000","tz":540},"subject":"External Authentication Server OAuth2.0 Grant Support","message":"External Authentication Server OAuth2.0 Grant Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint external-authentication-server-oauth2-grant-support\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f79b2aa53d48b533e4a2309dd920887477d84c37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f79b2aa53d48b533e4a2309dd920887477d84c37"}]},"branch":"refs/heads/master"},"e1aef0021ac0b1da49c57dd59f1c77336920fc1b":{"kind":"REWORK","_number":4,"created":"2022-12-28 04:34:31.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/4"}}},"commit":{"parents":[{"commit":"5b23a88d2832f8c39991e834ee30c415ea885803","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5b23a88d2832f8c39991e834ee30c415ea885803"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 04:34:19.000000000","tz":540},"subject":"External Authentication Server OAuth2.0 Grant Support","message":"External Authentication Server OAuth2.0 Grant Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint external-authentication-server-oauth2-grant-support\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e1aef0021ac0b1da49c57dd59f1c77336920fc1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e1aef0021ac0b1da49c57dd59f1c77336920fc1b"}]},"branch":"refs/heads/master"},"a30a8287acbba73718ac08657d77f2d7622c27bb":{"kind":"REWORK","_number":5,"created":"2022-12-28 06:42:00.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/5"}}},"commit":{"parents":[{"commit":"5b23a88d2832f8c39991e834ee30c415ea885803","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5b23a88d2832f8c39991e834ee30c415ea885803"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 06:41:51.000000000","tz":540},"subject":"External Authentication Server OAuth2.0 Grant Support","message":"External Authentication Server OAuth2.0 Grant Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint external-authentication-server-oauth2-grant-support\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a30a8287acbba73718ac08657d77f2d7622c27bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a30a8287acbba73718ac08657d77f2d7622c27bb"}]},"branch":"refs/heads/master"},"e2cd2944cbd92f3fde3293a8eebc6134b095d065":{"kind":"REWORK","_number":6,"created":"2023-01-20 00:48:31.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/6"}}},"commit":{"parents":[{"commit":"5b23a88d2832f8c39991e834ee30c415ea885803","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5b23a88d2832f8c39991e834ee30c415ea885803"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-01-20 00:47:30.000000000","tz":540},"subject":"External Authentication Server OAuth2.0 Grant Support","message":"External Authentication Server OAuth2.0 Grant Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e2cd2944cbd92f3fde3293a8eebc6134b095d065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e2cd2944cbd92f3fde3293a8eebc6134b095d065"}]},"branch":"refs/heads/master"},"c83c26491578b8dafb51ee57bdba60e35ca83dd8":{"kind":"REWORK","_number":7,"created":"2023-02-07 05:52:16.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/7"}}},"commit":{"parents":[{"commit":"5b23a88d2832f8c39991e834ee30c415ea885803","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5b23a88d2832f8c39991e834ee30c415ea885803"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-02-07 05:48:58.000000000","tz":540},"subject":"External Authentication Server OAuth2.0 Grant Support","message":"External Authentication Server OAuth2.0 Grant Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c83c26491578b8dafb51ee57bdba60e35ca83dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c83c26491578b8dafb51ee57bdba60e35ca83dd8"}]},"branch":"refs/heads/master"},"207088fd743c2d51acea783bb0fc1d3f28a25f98":{"kind":"REWORK","_number":8,"created":"2023-02-22 04:51:41.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/8"}}},"commit":{"parents":[{"commit":"5b23a88d2832f8c39991e834ee30c415ea885803","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5b23a88d2832f8c39991e834ee30c415ea885803"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-02-22 04:51:28.000000000","tz":540},"subject":"External Authentication Server OAuth2.0 Grant Support","message":"External Authentication Server OAuth2.0 Grant Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/207088fd743c2d51acea783bb0fc1d3f28a25f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/207088fd743c2d51acea783bb0fc1d3f28a25f98"}]},"branch":"refs/heads/master"},"2e06e3910a4920b80a9c78470da6ae6079c3b3c5":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-03-03 00:43:47.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/9"}}},"commit":{"parents":[{"commit":"5b23a88d2832f8c39991e834ee30c415ea885803","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5b23a88d2832f8c39991e834ee30c415ea885803"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-03-03 00:43:54.000000000","tz":540},"subject":"External OAuth2.0 Authorization Server Support","message":"External OAuth2.0 Authorization Server Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2e06e3910a4920b80a9c78470da6ae6079c3b3c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2e06e3910a4920b80a9c78470da6ae6079c3b3c5"}]},"branch":"refs/heads/master"},"8e3435ee670b0ac97f5e248d37eaa96b737f73e1":{"kind":"REWORK","_number":10,"created":"2023-03-03 06:49:47.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/10"}}},"commit":{"parents":[{"commit":"a59020fdab670314ac1ab3d0b77e89b352d7cf27","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a59020fdab670314ac1ab3d0b77e89b352d7cf27"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-03-03 06:49:53.000000000","tz":540},"subject":"External OAuth2.0 Authorization Server Support","message":"External OAuth2.0 Authorization Server Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/8e3435ee670b0ac97f5e248d37eaa96b737f73e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/8e3435ee670b0ac97f5e248d37eaa96b737f73e1"}]},"branch":"refs/heads/master"},"7c4aa6aa6edd5200ee69cf71e7e1bded055136c7":{"kind":"REWORK","_number":11,"created":"2023-03-03 07:53:15.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/11"}}},"commit":{"parents":[{"commit":"a59020fdab670314ac1ab3d0b77e89b352d7cf27","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a59020fdab670314ac1ab3d0b77e89b352d7cf27"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-03-03 07:53:32.000000000","tz":540},"subject":"External OAuth2.0 Authorization Server Support","message":"External OAuth2.0 Authorization Server Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/7c4aa6aa6edd5200ee69cf71e7e1bded055136c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/7c4aa6aa6edd5200ee69cf71e7e1bded055136c7"}]},"branch":"refs/heads/master"},"e30998137d7b8b01e1d97158af84a53208382617":{"kind":"REWORK","_number":12,"created":"2023-03-16 08:39:42.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/12"}}},"commit":{"parents":[{"commit":"a59020fdab670314ac1ab3d0b77e89b352d7cf27","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a59020fdab670314ac1ab3d0b77e89b352d7cf27"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-03-16 08:40:33.000000000","tz":540},"subject":"External OAuth2.0 Authorization Server Support","message":"External OAuth2.0 Authorization Server Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e30998137d7b8b01e1d97158af84a53208382617"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e30998137d7b8b01e1d97158af84a53208382617"}]},"branch":"refs/heads/master"},"d050f9fd9b50ed871593052110d821247fd3a9fb":{"kind":"REWORK","_number":13,"created":"2023-03-20 00:59:41.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/13"}}},"commit":{"parents":[{"commit":"a59020fdab670314ac1ab3d0b77e89b352d7cf27","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a59020fdab670314ac1ab3d0b77e89b352d7cf27"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-03-20 01:00:24.000000000","tz":540},"subject":"External OAuth2.0 Authorization Server Support","message":"External OAuth2.0 Authorization Server Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d050f9fd9b50ed871593052110d821247fd3a9fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d050f9fd9b50ed871593052110d821247fd3a9fb"}]},"branch":"refs/heads/master"},"3c9589b21ed61189d37e8d9987aaf5c8a83e004a":{"kind":"REWORK","_number":14,"created":"2023-03-20 01:30:17.000000000","uploader":{"_account_id":34634,"name":"Yonggen Sun","email":"sunyonggen@fujitsu.com","username":"sunyonggen"},"ref":"refs/changes/34/868734/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/14"}}},"commit":{"parents":[{"commit":"a59020fdab670314ac1ab3d0b77e89b352d7cf27","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a59020fdab670314ac1ab3d0b77e89b352d7cf27"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2023-03-20 01:30:44.000000000","tz":540},"subject":"External OAuth2.0 Authorization Server Support","message":"External OAuth2.0 Authorization Server Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3c9589b21ed61189d37e8d9987aaf5c8a83e004a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3c9589b21ed61189d37e8d9987aaf5c8a83e004a"}]},"branch":"refs/heads/master"},"b69d83146361829bef718312017bf77a3e125bf0":{"kind":"REWORK","_number":15,"created":"2023-08-28 07:50:22.000000000","uploader":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"ref":"refs/changes/34/868734/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/15"}}},"commit":{"parents":[{"commit":"a59020fdab670314ac1ab3d0b77e89b352d7cf27","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a59020fdab670314ac1ab3d0b77e89b352d7cf27"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","date":"2023-08-28 07:55:54.000000000","tz":0},"subject":"External OAuth2.0 Authorization Server Support","message":"External OAuth2.0 Authorization Server Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b69d83146361829bef718312017bf77a3e125bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b69d83146361829bef718312017bf77a3e125bf0"}]},"branch":"refs/heads/master"},"de15a610e160defb367b224258498727384d10a8":{"kind":"REWORK","_number":16,"created":"2023-08-30 13:25:28.000000000","uploader":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"ref":"refs/changes/34/868734/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/34/868734/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/34/868734/16"}}},"commit":{"parents":[{"commit":"a59020fdab670314ac1ab3d0b77e89b352d7cf27","subject":"OAuth 2.0 Mutual-TLS Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a59020fdab670314ac1ab3d0b77e89b352d7cf27"}]}],"author":{"name":"sunyonggen","email":"sunyonggen@fujitsu.com","date":"2022-12-28 03:35:59.000000000","tz":540},"committer":{"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","date":"2023-08-30 13:30:32.000000000","tz":0},"subject":"External OAuth2.0 Authorization Server Support","message":"External OAuth2.0 Authorization Server Support\n\nThe external_oauth2_token filter has been added for accepting or denying\nincoming requests containing OAuth 2.0 access tokens that are obtained\nfrom an External Authentication Server.\n\nDepends-On: https://review.opendev.org/c/openstack/keystoneauth/+/860614\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: I529c5b0c89933395b126e86651ef09368dd7e6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/de15a610e160defb367b224258498727384d10a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/de15a610e160defb367b224258498727384d10a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
