)]}'
{"id":"openstack%2Fkeystonemiddleware~873924","triplet_id":"openstack%2Fkeystonemiddleware~stable%2Fxena~Ice8e34440a5fe1baa370646ed70b5e085c4af70e","project":"openstack/keystonemiddleware","branch":"stable/xena","topic":"cherrypick-lp1987355-pgg14d79xd","attention_set":{},"removed_from_attention_set":{"34275":{"account":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"last_update":"2023-02-28 17:10:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice8e34440a5fe1baa370646ed70b5e085c4af70e","subject":"Remove cache invalidation when using expired token","status":"MERGED","created":"2023-02-15 16:09:34.000000000","updated":"2023-02-28 17:11:54.000000000","submitted":"2023-02-28 17:10:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"873924-cherrypick-lp1987355-pgg14d79xd","cherry_pick_of_change":860481,"cherry_pick_of_patch_set":3,"meta_rev_id":"0df2e6c44a7b207bf9ec876ec4b9c6b65fe3d1ce","_number":873924,"virtual_id_number":873924,"owner":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-28 17:10:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-02-28 15:15:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2023-02-28 15:02:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2023-02-28 15:31:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-02-28 15:31:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-15 17:52:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 15:02:15.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-02-28 15:15:33.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2023-02-28 15:31:28.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"bab10d77c318d7b4d866afae969d85c19e8b0288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"date":"2023-02-15 16:09:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"65ebd7f588bb0b8774de5aa35822569f257b1c55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 17:52:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1d3bb03a5ea42dbae0602582b1450cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b3fb8b756724fd3ba778a84fe70dfa9 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e2753bfc3814110affc710f84beb14f : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc4dc339203847a591102205ffb74b64 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94a6be33ac2348dc9300e8c289d07a54 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d5b639102044d3fa11714f5ee6b9c63 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e371fc47306e4a54a3c61039ea334f8b : SUCCESS in 8m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06cd994c2002415392ea5dab1e61386f : SUCCESS in 1h 34m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc75d3ec944244969a75e5dac2f12409 : FAILURE in 2m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a4425f35b07d1fba47f1cc0d221e2d82d4f1313","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-02-28 15:02:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fd3da02b16044fbf45c00f70c30f6808d8a89b8e","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-02-28 15:15:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f4f2de851a5e23cdd954b7aae8ce28088758fa1d","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-02-28 15:31:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1a3b376434b9c66a18e68beb78d0bd3f8daabc4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 15:32:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3fb34ff2a1aff303f34da88f7394d2eb9d91cb9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 17:10:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861f885ed19d400fb30451c785a667a1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d384e0fb27944e688ad31b9788f04ca1 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e34f8b577ac410da5741386ba68adc7 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffed6c82a264466f911de03dc21a2af0 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4a4481552b749e784482af1d9998137 : SUCCESS in 5m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99c775b260f141bfa2c577f5fab393c1 : SUCCESS in 1h 18m 31s","accounts_in_message":[],"_revision_number":1},{"id":"4c878da5f4f579d7b0212446b858d0bb6e1130d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 17:10:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0df2e6c44a7b207bf9ec876ec4b9c6b65fe3d1ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 17:11:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c21aaa22cc74c4c962d25693a99105d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e748c3714bd34fccba737e495742d4da : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11eef7db7591d12acbc9d6e449a871ce21b76b94","revisions":{"11eef7db7591d12acbc9d6e449a871ce21b76b94":{"kind":"REWORK","_number":1,"created":"2023-02-15 16:09:34.000000000","uploader":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"ref":"refs/changes/24/873924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/24/873924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/873924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/873924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/873924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/873924/1"}}},"commit":{"parents":[{"commit":"5c318d603b2c27d98f57a9512225ed6ebf7108cd","subject":"Xena: Update classifiers according to tested runtimes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5c318d603b2c27d98f57a9512225ed6ebf7108cd"}]}],"author":{"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","date":"2022-10-05 17:41:06.000000000","tz":-180},"committer":{"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","date":"2023-02-15 16:09:34.000000000","tz":0},"subject":"Remove cache invalidation when using expired token","message":"Remove cache invalidation when using expired token\n\nThis can create a race condition for long running services that reuse\ntheir token (eg. Kubernetes Cinder CSI plugin) in this case for\nexample:\n\n1 [user] Asks nova to attach a volume to a server\n2 ...the user\u0027s token expires\n3 [user] Asks cinder if the volume has been attached\n4 [nova] Asks cinder to attach the volume\n\nIn step 3 the token is marked as invalid in the cache and step 4 fails\neven if allow_expired is true\n\nCloses-Bug: #1987355\nChange-Id: Ice8e34440a5fe1baa370646ed70b5e085c4af70e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/11eef7db7591d12acbc9d6e449a871ce21b76b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/11eef7db7591d12acbc9d6e449a871ce21b76b94"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[]}
