)]}'
{"id":"openstack%2Fkeystonemiddleware~875107","triplet_id":"openstack%2Fkeystonemiddleware~stable%2F2023.1~I14c91b2156803a02db67f05d352e70229e438282","project":"openstack/keystonemiddleware","branch":"stable/2023.1","topic":"create-2023.1","hashtags":[],"change_id":"I14c91b2156803a02db67f05d352e70229e438282","subject":"Update .gitreview for stable/2023.1","status":"MERGED","created":"2023-02-24 15:10:43.000000000","updated":"2023-10-09 08:29:01.000000000","submitted":"2023-10-09 08:27:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875107-create-2023.1","meta_rev_id":"057a15838083405ef17500b106b4a08a0d8ea3a7","_number":875107,"virtual_id_number":875107,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-09 08:27:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-02-27 14:49:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-03-24 15:43:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2023-10-06 14:26:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-10-09 07:05:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 15:34:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 14:49:07.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-03-17 15:30:16.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"CC"},{"updated":"2023-03-24 15:43:13.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-06-26 09:35:41.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"CC"},{"updated":"2023-10-06 14:26:45.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"b9d58c95e7ea46a2b3be557865891876bab883c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-02-24 15:10:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7759d2c3b4393330bd0491e1a3435d20ededfd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:34:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/078845609ed44c05a2b43ec95520d623\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23acf826a0904e80979f0dcdd468de93 : FAILURE in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62f40e4120b485ea99227dfd1958344 : SUCCESS in 2m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a69721cc129e477e92ae9fd305555049 : FAILURE in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04672b1ea52043b3b0f962c98c3477ec : FAILURE in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c08c2d8fe2e44cdbe20b119ccfe6c13 : FAILURE in 3m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a73305f6a84f409ca35f71e299809e52 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0ca14354ec5ace6ed0254f59367356da4592a037","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-27 14:49:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e452d38d2df3b82781dd272eb83aa7de5497b2e5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-27 14:50:03.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0fd5ba1936e07d0a08d222d4173195e01cbaec03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:16:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a62236ef8fb4765bc8ecce1cc5bbc52\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46a7244263df4f1a82c17d21bc291c84 : FAILURE in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417028c049044401bed6334f1750adec : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/410d61e72b4f4ad68ef29150c3adde14 : FAILURE in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b15b793233a74ea5b8ffbbeb3840cfff : FAILURE in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1b314cb91804617a11e799268c0bd09 : FAILURE in 3m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc11e9c99fb42729b8dbeb68460f6f8 : FAILURE in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"74d547c4b299ff7078913ca0e90891db8aa63e57","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-03-17 15:30:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9abc2c68b7f1877fd874ef0859c3d378cd941f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 17:10:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6d69f051e2a43199acf031fab2233e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bca8510187dd42058620b3533e086853 : FAILURE in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b61760df9ec46dfa2323ea5b5f66fe4 : FAILURE in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30386db2833a4b279a7af43bce8b6987 : FAILURE in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/200414e251084744bede05f1241089d9 : FAILURE in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9813606a39b45a29359565c4f64de82 : FAILURE in 3m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4317e97528f246e4ad95a04fb4887ab5 : FAILURE in 4m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e64c88ecb62446ba58de9a768b0fec7 : SUCCESS in 1h 30m 15s","accounts_in_message":[],"_revision_number":1},{"id":"52e560d1ffb5d21589588e77c4f01920a29ca07e","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-03-24 15:43:13.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"198eb6d4d604ff654aee0c1e91de10eee5280ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 16:50:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69b44632cc7747f19f02e17ba764bda6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47ec81a03e994acb9acdc212d72a66e1 : FAILURE in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1954b0e4dc7e4eaca880cdaa8b7b4176 : FAILURE in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f01e121bda54284b0f9885a8aa14795 : FAILURE in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60a47d91bfc44ac5b7fdcf7855b235ec : FAILURE in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27214fd2ef7f411d89510600f08ff754 : FAILURE in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04968325d214141a8b535b4ebc5d788 : FAILURE in 4m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5becd4b0498b4e1e92e411740b0b873e : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cecee7acb4884f21fbec601f7d0275ecc1c2334a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-05-26 14:28:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b30fb389333de52af2f4471a5b680fa0e200805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 15:52:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a2d3e4dbfa5466fba35b99f7f613965\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28f8ed86f0e84876ae7bf0c7377ac73a : FAILURE in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091789ec08bd499cba716a423c0be673 : FAILURE in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ad3e87fdc1b4521807255e6dd8cc748 : FAILURE in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1efe71e0c7034aababbc5b77a25a2706 : FAILURE in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d96327bbb8a496b85ddf95d7ac06d30 : FAILURE in 2m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c735d25ace3498eb2e1d99f1244d8cb : FAILURE in 4m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d51729d6cc94a33baa72e11205da347 : SUCCESS in 1h 21m 49s","accounts_in_message":[],"_revision_number":1},{"id":"85e618f356a9fdfac95d5af68f8369cbc7d71f4c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-06-26 09:35:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7f98608cdd92533e8af63f47e710329e7d4a064","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 11:14:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd224b7e11214444b0aecc51fee777a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/779e1ed1fb054daa9e4751da33ab2540 : FAILURE in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee83a17a61840b2b54144c5cd0dde54 : FAILURE in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dfc62da28e0461482dbe9ade7e27c97 : FAILURE in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d4e37df34d448bda6b113863ba3d435 : FAILURE in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34ad6120adbf449c829bc0644cded505 : FAILURE in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25b4a10d9f834c4797623e88992d95ce : FAILURE in 6m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e20ca8904354ba9b596d3b4695ec388 : SUCCESS in 1h 30m 35s","accounts_in_message":[],"_revision_number":1},{"id":"889c4bc5e2522fe9bd9234f84e68e8ffcf59d5e7","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-10-06 14:26:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8fd14b589460127af1c94468a6f72adc130a788a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 15:55:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81258f7e49744211b6d0ecfb87759252\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b37df2f09e4b4b9ca308488b7fa9820b : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59bab02406cb48a59821e09b97bfef1b : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17eb8fe751a04e90b82d52f952e34a61 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40a6fb2688c64088a129a07286ade210 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dae1f0b5a794e61a71af5836f6377e2 : SUCCESS in 5m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05a158bb0a6443dba9900a5d2f6056c1 : SUCCESS in 1h 23m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5a2f6a04176078186d0f29f53383f057b77d71e8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-10-09 07:05:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c4bafc9cdb777fb41dd04b4ace6fbb074067721","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 07:05:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ae0d51f318954b2f0f73dd8794572846e883764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 08:27:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d998f2ecb65242da9a5cf85424b82ac3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c59847be128f47f7868a4b4a441064a9 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a185c81c1814b06a37bbd4fde56df30 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf114b5701cf41fdaa3157eff832a39b : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/add31216cdbf4a29b5ad7832c4559f45 : SUCCESS in 6m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bddc10e65734688bf18189cfe66d0b2 : SUCCESS in 1h 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"bacf6b7a6dac27b277f0ec7e8d21e2505f3fab8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 08:27:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"057a15838083405ef17500b106b4a08a0d8ea3a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 08:29:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93d1be13ba34479baaf6ee8cbaccd362\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1932c5f8bcb348f391ccf7ad3440a028 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"404b44320f7c7f791f6d14158b0992e0ffcc9306","revisions":{"404b44320f7c7f791f6d14158b0992e0ffcc9306":{"kind":"REWORK","_number":1,"created":"2023-02-24 15:10:43.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/07/875107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/07/875107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/07/875107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/07/875107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/07/875107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/07/875107/1"}}},"commit":{"parents":[{"commit":"e05466c5f439cd05482f109e6eb97a50ba156698","subject":"Remove cache invalidation when using expired token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e05466c5f439cd05482f109e6eb97a50ba156698"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-02-24 15:10:42.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-02-24 15:10:42.000000000","tz":0},"subject":"Update .gitreview for stable/2023.1","message":"Update .gitreview for stable/2023.1\n\nChange-Id: I14c91b2156803a02db67f05d352e70229e438282\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/404b44320f7c7f791f6d14158b0992e0ffcc9306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/404b44320f7c7f791f6d14158b0992e0ffcc9306"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
