)]}'
{"id":"openstack%2Fkeystonemiddleware~875108","triplet_id":"openstack%2Fkeystonemiddleware~stable%2F2023.1~I6db444f75a16085df5eb0f0fd94ba197eb265657","project":"openstack/keystonemiddleware","branch":"stable/2023.1","topic":"create-2023.1","hashtags":[],"change_id":"I6db444f75a16085df5eb0f0fd94ba197eb265657","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","status":"MERGED","created":"2023-02-24 15:10:44.000000000","updated":"2024-01-31 15:32:34.000000000","submitted":"2024-01-31 15:30:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"875108-create-2023.1","meta_rev_id":"d2ee772c865abc5aa583734c8918eb128f821ee2","_number":875108,"virtual_id_number":875108,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-31 15:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-02-27 14:49:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-06-27 22:01:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2024-01-31 14:36:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"date":"2023-02-27 14:50:18.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2024-01-31 14:36:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-24 15:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-27 14:49:27.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2023-06-27 22:01:21.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-01-31 12:58:28.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"CC"},{"updated":"2024-01-31 14:36:51.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"5c4aa9b48cd437a1f74da304da5b7d670632ddbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-02-24 15:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bcf442fc006fb1c75bd5fa682bf8511a5902411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 15:44:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74c9d7cf62354b16bcda388a50b29c80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/315759549cae49519e5babb56847417b : FAILURE in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7bd1c4aa4fb45ea8fb6e2ca3f54735e : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/269759a90d4545bd9f47069e8948e9e2 : FAILURE in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/107906c134e34e65bdf583cc99c4a8c3 : FAILURE in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d8ae593dca644b7993d023fdf192dbe : FAILURE in 2m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c915038cadd4617a7af6a4759364def : FAILURE in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2298872c0dd145fcab6ca5d46c9928b5 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"76845f0a51bcef3e0a14d56d717d4b46c17160df","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-27 14:49:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"164a562424bd917f6af956e9273c97df4bdc94d1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-02-27 14:50:18.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51460e76901ad8376212736685cef3ec765237e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 15:13:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d22c0908e8fb4c66aa2977b0048f7e0f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a749e32535c64c4ca1f7ba9b29a92c13 : FAILURE in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14ca7801f92846bf98d908b47eaf418d : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb325242aa63464d8e01841cb143ac89 : FAILURE in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52642f43a45045fc83f12f80567bf06a : FAILURE in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5785e9e5b3964dc9952183538e42c410 : FAILURE in 2m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b384df72b984edb963f2b73b46ba3b8 : FAILURE in 3m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2112fdfde2e4ee2b6f95504365b588c : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c57536fb91d8f1e19beb5fc3c8538d1f8e2c26e7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2023-06-27 22:01:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"288fa497ba38f303cfcdf21727ab0fb0f838c4f1","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-01-31 12:58:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"89832b48d4554126ac4ef7d6f9ee3c6310519729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 14:34:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66f7e116cd1f4329a85107d28714fa2a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee35fbc0d56045739f35da31880a0961 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47efd33a3f744a86b6852c7b9c3fc83e : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8ed16c75251452f83174a57a36b5462 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2279f407c93840aabd019ea6145188cd : SUCCESS in 5m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67f306dd3278469dbd40627a3e5c882d : SUCCESS in 1h 30m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/212b995dffd2458cab0fa9952deca07a : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"dd7652b870bfda8c09e2ef16d47438c34e2a6332","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-01-31 14:36:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c248a0aacd8b047f72c9d79a8c7da58a9e2e7f6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 14:37:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd7679bd02a213811455069ea891df52ce864634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 15:30:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/189376e10f4d4187bb16229967e277ce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69920a2c738143c7b83c46801ff17c3a : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4490b55a17fc48dba9037683673b824e : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/240bdf43eda8411d8c8930adb8f588ae : SUCCESS in 5m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae754f5cdd43430a8fa20649d97ad05a : SUCCESS in 52m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cb0c95eb5a241a491d9ebdce21848d4 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8396126b4afa5fc0529c3f5e3c4612779b225bb0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 15:30:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d2ee772c865abc5aa583734c8918eb128f821ee2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 15:32:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e181bfd4d9b4fb59a5c83341765bb6e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df0cfbd8f83410e885208e3f3c6fc50 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a80a00a7d0c249078f483efa9a9717b1 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4faab570564223065b4786db054384ebc65d21ae","revisions":{"4faab570564223065b4786db054384ebc65d21ae":{"kind":"REWORK","_number":1,"created":"2023-02-24 15:10:44.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/08/875108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/08/875108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/875108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/875108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/875108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/875108/1"}}},"commit":{"parents":[{"commit":"404b44320f7c7f791f6d14158b0992e0ffcc9306","subject":"Update .gitreview for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/404b44320f7c7f791f6d14158b0992e0ffcc9306"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-02-24 15:10:43.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-02-24 15:10:43.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2023.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2023.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I6db444f75a16085df5eb0f0fd94ba197eb265657\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/4faab570564223065b4786db054384ebc65d21ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/4faab570564223065b4786db054384ebc65d21ae"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
