)]}'
{"id":"openstack%2Fkeystonemiddleware~884778","triplet_id":"openstack%2Fkeystonemiddleware~stable%2Fussuri~I2092993dcfdb67a646708941f610019665b7a3b0","project":"openstack/keystonemiddleware","branch":"stable/ussuri","topic":"lower-contraints-ussuri","attention_set":{},"removed_from_attention_set":{"34275":{"account":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"last_update":"2023-06-20 17:38:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2092993dcfdb67a646708941f610019665b7a3b0","subject":"Make openstack-tox-lower-constraints non-voting","status":"MERGED","created":"2023-05-30 20:04:33.000000000","updated":"2023-06-20 17:41:04.000000000","submitted":"2023-06-20 17:38:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884778-lower-contraints-ussuri","meta_rev_id":"c672c5f41ae98435db7241f1b0f6363d00cc56d0","_number":884778,"virtual_id_number":884778,"owner":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-20 17:38:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-06-20 15:44:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2023-06-02 14:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2023-06-02 14:55:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2023-06-20 15:44:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 21:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-02 14:54:24.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2023-06-02 14:55:04.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-06-20 15:44:00.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"}],"messages":[{"id":"d5f647782bfc1392bab63893c1cdd5001e00f86b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"date":"2023-05-30 20:04:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6376b380fa5906b13bc2a2dcdd4f325ded3801b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 21:22:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/555781a1b424404e965e9baef867fbff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22753ae64549494cb5ba8a8bdcdf4b3e : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97509ca8e91843158d3a0ba574d460fa : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5151f8f743748f9a0a7ed442adc215d : SUCCESS in 2m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8177e3b129b4777aaf5734e88aa08f9 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d99741851ad341efb31dad2657d4ba5d : SUCCESS in 5m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a3adb38d76543ce8ce762e3eca3b293 : SUCCESS in 1h 17m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b778e25c4ce41aea64c9b1c4005556d : FAILURE in 2m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c3ed91c7fa8d256b410f08d033461510b12c46d","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-06-02 14:54:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"974b79b922914b2b5e92315f2102c34622604519","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-06-02 14:55:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"431defd65c6e153e322643dab2a07d51ff166e96","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-06-20 15:44:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac710eae4903c85a00b42e7a8974dc1c929b3ff6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 15:44:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b297fb4dd396febccdfaaadab5913626d636484","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 17:38:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a1604ace3f54130ac5cde5752cc3968\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9246547f161496fa908973d7d643578 : SUCCESS in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec52993b13f149efad4bc36861ebbae6 : SUCCESS in 7m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f1143cefd59645bfaf2ff368cbe9ee51 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3da40d3ddf445ada4583f4d81383532 : SUCCESS in 9m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b3c23524fb44607a213d4c47c2893ec : SUCCESS in 1h 41m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e6b74a3b6dbd70b3a15515da2aad601a419f655b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 17:38:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c672c5f41ae98435db7241f1b0f6363d00cc56d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 17:41:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7636544f4cb42f9b1af35e7e37de5df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7e3734481d42a386151da6484ae451 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fc8233b106cdcc7cfb860a903a491bbff17f047","revisions":{"2fc8233b106cdcc7cfb860a903a491bbff17f047":{"kind":"REWORK","_number":1,"created":"2023-05-30 20:04:33.000000000","uploader":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"ref":"refs/changes/78/884778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/78/884778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/884778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/884778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/884778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/884778/1"}}},"commit":{"parents":[{"commit":"654d31a1bd32968bfa008a117a60b58e18d68df8","subject":"Merge \"Change the default Identity endpoint to internal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/654d31a1bd32968bfa008a117a60b58e18d68df8"}]}],"author":{"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","date":"2023-05-30 19:53:36.000000000","tz":-180},"committer":{"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","date":"2023-05-30 20:04:18.000000000","tz":-180},"subject":"Make openstack-tox-lower-constraints non-voting","message":"Make openstack-tox-lower-constraints non-voting\n\nThis test is non-voting in Wallaby and was removed from Xena onwards.\nCurrently is failing for all backports.\n\nChange-Id: I2092993dcfdb67a646708941f610019665b7a3b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2fc8233b106cdcc7cfb860a903a491bbff17f047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2fc8233b106cdcc7cfb860a903a491bbff17f047"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
