)]}'
{"id":"openstack%2Fkeystonemiddleware~884786","triplet_id":"openstack%2Fkeystonemiddleware~stable%2Fvictoria~Ib7c1050ad0ce266d62879b127e643f730c308aa1","project":"openstack/keystonemiddleware","branch":"stable/victoria","topic":"lower-constraints-victoria","attention_set":{},"removed_from_attention_set":{"34275":{"account":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"last_update":"2024-03-26 18:21:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib7c1050ad0ce266d62879b127e643f730c308aa1","subject":"Make openstack-tox-lower-constraints non-voting","status":"ABANDONED","created":"2023-05-30 21:59:43.000000000","updated":"2024-03-26 18:21:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f7c687847210b134274f3073278fb497966ecd08","_number":884786,"virtual_id_number":884786,"owner":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-30 23:10:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 23:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-26 18:21:10.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"0e0a7ae0d28ca71a751f38db58eba1e3e0e2077d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"date":"2023-05-30 21:59:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"589e7e00b5749d67c4bd36fc6af8ab84fa1e40d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 23:10:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffb337c4d32e44feba6b453685f99a4a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3d72a77f32c4d23bae8783f3128ca9d : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ca3f80caee547bcb6e706c757d8a956 : FAILURE in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/444d6bd31da24fa5be2d7601e711c27c : SUCCESS in 2m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc1bbac6e93a4bdb9f53c8527f7277a4 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a68eb7976ce4a6c99e6efc6f1dcedfe : SUCCESS in 4m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6ef18b8946d4425bd3ddef6cdbd2821 : SUCCESS in 1h 09m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa5a0f83b1f94944a972494be52cd37d : FAILURE in 2m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7c687847210b134274f3073278fb497966ecd08","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-26 18:21:10.000000000","message":"Abandoned\n\nstable/victoria branch of openstack/keystonemiddleware is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/victoria if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37ca859a81db0183b537b019e1e6b2e2307b27b3","revisions":{"37ca859a81db0183b537b019e1e6b2e2307b27b3":{"kind":"REWORK","_number":1,"created":"2023-05-30 21:59:43.000000000","uploader":{"_account_id":34275,"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","username":"drencrom"},"ref":"refs/changes/86/884786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/86/884786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/86/884786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/86/884786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/86/884786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/86/884786/1"}}},"commit":{"parents":[{"commit":"654d31a1bd32968bfa008a117a60b58e18d68df8","subject":"Merge \"Change the default Identity endpoint to internal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/654d31a1bd32968bfa008a117a60b58e18d68df8"}]}],"author":{"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","date":"2023-05-30 21:12:46.000000000","tz":-180},"committer":{"name":"Jorge Merlino","email":"jorge.merlino@canonical.com","date":"2023-05-30 21:59:41.000000000","tz":-180},"subject":"Make openstack-tox-lower-constraints non-voting","message":"Make openstack-tox-lower-constraints non-voting\n\nThis test is non-voting in Wallaby and was removed from Xena onwards.\nCurrently is failing for all backports.\n\nChange-Id: Ib7c1050ad0ce266d62879b127e643f730c308aa1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/37ca859a81db0183b537b019e1e6b2e2307b27b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/37ca859a81db0183b537b019e1e6b2e2307b27b3"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
