)]}'
{"id":"openstack%2Fkeystonemiddleware~887087","triplet_id":"openstack%2Fkeystonemiddleware~master~I615be3453db37588edf98a46ce484efc5e051f11","project":"openstack/keystonemiddleware","branch":"master","topic":"tox-4","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-06-28 19:14:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I615be3453db37588edf98a46ce484efc5e051f11","subject":"Bump hacking to 6.0.x","status":"MERGED","created":"2023-06-27 17:03:42.000000000","updated":"2023-06-28 19:15:27.000000000","submitted":"2023-06-28 19:14:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887087-tox-4","meta_rev_id":"228e03553831379641837d4c663aa71557aa772d","_number":887087,"virtual_id_number":887087,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-28 19:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2023-06-27 19:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2023-06-28 12:06:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-06-28 12:06:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-27 18:51:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-27 19:31:37.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-06-28 12:06:41.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"944497f26eaff8ea1fbbe5d735f86d207a598202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-06-27 17:03:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f2d2ac8dd9a75bbaf382947aa39cf8348c5690e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 18:51:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1519267c427f4f1fb074b1b79551a2d0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c28f1801d3f4795a5ee5cbd095ee0cb : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/566edd8f6af74db294f3111c963d389f : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3b645108f964afeb8595fe1e5502e43 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f27ca27ec444631b094cb842c6b1a5a : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4166441786e2434aa34cf9f87235eed1 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ff744b077ce43e19f8db2f8b84fe147 : SUCCESS in 6m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/286dea5ea2f348c9a313f675607aebc4 : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/adf242b2b7f744f3a36d5fe38673b242 : SUCCESS in 1h 41m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d7c85e6ad9a4a57ac7abd73fc7fa350 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"dcca2184f2379bb8189de61bb907b8f2a25e20ad","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-06-27 19:31:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5434cb079a56f16bdd68e6c2e126fd25a849541b","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-06-28 12:06:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e5221a0244eec552105d3ff006228c78fdf7e1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 12:07:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea240f46d4259566ad8084ae0ea746cc0a49c1e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 13:58:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce7ca65293f842fcbaad6053ea053817\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d0efa88f446499d92dd045e0502788c : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a5b188da9464135b8ba1224b038e6bb : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da46b023bac84ba9a312a7ba940eda5d : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f4d2e14bf04daf9fefc365a70d6405 : SUCCESS in 6m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8e82d842a7b24ecf81e3b943e728b8ea : SUCCESS in 2m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e6e66b3b54041f392bef58701fab0ad : FAILURE in 1h 46m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/574db34de9bd4862b545d386efce9f88 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e9ffa739304ec29dd9053fa4c83f940ada40a488","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-06-28 15:05:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03cf0c7fb4f51d449c9ca641be28f0092193523e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 17:19:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af6c19358f0246cd9c3c35dc0cb5ec22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/960f629e7fb44e7f939f0030345e2e86 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9638031666d24e258803ef73e129cad6 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c040a216117a4517b6833f84891a49ec : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d79a07fbe5f74b8e8faf14833c44c0d3 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a38cf385f3d74dd0a5d697418428eabf : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5d0bf801be467db1c532bf44882fc3 : SUCCESS in 4m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/21bbf1edcf2f4088bd37a3185dfead03 : SUCCESS in 1m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8d226a2bc6244d28fe6710d9957da85 : SUCCESS in 2h 04m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f93caae482647e790409dbca42bf05e : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"45360d0361f6ba7e0d2486466bbb85e3ce421445","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 17:19:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b4c0820d624844f01a6927156ab4199f5776037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 19:14:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb0219e6d7de4c1aacc57d1561b1f787\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c111d387b0346d585fe7c2b4592b941 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2386e42b722f4e9f8c78ecfd4aef79ae : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8de41b3e79247aeb7545e6176ab7ed2 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc811a8d2bd4ad0ac7bba476910aaf9 : SUCCESS in 5m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a3f65a335f04d49a9250dacbe3fe4f9 : SUCCESS in 2m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f2948ca1c36438db9bbd2ccf22aeb00 : SUCCESS in 1h 47m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c48c6ce43994e0f9f507db902e60875 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3fc83e5d3bf77e8728e0df875b0e731683337a17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 19:14:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"228e03553831379641837d4c663aa71557aa772d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 19:15:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58d89471e2434e319f84047b3db484e6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f23b2c9e82e49d68fcf7a6af1e66faa : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04e3f1e754874251913021c2b46eece4 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f48ff3e8d3c7d3311a58442b8cf71d6e3658620","revisions":{"0f48ff3e8d3c7d3311a58442b8cf71d6e3658620":{"kind":"REWORK","_number":1,"created":"2023-06-27 17:03:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/87/887087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/87/887087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/87/887087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/87/887087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/87/887087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/87/887087/1"}}},"commit":{"parents":[{"commit":"3f12bc6e3363f3c3898cb01217f3993e9c5dc0fe","subject":"tox: Trivial formatting changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3f12bc6e3363f3c3898cb01217f3993e9c5dc0fe"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-27 16:29:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-27 17:03:30.000000000","tz":60},"subject":"Bump hacking to 6.0.x","message":"Bump hacking to 6.0.x\n\nA note about dependency ordering is removed from the requirements file:\nthis is no longer true with the dependency resolver introduced with pip\n20.3.\n\nChange-Id: I615be3453db37588edf98a46ce484efc5e051f11\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0f48ff3e8d3c7d3311a58442b8cf71d6e3658620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0f48ff3e8d3c7d3311a58442b8cf71d6e3658620"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
