)]}'
{"id":"openstack%2Fkeystonemiddleware~887189","triplet_id":"openstack%2Fkeystonemiddleware~stable%2F2023.1~I3b8263afbf0ccee88ceaac2040d5ad274f22d74a","project":"openstack/keystonemiddleware","branch":"stable/2023.1","topic":"tox-4-stable/2023.1","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-07-07 15:57:55.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-07-07 15:57:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3b8263afbf0ccee88ceaac2040d5ad274f22d74a","subject":"Make tox.ini tox 4.0.0 compatible/fix gate","status":"MERGED","created":"2023-06-30 19:25:12.000000000","updated":"2023-07-07 15:58:58.000000000","submitted":"2023-07-07 15:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887189-tox-4-stable/2023.1","cherry_pick_of_change":887085,"cherry_pick_of_patch_set":1,"meta_rev_id":"e2a5385612dd9c7665e6f7cab325d9ee4f1894f7","_number":887189,"virtual_id_number":887189,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-07 15:57:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2023-07-07 14:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2023-07-07 14:10:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2023-07-07 14:11:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-30 19:25:12.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2023-06-30 21:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-03 11:31:26.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-07-07 14:11:47.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"85b988318ecf7113561e3313bf540a92e28473f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-06-30 19:25:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2ee90e8f2c232fab982523fcc92b1342976b6b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-30 21:04:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3ce2ae84ddc48769497d58186e9f582\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d4270f59255406781d57532f3bc0945 : FAILURE in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c221905559de4cd48e8ba27ea676fbfe : FAILURE in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b15549914ea44ed48827b419fc844db4 : FAILURE in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c36ccb104239426d8d8435beca4a5f30 : FAILURE in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf7f4ff2bd28484ab397282c0ebcfa72 : FAILURE in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61831e703b504f9da28dec8239e1846d : SUCCESS in 5m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4ac2d1ea4b24427969ec248236c75a9 : SUCCESS in 1h 28m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc69bd88da8e45f8a8f74fab269e9a39 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0e9a03789c3779768aab5264d4bfaa0b70b3b1a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-03 11:31:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e0c23b087d8da1c91fe994446a23be00483c7026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 12:49:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea38e36bb6004ac8b93d09a91bef002a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e51953d6a2674f6e9d67df853e0e9170 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331b04fb649a4fb9b6355fdaaf592769 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a12a1673fb614eb696ccd3717a672a4c : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77ce52987f224bd6bab99a507b643f44 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9fe54eaf4cd4bd7bd729d67e1923668 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/027acb8935c146d79e4430151120323d : SUCCESS in 5m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6647840743ad4732ad2c6b6f31fa7d05 : SUCCESS in 1m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8605079af281454eb9e44ded09ce396f : SUCCESS in 1h 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea059cab6a9e4c028dc1492ed33992f3 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8160b4ffd5031ebc1df1f86f8e0307b8a0fff16b","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-07-07 14:10:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4c407274a8b7bf608fcf4055fc01e94ded6759f9","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-07-07 14:11:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"659299238a79668a10150dc10f2e96337550f8b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 14:12:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"064ad35b0d49d9b2c60e67e977f9f8aa7a2a7f1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 15:57:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b99490c8e754464bb549d5f29a28c57\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/657c68427dde4a7c94743c77e60f0bc7 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/010de2c0d76144768d28595ff31bff2b : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c10eb9b76e9f46408aa144d8349ef31d : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7749e9ffd2ca464f94f98ee68a14dfcc : SUCCESS in 5m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/54cc1b0049ad4f17a9d8c0d4453b3b43 : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91ad4b682adc4c52b509baf30111185a : SUCCESS in 1h 38m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7b4630aee184216bb7caf0f01363663 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"5bb8a361ce6b5518c4d397b9c819462d811acff0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 15:57:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e2a5385612dd9c7665e6f7cab325d9ee4f1894f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 15:58:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a07c3e847f7f466286e90af70515ea21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a32263a4c43409ebfc0f1445d58c012 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f22cd3c21d7d49a6ba6d1a4f1f564838 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ce29fcfb4cc76d890856d96def2e888ffee26fc2","revisions":{"6ad1c2ab199ac0dcd5464e3a0348fdeca8e183ae":{"kind":"REWORK","_number":1,"created":"2023-06-30 19:25:12.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/89/887189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/89/887189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/89/887189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/89/887189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/89/887189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/89/887189/1"}}},"commit":{"parents":[{"commit":"e05466c5f439cd05482f109e6eb97a50ba156698","subject":"Remove cache invalidation when using expired token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e05466c5f439cd05482f109e6eb97a50ba156698"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-27 16:16:47.000000000","tz":60},"committer":{"name":"David Wilde","email":"dwilde@redhat.com","date":"2023-06-30 19:25:12.000000000","tz":0},"subject":"Make tox.ini tox 4.0.0 compatible/fix gate","message":"Make tox.ini tox 4.0.0 compatible/fix gate\n\n* Removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. Without this, our entrypoints are not available.\n\n* Removed basepython \u003d python3 as we assume all developer switched to\n  python3 in their env already\n\n* Removed ignore_basepython_conflict \u003d True as without the basepython\n  definition generative targets now work without conflict\n\nSee [1] for a similar change made to placement.\n\nIt is also necessary to fix issues with the gate. For reasons that I\nhave yet to grok, a mock of \u0027requests.request\u0027 used in some test is no\nlonger functioning as expected. My guess is that something is now\nimporting requests before us and interfering with the mock but never\nmind - we can easily bypass the issue by mocking \u0027requests.post\u0027\ninstead.\n\n[1] https://review.opendev.org/c/openstack/placement/+/868418/\n\nChange-Id: I3b8263afbf0ccee88ceaac2040d5ad274f22d74a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 626df3a5e9e9d4fbea2649a9cfa0048083bdb88b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6ad1c2ab199ac0dcd5464e3a0348fdeca8e183ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6ad1c2ab199ac0dcd5464e3a0348fdeca8e183ae"}]},"branch":"refs/heads/stable/2023.1"},"ce29fcfb4cc76d890856d96def2e888ffee26fc2":{"kind":"REWORK","_number":2,"created":"2023-07-03 11:31:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/887189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/89/887189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/89/887189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/89/887189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/89/887189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/89/887189/2"}}},"commit":{"parents":[{"commit":"e05466c5f439cd05482f109e6eb97a50ba156698","subject":"Remove cache invalidation when using expired token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e05466c5f439cd05482f109e6eb97a50ba156698"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-06-27 16:16:47.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-03 11:29:45.000000000","tz":60},"subject":"Make tox.ini tox 4.0.0 compatible/fix gate","message":"Make tox.ini tox 4.0.0 compatible/fix gate\n\n* Removed skipsdist\u003dTrue to make sure placement available in the virtual\n  env. Without this, our entrypoints are not available.\n\n* Removed basepython \u003d python3 as we assume all developer switched to\n  python3 in their env already\n\n* Removed ignore_basepython_conflict \u003d True as without the basepython\n  definition generative targets now work without conflict\n\nSee [1] for a similar change made to placement.\n\nIt is also necessary to fix issues with the gate. For reasons that I\nhave yet to grok, a mock of \u0027requests.request\u0027 used in some test is no\nlonger functioning as expected. My guess is that something is now\nimporting requests before us and interfering with the mock but never\nmind - we can easily bypass the issue by mocking \u0027requests.post\u0027\ninstead.\n\nChanges:\n  keystonemiddleware/tests/unit/test_ec2_token_middleware.py\n  test-requirements.txt\n\nNOTE(stephenfin): It is necessary to revert the test changes since they\ndon\u0027t apply to this branch. It is also necessary to cap bandit since we\ncan\u0027t fix the timeout errors it is warning about without a feature\nbackport.\n\n[1] https://review.opendev.org/c/openstack/placement/+/868418/\n\nChange-Id: I3b8263afbf0ccee88ceaac2040d5ad274f22d74a\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 626df3a5e9e9d4fbea2649a9cfa0048083bdb88b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ce29fcfb4cc76d890856d96def2e888ffee26fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ce29fcfb4cc76d890856d96def2e888ffee26fc2"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
