)]}'
{"id":"openstack%2Fkeystonemiddleware~899911","triplet_id":"openstack%2Fkeystonemiddleware~master~Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2","project":"openstack/keystonemiddleware","branch":"master","topic":"bp/enhance-oauth2-interoperability","attention_set":{"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2025-01-15 15:09:12.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"33492":{"account":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"last_update":"2025-01-15 15:09:12.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2","subject":"Add FT for External OAuth2.0 Server Support","status":"NEW","created":"2023-11-02 05:13:10.000000000","updated":"2025-01-17 12:29:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8bc2d4e235db5381447418cddccafbb49936f332","_number":899911,"virtual_id_number":899911,"owner":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-16 05:52:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-01-15 15:09:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-01-17 12:29:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-02 05:13:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-21 11:18:42.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"state":"CC"},{"updated":"2025-01-15 15:09:12.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-01-17 12:29:25.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"1a832ce8ec35d93981f11d60a69f19a0f7a3e93c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-02 05:13:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26536ecdf02a5177da04c2164d945153d4f90cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 05:13:33.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/keystonemiddleware on branch master.  The error was:\n\n  Invalid Ansible variable name \u0027pre-run\u0027 for dictionary value @ data[\u0027vars\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: tacker-functional-ext-auth-base\n      parent: tacker-functional-devstack-multinode-legacy\n      description: |\n        Base job for tacker external oauth2 functional tests\n      roles:\n        - zuul: openstack/keystonemiddleware\n      vars:\n        # keycloak domain name\n        keycloak_domain_name: \"keycloak.host\"\n        # tacker domain name\n        tacker_domain_name: \"tacker.host\"\n        # barbican domain name\n        barbican_domain_name: \"barbican.host\"\n        # self-signed certs directory\n        local_cert_dir: \"/etc/ssl/localcerts\"\n        # ca certificate \u0026 private key\n        ca_key_file: \"ca.key\"\n        ca_cert_file: \"ca.pem\"\n        # keycloak certificate \u0026 private key\n        keycloak_key_file: \"keycloak.key\"\n        keycloak_cert_file: \"keycloak.pem\"\n        # barbican certificate \u0026 private key\n        barbican_key_file: \"barbican.key\"\n        barbican_cert_file: \"barbican.pem\"\n        # tacker certificate \u0026 private key\n        tacker_key_file: \"tacker.key\"\n        tacker_cert_file: \"tacker.pem\"\n  \n        # keycloak realm name\n        keycloak_realm_name: \"extauth\"\n  \n        pre-run: playbooks/extauth/pre.yaml\n        voting: false\n\n  in \"openstack/keystonemiddleware/.zuul.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"bf74c2e0985f7e01f8607181ea5c747943071d20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-02 05:24:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c84420353bf41c09c71855235b7b714dd3883706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 07:42:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/570ed92661b04d948d9a52ed94f5d903\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd0784e042284a8da317dd567d78d435 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2886779e6f294f7aa90bad1965dad98f : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0632f10aa6ee42819e14bf9597dca124 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32bcb82548fd4269a983c3be0c691c1e : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/009388d92c334d338c0d17b375b472ac : SUCCESS in 4m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9fb2fd33ae844e1917b7e47426c540d : SUCCESS in 1h 03m 03s\n- tacker-functional-ext-auth-tls-client https://zuul.opendev.org/t/openstack/build/d5f22b26aa5b41069195977d005197b4 : RETRY_LIMIT in 55m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df85f765f713cf9cd61b9f19a1994e6e0c3e0da5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-03 01:12:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4127509029e720eb3fbe4d2617fae57406f95080","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-03 01:15:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e61b48673f88b73026e0c0e8aa4e2b661b312b6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-03 01:55:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"af63ff87d0bb557e5faab069eeac75a59f31eace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-03 03:14:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8091b8203e004552f481d0755d1920d91aa0671a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 04:37:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12678325d1d04993b8d3162b85afb15a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad0add06d1a14811b1cd84631c675655 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/489001ec41ee4fdf9dc648d65a1f2554 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6273d0bf120b4c709bc6743a2a196938 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b42f1002eac04e53b84c9ca779326558 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e4299825dfa48c4a5d89fd021858980 : SUCCESS in 6m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcf6136146894014a0a596d041322af3 : SUCCESS in 1h 19m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52055240a15f4b36898d75f226048f4c : SUCCESS in 3m 26s\n- tacker-functional-ext-auth-tls-client https://zuul.opendev.org/t/openstack/build/75dbe038de9148ef80d4f87fdd7ca69a : SUCCESS in 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2fa4c188c068102dc0bc4a67d53df230e8c72c38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-03 06:16:59.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7c7db46fe1500814881712e8e9c6f106c519422b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-03 07:24:55.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d3104b61c2be934370bd0533ccb57f8a70cc293b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 08:33:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ca6ffb94e7e49ecb43ce5b34bec4f4b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32b8e40af5804489af760d96f261a7f0 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ed9dbb893b437492b707a54b33e489 : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76ec4545e3474fe3aae857a74e8e4490 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/34620c4eac2b489fa69e6d5f2ba63c21 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665f102da92b42d5bc133b539d328ed1 : SUCCESS in 4m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c40b0119eee4e38b4913194b7cfaddb : SUCCESS in 1h 07m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/974ffe72100f460aaa865d9361b1771f : SUCCESS in 3m 55s\n- tacker-functional-ext-auth-tls-client https://zuul.opendev.org/t/openstack/build/a06e164237ec401e93f9152fb1d9b1c4 : FAILURE in 31m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b7ee5cf1926c68bfdff1a8b20901f84716f32e6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-06 07:13:45.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"08bbef0763ec0c7ab82bf1f071f70ba13adeea40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 08:22:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cf29e975bee44fdbd79f4853e4c7a56\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cb742b174854704bcfa50bb26683420 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c98c7d0142a44c88b37012cf0c85eebe : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cf2cd3ce02f4c739b3726b519878397 : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5a547f0006d4e5a8dbe6d7c73abae7c : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc9e6a1cc38469bb917782df9d688ba : SUCCESS in 4m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52c6dc31e9f1405f813a9720ab17d211 : SUCCESS in 1h 07m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf6e4ce770904535ae7495c703f4d9aa : SUCCESS in 2m 58s\n- keystonemiddleware-ext-auth-functional-with-tacker https://zuul.opendev.org/t/openstack/build/b3540e610bb44420a81e9d9e5c967eb5 : FAILURE in 26m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"270a5e8e3cf425d49be60f1b011da24ec2239836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-07 04:47:23.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"4d8bc04907927eea0214c66eac7a795a9c9dc03d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 06:48:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/901c37abb2c64f2bab1980cdebbfe713\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba0577919ff44d10b7942c70ca14b99c : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7531f639dd074f1bb84e2edc890699ad : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ce4373efb0a431a9164dad7f15c8ed6 : SUCCESS in 3m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/884d4649dcd44ab5a0b1774cf64832b6 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e09616935a4c25ad592f41a4f83b43 : SUCCESS in 5m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24355dc1f030429c89730309b72753aa : SUCCESS in 1h 21m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/159e6d0c52254b40a94e8d630227cb0e : SUCCESS in 3m 59s\n- keystonemiddleware-ext-auth-functional-with-tacker https://zuul.opendev.org/t/openstack/build/ad076496e05941399b878a5e05dec7e3 : RETRY_LIMIT in 40m 05s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d4e0c4a99611d806aec54ad3d33bb38fbcfc1680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-09 00:45:20.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"44fb95e2eac49faca20b24d041d7bf1f982e3b89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-09 01:13:43.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d37b32a81aaf33e10eec5e794aa70f9e2d955018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 02:51:34.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5beeff77165f407bb3fe6c674712db1e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/225878e18e6141878b144d3fcfd78e1c : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3676d2679b49db8748971061d16dec : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2cc5f2308a24dee89aa00b55a0c0685 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e42ab6b7cde41ed83291d4ea181298e : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ab8cbbc94341de914e2a36d70b1f7a : SUCCESS in 5m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b81234e066a49b0930b9291a8d806bc : FAILURE in 1h 29m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41db64fa3838432b847293e2eac57340 : SUCCESS in 3m 21s\n- keystonemiddleware-ext-auth-functional-with-tacker https://zuul.opendev.org/t/openstack/build/1d49ebe75ca748ccb95cb16cb535e6a0 : FAILURE in 44m 52s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6a73652c82895a97d90df3124fe44f620c570279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-09 03:07:27.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"bc6a9d412fbbdb344788a694d69d09aca41e35e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 04:08:28.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7a9610fa6764d0094ef279cb4d5313e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b80bfa892ea24e5093f9c683afaaf099 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e317cd6436d14d7e9d9b2604c15cc42f : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/080fc1a84f914f22a60a31d9030efbaa : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f9829d884c24e149f6c0003bb42395b : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ee0792edbd47a1a9e2de7920890de0 : SUCCESS in 5m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83d3796d0778461baa63d34ba42e360d : FAILURE in 56m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d003f69b81364d6a810fc8303cbccf63 : SUCCESS in 3m 39s\n- keystonemiddleware-ext-auth-functional-with-tacker https://zuul.opendev.org/t/openstack/build/fb3e0cd4c089429fb22523195da46e58 : FAILURE in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"74a6bd13680b27cf03c020718a726608efb3a666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-09 06:07:58.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"741109790091226affe8a1f9bf50cb6b33f71c85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-09 07:11:36.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"3063bf7ff55fd827bc75a7302d52d07a6fb92cab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-09 08:16:54.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"6381fb3d597c5704490dac91b456c8d8ca72c416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 10:25:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b4cec901f844046a9425d5d8b7be047\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb63b46e69f3468a90f07d95d15f1ee6 : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78574748c48f4cdeb7e1ee56df98d378 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb53b9000ca14300a3e7496a8d005fe1 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25bd231735e44aa490ca695a0c1dd450 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf8cd0e0e50043fa9d3bdd93ccbed0ea : SUCCESS in 5m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/feab1ced6ccc4c14a03c86d285d14e43 : SUCCESS in 1h 09m 58s\n- keystonemiddleware-ext-auth-functional-with-tacker https://zuul.opendev.org/t/openstack/build/08391776d0b34226bcbfc2ed6b3871a5 : TIMED_OUT in 2h 02m 09s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"edcfe043a1231b8d3cdaae8e5928040a5988ebb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-10 07:54:04.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"76b2fb2c20acd06a6a3cd581c273c084c41ab384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 09:03:31.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9133330b7031418788ae4a3e25789b96\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b0fce60a66f46a9bde2672be358e655 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783e06a5d0b147c88ee78421930bea1e : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84d107289129400b8ed35a74751c0e5d : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67b02058c25344c9bde8f4c614551600 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f818e6da818a425ab0962403a06b3f51 : SUCCESS in 5m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2f0f3ff844f4841aaa3f389de567c73 : SUCCESS in 1h 08m 12s\n- keystonemiddleware-functional-tacker-tls-client-auth https://zuul.opendev.org/t/openstack/build/d918192769f647fda53d8f759f183a37 : FAILURE in 27m 14s (non-voting)\n- keystonemiddleware-functional-tacker-client-secret-basic https://zuul.opendev.org/t/openstack/build/162da117973d4be18e782f16da2c4aff : FAILURE in 42m 55s (non-voting)\n- keystonemiddleware-functional-tacker-client-secret-post https://zuul.opendev.org/t/openstack/build/b5a0a3122ac64514a1de21c5e1a0c784 : FAILURE in 43m 05s (non-voting)\n- keystonemiddleware-functional-tacker-private-key-jwt https://zuul.opendev.org/t/openstack/build/cafc0a9c38ba430db7104083965aafef : FAILURE in 42m 19s (non-voting)\n- keystonemiddleware-functional-tacker-client-secret-jwt https://zuul.opendev.org/t/openstack/build/ea0f3c875ec04a71a62022f601dc9a75 : FAILURE in 42m 32s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"c4dc6b078d0e0788f9cceb12099a60ec6e4197fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-13 05:06:09.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"17601a8b38268946ff87c59ff2bc815d1f631a2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-13 06:47:40.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"3bb48e4c2bb51428dc4d6c2b392ecdd81e825703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 09:02:04.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25207280dd334306af6b3fc8cf410687\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71a36d22c6f140d388b5df1da0dca087 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d355e82443ae44cca8aa9048959e18fd : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0603c2d100a9429da1dd78d07197b115 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff0ed16fc5a44f20aace43fe19c5d92f : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a5e553be8b4cc8a973d20e5e964a2d : SUCCESS in 7m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/656e9df677084b608aac6a70f1c403fa : SUCCESS in 1h 36m 44s\n- keystonemiddleware-functional-ext-oauth2-tacker-tls-client-auth https://zuul.opendev.org/t/openstack/build/d74d6ac3f59847469a842dfe5ba6a147 : RETRY_LIMIT in 28m 08s (non-voting)\n- keystonemiddleware-functional-ext-oauth2-tacker-client-secret-basic https://zuul.opendev.org/t/openstack/build/77200976531b4b69b1ce4acc1cf861a5 : FAILURE in 42m 27s (non-voting)\n- keystonemiddleware-functional-ext-oauth2-tacker-client-secret-post https://zuul.opendev.org/t/openstack/build/cc12b9b4e1bb445ab39cf8dc19de78cf : RETRY_LIMIT in 43m 37s (non-voting)\n- keystonemiddleware-functional-ext-oauth2-tacker-private-key-jwt https://zuul.opendev.org/t/openstack/build/e7b55e4485c64f56a5239e5c13d3ddfb : RETRY_LIMIT in 43m 49s (non-voting)\n- keystonemiddleware-functional-ext-oauth2-tacker-client-secret-jwt https://zuul.opendev.org/t/openstack/build/c199a71832164ecbadb1ca734156f79a : RETRY_LIMIT in 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"cafafd7fc5c3f35943475b6b6c88c3f19a194628","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-14 05:28:31.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"56a22278896dd5fa4495919666bf0aa36bcd36b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 07:09:06.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c668e9f1b3354090a449bb22190e72d0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e49234a49ccc4500aea8caff346e5e77 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c039428c67c45689dc887369312deef : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb78215e23cb4a7d9e55ef505584725d : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0dc878101858470983aef2a5f95eb76b : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6285f8f001bc4c64a91b273e9457d6aa : SUCCESS in 5m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ebddeaed27849f295288bc77e0c98d1 : SUCCESS in 1h 32m 53s\n- keystonemiddleware-functional-ext-oauth2-tacker-tls-client-auth https://zuul.opendev.org/t/openstack/build/f4d336f8190443eb90174916c655f729 : FAILURE in 30m 26s (non-voting)\n- keystonemiddleware-functional-ext-oauth2-tacker-client-secret-basic https://zuul.opendev.org/t/openstack/build/31862c1c0dde4157ad3d9ae83e588685 : FAILURE in 43m 50s (non-voting)\n- keystonemiddleware-functional-ext-oauth2-tacker-client-secret-post https://zuul.opendev.org/t/openstack/build/a68349a42918496792948b6d7d8e1ea5 : FAILURE in 32m 40s (non-voting)\n- keystonemiddleware-functional-ext-oauth2-tacker-private-key-jwt https://zuul.opendev.org/t/openstack/build/f1177e46ae0843fd87a17520b31517fb : FAILURE in 44m 36s (non-voting)\n- keystonemiddleware-functional-ext-oauth2-tacker-client-secret-jwt https://zuul.opendev.org/t/openstack/build/53c8c2e12b434631930d7d9716a49935 : FAILURE in 41m 57s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"5ab32cf89d27fd2a418085ea305cab8474fa9444","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-16 01:42:10.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"19e2c30e8341644610c7ff7a9125c74c220ffa09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 01:42:23.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d82bcb0708794ffca14b62c7aefaa6b1\n\n\nWarning:\n  Dependency cycle detected and project openstack/tacker doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":21},{"id":"962ce36917047595052ba98b2f68c085adeb5ce0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-16 01:53:30.000000000","message":"Uploaded patch set 22: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"42bd248602cd2dae3dcb7c05c8073813716f012b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 01:53:53.000000000","message":"Patch Set 22: Verified-1\n\n(6 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/keystonemiddleware on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027zuul_copy_output\u0027]\n\nThe problem appears in the the \"keystonemiddleware-functional-ext-oauth2-tacker-base\" job stanza:\n\n  job:\n      name: keystonemiddleware-functional-ext-oauth2-tacker-base\n      parent: devstack\n      description: |\n        Base job for keystonemiddleware external oauth2 functional tests for tacker\n      nodeset: openstack-single-node-jammy\n      required-projects:\n        - openstack/barbican\n        - openstack/heat\n  ...\n\n  in \"openstack/keystonemiddleware/.zuul.yaml@master\", line 1","accounts_in_message":[],"_revision_number":22},{"id":"54cc41fb9c03e6b5220dd5697f3c2f7012bfc4d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-16 01:57:47.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"2607d66bd81bc8a8630021f8c1fa57a10bc95210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 01:58:10.000000000","message":"Patch Set 23: Verified-1\n\n(6 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/keystonemiddleware on branch master.  The\nproblem was:\n\n  extra keys not allowed @ data[\u0027zuul_copy_output\u0027]\n\nThe problem appears in the the \"keystonemiddleware-functional-ext-oauth2-tacker-base\" job stanza:\n\n  job:\n      name: keystonemiddleware-functional-ext-oauth2-tacker-base\n      parent: devstack\n      description: |\n        Base job for keystonemiddleware external oauth2 functional tests for tacker\n      nodeset: openstack-single-node-jammy\n      required-projects:\n        - openstack/barbican\n        - openstack/heat\n  ...\n\n  in \"openstack/keystonemiddleware/.zuul.yaml@master\", line 1","accounts_in_message":[],"_revision_number":23},{"id":"e60a55562bdf9b16a4129cab1a6daadd924fbe6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-16 03:09:17.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"92f37fe9c49a644eccb81e37479cd270d3d1122a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-16 05:11:54.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"4990ecb2c72086a1ff69daf7102778b2526bbf65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-16 07:07:10.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"1f27eea625037321c9f0003f7c52a99d465bae0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 07:07:45.000000000","message":"Patch Set 26: Verified-1\n\n(6 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/keystonemiddleware on branch master.  The\nproblem was:\n\n  Unknown projects: review.opendev.org/openstack/tacker\n\nThe problem appears in the the \"keystonemiddleware-functional-ext-oauth2-tacker-base\" job stanza:\n\n  job:\n      name: keystonemiddleware-functional-ext-oauth2-tacker-base\n      parent: devstack\n      description: |\n        Base job for keystonemiddleware external oauth2 functional tests for tacker\n      nodeset: openstack-single-node-jammy\n      required-projects:\n        - openstack/barbican\n        - openstack/heat\n  ...\n\n  in \"openstack/keystonemiddleware/.zuul.yaml@master\", line 1","accounts_in_message":[],"_revision_number":26},{"id":"b0356107e4f601ff65e7d681d771ad1e41d73e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-16 07:23:32.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"adae6f631da5095d66970e4225b2de5ca1e2ea4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 09:28:43.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c64f77c1423d4f1f870d717a4787eda5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b33da76526194daabf2569d44ba57bfa : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91dd0b11b6da49309a2c1f297ac74969 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50152bf043014f32afedc5fa27c0aa44 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c267b3a2bcd46cfa548dc5080a1a5cd : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb7123f542d54701bb544eee0df7129c : SUCCESS in 5m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47906f41052a4528b0958579434eba6e : SUCCESS in 1h 37m 32s\n- keystonemiddleware-functional-ext-oauth2-tacker-tls-client-auth https://zuul.opendev.org/t/openstack/build/e7dcfa7a11024316a680e5dbab3bb1f5 : RETRY_LIMIT in 39m 42s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"7c0daa9210af2bf3ee5d253de748fbc138b3c59c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-11-28 02:39:19.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"7ea53442521112ce8b54d740d7b4765a28512069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 08:40:59.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afb5834c90cd490ea414b4fbec678760\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/580c3e40a9c144f3b800ec090a294630 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc32050b3b94dfab1836c67fd299e4d : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9c7f71783fc43c19a11ff00b42a8e8e : SUCCESS in 2m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20a9c486cd2049b3be24cb35fb2e4160 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f4a88b8a1f2431abb1f87e35d2e20e5 : SUCCESS in 4m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e846cfc1d07473aa931ab7dd4501039 : SUCCESS in 1h 29m 17s\n- tacker-functional-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/59a5d40108614be19407104b1875e315 : SUCCESS in 1h 14m 19s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"5ecc827e6b73ea5266a335c534fbaa73aa879a63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-12-07 01:13:48.000000000","message":"Patch Set 29: Patch Set 28 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"34b8e92120a5dd6175f0b2ecd250e9893a47abd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 02:34:21.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d547162d679f4b1ebdd63096e4ae7e9e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae2ab86abeb543298f385334adf0a799 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7d3aaa75aa4c90bba6516c97f1c627 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35364992b1c94f3d95bd6a14250e16a4 : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/066ce6696f834568a805491e00090eb9 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bcfc1e6c38c4908b5b6aec8fc7b286f : POST_FAILURE in 35m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a672d6df0f754cb1a4744a7b9a6e42d8 : SUCCESS in 1h 17m 14s\n- tacker-functional-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/93f671fe45f94272b66a6757fa610b75 : FAILURE in 1h 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"911e41656fa0e6e8003ff4ea73812fb2285fe199","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-12-07 02:52:40.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"35d23ef254e780603c0553bdf895e81047102974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 04:50:48.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e98b2b759b1445598322bf92db86e347\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3612b51321f845378ab8f9ac79bfe0ff : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/960772a124bd42b48845857ddd976a06 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/153618defd24400aa9e5bc22648dfe08 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cf72cca2154143429e55607865944206 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6cbaf444e0d4f4daf7a3076379805e5 : SUCCESS in 6m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6badc8fe0bb549d391ee89b048c1af81 : SUCCESS in 1h 28m 48s\n- tacker-functional-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/625b9dc6ba0e4f0bbf02da1583ae4076 : SUCCESS in 1h 23m 34s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"dcd9cb4e8adbc9e5cc8d2d6c387fddcf90710cfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-12-07 08:24:31.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"bd528c0a9eb66f831f45d4d9b82a3b4bdf65be2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 09:40:36.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d431c92f5b64ad49c9962b13f7d82fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07bfd092e8c24109b8edac5a2e51264e : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71532882a49443c985c4f2cc35367888 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e61e260c76284839aee46503bd3a7c99 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05abb8efa59b4f5bb86bbeb62025f116 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/affbcb4361614fbf8ce7e998c30c349f : SUCCESS in 5m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fcdd790309a488fab4295247ccdaf96 : SUCCESS in 1h 15m 01s\n- tacker-functional-devstack-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/4e81c1797e994987853ef644e92b120c : SUCCESS in 1h 03m 37s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"bba082f23f5403cda0401bb29839346e98a1388b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-12-13 01:34:45.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"26bbe5f07f1ead58563260fbbbb2865d8564002a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 02:52:44.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b529adbb11346fa933988edbf820858\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96683071725e4d149b49201c49f24d4b : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d78ec4251fe44953b40388c34c90aedd : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb7ee842fb444de7b03019d3b955b44b : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ea241c52ac74f13ade7eafda4b13992 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b8b84bc0d64966957927d1d26e5501 : SUCCESS in 5m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2309799b01af430d864c58e376076a3d : FAILURE in 1h 11m 21s\n- tacker-functional-devstack-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/06e15ba51f394663a4fcf02688f63b87 : SUCCESS in 1h 11m 06s (non-voting)\n- barbican-functional-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/308672b4b3fb46088a4bbfe562a367e7 : SUCCESS in 44m 22s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"a8e3e98dab4d4486300b4c03c1750477d87266e1","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-12-13 04:41:47.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"164203561fd171181b046ac245aac34391f15a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 05:00:27.000000000","message":"Patch Set 31:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/903269 is needed.","accounts_in_message":[],"_revision_number":31},{"id":"4566f407a3612506a2b6d4d5af58ee102c85b4c2","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-12-13 05:29:24.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"64b5ef6693330dd12a8bd5ec9182899c53829932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 06:45:50.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ac9b9832414501bf1dea18cd5897b1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3931f56c8bf41978b30076eefdba261 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb4d8c73d85643219eef4eaf7df4a838 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d74087c4349427481cdfb8d7bd9413c : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95941d0520a84247967ed0cddc9a0166 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73de136c60084e618d62416918931a34 : SUCCESS in 7m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cdc7d278bbbf4a849854a7071a0c6fb3 : SUCCESS in 1h 06m 54s\n- tacker-functional-devstack-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/15169028e0264caf948a541b2409ffe2 : SUCCESS in 1h 04m 46s (non-voting)\n- barbican-functional-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/7cc540f6bad5424996cad6b80680b1fa : SUCCESS in 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"c0f803fb8cb3760b8429d103224219d0bede5110","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-12-18 01:50:25.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"7db25d3f1d2b01d3946b0202f92b3347490c3fd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 03:22:18.000000000","message":"Patch Set 31:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a94acfdbf5b141f6af1e4b50afb17553\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfa1a411630a4d45a03ae1f30b150afd : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81a6b44507344115a39e93e0a59eac95 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e5c509d5b264228bbbdbc8d150776cf : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94c598bea898430c9265c22a3d3f1792 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cd26096fda24bcd9c83464982dd07e0 : SUCCESS in 7m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d9e79189e9ea495ebc1a455ab6e8a80c : SUCCESS in 1h 25m 20s\n- tacker-functional-devstack-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/d248d44f8ae54eccb5b4b01d7587124e : SUCCESS in 55m 18s (non-voting)\n- barbican-functional-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/6d6ccca9476e4769ba4a42546a07d5bd : SUCCESS in 37m 17s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"b5afd59c6fa2d21bc1ca3befd93aee9de9f7e396","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"date":"2023-12-27 07:57:56.000000000","message":"Patch Set 32: Patch Set 31 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"e001a5e3495e0b879f564c24ca5fd3476010d4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-27 09:08:31.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b34adc1a44374ef884d7ee99ab30a147\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57df0f63743848acad3b9fd7f110dd66 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05661ead62e94ccf9ed4ce24d35dbb0a : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1b212e636d8417197df38487b493b29 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b418171bdcfd4094bba86db623db5889 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ecb73dd972449cba3d422fd68bd855d : SUCCESS in 5m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ada6f77eebf943f7a61c649a4ae27946 : SUCCESS in 1h 05m 35s\n- tacker-functional-devstack-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/189a1995b61a4a4bbf052594bf11ac6b : SUCCESS in 1h 04m 38s (non-voting)\n- barbican-functional-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/a03dbb2b772d4842bb24069f38e552d8 : SUCCESS in 50m 07s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"c6bc21ddd4aa478240f8719c74985b73cdbab6f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2024-02-16 04:15:22.000000000","message":"Patch Set 33: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"1d2833839174b9d1ac1f48b0cd52b5b854f568e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 05:52:00.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed9395fc8c0c4d0f8432f27d4e41b1b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c2d6a72bc9e4fe390b89e8f7deefd4a : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62f3a08037f4c2f88b48cc58d7b63d3 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d09e8e6c3854b51ae9175e9dd1b5660 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/042ba25779864ee3abc45d8c08e40723 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/753fb221418b4be590d48399496a6509 : SUCCESS in 6m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68a09d9e19fd4f578876bfbe072c494f : SUCCESS in 1h 30m 18s\n- tacker-functional-devstack-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/bb57a280688b412d87e07b13ec68f76e : FAILURE in 45m 00s (non-voting)\n- barbican-functional-keystonemiddleware-ext-oauth2 https://zuul.opendev.org/t/openstack/build/e1c833877fa141789c69a67df347173a : SUCCESS in 38m 46s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"c10218e287a9ac8eb45031594b875d10f15a1490","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-01-15 15:09:12.000000000","message":"Patch Set 33: Code-Review+2","accounts_in_message":[],"_revision_number":33},{"id":"8bc2d4e235db5381447418cddccafbb49936f332","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-01-17 12:29:25.000000000","message":"Patch Set 33: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":33}],"current_revision_number":33,"current_revision":"f9e4fb837c7b65a862fd47d4f1b994f4e4afcd46","revisions":{"1f69f1210eb1ba8dfcec095b9f5c41c7e2537069":{"kind":"REWORK","_number":1,"created":"2023-11-02 05:13:10.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/1"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:15:06.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/1f69f1210eb1ba8dfcec095b9f5c41c7e2537069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/1f69f1210eb1ba8dfcec095b9f5c41c7e2537069"}]},"branch":"refs/heads/master"},"3b990dd5316dc42102baf56ac19ad0886e8e64d2":{"kind":"REWORK","_number":2,"created":"2023-11-02 05:24:16.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/2"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:26:06.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3b990dd5316dc42102baf56ac19ad0886e8e64d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3b990dd5316dc42102baf56ac19ad0886e8e64d2"}]},"branch":"refs/heads/master"},"3d60e033be2994e83e9c1ebe8db8c3edb63da0d2":{"kind":"REWORK","_number":3,"created":"2023-11-03 01:12:33.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/3"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-03 01:14:13.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3d60e033be2994e83e9c1ebe8db8c3edb63da0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3d60e033be2994e83e9c1ebe8db8c3edb63da0d2"}]},"branch":"refs/heads/master"},"61cf2f5aaa78410ecaf5632e67e29a981012264c":{"kind":"REWORK","_number":4,"created":"2023-11-03 01:15:42.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/4"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-03 01:17:32.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/61cf2f5aaa78410ecaf5632e67e29a981012264c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/61cf2f5aaa78410ecaf5632e67e29a981012264c"}]},"branch":"refs/heads/master"},"fe2a66c6434b699976f920907a1427b615782dc5":{"kind":"REWORK","_number":5,"created":"2023-11-03 01:55:04.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/5"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-03 01:56:50.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/fe2a66c6434b699976f920907a1427b615782dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/fe2a66c6434b699976f920907a1427b615782dc5"}]},"branch":"refs/heads/master"},"8d400779c9e9e80e584969344ec692aadd38a9ec":{"kind":"REWORK","_number":6,"created":"2023-11-03 03:14:36.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/6"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-03 03:16:16.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/8d400779c9e9e80e584969344ec692aadd38a9ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/8d400779c9e9e80e584969344ec692aadd38a9ec"}]},"branch":"refs/heads/master"},"fa0808ec74b79d707501454a0841ed178424591c":{"kind":"REWORK","_number":7,"created":"2023-11-03 06:16:59.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/7"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-03 06:18:44.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/fa0808ec74b79d707501454a0841ed178424591c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/fa0808ec74b79d707501454a0841ed178424591c"}]},"branch":"refs/heads/master"},"a179bd605c686747f663f4801555130e553e2938":{"kind":"REWORK","_number":8,"created":"2023-11-03 07:24:55.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/8"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-03 07:26:52.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a179bd605c686747f663f4801555130e553e2938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a179bd605c686747f663f4801555130e553e2938"}]},"branch":"refs/heads/master"},"abde501616082d3c8b61a77a6fe623d53726ca10":{"kind":"REWORK","_number":9,"created":"2023-11-06 07:13:45.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/9"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-06 07:15:38.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/abde501616082d3c8b61a77a6fe623d53726ca10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/abde501616082d3c8b61a77a6fe623d53726ca10"}]},"branch":"refs/heads/master"},"db377da58891ac8cfd20d54e4d6cd948764903f7":{"kind":"REWORK","_number":10,"created":"2023-11-07 04:47:23.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/10"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-07 04:49:18.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/db377da58891ac8cfd20d54e4d6cd948764903f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/db377da58891ac8cfd20d54e4d6cd948764903f7"}]},"branch":"refs/heads/master"},"5c2e910783e183526451a8dae15f599790ba0faf":{"kind":"REWORK","_number":11,"created":"2023-11-09 00:45:20.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/11"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-09 00:46:36.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5c2e910783e183526451a8dae15f599790ba0faf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5c2e910783e183526451a8dae15f599790ba0faf"}]},"branch":"refs/heads/master"},"a07d211c4eb01e54887b32aa3dc81b92ca45dd41":{"kind":"REWORK","_number":12,"created":"2023-11-09 01:13:43.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/12"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-09 01:15:39.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a07d211c4eb01e54887b32aa3dc81b92ca45dd41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a07d211c4eb01e54887b32aa3dc81b92ca45dd41"}]},"branch":"refs/heads/master"},"0e9133a0fa3236a0b4230fc510d5d9160a6693c6":{"kind":"REWORK","_number":13,"created":"2023-11-09 03:07:27.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/13"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-09 03:09:16.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0e9133a0fa3236a0b4230fc510d5d9160a6693c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0e9133a0fa3236a0b4230fc510d5d9160a6693c6"}]},"branch":"refs/heads/master"},"e0f3e947c4f60613fabf27a44b85f87b5753a3d9":{"kind":"REWORK","_number":14,"created":"2023-11-09 06:07:58.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/14"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-09 06:09:56.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e0f3e947c4f60613fabf27a44b85f87b5753a3d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e0f3e947c4f60613fabf27a44b85f87b5753a3d9"}]},"branch":"refs/heads/master"},"d0a1674fb791b07e6ef9507a34dd3856b87a6cab":{"kind":"REWORK","_number":15,"created":"2023-11-09 07:11:36.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/15"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-09 07:13:31.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d0a1674fb791b07e6ef9507a34dd3856b87a6cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d0a1674fb791b07e6ef9507a34dd3856b87a6cab"}]},"branch":"refs/heads/master"},"6744cb92c464c7957596951ff2526a4b2ad69a8f":{"kind":"REWORK","_number":16,"created":"2023-11-09 08:16:54.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/16"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-09 08:18:51.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6744cb92c464c7957596951ff2526a4b2ad69a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6744cb92c464c7957596951ff2526a4b2ad69a8f"}]},"branch":"refs/heads/master"},"b2322f1661c7568d14c14a21f60f46cba00b11bf":{"kind":"REWORK","_number":17,"created":"2023-11-10 07:54:04.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/17"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-10 07:55:26.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b2322f1661c7568d14c14a21f60f46cba00b11bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b2322f1661c7568d14c14a21f60f46cba00b11bf"}]},"branch":"refs/heads/master"},"ee22aafdca4f8a7438961464fd068a98abc6e71b":{"kind":"REWORK","_number":18,"created":"2023-11-13 05:06:09.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/18"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-13 05:07:46.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ee22aafdca4f8a7438961464fd068a98abc6e71b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ee22aafdca4f8a7438961464fd068a98abc6e71b"}]},"branch":"refs/heads/master"},"f1eebffd026cca9ab7b421c9398f0cae0e8202ca":{"kind":"REWORK","_number":19,"created":"2023-11-13 06:47:40.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/19"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-13 06:49:44.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f1eebffd026cca9ab7b421c9398f0cae0e8202ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f1eebffd026cca9ab7b421c9398f0cae0e8202ca"}]},"branch":"refs/heads/master"},"939cf4e7cc9a16e2bf48c9c4115338ef1bbe1324":{"kind":"REWORK","_number":20,"created":"2023-11-14 05:28:31.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/20"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-14 05:30:04.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/939cf4e7cc9a16e2bf48c9c4115338ef1bbe1324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/939cf4e7cc9a16e2bf48c9c4115338ef1bbe1324"}]},"branch":"refs/heads/master"},"6a833643d5a13876db8165a136c8ee5b80b052ef":{"kind":"REWORK","_number":21,"created":"2023-11-16 01:42:10.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/21"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-16 01:43:21.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nDepends-On: https://review.opendev.org/c/openstack/tacker/+/900971\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6a833643d5a13876db8165a136c8ee5b80b052ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6a833643d5a13876db8165a136c8ee5b80b052ef"}]},"branch":"refs/heads/master"},"99742ed2f118a03982742d2a6e35dcd889e0b357":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2023-11-16 01:53:30.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/22"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-16 01:55:29.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/99742ed2f118a03982742d2a6e35dcd889e0b357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/99742ed2f118a03982742d2a6e35dcd889e0b357"}]},"branch":"refs/heads/master"},"146483e3c929f2187db3a441a738cbe80d225fd0":{"kind":"REWORK","_number":23,"created":"2023-11-16 01:57:47.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/23"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-16 01:59:00.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/146483e3c929f2187db3a441a738cbe80d225fd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/146483e3c929f2187db3a441a738cbe80d225fd0"}]},"branch":"refs/heads/master"},"d793e9f09b66cfdd3253243513114dd0f25e36e6":{"kind":"REWORK","_number":24,"created":"2023-11-16 03:09:17.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/24"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-16 03:11:23.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d793e9f09b66cfdd3253243513114dd0f25e36e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d793e9f09b66cfdd3253243513114dd0f25e36e6"}]},"branch":"refs/heads/master"},"c650b6130e5e13061c986708d59be42df6548ee2":{"kind":"REWORK","_number":25,"created":"2023-11-16 05:11:54.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/25"}}},"commit":{"parents":[{"commit":"53b4cb21ad1100cb50c1ea3e49580068183ddddf","subject":"External OAuth2.0 Authorization Server Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53b4cb21ad1100cb50c1ea3e49580068183ddddf"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-16 05:13:26.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c650b6130e5e13061c986708d59be42df6548ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c650b6130e5e13061c986708d59be42df6548ee2"}]},"branch":"refs/heads/master"},"06d48f9a8d8aed873558a87bfdbe796861a78468":{"kind":"REWORK","_number":26,"created":"2023-11-16 07:07:10.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/26"}}},"commit":{"parents":[{"commit":"9ade074c9b665926e939c9b6b45bebb1e307b2f1","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9ade074c9b665926e939c9b6b45bebb1e307b2f1"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-16 07:08:49.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/06d48f9a8d8aed873558a87bfdbe796861a78468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/06d48f9a8d8aed873558a87bfdbe796861a78468"}]},"branch":"refs/heads/master"},"b59bc1d63b27adadedea13eb58558e21b0dc5e57":{"kind":"REWORK","_number":27,"created":"2023-11-16 07:23:32.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/27"}}},"commit":{"parents":[{"commit":"9ade074c9b665926e939c9b6b45bebb1e307b2f1","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9ade074c9b665926e939c9b6b45bebb1e307b2f1"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-16 07:25:33.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b59bc1d63b27adadedea13eb58558e21b0dc5e57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b59bc1d63b27adadedea13eb58558e21b0dc5e57"}]},"branch":"refs/heads/master"},"d217d84c2653e0a2f3913678c05bb14b047f0903":{"kind":"REWORK","_number":28,"created":"2023-11-28 02:39:19.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/28"}}},"commit":{"parents":[{"commit":"9ade074c9b665926e939c9b6b45bebb1e307b2f1","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9ade074c9b665926e939c9b6b45bebb1e307b2f1"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-28 02:39:26.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nDepends-On: https://review.opendev.org/c/openstack/tacker/+/900971\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d217d84c2653e0a2f3913678c05bb14b047f0903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d217d84c2653e0a2f3913678c05bb14b047f0903"}]},"branch":"refs/heads/master"},"58b0d8c7500a5cfa379e341fdde93e9244d52e7b":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2023-12-07 01:13:48.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/29"}}},"commit":{"parents":[{"commit":"9073ee4c34f6ba5a7c8b4fd6e3ade2dd90937b3d","subject":"Merge \"Update master for stable/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9073ee4c34f6ba5a7c8b4fd6e3ade2dd90937b3d"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2023-12-07 01:13:48.000000000","tz":0},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nDepends-On: https://review.opendev.org/c/openstack/tacker/+/900971\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/58b0d8c7500a5cfa379e341fdde93e9244d52e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/58b0d8c7500a5cfa379e341fdde93e9244d52e7b"}]},"branch":"refs/heads/master","description":"Rebase"},"96f45c22e227a2ee400d149807aaf5fdb4a996b1":{"kind":"REWORK","_number":30,"created":"2023-12-07 08:24:31.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/30"}}},"commit":{"parents":[{"commit":"9ade074c9b665926e939c9b6b45bebb1e307b2f1","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9ade074c9b665926e939c9b6b45bebb1e307b2f1"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-12-07 08:26:21.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nDepends-On: https://review.opendev.org/c/openstack/tacker/+/900971\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/96f45c22e227a2ee400d149807aaf5fdb4a996b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/96f45c22e227a2ee400d149807aaf5fdb4a996b1"}]},"branch":"refs/heads/master"},"bce2d99ae5fedbdf26a4eb0df01182a2082652f6":{"kind":"REWORK","_number":31,"created":"2023-12-13 01:34:45.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/31"}}},"commit":{"parents":[{"commit":"9ade074c9b665926e939c9b6b45bebb1e307b2f1","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9ade074c9b665926e939c9b6b45bebb1e307b2f1"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-12-13 01:36:12.000000000","tz":540},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nDepends-On: https://review.opendev.org/c/openstack/tacker/+/900971\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/903269\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/bce2d99ae5fedbdf26a4eb0df01182a2082652f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/bce2d99ae5fedbdf26a4eb0df01182a2082652f6"}]},"branch":"refs/heads/master"},"f6ccf047b595afcbaa3019ce6ef8fe3051997d4c":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2023-12-27 07:57:56.000000000","uploader":{"_account_id":33492,"name":"Yi Feng","email":"fengyi@fujitsu.com","username":"YiFeng"},"ref":"refs/changes/11/899911/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/32"}}},"commit":{"parents":[{"commit":"9073ee4c34f6ba5a7c8b4fd6e3ade2dd90937b3d","subject":"Merge \"Update master for stable/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9073ee4c34f6ba5a7c8b4fd6e3ade2dd90937b3d"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"Yi Feng","email":"fengyi@fujitsu.com","date":"2023-12-27 07:57:56.000000000","tz":0},"subject":"[WIP] Add FT for External OAuth2.0 Server Support","message":"[WIP] Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nDepends-On: https://review.opendev.org/c/openstack/tacker/+/900971\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/903269\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f6ccf047b595afcbaa3019ce6ef8fe3051997d4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f6ccf047b595afcbaa3019ce6ef8fe3051997d4c"}]},"branch":"refs/heads/master","description":"Rebase"},"f9e4fb837c7b65a862fd47d4f1b994f4e4afcd46":{"kind":"NO_CODE_CHANGE","_number":33,"created":"2024-02-16 04:15:22.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/11/899911/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/899911/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/899911/33"}}},"commit":{"parents":[{"commit":"9073ee4c34f6ba5a7c8b4fd6e3ade2dd90937b3d","subject":"Merge \"Update master for stable/zed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9073ee4c34f6ba5a7c8b4fd6e3ade2dd90937b3d"}]}],"author":{"name":"YiFeng","email":"fengyi@fujitsu.com","date":"2023-11-02 05:10:21.000000000","tz":540},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2024-02-16 04:15:22.000000000","tz":0},"subject":"Add FT for External OAuth2.0 Server Support","message":"Add FT for External OAuth2.0 Server Support\n\nTwo automated test jobs have been added to verify the external\nOAuth 2.0 authentication function through the test code of the\nbarbican/tacker service.\n\nDepends-On: https://review.opendev.org/c/openstack/tacker/+/900971\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/903269\nImplements: blueprint enhance-oauth2-interoperability\nChange-Id: Iac1e07291ab36e8c2ac33cee9a2ac96cb0a0c4c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f9e4fb837c7b65a862fd47d4f1b994f4e4afcd46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f9e4fb837c7b65a862fd47d4f1b994f4e4afcd46"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
