)]}'
{"id":"openstack%2Fkeystonemiddleware~926457","triplet_id":"openstack%2Fkeystonemiddleware~master~I84e21a63b15ef632f55ce30cfdb516a8ab0fac7a","project":"openstack/keystonemiddleware","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-16 17:51:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I84e21a63b15ef632f55ce30cfdb516a8ab0fac7a","subject":"Remove deprecated AuthTokenPlugin","status":"ABANDONED","created":"2024-08-16 03:36:37.000000000","updated":"2025-11-16 17:51:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a0033799b8f3b352aa95c750d46505f09affe75e","_number":926457,"virtual_id_number":926457,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-09 15:29:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-16 05:16:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3704d816ceda8b38f95490616f2bc3a38c5f99cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-16 03:36:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2152ed2f1492a6cd7660d6c3eb0731d29deb6b69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 05:16:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb04b153703d405bb0100f91ef6e1802\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/098b6f8322d0440fa8cf065e9292fab4 : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6f6f37fd6546fd89a9b22eda07aee3 : FAILURE in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3921439ce2d246f38c8bd597aa814168 : FAILURE in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89ae763fa3e54ea0b4d0691c80f3496a : FAILURE in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4c65522858f4f3da373b9f040800b55 : FAILURE in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c174e5227645c780441a63d453122e : SUCCESS in 6m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be6e419a83054028927601ff2cee86a0 : FAILURE in 1h 38m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b9141f19bd745e3b806a625387e30eb : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ea75cd7bc4d80751b8b967416eeabf68fdb298ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-16 07:50:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4be823bc77d8fab9f7ec0a1e43592f16f7484ef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-16 08:16:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"14a234a1f0780ed38d7ecf5ef51a309991b687b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 09:51:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59b84a7f42024000953e217c6a9c8be5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61f0a071aa094bc7a2b530172602be91 : FAILURE in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6bf8ff533a24e8bb2c9f648a5d9ea36 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d29c1398879248eab560ed5456b039b2 : FAILURE in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/232c566a573d4696bf836a2cbc3df008 : FAILURE in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43a1e2ebc8c04229bcfddce19f57d2d7 : FAILURE in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d065408b974b5ba3326b05b48d3d53 : SUCCESS in 5m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15e1942c0636471aa21f6076b54039f4 : SUCCESS in 1h 34m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f7f19668284437cbc23d86a1ba2c43c : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"2f3239e131f7a8864c8894861192f3f01c04051f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-12-09 13:08:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"31ff5f3e7f680b09a31c01a650700f514c2d1519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-09 15:29:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/834232d4d30e43fc99cfc56de697fe74\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f71c6cf73a664480804ace29087e3348 : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/799451638654477d8602d0e266830974 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f05d59b558140049af29aa2d21d0a5d : FAILURE in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f936bbb0f7b9457aa70b445066c6fb14 : FAILURE in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1160e0b22c4621896a575e6018ca82 : SUCCESS in 5m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e49db47f92945e09849f13d0af85536 : SUCCESS in 1h 38m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70289339a6a14fd3a0a15567dd2ce66f : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"a0033799b8f3b352aa95c750d46505f09affe75e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-16 17:51:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"970d9b2c3d2255d8d9b2548d9c6ff9d81c038981","revisions":{"2c191d07f348f94500e963db46796196dbe9d03c":{"kind":"REWORK","_number":1,"created":"2024-08-16 03:36:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/926457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/57/926457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/1"}}},"commit":{"parents":[{"commit":"ebb842abf01745be4dbd280764faee6d32b2a76b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ebb842abf01745be4dbd280764faee6d32b2a76b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-16 03:28:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-16 03:36:08.000000000","tz":540},"subject":"Remove deprecated AuthTokenPlugin","message":"Remove deprecated AuthTokenPlugin\n\nIt was deprecated during Mitaka cycle so can be removed now.\n\nStill WIP\n\nChange-Id: I84e21a63b15ef632f55ce30cfdb516a8ab0fac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2c191d07f348f94500e963db46796196dbe9d03c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2c191d07f348f94500e963db46796196dbe9d03c"}]},"branch":"refs/heads/master"},"d5d656a03bbb28903623682e43cd564a4f543de5":{"kind":"REWORK","_number":2,"created":"2024-08-16 07:50:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/926457/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/57/926457/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/2"}}},"commit":{"parents":[{"commit":"ebb842abf01745be4dbd280764faee6d32b2a76b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ebb842abf01745be4dbd280764faee6d32b2a76b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-16 03:28:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-16 07:48:13.000000000","tz":540},"subject":"Remove deprecated AuthTokenPlugin","message":"Remove deprecated AuthTokenPlugin\n\nIt was deprecated during Mitaka cycle so can be removed now.\n\nStill WIP\n\nChange-Id: I84e21a63b15ef632f55ce30cfdb516a8ab0fac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d5d656a03bbb28903623682e43cd564a4f543de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d5d656a03bbb28903623682e43cd564a4f543de5"}]},"branch":"refs/heads/master"},"99ce2e7f3962a7b909656fd33dc8220f40d12cd7":{"kind":"REWORK","_number":3,"created":"2024-08-16 08:16:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/926457/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/57/926457/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/3"}}},"commit":{"parents":[{"commit":"ebb842abf01745be4dbd280764faee6d32b2a76b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ebb842abf01745be4dbd280764faee6d32b2a76b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-16 03:28:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-16 08:16:24.000000000","tz":540},"subject":"Remove deprecated AuthTokenPlugin","message":"Remove deprecated AuthTokenPlugin\n\nIt was deprecated during Mitaka cycle so can be removed now.\n\nStill WIP\n\nChange-Id: I84e21a63b15ef632f55ce30cfdb516a8ab0fac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/99ce2e7f3962a7b909656fd33dc8220f40d12cd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/99ce2e7f3962a7b909656fd33dc8220f40d12cd7"}]},"branch":"refs/heads/master"},"970d9b2c3d2255d8d9b2548d9c6ff9d81c038981":{"kind":"REWORK","_number":4,"created":"2024-12-09 13:08:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/926457/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/57/926457/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/57/926457/4"}}},"commit":{"parents":[{"commit":"5c86dc240818bd69cabc506cc6e08b2df1015cd5","subject":"Merge \"Use oslo.utils to escape IPv6 address\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5c86dc240818bd69cabc506cc6e08b2df1015cd5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-09 13:08:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-12-09 13:08:12.000000000","tz":540},"subject":"Remove deprecated AuthTokenPlugin","message":"Remove deprecated AuthTokenPlugin\n\nIt was deprecated during Mitaka cycle so can be removed now.\n\nStill WIP\n\nChange-Id: I84e21a63b15ef632f55ce30cfdb516a8ab0fac7a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/970d9b2c3d2255d8d9b2548d9c6ff9d81c038981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/970d9b2c3d2255d8d9b2548d9c6ff9d81c038981"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
