)]}'
{"id":"openstack%2Fkeystonemiddleware~938290","triplet_id":"openstack%2Fkeystonemiddleware~master~I08a94cb31ecb230ed6e6310e4fe33f9f57323ce0","project":"openstack/keystonemiddleware","branch":"master","topic":"fix-memcached-dependencies-doc-bug","attention_set":{},"removed_from_attention_set":{"37432":{"account":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"last_update":"2025-01-29 18:32:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I08a94cb31ecb230ed6e6310e4fe33f9f57323ce0","subject":"Fix memcached dependencies\u0027s doc bug","status":"MERGED","created":"2024-12-27 15:19:07.000000000","updated":"2025-01-29 18:34:24.000000000","submitted":"2025-01-29 18:32:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938290-fix-memcached-dependencies-doc-bug","meta_rev_id":"04a6e240d0295271d8a3f6b2a4fb6d81838af1b1","_number":938290,"virtual_id_number":938290,"owner":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-29 18:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-01-17 14:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-01-29 18:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-24 03:16:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2025-01-29 18:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-27 15:27:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-17 14:56:46.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-01-24 03:16:33.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2025-01-29 18:25:03.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"5e4c083b62914c36bdbac6b325beeb3b2d866b82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2024-12-27 15:19:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec121bd0f5fb23227b4e1afa99ddec3e69d9ce07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-27 15:27:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a45a8da5e3b74030b2fe108fa031391a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c72e5427c7046f2a1d6b71f3787ab66 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6abc0e4682e4f5da8fffb2dfb8ec36d : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3db6cc4ea1cd0c3031f36b0056ecda8549434299","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-01-17 14:56:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6ceee974160fd76d960de787ffae266b4fb2967","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-01-24 03:16:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"92efaf085e4c6850966e41c479f3e3576321aed7","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-29 18:25:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d60c371b079528e1a318e67e515b333fb3fcc848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 18:25:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95a77d38faff4991a8be03b4b1164dc4431cb6b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 18:32:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec51f3a005844b68e143cf3f36c8e49\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0793ff151644f509f9acac7c27385af : SUCCESS in 2m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08be193281344971b6b37fe5dfa15150 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e7c37921581de3d7b489461d30d267b9c96f0f3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 18:32:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"04a6e240d0295271d8a3f6b2a4fb6d81838af1b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 18:34:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7608715b75a14023937b2b4b98a4b59f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e5ce037d6c24a3998765578e03628ba : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef54c0ed83ea33261f1872e27753a76fe8552916","revisions":{"ef54c0ed83ea33261f1872e27753a76fe8552916":{"kind":"REWORK","_number":1,"created":"2024-12-27 15:19:07.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/90/938290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/90/938290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/90/938290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/90/938290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/90/938290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/90/938290/1"}}},"commit":{"parents":[{"commit":"5c86dc240818bd69cabc506cc6e08b2df1015cd5","subject":"Merge \"Use oslo.utils to escape IPv6 address\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5c86dc240818bd69cabc506cc6e08b2df1015cd5"}]}],"author":{"name":"Watanabe Koya","email":"koywatan@yahoo-corp.jp","date":"2024-12-27 15:11:32.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2024-12-27 15:13:25.000000000","tz":540},"subject":"Fix memcached dependencies\u0027s doc bug","message":"Fix memcached dependencies\u0027s doc bug\n\nThe library has been switched from pycrypto to cryptography.\nThe document was outdated, so it should be updated.\n\nChange-Id: I08a94cb31ecb230ed6e6310e4fe33f9f57323ce0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ef54c0ed83ea33261f1872e27753a76fe8552916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ef54c0ed83ea33261f1872e27753a76fe8552916"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
