)]}'
{"id":"openstack%2Fkeystonemiddleware~940278","triplet_id":"openstack%2Fkeystonemiddleware~master~Ie9a1f273019a1f0e7a4573e1112bbfd4330690db","project":"openstack/keystonemiddleware","branch":"master","topic":"add-cryptography-package","attention_set":{},"removed_from_attention_set":{"37432":{"account":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"last_update":"2025-12-16 13:45:26.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-11-24 09:04:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8878":{"account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"last_update":"2025-11-24 04:59:46.000000000","reason":"\u003cGERRIT_ACCOUNT_8878\u003e replied on the change","reason_account":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}}},"hashtags":[],"change_id":"Ie9a1f273019a1f0e7a4573e1112bbfd4330690db","subject":"Add cryptography package as an optional dependency","status":"MERGED","created":"2025-01-28 01:58:05.000000000","updated":"2025-12-16 13:46:30.000000000","submitted":"2025-12-16 13:45:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940278-add-cryptography-package","meta_rev_id":"09f7dae51b4eb22d3e0e5258bb0d64e5fb9cddaf","_number":940278,"virtual_id_number":940278,"owner":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-16 13:45:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-11-24 09:04:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-12-16 12:28:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-12-10 08:35:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":1,"date":"2025-11-24 04:59:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-12-16 12:28:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-28 03:27:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-01 16:14:44.000000000","updated_by":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"reviewer":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"state":"REVIEWER"},{"updated":"2025-11-16 13:13:10.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-12-10 08:35:19.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-12-16 12:28:23.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"7f1fc27e8dfb1a41a850249b8b36e1414feb4c37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-01-28 01:58:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1a88470dd06fc258aa0a0103754618b669ef5f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 03:27:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9738475582a4ffea589e26483b31fab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fc6eb44c0124a8e8cc4c0422da0edae : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5247586e0304332bbfbdaf9b9fe96ad : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f323096ace1c4bb8b57eef23a96dd247 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09cd8a7bb58049c8a4b23291e54f77c0 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0680b4a50248ba908c4280a071d785 : SUCCESS in 4m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6da256821499414baebebcab755d5e56 : SUCCESS in 1m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fff0c2e20cde4af5be797b31d8deeb84 : SUCCESS in 1h 28m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ed4fa8e31eccdae1aeb7cdb25f96072a68965d58","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-04-01 16:14:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a9f0b914551c0eb292a5906722168e5b730a45e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-16 13:13:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9a9c786eae0f79f2b5d1d2b6e5b93435caadbbc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-16 13:14:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"594279429d9e02459063dad50018df6b4e3265dd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-17 02:50:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2ce8238b96b7155a8497fd08bb0480c2757c2d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-11-22 07:50:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"eb41c8f99211d35c0ea0c95374b9ea249a2d1812","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-11-22 07:58:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3d2e2db4e2e28bb669cce2b9594cce3f9ca945e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-22 09:07:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afca76efc84645069985aa06d4716172\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e88591de4d140a391db535790aaabe1 : SUCCESS in 1m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5882f00ea21444583f6d05f42bfa324 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba29b94b1094418db73d798fa2e4f124 : SUCCESS in 2m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94b75ba3dddd41f78bf176f2725547b6 : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/518b4fcb13ee4d92b594f949b97d072f : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e0ca6612a854eb688d3beb06c455ebb : SUCCESS in 6m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0a14f437b5342ee8bfccec780c44036 : SUCCESS in 1h 16m 08s","accounts_in_message":[],"_revision_number":2},{"id":"07ebb5e908065eac5619a7583001bbd8a967a837","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-11-23 14:43:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6a93e12ba5af96f441a9c533b9d5801f3a043d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-11-23 23:49:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"503f962de68571cf1a30d2a9b641826931921e60","author":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"date":"2025-11-23 23:51:20.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c17e168d869cf504ee07face5b24516f62d9c08e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 00:43:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c0c159f76ca4e38995b65725ed8b24c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d2e83e34bde4998925980ea6b02bbd0 : SUCCESS in 1m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f26d50267fe4853b66fd51f194fc9e8 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b03c2d32566d4b33812ff5d5ac060a9a : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebde387672784f95809dc9b06087ff27 : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b41ce8a3056459bb888d042ae67dcb2 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/debefe76d87d4b8fae3b7c9b6815198f : SUCCESS in 7m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ec92adb48f845d789d8f2b131eda091 : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":3},{"id":"86386b8da4a445b8d78059980edbd218ba1437f1","author":{"_account_id":8878,"name":"Masahito Muroi","email":"masahito.muroi@g.softbank.co.jp","username":"masa"},"date":"2025-11-24 04:59:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"10b5c65c36e2ed545e08d0bcde76f9172522d7de","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-24 09:04:55.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"780b5ec327881aa62675951f4ee09b2c52c64b56","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-12-10 08:35:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6c717e0263dd6de91e4f1d59e0411b5bcd705d14","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-12-16 12:28:23.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"edd9a3e4d7fdd0018b918d1e6fe567e69d9039af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:29:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"466ff863566044124e5b29a1b439da6861313e4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 13:45:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44e538538edf48319272ea67792d7da9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/31c1843fdcce4effb9bc9ff4be49a928 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e25432623545188f742cff401c100e : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94ac249a01664281960626d9efa0ecb8 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d009b7a97e564a71be5bfdf2c4e3c0fd : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f123be477c3343d6bb090f61dbc3f58e : SUCCESS in 6m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67748c1e84224de2942ffa96f0393831 : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":3},{"id":"efd3f0bec93814a566fb7a8f512b99dbdae0b6be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 13:45:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"09f7dae51b4eb22d3e0e5258bb0d64e5fb9cddaf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 13:46:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ca125fc8fd445459cd2a5e276ecc05c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40f6772e82e041ddb98f927f0a1bf23b : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9f16c76a6029595cb38c04656f13db834de9a021","revisions":{"017c2f5cfbbf22c50859516bf655afdd44191a33":{"kind":"REWORK","_number":1,"created":"2025-01-28 01:58:05.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/78/940278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/78/940278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/1"}}},"commit":{"parents":[{"commit":"5c86dc240818bd69cabc506cc6e08b2df1015cd5","subject":"Merge \"Use oslo.utils to escape IPv6 address\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5c86dc240818bd69cabc506cc6e08b2df1015cd5"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-01-28 01:57:02.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-01-28 01:57:02.000000000","tz":540},"subject":"Add cryptography package as a dependency","message":"Add cryptography package as a dependency\n\nThe cryptography package is used for encryption, but it wasn\u0027t included in the requirements.\n\nChange-Id: Ie9a1f273019a1f0e7a4573e1112bbfd4330690db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/017c2f5cfbbf22c50859516bf655afdd44191a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/017c2f5cfbbf22c50859516bf655afdd44191a33"}]},"branch":"refs/heads/master"},"11e9c91f8440954ab93f697f0d10982ee716afe5":{"kind":"REWORK","_number":2,"created":"2025-11-22 07:50:56.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/78/940278/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/78/940278/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/2"}}},"commit":{"parents":[{"commit":"5c86dc240818bd69cabc506cc6e08b2df1015cd5","subject":"Merge \"Use oslo.utils to escape IPv6 address\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5c86dc240818bd69cabc506cc6e08b2df1015cd5"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-01-28 01:57:02.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-11-22 07:50:31.000000000","tz":540},"subject":"Add cryptography package as an optional dependency","message":"Add cryptography package as an optional dependency\n\nThe cryptography package is required for certain memcache encryption\nfeature (i.e., \u0027memcache_security_strategy \u003d ENCRYPT\u0027).\n\nInstead of a mandatory requirement, it has been added as a optional\ndependency to allow users to install it only when needed\n(e.g., \u0027pip install keystonemiddleware[memcache_encryption]\u0027).\n\nChange-Id: Ie9a1f273019a1f0e7a4573e1112bbfd4330690db\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/11e9c91f8440954ab93f697f0d10982ee716afe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/11e9c91f8440954ab93f697f0d10982ee716afe5"}]},"branch":"refs/heads/master"},"9f16c76a6029595cb38c04656f13db834de9a021":{"kind":"REWORK","_number":3,"created":"2025-11-23 23:49:46.000000000","uploader":{"_account_id":37432,"name":"Koya Watanabe","display_name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","username":"koya-watanabe"},"ref":"refs/changes/78/940278/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/78/940278/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/940278/3"}}},"commit":{"parents":[{"commit":"5c86dc240818bd69cabc506cc6e08b2df1015cd5","subject":"Merge \"Use oslo.utils to escape IPv6 address\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5c86dc240818bd69cabc506cc6e08b2df1015cd5"}]}],"author":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-01-28 01:57:02.000000000","tz":540},"committer":{"name":"Koya Watanabe","email":"koywatan@lycorp.co.jp","date":"2025-11-23 23:48:59.000000000","tz":540},"subject":"Add cryptography package as an optional dependency","message":"Add cryptography package as an optional dependency\n\nThe cryptography package is required for certain memcache encryption\nfeature (i.e., \u0027memcache_security_strategy \u003d ENCRYPT\u0027).\n\nInstead of a mandatory requirement, it has been added as a optional\ndependency to allow users to install it only when needed\n(e.g., \u0027pip install keystonemiddleware[memcache_encryption]\u0027).\n\nChange-Id: Ie9a1f273019a1f0e7a4573e1112bbfd4330690db\nSigned-off-by: Koya Watanabe \u003ckoywatan@lycorp.co.jp\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9f16c76a6029595cb38c04656f13db834de9a021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9f16c76a6029595cb38c04656f13db834de9a021"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
