)]}'
{"id":"openstack%2Fkeystonemiddleware~947683","triplet_id":"openstack%2Fkeystonemiddleware~master~I8d237c340dd9f87ec68887fc52deb3a4e35f8149","project":"openstack/keystonemiddleware","branch":"master","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2025-06-13 14:06:49.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-13 15:54:27.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2025-06-13 15:54:27.000000000","reason":"Change was submitted"},"30126":{"account":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"last_update":"2025-06-13 15:54:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d237c340dd9f87ec68887fc52deb3a4e35f8149","subject":"Strip inet(6) prefix","status":"MERGED","created":"2025-04-19 03:05:54.000000000","updated":"2025-06-24 03:36:26.000000000","submitted":"2025-06-13 15:54:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947683","meta_rev_id":"3153d39ee551130318d6f3433c974dd4fab2945d","_number":947683,"virtual_id_number":947683,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-13 15:54:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-06-13 14:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-10 13:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2025-06-13 14:06:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-19 03:08:01.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-04-19 03:08:01.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-04-19 03:08:01.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"state":"REVIEWER"},{"updated":"2025-04-19 04:55:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-10 13:20:11.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"60c79c86c21cc5c73f5cce5448c09a3c751fb78b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-19 03:05:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3a9fa2d413fe02628739db309452fdae991ac9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-19 03:06:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c01c51566887711a4e17965c354f0ee14c408c43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-19 03:07:22.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c96ba6ba88cc08ed37036c7d0a6d287a9bdbffd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-19 04:55:38.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/192db24894464e9bb4db092b0b4fb0b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5307bfcd1926447c8fdbd584d35555d2 : FAILURE in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/952c881823ec4fc2b5cbe06857d2775b : FAILURE in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14100b5fd7a3452991b9de7f3d3c1892 : FAILURE in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/561afe29eb384e99a2a0e59723fa4d08 : FAILURE in 3m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bf79f8a7f234ab3ac104595789e1308 : FAILURE in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/845e7b3e76c64925807af3e03104c01e : SUCCESS in 5m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d75555ffcc64fcda0bd43c3b5049cce : SUCCESS in 1h 43m 47s","accounts_in_message":[],"_revision_number":3},{"id":"c7dfdfcba423fc325ef73b63238f5be91b33bd9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-19 12:10:13.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"cac55222ae2735b4dfdeb4d546cdeb53fe0f8dd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-19 13:26:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5361f673e7d346baba1305e34de2d2b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd50208f466d4a48ad3163bba6777b50 : FAILURE in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4a99f87ef64fb19d6d147b0daefab4 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02149a145a75425a94cc3f12e3503e8c : FAILURE in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/196654197dad42199fa459bda045a891 : FAILURE in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cd30163ef4a451ebaab1bea72e36376 : FAILURE in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be781295b1c4ba282d4e28b6ce6de0e : SUCCESS in 6m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2fd7e99e7582439396a0efbd3a2926a0 : SUCCESS in 1h 14m 25s","accounts_in_message":[],"_revision_number":4},{"id":"64c823ed58f08373823694f97209f958f109d1c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-19 13:49:36.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"657c425b604ca39bf6fc66616a10d569e18447f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-19 15:20:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/761c60b4904f404d981a4ffeecd28727\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0dc0d58412e46d581d05cef78459980 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de666fdd2159430a97a8864bbcfb5fea : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ace55f5faa1a440ba7dba4cc6c5393da : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe2b9ba4b9c249d29e3a40c9e22fde93 : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bfa840749de14b1c95114c0a0744a65a : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d280a4b48ff54c76a5158ad22c0ecc24 : SUCCESS in 5m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c19ed244a0c43f092dcfdad1909b4bb : SUCCESS in 1h 30m 22s","accounts_in_message":[],"_revision_number":5},{"id":"2d025ddcf58f865cdd8c8409c00d5270ec42132e","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-06-10 13:20:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c7a2eeb116bde5917fb85301c6e5cbac6642fd42","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-06-13 14:06:49.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"efbb4661d1c24da608a2a7ba632f81f78a2e2cd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 14:08:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6aecb67cad13b06feb8dcddc804986717c40da22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 15:54:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67d0de60c01347a79f9fbd74bd916dfb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a253a68a4157468d83270ced5684666c : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c09eb6cf5fc946229e416565bd2a9904 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f31848b2792f4b80be5b7fa3a7b2656c : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9feb868f19f4eb49980cd1071e38d78 : SUCCESS in 5m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5a3090b4b994e31a76ac85e6e70ca1f : SUCCESS in 1h 39m 31s","accounts_in_message":[],"_revision_number":5},{"id":"ef7bccc2acf776c6797c8432bf0093bfd8d2ca17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 15:54:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"3cd2aa97ec40d5877fc9d3b5c425a67e5d00d5a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 15:55:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d240d0dcf7e64a08b33c26a71dce68b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1111930b37d4c0195ca1cfa9d7509ae : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5},{"id":"3153d39ee551130318d6f3433c974dd4fab2945d","tag":"autogenerated:gerrit:revert","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2025-06-24 03:36:26.000000000","message":"Created a revert of this change as Ibb60eb40d20cc5b37839daf1c6849fd5c49cabdd","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"87b4339e89f22794298d1716f20bd699815bcbd3","revisions":{"bac1f221b37ac3359290ae880a2d72cc82434e78":{"kind":"REWORK","_number":1,"created":"2025-04-19 03:05:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/947683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/83/947683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/1"}}},"commit":{"parents":[{"commit":"21782d332b6ceb781263e0c66152ea9d2c0692b0","subject":"Apply upper constraints to build documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/21782d332b6ceb781263e0c66152ea9d2c0692b0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 03:02:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 03:05:36.000000000","tz":540},"subject":"Strip inet6 prefix","message":"Strip inet6 prefix\n\nSince python-memcached was replaced by pymemcache, the inet6 prefix is\nno longer needed, and actually should NOT be added when IPv6 is used.\n\nChange-Id: I8d237c340dd9f87ec68887fc52deb3a4e35f8149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/bac1f221b37ac3359290ae880a2d72cc82434e78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/bac1f221b37ac3359290ae880a2d72cc82434e78"}]},"branch":"refs/heads/master"},"22b4ce6ffed61829a91cb1c95a7550f8a92d5411":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-19 03:06:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/947683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/83/947683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/2"}}},"commit":{"parents":[{"commit":"21782d332b6ceb781263e0c66152ea9d2c0692b0","subject":"Apply upper constraints to build documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/21782d332b6ceb781263e0c66152ea9d2c0692b0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 03:02:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 03:06:57.000000000","tz":0},"subject":"Strip inet(6) prefix","message":"Strip inet(6) prefix\n\nSince python-memcached was replaced by pymemcache, the inet prefix is\nno longer needed, and actually should NOT be added.\n\nStrip the prefix internally to support the values previously supported.\n\nChange-Id: I8d237c340dd9f87ec68887fc52deb3a4e35f8149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/22b4ce6ffed61829a91cb1c95a7550f8a92d5411"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/22b4ce6ffed61829a91cb1c95a7550f8a92d5411"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b13f85c8b1854a90c1f686680174f2ad991b4598":{"kind":"REWORK","_number":3,"created":"2025-04-19 03:07:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/947683/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/83/947683/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/3"}}},"commit":{"parents":[{"commit":"21782d332b6ceb781263e0c66152ea9d2c0692b0","subject":"Apply upper constraints to build documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/21782d332b6ceb781263e0c66152ea9d2c0692b0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 03:02:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 03:07:22.000000000","tz":0},"subject":"Strip inet(6) prefix","message":"Strip inet(6) prefix\n\nSince python-memcached was replaced by pymemcache, the inet prefix is\nno longer needed, and actually should NOT be added.\n\nStrip the prefix internally to support the values previously supported.\n\nChange-Id: I8d237c340dd9f87ec68887fc52deb3a4e35f8149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b13f85c8b1854a90c1f686680174f2ad991b4598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b13f85c8b1854a90c1f686680174f2ad991b4598"}]},"branch":"refs/heads/master"},"b4ec55691d19fad5611bd96224a1e7514b225d5c":{"kind":"REWORK","_number":4,"created":"2025-04-19 12:10:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/947683/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/83/947683/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/4"}}},"commit":{"parents":[{"commit":"21782d332b6ceb781263e0c66152ea9d2c0692b0","subject":"Apply upper constraints to build documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/21782d332b6ceb781263e0c66152ea9d2c0692b0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 03:02:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 12:10:12.000000000","tz":0},"subject":"Strip inet(6) prefix","message":"Strip inet(6) prefix\n\nSince python-memcached was replaced by pymemcache, the inet prefix is\nno longer needed, and actually should NOT be added.\n\nStrip the prefix internally to support the values previously supported.\n\nChange-Id: I8d237c340dd9f87ec68887fc52deb3a4e35f8149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b4ec55691d19fad5611bd96224a1e7514b225d5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b4ec55691d19fad5611bd96224a1e7514b225d5c"}]},"branch":"refs/heads/master"},"87b4339e89f22794298d1716f20bd699815bcbd3":{"kind":"REWORK","_number":5,"created":"2025-04-19 13:49:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/947683/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/83/947683/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/83/947683/5"}}},"commit":{"parents":[{"commit":"21782d332b6ceb781263e0c66152ea9d2c0692b0","subject":"Apply upper constraints to build documentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/21782d332b6ceb781263e0c66152ea9d2c0692b0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 03:02:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-04-19 13:48:22.000000000","tz":540},"subject":"Strip inet(6) prefix","message":"Strip inet(6) prefix\n\nSince python-memcached was replaced by pymemcache, the inet prefix is\nno longer needed, and actually should NOT be added.\n\nStrip the prefix internally to support the values previously supported.\n\nChange-Id: I8d237c340dd9f87ec68887fc52deb3a4e35f8149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/87b4339e89f22794298d1716f20bd699815bcbd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/87b4339e89f22794298d1716f20bd699815bcbd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
