)]}'
{"id":"openstack%2Fkeystonemiddleware~953168","triplet_id":"openstack%2Fkeystonemiddleware~master~Ibb60eb40d20cc5b37839daf1c6849fd5c49cabdd","project":"openstack/keystonemiddleware","branch":"master","topic":"mtls-oslo-cache","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2025-06-25 12:35:57.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-02 13:57:01.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2025-06-25 13:41:38.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-07-02 13:57:01.000000000","reason":"Change was submitted"},"30126":{"account":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"last_update":"2025-07-02 13:57:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibb60eb40d20cc5b37839daf1c6849fd5c49cabdd","subject":"Revert \"Strip inet(6) prefix\"","status":"MERGED","created":"2025-06-24 03:36:26.000000000","updated":"2025-07-02 13:57:55.000000000","submitted":"2025-07-02 13:57:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":947683,"submission_id":"953168-mtls-oslo-cache","meta_rev_id":"38d49204a5225623ec444e78501fa6e39516bc1c","_number":953168,"virtual_id_number":953168,"owner":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-02 13:57:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-06-25 13:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-07-02 13:03:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2025-07-02 13:03:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-24 03:36:26.000000000","updated_by":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-06-24 03:36:26.000000000","updated_by":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-06-24 03:36:26.000000000","updated_by":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-06-24 03:36:26.000000000","updated_by":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-06-24 03:36:26.000000000","updated_by":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0615214660ef4feff565ae1e0017a863c0df713","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2025-06-24 03:36:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6091e7c0b31b811dfd11d1f858d9e9acbc5dca8f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2025-06-24 03:37:55.000000000","message":"Topic set to mtls-oslo-cache","accounts_in_message":[],"_revision_number":1},{"id":"ad4cbe78b3d66bf384239eb931b9e58a6a24da8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 05:18:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9a33252af6e4cc58496ec1065bc18a6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b010d9617b549da8269e47f3414e1cd : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/469a3e6886c24e079554630b4cd9fe06 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b73dab5d22644ecaafb768f7f40d0b42 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7f2de5664c84742a0e9c30075ce58ed : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/26ed0cca786d4a18b8f784c585511b86 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67da7ea9712e4414b9ff23d19748a118 : SUCCESS in 3m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/524aba4ce4ce4d86a457eb3a4579c86a : SUCCESS in 1h 37m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cd32b202bc4f1ccc0546045d7ad7e0fd6fc15299","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-06-25 12:35:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37d135b7adccd7403fc2e770ccea49ddd0092100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2025-06-25 13:41:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"da61e06826f62a24cfe82116cc40d5927db28eb3","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-06-25 13:41:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nCopied votes on follow-up patch sets have been updated:\n* Code-Review+2 has been copied to patch set 2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\").","accounts_in_message":[],"_revision_number":1},{"id":"23038b65714672b5d7acac209894205923779d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 15:24:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2abdc069bb754bccbce8d8ada987364c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38db7b11ae284f1c9a1670920fd1d61e : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c991b516c4d743e69a14925e4c5bd31e : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b83a78d9c7fb4723a6a0b8302908c485 : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19084b67089c4361a2cc6c1001426fb2 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b3434f4b954493982240f53cdb61418 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d958aeb8dfe41518686ed4c8e726304 : SUCCESS in 9m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60f8722a75d3490ab378e3b6172f445d : SUCCESS in 1h 37m 38s","accounts_in_message":[],"_revision_number":2},{"id":"fb135c4bc5d811fae82800661b2bff009497e3e1","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-07-02 13:03:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"85749bca918af97106ace79a9c3fb01d1e0eb1ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 13:03:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e001356cda923803183f8a4eeecd544714ba56b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 13:57:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4993a059483402fb57579470e622860\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05456036b6344b9383ae813bb6e5094f : SUCCESS in 1m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96d223aab4ca4063bb6e13a9ef288525 : SUCCESS in 1m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d33c50506f414721896145cbeee68191 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4666eecd55a4d24bfbb3b0211049ae8 : SUCCESS in 3m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c19a143db2443778ad2e8512dafc0c6 : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":2},{"id":"6c7ebe35d1a513d4cf92dc85a190f1cf745a860c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 13:57:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"38d49204a5225623ec444e78501fa6e39516bc1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 13:57:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e263254ab0444a10be54b255ab584cd2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33d7fde0f7149b4a26df52d86d5f16a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b32288ce771a92028185a92b9ceb9f0a815c37b9","revisions":{"6b5d5cfd484e6acd035dd736dddc746d4910a0b8":{"kind":"REWORK","_number":1,"created":"2025-06-24 03:36:26.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/68/953168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/68/953168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/68/953168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/68/953168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/68/953168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/68/953168/1"}}},"commit":{"parents":[{"commit":"87b4339e89f22794298d1716f20bd699815bcbd3","subject":"Strip inet(6) prefix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/87b4339e89f22794298d1716f20bd699815bcbd3"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2025-06-24 03:36:26.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-06-24 03:36:26.000000000","tz":0},"subject":"Revert \"Strip inet(6) prefix\"","message":"Revert \"Strip inet(6) prefix\"\n\nThis reverts commit 87b4339e89f22794298d1716f20bd699815bcbd3.\n\nReason for revert: reverting I8a361af1658722eb2eac5592504925347a57020c\n\nChange-Id: Ibb60eb40d20cc5b37839daf1c6849fd5c49cabdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6b5d5cfd484e6acd035dd736dddc746d4910a0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6b5d5cfd484e6acd035dd736dddc746d4910a0b8"}]},"branch":"refs/heads/master"},"b32288ce771a92028185a92b9ceb9f0a815c37b9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-25 13:41:35.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/68/953168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/68/953168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/68/953168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/68/953168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/68/953168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/68/953168/2"}}},"commit":{"parents":[{"commit":"a93d2951df70e44170f40e48ca723d31d9c7a889","subject":"Merge \"Add TLS support to MemcacheClientPool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/a93d2951df70e44170f40e48ca723d31d9c7a889"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2025-06-24 03:36:26.000000000","tz":0},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2025-06-25 13:41:35.000000000","tz":0},"subject":"Revert \"Strip inet(6) prefix\"","message":"Revert \"Strip inet(6) prefix\"\n\nThis reverts commit 87b4339e89f22794298d1716f20bd699815bcbd3.\n\nReason for revert: reverting I8a361af1658722eb2eac5592504925347a57020c\n\nChange-Id: Ibb60eb40d20cc5b37839daf1c6849fd5c49cabdd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b32288ce771a92028185a92b9ceb9f0a815c37b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b32288ce771a92028185a92b9ceb9f0a815c37b9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
