)]}'
{"id":"openstack%2Fkeystonemiddleware~979374","triplet_id":"openstack%2Fkeystonemiddleware~master~Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3","project":"openstack/keystonemiddleware","branch":"master","topic":"cache-opts","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-25 22:42:42.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-03-07 11:34:47.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3","subject":"Add oslo.cache backend support","status":"NEW","created":"2026-03-07 11:22:51.000000000","updated":"2026-08-26 09:34:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"72a7a108be5e02c451733c691790c97747d4f048","_number":979374,"virtual_id_number":979374,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-26 09:34:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-07 11:34:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-25 22:42:42.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"5f6f7f839d214cf0cda7b45bc786f4899446ac5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-07 11:22:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f04d68a18d18638ecb49a1eb38903428c3b931d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 11:34:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60289f2ae0e04a73880549802b4f1804\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/104dcf6b585e4b61937c20dab62da5be : FAILURE in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c23358cf5d4bdf8621740f06361242 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f3b78bd54e84fe1a36c45c794501ec1 : FAILURE in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/834d7484b46b4c479ef526adf5330765 : FAILURE in 3m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7bd9a154998a4091926dacb6032c505e : FAILURE in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d13e637aebbe4b96b072acee501c4468 : FAILURE in 5m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b162d45dbed4b7c8dafd04e11518e50 : FAILURE in 8m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fdf6401043261a75158a32b655db4f4c665dd119","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-07 12:00:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dd163dc6c3483e74b433585e7cdab0b1ed3bf945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-07 12:43:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"db388ba8cbbb817367aaecfad88e3d995e5dbfd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 14:32:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79ec797b3b5f4161bf606f9b865e1e5c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1513e5ac08634e179557527c0edeb95f : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc070b3b839b47a980c52b8eb3a17a91 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/294a2358ff074e1fa6c326a0793d14d9 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f1e5706408a4adeb45b93025f8d2a2f : SUCCESS in 3m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e473f25f95424c069542faf5bf4539cf : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e798aa6886642c8a670e1f85d568dbd : SUCCESS in 5m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ff709bfaa3b4626b66be515a8ddf13f : SUCCESS in 1h 44m 31s","accounts_in_message":[],"_revision_number":3},{"id":"2ac305a170e785a2a11f6a9e0c342a444ff760a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-07 15:06:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b625ed2a96501bdce00e42fc87d78e3de98a965b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-07 15:58:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4dc8614447045be8f1b3f50ee0a2313\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d11094227134524bbea2f07adea49c1 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ae6e9cba53485d84b12f18a4778bd3 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5616e7bc764483a82322fe0e96aa575 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae56db18bc024daaa15638e55d2c2d3e : SUCCESS in 3m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b07cd6eac1ee453ea972ed237eb086b7 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98d1bdffc0b4496b9342c341075620c0 : SUCCESS in 5m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6266056b3204962a8e40425e4bf0b68 : SUCCESS in 51m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98c879da4cb14e1ca809edbeb81338b6 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":4},{"id":"24f3cc6dafe27b5ad258127034299593451a296d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 02:54:45.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"28cd0081c732f772ebb42a9861670ea5d35f2ea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 03:06:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fc9ba96a2bd865111ac1945f0f369d34fb9e8718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-08 04:24:10.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24259bec4f2c49629dccb1c16982d41c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29f842c97f22421bbeec46aae93921ff : SUCCESS in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a35fe01343bf42c6981211e14e1843ae : FAILURE in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1af2b634179d449a96b6ff96169f160b : SUCCESS in 6m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1c5c0f7d7bbe4fa08e133aee6ef36bb8 : SUCCESS in 4m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34b3a0fc1e594ef899d20af8d659f640 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8852c4f0411480496cb810bb0f2f3b9 : SUCCESS in 8m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa6d4751815846fc88e3156df2fbdc26 : SUCCESS in 1h 16m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90ad42d8db85484f94462c640d3bbe82 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":6},{"id":"0bef173da3445a5c7ced297aef5c9638f1a22dd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 04:41:02.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"81a0bb4589b1c6b3a68f8c64b6b74e1cc047b333","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 04:46:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5fc3e730001df3ba9e533d5d4869a390031f5f62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 04:50:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6da7e4b13229c01a183ac5e407d8e8c155d705fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-08 06:48:50.000000000","message":"Patch Set 9: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8794b53564c649679f422fa88abea171\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e2189ca18e2441b83699eb22264cd22 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f20b119d9545cda2778d1845583354 : FAILURE in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca13302461f2462885094d11e8d87668 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1b49a739afc4be395f5f1ec9b45819d : SUCCESS in 2m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b148d3342334a8392dd8e20f970d286 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e9576b4eed45c6850d7fbfff08f9e8 : SUCCESS in 3m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c84d6c2407a4702b6d4de4d7d1bd956 : SUCCESS in 1h 52m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/939d679a018149878df83e9913331569 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":9},{"id":"b5c9eaf348dfb63a588bd9e47e19739d077536c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 06:58:13.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"829a75657af825989af26f468fe465b6f988a40c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-08 08:56:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e29379d1cb149e787dd75c90c9e8c04\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bed08ca7b734d0698532db48fe5d487 : SUCCESS in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38f9a1080e514d4584dafdf7d3373aa7 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/758f0256725a4b18b01d1dd8cc38b9c9 : SUCCESS in 3m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a1b9676c6b540ac8fffbdd453e52300 : SUCCESS in 5m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/85a3d484603b4950984bd5f3522a6cfe : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53f146afb8564081a04a0f64cb4ff361 : SUCCESS in 8m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25a1e1dac8324a9081253da69e7ecf00 : SUCCESS in 1h 51m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa0e8d55ed7f482f8c41c62c89e94925 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":10},{"id":"1a4a45418466acae74a298c24ca720b26bbc255a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-17 14:34:27.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"b709173e4a4a5d6efcea361114404c54d2f1cdc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-17 15:36:28.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"7c65c09a195184665bbdff63cd6b86a30589a821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-17 15:47:37.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"95052b6f38204a5d873b0d99e7d059e8984fcc31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 17:20:48.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d4ec00b88bf44239aa5b90abc345ea9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eceeaf886a3543abbccec5089d8b048a : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/041f22304a074dc69a987d32eeaa6e9f : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b784a64cc38f44a2b0065c61a7f72a48 : SUCCESS in 4m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8ba0349ef8c4299949a5875856b1b0d : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c91be8175be94987a0a243c3419310fa : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063c9f08ddb1443990b22548aca2b33e : SUCCESS in 10m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2513c689c98542f99c8519c2335a006c : SUCCESS in 1h 27m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cde25fbf3fee4d6ca808d512f07b335f : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":13},{"id":"854c0caf680bf3767bcd103e847c84255c73e94a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-26 02:06:12.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"7544c110fe80520e4b0db6a2938084ebf9e00b62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 05:03:03.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ed9fe7b1cf24a319392f6a46971bd59\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ec8cf31777c4ed28654f16231fe95e0 : FAILURE in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b9ffe59c7844f8a84f2bdc4615c4564 : FAILURE in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e3fd26b97ce45e385c72cad6e743ee7 : FAILURE in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6cfa8a6d437640e9ae5233002d4e4eb8 : FAILURE in 2m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5bd9aea292004ccabf841cec4e063fe5 : FAILURE in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddbbc85e874b4b1f8d30000e7ba394d6 : SUCCESS in 4m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b8462f1a4084159b0a2a55abfe6a275 : SUCCESS in 1h 52m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14ac42293ddc4047b6631d6af017c823 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":14},{"id":"c0a9ec0410d22bc85a9b8f64b3f0b094723d00ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-26 06:32:02.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"860468898bcca3ac099c47141fcb72b2ef40c0ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 07:46:53.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/898441729e6d454785c4099f4441462d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55b8819341ea4a70b22b295b2decaba9 : FAILURE in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed3efd42b3e4903b9522db6a02d4dec : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12f858b3195849d0b6db59ffece90002 : FAILURE in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70bdef0ff57346b4b9b30654786b8f27 : FAILURE in 2m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/84b053695d0e4379bf55d927eafa9b92 : FAILURE in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/227b2837afef4eb7bc637665bef0ac3e : SUCCESS in 5m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/832e18f863a24ee3a806414d2cdda300 : SUCCESS in 1h 00m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/575526e22e2e49c18d90c2973a4cd412 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":15},{"id":"3c359da249965ca416efd9e018783f471de1329a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-26 07:50:45.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"72a7a108be5e02c451733c691790c97747d4f048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:34:55.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02c789ad263a4f3e8c508aaf59b51b03\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7845b83ea49745d28195eebf63a2b099 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194c715b77904de0a47367f7035ccbe1 : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7831d54ddcc540178e2d26e0a2555c05 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ebe955fe32a41299b320a805050b373 : SUCCESS in 3m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/977965e1726b416697b8b3a9001b3103 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a379ef1309e649bba7d696308554191f : SUCCESS in 8m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7ae334d764b4aa1af305e7a5445dc80 : SUCCESS in 1h 37m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2bdbbc23b604d4c99878d29710d965a : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"e0745ad3c4e022ac5fe51469dc907788c77ad6d4","revisions":{"f34de79a5338a44791d1eb7ae9c3ba31d4738591":{"kind":"REWORK","_number":1,"created":"2026-03-07 11:22:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/1"}}},"commit":{"parents":[{"commit":"fe744ab96ff7df113b0f3ddbae6eb15bb37d1325","subject":"Merge \"Remove ec2_token and s3_token middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/fe744ab96ff7df113b0f3ddbae6eb15bb37d1325"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:43.000000000","tz":540},"subject":"wip: Add oslo.cache backend support","message":"wip: Add oslo.cache backend support\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f34de79a5338a44791d1eb7ae9c3ba31d4738591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f34de79a5338a44791d1eb7ae9c3ba31d4738591"}]},"branch":"refs/heads/master"},"d443784c55da0a876d51b939ba4d7266c213cc4a":{"kind":"REWORK","_number":2,"created":"2026-03-07 12:00:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/2"}}},"commit":{"parents":[{"commit":"63aa04dce620be420d02ef1877680615ccf50a7e","subject":"Drop unused or unnecessary oslo.cache call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/63aa04dce620be420d02ef1877680615ccf50a7e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 12:00:40.000000000","tz":540},"subject":"wip: Add oslo.cache backend support","message":"wip: Add oslo.cache backend support\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d443784c55da0a876d51b939ba4d7266c213cc4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d443784c55da0a876d51b939ba4d7266c213cc4a"}]},"branch":"refs/heads/master"},"c7c952207edc5c2e9c0154d68c4bedab5088a2aa":{"kind":"REWORK","_number":3,"created":"2026-03-07 12:43:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/3"}}},"commit":{"parents":[{"commit":"d54dc81c4040d93f25d7c8168f66c1dd1a9bac95","subject":"Drop unused or unnecessary oslo.cache call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d54dc81c4040d93f25d7c8168f66c1dd1a9bac95"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 12:42:28.000000000","tz":540},"subject":"wip: Add oslo.cache backend support","message":"wip: Add oslo.cache backend support\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c7c952207edc5c2e9c0154d68c4bedab5088a2aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c7c952207edc5c2e9c0154d68c4bedab5088a2aa"}]},"branch":"refs/heads/master"},"decfb50877a15f87c62e538f451cd5b5fa521ce3":{"kind":"REWORK","_number":4,"created":"2026-03-07 15:06:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/4"}}},"commit":{"parents":[{"commit":"42be67e2e611cff343f16b2e95ac059cce65ad4c","subject":"Drop unused or unnecessary oslo.cache call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/42be67e2e611cff343f16b2e95ac059cce65ad4c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 15:05:01.000000000","tz":540},"subject":"wip: Add oslo.cache backend support","message":"wip: Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/decfb50877a15f87c62e538f451cd5b5fa521ce3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/decfb50877a15f87c62e538f451cd5b5fa521ce3"}]},"branch":"refs/heads/master"},"ce79319e52e11f7659987c0584533a8d35fea180":{"kind":"REWORK","_number":5,"created":"2026-03-08 02:54:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/5"}}},"commit":{"parents":[{"commit":"2daa373e3d88310e63a727294851229348cefd62","subject":"Un-deprecate memcache_use_advanced_pool \u003d False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2daa373e3d88310e63a727294851229348cefd62"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 02:54:33.000000000","tz":540},"subject":"wip: Add oslo.cache backend support","message":"wip: Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ce79319e52e11f7659987c0584533a8d35fea180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/ce79319e52e11f7659987c0584533a8d35fea180"}]},"branch":"refs/heads/master"},"22529a8cd79a45dcf5e31daf9a998a157b63ee9e":{"kind":"REWORK","_number":6,"created":"2026-03-08 03:06:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/6"}}},"commit":{"parents":[{"commit":"2daa373e3d88310e63a727294851229348cefd62","subject":"Un-deprecate memcache_use_advanced_pool \u003d False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2daa373e3d88310e63a727294851229348cefd62"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 03:05:44.000000000","tz":540},"subject":"wip: Add oslo.cache backend support","message":"wip: Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/22529a8cd79a45dcf5e31daf9a998a157b63ee9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/22529a8cd79a45dcf5e31daf9a998a157b63ee9e"}]},"branch":"refs/heads/master"},"1ff5b1e3fa5a272a663e09c43d12bb9b57abaeca":{"kind":"REWORK","_number":7,"created":"2026-03-08 04:41:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/7"}}},"commit":{"parents":[{"commit":"e29117848900bfdabac2c58ac2b7b9da0d1b211a","subject":"Install memcached","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e29117848900bfdabac2c58ac2b7b9da0d1b211a"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 04:40:52.000000000","tz":540},"subject":"wip: Add oslo.cache backend support","message":"wip: Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/1ff5b1e3fa5a272a663e09c43d12bb9b57abaeca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/1ff5b1e3fa5a272a663e09c43d12bb9b57abaeca"}]},"branch":"refs/heads/master"},"c7dfd2c13f2f3801c71d3bc9111ef418ceafedbc":{"kind":"REWORK","_number":8,"created":"2026-03-08 04:46:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/8"}}},"commit":{"parents":[{"commit":"e0bd7065f620b7ec170c11d5d098059b0874a17d","subject":"Drop redundant encode/decode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e0bd7065f620b7ec170c11d5d098059b0874a17d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 04:45:33.000000000","tz":540},"subject":"wip: Add oslo.cache backend support","message":"wip: Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c7dfd2c13f2f3801c71d3bc9111ef418ceafedbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c7dfd2c13f2f3801c71d3bc9111ef418ceafedbc"}]},"branch":"refs/heads/master"},"d25f1080fb3eab7c87108f20b9345193acb1094d":{"kind":"REWORK","_number":9,"created":"2026-03-08 04:50:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/9"}}},"commit":{"parents":[{"commit":"b61e3839d1cf0839ade264df21be737d91c5ae4e","subject":"Drop redundant encode/decode check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b61e3839d1cf0839ade264df21be737d91c5ae4e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 04:49:56.000000000","tz":540},"subject":"wip: Add oslo.cache backend support","message":"wip: Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d25f1080fb3eab7c87108f20b9345193acb1094d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d25f1080fb3eab7c87108f20b9345193acb1094d"}]},"branch":"refs/heads/master"},"c1f53c8d782f9881365db15ac7cf02c481f9c612":{"kind":"REWORK","_number":10,"created":"2026-03-08 06:58:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/10"}}},"commit":{"parents":[{"commit":"b61e3839d1cf0839ade264df21be737d91c5ae4e","subject":"Drop redundant encode/decode check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b61e3839d1cf0839ade264df21be737d91c5ae4e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 06:57:57.000000000","tz":540},"subject":"Add oslo.cache backend support","message":"Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c1f53c8d782f9881365db15ac7cf02c481f9c612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c1f53c8d782f9881365db15ac7cf02c481f9c612"}]},"branch":"refs/heads/master"},"f48f09f433609363d8f93faf1a0d58c8b62f3a6e":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-07-17 14:34:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/11"}}},"commit":{"parents":[{"commit":"58552e42d531edbb2f462011cca3b47674870612","subject":"Drop redundant encode/decode check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/58552e42d531edbb2f462011cca3b47674870612"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 14:34:27.000000000","tz":0},"subject":"Add oslo.cache backend support","message":"Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f48f09f433609363d8f93faf1a0d58c8b62f3a6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f48f09f433609363d8f93faf1a0d58c8b62f3a6e"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"b61e3839d1cf0839ade264df21be737d91c5ae4e","ours":"c1f53c8d782f9881365db15ac7cf02c481f9c612","theirs":"58552e42d531edbb2f462011cca3b47674870612","merge_strategy":"recursive","contains_conflicts":false}},"224e30643e6c9f6c4142204313bc0c6f73274f48":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-17 15:36:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/12"}}},"commit":{"parents":[{"commit":"0235e75e4a5e43baf7a60969fd7a269e496f546c","subject":"Drop redundant encode/decode check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0235e75e4a5e43baf7a60969fd7a269e496f546c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 15:36:12.000000000","tz":540},"subject":"Add oslo.cache backend support","message":"Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/224e30643e6c9f6c4142204313bc0c6f73274f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/224e30643e6c9f6c4142204313bc0c6f73274f48"}]},"branch":"refs/heads/master"},"4e7aeea7e79ca1387211e9cebf327d6ee1477559":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-07-17 15:47:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/13"}}},"commit":{"parents":[{"commit":"e1411e1e147f12eb465f421950f9f5fbe0b5a213","subject":"Drop redundant encode/decode check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e1411e1e147f12eb465f421950f9f5fbe0b5a213"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 15:47:27.000000000","tz":540},"subject":"Add oslo.cache backend support","message":"Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/4e7aeea7e79ca1387211e9cebf327d6ee1477559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/4e7aeea7e79ca1387211e9cebf327d6ee1477559"}]},"branch":"refs/heads/master"},"1bab4a6fae955dc7ba3d9c9b1a81c630d7dad873":{"kind":"REWORK","_number":14,"created":"2026-08-26 02:06:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/14"}}},"commit":{"parents":[{"commit":"31b5f70a24da343df236b2bef3d7054f5ecf8b9b","subject":"Drop redundant encode/decode check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/31b5f70a24da343df236b2bef3d7054f5ecf8b9b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-26 02:05:52.000000000","tz":540},"subject":"Add oslo.cache backend support","message":"Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/1bab4a6fae955dc7ba3d9c9b1a81c630d7dad873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/1bab4a6fae955dc7ba3d9c9b1a81c630d7dad873"}]},"branch":"refs/heads/master"},"e47346e37987f5395426dd9e78b64a5ec0da3755":{"kind":"REWORK","_number":15,"created":"2026-08-26 06:32:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/15"}}},"commit":{"parents":[{"commit":"31b5f70a24da343df236b2bef3d7054f5ecf8b9b","subject":"Drop redundant encode/decode check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/31b5f70a24da343df236b2bef3d7054f5ecf8b9b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-26 06:31:18.000000000","tz":540},"subject":"Add oslo.cache backend support","message":"Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e47346e37987f5395426dd9e78b64a5ec0da3755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e47346e37987f5395426dd9e78b64a5ec0da3755"}]},"branch":"refs/heads/master"},"e0745ad3c4e022ac5fe51469dc907788c77ad6d4":{"kind":"REWORK","_number":16,"created":"2026-08-26 07:50:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/979374/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/74/979374/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/74/979374/16"}}},"commit":{"parents":[{"commit":"31b5f70a24da343df236b2bef3d7054f5ecf8b9b","subject":"Drop redundant encode/decode check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/31b5f70a24da343df236b2bef3d7054f5ecf8b9b"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-07 11:22:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-26 07:50:33.000000000","tz":540},"subject":"Add oslo.cache backend support","message":"Add oslo.cache backend support\n\nThe oslo.cache library is a library widely used in multiple OpenStack\nprojects. Add mechanism to use it to set up cache backend. This allows\nus to use new backends without implementing support for these\nindividually within keystonemiddleware.\n\nChange-Id: Idb28a38312b7bf5de4622b2a3a35dbe4fe7c9da3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e0745ad3c4e022ac5fe51469dc907788c77ad6d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e0745ad3c4e022ac5fe51469dc907788c77ad6d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
