)]}'
{"id":"openstack%2Fkeystonemiddleware~979411","triplet_id":"openstack%2Fkeystonemiddleware~master~Id0451b8dd2b06dabba69cc385498ecc96d2074f4","project":"openstack/keystonemiddleware","branch":"master","topic":"cache-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-05 08:37:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id0451b8dd2b06dabba69cc385498ecc96d2074f4","subject":"Install memcached","status":"MERGED","created":"2026-03-08 04:41:02.000000000","updated":"2026-09-05 08:37:06.000000000","submitted":"2026-09-05 08:37:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979411-cache-opts","meta_rev_id":"0c51eff7b747faf2c4be6eb7ceb393fc31794612","_number":979411,"virtual_id_number":979411,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-05 08:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2026-09-02 20:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2026-08-26 02:06:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2026-09-02 20:14:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-08 06:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-25 22:40:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-02 20:14:51.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"e0c5184794ae0e8abd1dcf547289d17dff6a473e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-08 04:41:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a25aa01bff9586d55d5a2e7bfdc299a48eb3c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-08 06:41:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab5006b7fe5441518d68ec625c075c72\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/112c4168005e4e8ba0e68a151221d528 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7701c5b284f9447aa89e3204a116c094 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59b39110800d40618669c7897bd1f355 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d63b6ce1d5d43e2adf3ee142ff41f83 : SUCCESS in 2m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d2c35118a04249b59f47525f9d1ac052 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56deb4fffc1e46cc9b6d383614c7d30e : SUCCESS in 8m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/04b3d65efd75493abf5f423a1cb0853e : SUCCESS in 1h 54m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f40a6ff00e94941ae7df9c7567b2200 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1d6c871bf7db4430b16554977fdbce738aae7456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-17 14:34:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ea86fc2a4de9665f3e6f4a219aa0ec5bacfb1325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-17 15:36:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"87fae514cd8bab3fbc875d9aa775e12c5b2cbef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-17 15:47:37.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"426d3f8e4548a6f204454d62eac4eb1b846a8ff2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-17 17:31:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efc8f4677d304cec8032eba1e98584ee\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d5da532f80b43939f967681f37cab19 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cc387ece50140dd8eff8deaa587be38 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f63b6e41d65415bb237193df4702682 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b2de9f1100f4943842c1ec85cfba9f8 : SUCCESS in 4m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7b8adc2b406f424e821d967f4ec442fa : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26035c4e3ff349abaa6dfc68c0bc32cc : SUCCESS in 7m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/401f91e35e5c44c5a739eb08934ad1fe : SUCCESS in 1h 27m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34558f87174b4042b4090811ca2cb67d : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":4},{"id":"e1fffae40222929878b17bf7c66680a33ce7b776","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-25 22:40:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f046f17ea51b0ab6632182557dc29794983f57df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-26 02:06:12.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"1209992d9f297ffeb93f2304c81a33cbd06d2501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 04:22:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a9ab21e8b444dfb8ce02c000b44767a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/961343dd4b0747f9814d22cb088218af : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f41a978eac4d309178e9fdc2c0cf29 : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f973001577ac4437abafcfe45cfbc445 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/265f5f01e0e6434e991e0b266fe35066 : SUCCESS in 4m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bba11c49de1849b7800a562db0902456 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ced4a25b464bfe9c574f5a9c8646ba : SUCCESS in 7m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e72a5a4f39f485289f3b89448fc9aa2 : SUCCESS in 1h 31m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afe4f1cf518944959edf22cd7399aaed : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":5},{"id":"e4f91e4fc11d71371b2be34407f7d617407ec67a","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2026-09-02 20:14:51.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"54a54d2b427012b9df8fd041ad5b510a54abec29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 06:55:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"021bebb0416bcb9d5a9d2e984f06c7fa85a01f37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 08:37:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94621e8605574cfeab24d0852194af48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e83ba0ba6340bdb8f98e8109642e40 : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/704d8c231f1f4a9eaa8b9c00b2cc6241 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68e11e8e097c4ac1bb2d80ffda0b5de8 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81587be6d65d42778103098fdcb789c7 : SUCCESS in 6m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e0441304c4241d79daf55d50ca00dd2 : SUCCESS in 46m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eda98cd714ae429aab347c7dc19a873c : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":5},{"id":"0c51eff7b747faf2c4be6eb7ceb393fc31794612","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 08:37:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c4cb4b94e89f20f77b71c4d201615c6c6984e3ce","revisions":{"e29117848900bfdabac2c58ac2b7b9da0d1b211a":{"kind":"REWORK","_number":1,"created":"2026-03-08 04:41:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/979411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/979411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/1"}}},"commit":{"parents":[{"commit":"2daa373e3d88310e63a727294851229348cefd62","subject":"Un-deprecate memcache_use_advanced_pool \u003d False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2daa373e3d88310e63a727294851229348cefd62"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 03:53:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 03:54:09.000000000","tz":540},"subject":"Install memcached","message":"Install memcached\n\n... to run live cache tests in CI.\n\nChange-Id: Id0451b8dd2b06dabba69cc385498ecc96d2074f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e29117848900bfdabac2c58ac2b7b9da0d1b211a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e29117848900bfdabac2c58ac2b7b9da0d1b211a"}]},"branch":"refs/heads/master"},"35d864b4355908b8643c555aaaff0cad512d9e9a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-17 14:34:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/979411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/979411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/2"}}},"commit":{"parents":[{"commit":"0285953e8e4890dc1ac987592617d77cb11ee105","subject":"Un-deprecate memcache_use_advanced_pool \u003d False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0285953e8e4890dc1ac987592617d77cb11ee105"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 03:53:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 14:34:17.000000000","tz":0},"subject":"Install memcached","message":"Install memcached\n\n... to run live cache tests in CI.\n\nChange-Id: Id0451b8dd2b06dabba69cc385498ecc96d2074f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/35d864b4355908b8643c555aaaff0cad512d9e9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/35d864b4355908b8643c555aaaff0cad512d9e9a"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"2daa373e3d88310e63a727294851229348cefd62","ours":"e29117848900bfdabac2c58ac2b7b9da0d1b211a","theirs":"0285953e8e4890dc1ac987592617d77cb11ee105","merge_strategy":"recursive","contains_conflicts":false}},"8d053ddeea463789de9910c0e4ee6d4dced12947":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-07-17 15:36:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/979411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/979411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/3"}}},"commit":{"parents":[{"commit":"53af4c1474782221ff3fcf2eb3fe728e4d125fcd","subject":"Un-deprecate memcache_use_advanced_pool \u003d False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/53af4c1474782221ff3fcf2eb3fe728e4d125fcd"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 03:53:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 15:36:12.000000000","tz":540},"subject":"Install memcached","message":"Install memcached\n\n... to run live cache tests in CI.\n\nChange-Id: Id0451b8dd2b06dabba69cc385498ecc96d2074f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/8d053ddeea463789de9910c0e4ee6d4dced12947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/8d053ddeea463789de9910c0e4ee6d4dced12947"}]},"branch":"refs/heads/master"},"bd64fbcf6cadaba54cda017dad985aadc963f4cc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-17 15:47:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/979411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/979411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/4"}}},"commit":{"parents":[{"commit":"2549b5f7d7ec28f63ecdd27abb3e838481e057c0","subject":"Un-deprecate memcache_use_advanced_pool \u003d False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2549b5f7d7ec28f63ecdd27abb3e838481e057c0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 03:53:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-17 15:47:27.000000000","tz":540},"subject":"Install memcached","message":"Install memcached\n\n... to run live cache tests in CI.\n\nChange-Id: Id0451b8dd2b06dabba69cc385498ecc96d2074f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/bd64fbcf6cadaba54cda017dad985aadc963f4cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/bd64fbcf6cadaba54cda017dad985aadc963f4cc"}]},"branch":"refs/heads/master"},"c4cb4b94e89f20f77b71c4d201615c6c6984e3ce":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-08-26 02:06:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/979411/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/11/979411/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/11/979411/5"}}},"commit":{"parents":[{"commit":"6e3fd611befd36d31998f4e39c2c668ab5147aae","subject":"Un-deprecate memcache_use_advanced_pool \u003d False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6e3fd611befd36d31998f4e39c2c668ab5147aae"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-08 03:53:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-26 02:05:17.000000000","tz":540},"subject":"Install memcached","message":"Install memcached\n\n... to run live cache tests in CI.\n\nChange-Id: Id0451b8dd2b06dabba69cc385498ecc96d2074f4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c4cb4b94e89f20f77b71c4d201615c6c6984e3ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c4cb4b94e89f20f77b71c4d201615c6c6984e3ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
