)]}'
{"id":"openstack%2Fkeystone~1002412","triplet_id":"openstack%2Fkeystone~stable%2F2026.1~Ice4a656e8deefd8538e10ad8d8d1fb12a3e3d12d","project":"openstack/keystone","branch":"stable/2026.1","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2026-09-01 16:34:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice4a656e8deefd8538e10ad8d8d1fb12a3e3d12d","subject":"Prevent unauthorized project-scoped assignment list","status":"MERGED","created":"2026-08-25 23:26:42.000000000","updated":"2026-09-01 16:36:19.000000000","submitted":"2026-09-01 16:34:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002412","meta_rev_id":"0561511933991acd47b026346502ccc3bd4c07da","_number":1002412,"virtual_id_number":1002412,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 16:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2026-09-01 13:55:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2026-08-31 13:46:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2026-09-01 13:55:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2026-08-31 13:46:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-25 23:26:42.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"},{"updated":"2026-08-26 00:38:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 13:46:43.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-08-31 16:39:50.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-09-01 13:55:31.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"79e5e07140ed7c7e424f015779d58d7903905d7f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-08-25 23:26:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"855903843aaebda84fbd18db77929da5ab04743c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-08-25 23:51:20.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"fb23dc117f8e66b30110d0cb5a7c66661ff94f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 00:38:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5fe78269a7884766a9cb99e222a84a2e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b36e3e0fc1a24900a9d8ef67c4047541 : SUCCESS in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c78945134de34530be01f83b5386d0b6 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecc3736da2d3490d98cbb8cf38802a00 : SUCCESS in 11m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05fe741e404245d788011300524d7d43 : SUCCESS in 12m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/987b8385740344f9a9c9dfc55c2316bc : SUCCESS in 17m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90b6adaf6584ed8a33f03a3aa543568 : SUCCESS in 14m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/da3216e53bf843a2be49a1fce97e2a82 : SUCCESS in 1h 05m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/021376c9ed064362922262a9b5dbec7d : SUCCESS in 58m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7ac30e9a3a34e37b00626bc4b86233f : SUCCESS in 10m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4230a6e9a23e470fa6825190224248cd : SUCCESS in 13m 11s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/735e88a8353440daa3488efc7d8a583e : SUCCESS in 36m 17s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/fa82ad5843004c569c6d966c5fdb19e8 : FAILURE in 11m 57s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/b941a23ea6e24fdeb2a7c9143ead0a2a : FAILURE in 26m 56s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/bab97cefb2404f8f9412543ebbf244af : SUCCESS in 19m 48s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/e1bdff7cc8974a75ae9a84cbbc494f71 : SUCCESS in 25m 02s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4b07bd718273499a8716f2bb528839bd : SUCCESS in 29m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46fccfb3825e4c1eba4767a455c2f522 : SUCCESS in 45m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/9124f1d528a848e9874a517959e57db5 : FAILURE in 20m 38s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a37f63fb0290456cbcbbf1c1ad61d691 : SUCCESS in 7m 48s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"230ede9e85829035c9e18538737e14053ae82ced","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-08-27 11:48:22.000000000","message":"Patch Set 1:\n\nrecheck - protection-functional skip fixed by 1002440 (merged)","accounts_in_message":[],"_revision_number":1},{"id":"addee0ddd2f6592150717713b0b9be66b2d6f528","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 13:29:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aedc17bae7e341f58dc24fdcec5a66b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdcee64ffd4248b1a4a935bec8c6a7aa : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9ddbfd7ea741929eed4ee11f3dd7b6 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd1a1b34ede54527884a1dc89af15b68 : SUCCESS in 12m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/032b06acf8ed423b88179b2f23f8c2cc : SUCCESS in 7m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4a3bc9eaf2af40f988418f81e6672231 : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29575a65193843d1a596761d61d07a6d : SUCCESS in 13m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/b8b7d93e329945378001824eea74e8b6 : SUCCESS in 33m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8fd15b8996d745388c00ded44e9fac35 : SUCCESS in 1h 36m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94f4dd62eadb47a09e657bb6da882930 : SUCCESS in 13m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2fe3ca393df64d89bc45767cbd7b8ccd : SUCCESS in 9m 42s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/42376315041d458f9d6f860cf9376675 : SUCCESS in 16m 31s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/5c29c00b46ff466aa8ea14f5fa958d33 : FAILURE in 22m 30s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/96047ebc75734eaf8bfba6b30ab6ab8b : FAILURE in 15m 29s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/7914dfe7fac94e0bb7b5882d74051424 : SUCCESS in 32m 46s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/b87c06ff52414b1ca315686667a6a7bb : SUCCESS in 26m 02s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8afe4dd296994e2e81b524c5dc341075 : SUCCESS in 46m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9c38ce46f62400ebafb59c8e076b413 : SUCCESS in 1h 04m 34s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/42f8056d3005451fa0b203484536ac39 : SUCCESS in 41m 46s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/8b4a358dd5c9493492c12a2099857017 : SUCCESS in 4m 10s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"180049d02672a4a93c63cf285ebba41927eb5f63","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-08-31 13:46:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f412bc6554d05d927bfd88457abb8457b15bc5b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 13:47:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ac8162f7287516977c7ca65168ea21b90b9137c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:16:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/308706f4c3bf4e0d92d17191de19d492\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9add8b0b1634b4fbe6e0bbfd387eabd : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a04b74d8c7c40428a0eec6617da4963 : SUCCESS in 10m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20064e161c684d1cb0c0523f35681955 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5806b3c8584a446cab67c7d801d31e32 : SUCCESS in 16m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/64a5ee30a5474f408750ae38d341c633 : SUCCESS in 33m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9b2a3e3317e4c4abb06bf8a6ba12f3d : SUCCESS in 1h 47m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60b64c0b44c04c148a5a0daa30c4c1e1 : SUCCESS in 5m 45s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/12a7bce496d240b68b6f31430a014361 : SUCCESS in 13m 38s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/d5d4c3f7effc4952a9cf4a927cb70ca4 : SUCCESS in 22m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ea08efcbb894c88b57daec780b31838 : SUCCESS in 36m 23s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/12aa717c768a4fbebb0b14da01aee462 : FAILURE in 18m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fba52c71cf53b14fe85183e2046427e4225b1fb9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-08-31 16:39:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"245624a13b23ac64adfb9c8847d24bda75e06dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 18:30:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/559febbb486b423795cdf57e659acfd7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d7da4d7b24941eda067b19caa403c73 : SUCCESS in 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b086f589d02245cebb2b4bcd9ebdd9da : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a464a8be28b4462971503c5ef544c7f : SUCCESS in 16m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d81a81c60e724c22886130453f3b2c37 : SUCCESS in 14m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a3b4e54c76a540039e64dae27cd746bd : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02579489c10f4cb69d2426afc9a2b8f1 : SUCCESS in 12m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/6b682cc4830947c2a0728258bd37be9f : SUCCESS in 40m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/297b3e2975624a47a9c1fe0e8f738abb : SUCCESS in 1h 05m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58a3cb166bbf41e3b1292e9df9f9ba8e : SUCCESS in 10m 42s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0041ff2ac29b49609cb2537a07c6f67f : SUCCESS in 9m 32s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/5e17636f21d041b898e27661649bf084 : SUCCESS in 31m 26s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/9f37d16ea4c140088c07e2efc3bf08e7 : FAILURE in 16m 37s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/bec7d7afb3714dbc84ce36621ab6a512 : FAILURE in 18m 53s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/5bbf6b380b4d4a2c9497b5a9c252db96 : SUCCESS in 33m 26s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/469eea57ad62461caf4b6452e3fd4d66 : SUCCESS in 35m 01s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1fbffe642d0a4bab85047d24f78e631a : SUCCESS in 37m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1069cfb4f06d4d84b61dcaf0a088c45e : SUCCESS in 59m 59s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/5e955e25f97c4318a6ea13f0ff169d51 : FAILURE in 37m 48s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/effc52d2b9d340619ec147b8330cd01a : SUCCESS in 7m 22s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"b9ece2eb6b64edbb8d8b025f103a205523e59b58","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-01 09:51:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07b215ed9c9e25d6ad73a7e2e34e73b2e1e67a70","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-01 09:52:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5a328f3c1fc2882b76f63df74a1a8f37daf88d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 11:18:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c49133551cd41b88ae896262bbf4b1b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24291f0701784bea840e7ee742650582 : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a07cd65af74b4f8f9b72e073dddaf2 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f02f9ff8b044f6abecc35dc6d5d159b : SUCCESS in 15m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9213e5d227e4ed38848d8685e4f13ba : SUCCESS in 10m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d9557b2d5f58434dae040dccf8e3ffe6 : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3327a5fe3e894ef0a4fe28230d844b7d : SUCCESS in 13m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/2c87d493c4714cc28e80e5aa9d68644d : SUCCESS in 1h 02m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9474f1605a08415cb579120bf4c9cbbb : SUCCESS in 1h 21m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f02b19c7cc14685838d168dcb8130b5 : SUCCESS in 13m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/472d039f685a4bf7bee89dbb066c3c35 : SUCCESS in 11m 39s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/1d22ce1133dd4ee5b6e743b37ddefe7b : SUCCESS in 22m 28s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/e20c46a126f949a6ad782c8e37c77208 : FAILURE in 18m 10s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/c8dfbe27ad464acbb8443ca172367b10 : FAILURE in 14m 41s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/bc3ffb7606f14f3a92a5c52633c812ac : SUCCESS in 30m 06s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/905574847ab74999bc17fe30700aeec3 : SUCCESS in 30m 34s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/000d77b98ae6481cbb556bc7bb8297c6 : SUCCESS in 41m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e7c0f62c58940e4b87e9e749b36cbbe : SUCCESS in 56m 00s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/4be93be4d3344b0fa1d924ac0749e10d : SUCCESS in 45m 17s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f5d6af18d33943e99df698eac6d727c6 : SUCCESS in 4m 15s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"f0146003b9e951f73570309b5abdcd315152e19d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 11:19:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22fa5d03afd7134a0be6f8149ca44133a90e7a50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 13:05:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1928609943bc45079c4ef8afd1a9d44e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15c9394f72944551832b80c3e2e6dad5 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f9b1f5ddb4f4d2bbeca35dba098865b : SUCCESS in 14m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6bd11e010bb749b396bb01b48902fd93 : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37279bd4d56485abdfa172016d4fec6 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/e654aea39e324f20a29ae6fe47224a12 : SUCCESS in 1h 02m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f72a8522ecc48f4addb56c7d1171a29 : SUCCESS in 1h 41m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccc91752d2514cb9859973c1303e16ef : SUCCESS in 14m 12s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/7732be6d2a7e43949d7c13fb3af5ac28 : SUCCESS in 31m 28s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/26bdf9103ee240e495c9a76f5a84ff0e : SUCCESS in 31m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5199547f5e1b4c99b9143cba58d4151f : SUCCESS in 1h 01m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a9b1f89088ce4a82ae9bfc83099c5f97 : FAILURE in 43m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9a5b50fda016e8f99c2b24a278509cdb8dc6d6ab","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2026-09-01 13:55:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a575515b12cce3d38bab15f07c0edce2a8f04613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:04:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c0016e99ad746a6b19e73bbf1250ce4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b53b26bc8c0b4ad7bafb86f30eb416eb : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c19f750f96d49aca59732385608bf70 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6c9988cc7d44226967846b1ad463e72 : SUCCESS in 12m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdcec44f47fa4000bb8c386e639ee14c : SUCCESS in 10m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/84e97008df6f4aafabf2b452632b40c5 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27671812d983400896f5f53943ac78fe : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/10a3b0df32fe4d10b61a7d910935d583 : SUCCESS in 1h 02m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b36f298a75b44d87be5c32cfb34a6020 : SUCCESS in 1h 07m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2ec68e810de4b04a90d07735783ac3e : SUCCESS in 5m 29s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/24af4727dce74ce48eabf382df077a44 : SUCCESS in 9m 54s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/5b8263a769a24302bc5c3560ece8b0dd : SUCCESS in 33m 59s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/8dca75309eab4d35ba6d6f829fdb996d : FAILURE in 17m 08s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/1557c4aabf56424a9f3b624d05034c02 : FAILURE in 27m 20s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/7666136f812848f19a54c072b01a0294 : SUCCESS in 28m 36s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/82067719c53b47718a3ad4de90318c2e : SUCCESS in 24m 22s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/bb23de054543451fb205e64971016d44 : SUCCESS in 36m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e68ee62104cb461087828e0b10ef0af3 : SUCCESS in 33m 48s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/131e7ad8224d44599fc83c4a986518d9 : SUCCESS in 42m 04s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/990e9f4c065e4382a56fd75730f73325 : SUCCESS in 7m 45s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"8f466bb44bcda1625806bd36b71df70656f80c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:05:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0a13dbc9b39ed3bdad3531f28186aa6d851060c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 16:34:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b43149e66bb343e89db463a03e20385b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/240e6f39ce6d418bb30688c40f0545da : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab25dfa63b6d4c67b3a55acb1661c174 : SUCCESS in 9m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1078f5821a034e1d9912b4abcd40ca43 : SUCCESS in 10m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c081213095f94b4095402a0f31792929 : SUCCESS in 7m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/c7226d7927c54af5ba7ae94f99dac61b : SUCCESS in 46m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25d6741e28704d2ebf250b85b2f5acfa : SUCCESS in 1h 01m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3b266fe3f704a4e97d20204dd713d65 : SUCCESS in 13m 35s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/c2e13c6b806f4f06b7b3e86a8a5547bb : SUCCESS in 20m 55s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/9e3af63441e54addb0d18552913c6c37 : SUCCESS in 28m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7daa790388d447429945d1c91104a4ac : SUCCESS in 1h 00m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/baf830d6e5044e14a32dacf7301f9ed3 : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2a321242bfeed6744b46ef27fd549e0a15d64565","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 16:34:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0561511933991acd47b026346502ccc3bd4c07da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 16:36:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b3610b471af49bc90d1ee418daeaeb6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50907b1147744066a19f0d052ecd6619 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af8e68c91c574669a1851ebaace610fc : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76db7d798898524ac3a505896e51b5cc22360855","revisions":{"76db7d798898524ac3a505896e51b5cc22360855":{"kind":"REWORK","_number":1,"created":"2026-08-25 23:26:42.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/12/1002412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/1002412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/1002412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/1002412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/1002412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/1002412/1"}}},"commit":{"parents":[{"commit":"7b9f39226744b8db50a6d702cc5f715fc9c17f37","subject":"Merge \"Prevent RBAC policy bypass via JSON body and query filters (CVE-2026-42999)\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7b9f39226744b8db50a6d702cc5f715fc9c17f37"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2026-08-04 11:51:03.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2026-08-25 23:26:24.000000000","tz":120},"subject":"Prevent unauthorized project-scoped assignment list","message":"Prevent unauthorized project-scoped assignment list\n\nAny user holding `role:reader` on any project could list all role\nassignments under any domain by passing a domain ID as\n`scope.project.id` to `GET /v3/role_assignments?include_subtree`.\nDomain projects store `domain_id\u003dnull`, which matched the `null`\n`domain_id` of any project-scoped token in the oslo.policy string\ncomparison, bypassing the domain-reader restriction.  An explicit `not\nNone:%(target.domain_id)s` guard has been added to the affected policy\nrules.\n\nCloses-Bug: #2154645\nChange-Id: Ice4a656e8deefd8538e10ad8d8d1fb12a3e3d12d\nCo-Authored-By: Grzegorz Grasza \u003cxek@redhat.com\u003e\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\n(cherry picked from commit eb27f4e309708746a8e7a5f0ad7e9a510e47ff0b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/76db7d798898524ac3a505896e51b5cc22360855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/76db7d798898524ac3a505896e51b5cc22360855"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
