)]}'
{"id":"openstack%2Fkeystone~1002414","triplet_id":"openstack%2Fkeystone~stable%2F2025.1~Ice4a656e8deefd8538e10ad8d8d1fb12a3e3d12d","project":"openstack/keystone","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2026-08-31 16:24:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice4a656e8deefd8538e10ad8d8d1fb12a3e3d12d","subject":"Prevent unauthorized project-scoped assignment list","status":"MERGED","created":"2026-08-25 23:26:48.000000000","updated":"2026-08-31 16:25:41.000000000","submitted":"2026-08-31 16:24:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002414","meta_rev_id":"76e083db73bdebc30d444d25155fbfd68021aea7","_number":1002414,"virtual_id_number":1002414,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-31 16:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2026-08-31 13:47:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2026-08-31 13:47:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-25 23:26:48.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"},{"updated":"2026-08-26 01:10:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-31 13:47:07.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"b0c3de951f5eb8c9809d49d08562d54aeab5d784","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-08-25 23:26:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e2d6f0f23a80a0ef4808d8dbe13ab710414aa87","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-08-25 23:51:28.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"3914b8cab229254c84bd57313190ba482bc733f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 01:10:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a415709fc841e7bbe7369c45169798\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89ec6b1db7994472929b2c249397ebd2 : SUCCESS in 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994ff69e810549188b406bec4494fcb6 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e39070c84a64d54b597b85c12f157c6 : SUCCESS in 13m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/47430961069948a9b4096d6494345d2f : SUCCESS in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ec95de7466474eb7f43728b2961d4f : SUCCESS in 14m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/720a2610f36f4da69cc1d55131c78d93 : SUCCESS in 13m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/cd58e499e2174e088549c437396bb7e1 : SUCCESS in 29m 01s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f501679c3d5e4c9881a5fcac90b1e1ed : FAILURE in 32m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ffd1cd93e9c4f7db2623d5f208de471 : FAILURE in 14m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/7ee54cf8b8734bb39df79df8308854e1 : SUCCESS in 27m 53s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d5e7ec202be640cdbb4bf9682eac688b : SUCCESS in 15m 49s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7a3474ee593a4013a0228210e7ec6708 : SUCCESS in 37m 07s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ad81c6f49714a7d98500f63a30a2534 : SUCCESS in 1h 40m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/68d71bc38a93429ca1c2067ecb037c03 : FAILURE in 53m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b66fd5ab72044b5fb2c53871572418d7 : SUCCESS in 42m 00s\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/9905d0943e3a4337b158d45d5af51383 : SUCCESS in 37m 47s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-dsvm-py3-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-dsvm-py3-functional-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"97fe93052deb117b86aca8b6613e0a2b01c9ed82","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-08-31 13:47:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b94aee5bf0563b287f79f9b67e67fb6498088839","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 13:47:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8da6029fbd7967fd261615b905ef6f3cb0961955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:24:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee12eb719aa2458bb9e5aa118764dc8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67104ca7e1804bc9ac313938ca3759ac : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a179b9724d448d9ba4532311e9bf13e : SUCCESS in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e0cc5a07aa94eab9f37b374e4026140 : SUCCESS in 9m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa187534c3134dccb41b2190481f3c97 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f30a790c161f4e0d8d5cf51ebe80ed36 : SUCCESS in 5m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6f45180b3e1c43ccb6b065dca2b72377 : SUCCESS in 25m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb3d2112c35b47bba899475256892452 : SUCCESS in 1h 38m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/7439e30b925341c0bc6ec5d640a0608e : FAILURE in 30m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88ce82d96ed3477b962af3beb49c67d0 : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b8188adbd5e38e4fa96e34c1c99580978e5c1e88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:24:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"76e083db73bdebc30d444d25155fbfd68021aea7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 16:25:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a504f0a9ce2491c8bb8a70218638baa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2381b1129fd54f20a0d586e586c977e9 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc330e1423d34037a1236a74126cb4be : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"27b649b1155bc7e7e518e4e606057e30c3ace4b2","revisions":{"27b649b1155bc7e7e518e4e606057e30c3ace4b2":{"kind":"REWORK","_number":1,"created":"2026-08-25 23:26:48.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/14/1002414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/14/1002414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/1002414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/1002414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/1002414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/14/1002414/1"}}},"commit":{"parents":[{"commit":"b896b71557ce8370c442fe05818a74d7a9a623a8","subject":"Merge \"Fix project policy allowing unauthorized access to root domains\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b896b71557ce8370c442fe05818a74d7a9a623a8"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2026-08-04 11:51:03.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2026-08-25 23:26:24.000000000","tz":120},"subject":"Prevent unauthorized project-scoped assignment list","message":"Prevent unauthorized project-scoped assignment list\n\nAny user holding `role:reader` on any project could list all role\nassignments under any domain by passing a domain ID as\n`scope.project.id` to `GET /v3/role_assignments?include_subtree`.\nDomain projects store `domain_id\u003dnull`, which matched the `null`\n`domain_id` of any project-scoped token in the oslo.policy string\ncomparison, bypassing the domain-reader restriction.  An explicit `not\nNone:%(target.domain_id)s` guard has been added to the affected policy\nrules.\n\nCloses-Bug: #2154645\nChange-Id: Ice4a656e8deefd8538e10ad8d8d1fb12a3e3d12d\nCo-Authored-By: Grzegorz Grasza \u003cxek@redhat.com\u003e\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\n(cherry picked from commit eb27f4e309708746a8e7a5f0ad7e9a510e47ff0b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/27b649b1155bc7e7e518e4e606057e30c3ace4b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/27b649b1155bc7e7e518e4e606057e30c3ace4b2"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
