)]}'
{"id":"openstack%2Fkeystone~1004177","triplet_id":"openstack%2Fkeystone~master~I2df254a684065c14598e8da1faa085801d54b169","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 02:11:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I2df254a684065c14598e8da1faa085801d54b169","subject":"Use common environment classifier","status":"MERGED","created":"2026-09-05 13:28:19.000000000","updated":"2026-09-10 05:47:10.000000000","submitted":"2026-09-10 05:45:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004177","meta_rev_id":"86a468b6cf89398d314434eb4c95d250e02c77e8","_number":1004177,"virtual_id_number":1004177,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 05:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2026-09-05 15:01:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2026-09-09 15:41:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 13:59:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-05 15:01:28.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-09-05 15:11:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e7ec794b8c92a8bf15b9d0f3bb92c3df25ca737e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 13:28:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10a1ffb6a45a720fa48ca879f7b6084addeb0e17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 13:59:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/667da2fcd0e84a179bb58fe278fe998f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f670c5ac2acf42029259d8bce557c8e4 : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8df56ae03f0b4ccfa184607861a1809d : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/eb18f2083cc1478dbbd0d4ea525064be : POST_FAILURE in 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6daa4a5a2663185ccae6844db9c25ff51aa1e21","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-05 15:01:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c479097a019f624c18edbdd4ec816e4508151e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 15:11:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8df809f619d4d5dbcf7f4e601285da0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3619c16c07b6476a9edf365802f754ed : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1943a31541d7436484222364ed088f0c : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9386bff71ad4bd89f26175aff20ffd1 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f824e024feca4cc6909e28c37df75ad6 : SUCCESS in 12m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4708f4106eca4fbcaad16d0a5725c754 : SUCCESS in 11m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be8b0e9bd96c46949e4ed18ad6fa9f2c : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9740c6f72fd8499b9184cb12eddb16c5 : SUCCESS in 12m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d6331208c0704c678e2867a4362f4a20 : SUCCESS in 2m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/8d2b73c2a5354be38f03ee2ccc64dd35 : SUCCESS in 1h 02m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a17fda5225a4f3c9486284f582e85f6 : SUCCESS in 1h 37m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bae6fb659ff145cc9a048ab1be593399 : SUCCESS in 12m 02s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8247214952cf4c3197e15cb66dbe7bde : FAILURE in 13m 55s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/b52edf22bd244c51a514f390f583ada5 : SUCCESS in 52m 22s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/37a08e79b81a436392d776cde1f84628 : FAILURE in 19m 42s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/2d1406663ae74e46a8cb25d4c98de11e : FAILURE in 23m 34s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/f9e72ccec5564d608adf4cd496205860 : SUCCESS in 1h 03m 41s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/960d27170a364c589401b1a58722bc2d : SUCCESS in 49m 33s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3087fa15e85541e69ace3ca1d2199327 : SUCCESS in 18m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5914c07af2a14cf69de9eebf964ff540 : SUCCESS in 57m 09s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7151c026eae2435b953439a64c16c5e6 : SUCCESS in 41m 34s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/adb6065aa7694b67823804af2d8ae373 : SUCCESS in 6m 52s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"188c745822ec60137b77484dc58f5ab2c3c3f436","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-09 15:41:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"12d617c80e7388220406e3ca071efd7c8ec113d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:42:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a3e0c6c08f234132672efdc005a080d32275b0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:28:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f59e5b519ee84afb882fa03e26b76f06\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a8f1ce903f9048ffb582ee6fde32c190 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09f62de4a91e4ceabd7581821c352a81 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7115a777278b4732b05a77cba3b1837a : SUCCESS in 13m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17612e8fb9fc4cc4afe83514b4f86c12 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd9cd8847f16479cb9213a2b11f45585 : SUCCESS in 7m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f5c00b0702e427eba0735bcb01d9259 : SUCCESS in 4m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/9ceb5153a0944cc6963b22aaaa903b2f : SUCCESS in 1h 06m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b52d1112aa14171bec5f5f14d233f5c : SUCCESS in 1h 49m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/072e7de5f5ca417795b6febada5d706f : SUCCESS in 15m 25s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/896f5d9129c4478aabebed24a44d213d : SUCCESS in 51m 49s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/822e8f23a21c49b2b5d53be42959e6ce : SUCCESS in 51m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/705de5bfd0dd4f2c9b6dade1e91cbc26 : SUCCESS in 1h 21m 49s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6458dd87d8a64b3ca358478dc5b245cf : FAILURE in 39m 28s","accounts_in_message":[],"_revision_number":1},{"id":"51639fdc68721348bc7b255cfdcc09e95ad07580","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-10 02:11:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9794fddc1f05a5c180fd05081101da6a6d850a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 03:37:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7dd88d9cae44daaa0e7586656d8a58b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0743ce135bc048ad9cd44812c5f7eb0a : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2c4f2d80304c4f1cb498ec40e7562a1e : SUCCESS in 22m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/be461ce4dfb7416b8279a52cec29b377 : SUCCESS in 36m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31f7a96611b06cbd8fbdff2f7b99380ea5f4c59c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 04:17:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39b2e3725d9141558fe9658618141a7b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/edfd3d1f85ea4221a8571ff35fb30015 : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c8170005c104e6c95e1808d05dacdae : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d75a7985fb44bc8f864d3afa5a6e16 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d797b23a0a28423282eecc1ec0269273 : SUCCESS in 11m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acc6342e85de400080211299340067fb : SUCCESS in 8m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/176099fb37974fa69ed57ac11157a510 : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/618c2b017f7642658168e1b514f2e676 : SUCCESS in 11m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6a29fb51d513424db852c3b80076dc59 : SUCCESS in 3m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/6a531fffce8b4b55916460d8cf90ad25 : SUCCESS in 1h 15m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/014cd12cead14ff48f636c30fb90c61d : SUCCESS in 48m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d73f8cb981f340088a0414e3e76204d2 : SUCCESS in 7m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/adbb98f30d8e405c996cef30941a0193 : SUCCESS in 11m 03s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/1744afe5908e456ea86a7a790bbe01d7 : SUCCESS in 27m 03s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/fe0a91a27b0e4ef0860e233dcce430ff : FAILURE in 12m 28s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/b7b9a0d0b37e4087add472e196dc7aac : FAILURE in 19m 56s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/e308ef5da75e48f4983b8b79a8d4e5c3 : SUCCESS in 47m 46s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/0f4efc703f9d4ba7986ba709aee9d013 : SUCCESS in 22m 22s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ef7622b50c8c4ba4bb80088e3f051f84 : SUCCESS in 23m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d50b8728e55d4dc8816b40c4741d6560 : SUCCESS in 56m 48s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/66ad353584a045e1a77612597a75961f : SUCCESS in 22m 20s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5232c437bac54cbf9e46fd68f50ddc06 : SUCCESS in 7m 42s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"b2dcd36f8edfda0b614e33bd39a650341c6100df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 04:18:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cf6d7f2409e11c814e9986d2215fc85706c0bc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 05:45:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ae9c4f423a4edc9f1faf3f8b6cf728\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2029462f3c147439c7d3594f19df0a8 : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb8c1dd37f1477eb6c05c129e50c181 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d80acf61d1c49e984b5c81dad1a4002 : SUCCESS in 12m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a703233e2cc542f2be493eaa35d933f7 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8533f69c26354d9a9e49b19000309635 : SUCCESS in 13m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d6c8c9978b6948c1b8c415c37963b5d8 : SUCCESS in 2m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/000a2dd8c24b4264a8d5ee585464db75 : SUCCESS in 30m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b9be6949e584f70baebcc0c5e02f1d7 : SUCCESS in 1h 27m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b27e309fa6a4801a7b192309dc6a28d : SUCCESS in 10m 01s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/d448d54a4e4f4f8f9aa1e43b09be913b : SUCCESS in 33m 41s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/5a4d8c7578bb4af48979520e677246be : SUCCESS in 50m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2a3df09784c4dcea7ed516a328134a4 : SUCCESS in 57m 42s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/29576a7eb93345dfa528989406c78e46 : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bbca017c8b4ed03b9b75197df4d5f49c9b535626","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 05:45:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86a468b6cf89398d314434eb4c95d250e02c77e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 05:47:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d0dbaf2b9c4d90bd0f80317fd35e14\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d51f5debe4a3443288eb3f7473fd9a78 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c45001999dcafc5af81c44197c8feddfbf868a5","revisions":{"3c45001999dcafc5af81c44197c8feddfbf868a5":{"kind":"REWORK","_number":1,"created":"2026-09-05 13:28:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/77/1004177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/77/1004177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/77/1004177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/77/1004177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/77/1004177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/77/1004177/1"}}},"commit":{"parents":[{"commit":"16afc813b7e6de727d8a91e065d7824b06e32925","subject":"Merge \"Abandon server-side paged search cursor when sizelimit is reached\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/16afc813b7e6de727d8a91e065d7824b06e32925"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 13:28:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 13:28:04.000000000","tz":540},"subject":"Use common environment classifier","message":"Use common environment classifier\n\nThe \"Environment :: OpenStack\" classifier is commonly used in most of\nthe OpenStack projects.\n\nChange-Id: I2df254a684065c14598e8da1faa085801d54b169\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3c45001999dcafc5af81c44197c8feddfbf868a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3c45001999dcafc5af81c44197c8feddfbf868a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
