)]}'
{"id":"openstack%2Fkeystone~1004178","triplet_id":"openstack%2Fkeystone~master~Ia2ff0b477fd1a0089b25173cbf54713f04b6bd0e","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 15:33:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia2ff0b477fd1a0089b25173cbf54713f04b6bd0e","subject":"Remove Python 3.10 support","status":"MERGED","created":"2026-09-05 13:30:04.000000000","updated":"2026-09-10 15:35:38.000000000","submitted":"2026-09-10 15:33:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004178","meta_rev_id":"f961a4d78193c2c905e321eb81cc772f70c4fd07","_number":1004178,"virtual_id_number":1004178,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-10 15:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2026-09-10 13:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2026-09-09 15:27:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2026-09-10 13:38:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 14:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-05 14:47:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 15:27:25.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-09-10 13:38:55.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"f0f044e5e96879daf67f806899879fc5d2c47d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 13:30:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55d302826e0d8340e836fea1bd8cc74943d42d2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 14:00:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aaa7bbb29264f9c847861a3876090a2\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f7f3b618b5944319b5f20666d413446a : SUCCESS in 17m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e4a5b9d63c314676a350b0e71d144293 : SUCCESS in 19m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5ff53009475a4ddf84bbc94d51016fa9 : SUCCESS in 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cbacc7450b62e95303996e785f73bdc2d664aac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 14:47:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fded328110b4d6f98fe450396dc4691\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3b8e1c778c1e4cf3859edaf6eb7bec84 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acd18f3c79d64f4ba5964967c059b625 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3c7056cdd34299ab6a81feedeeb4cc : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9eaab6b0a9de447a8596f8d84dc8206e : SUCCESS in 6m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dfd591816c61438580bfce625b0ff331 : SUCCESS in 11m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/32b6e16036c44cf4b0485f4d7aefdbf3 : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7994cd3d5fd447d5a2ea57950cfe973e : SUCCESS in 6m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/035659314a264d7db5ce62d966f1f9ef : SUCCESS in 3m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/cf99735da4e44f5e84a4933d99708b52 : SUCCESS in 43m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1651d86b227e4eb0a667e36b0c268fc1 : SUCCESS in 1h 15m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec4b772dc18f4d2fab2b1c82b987eb7a : SUCCESS in 11m 11s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/86cef5783222489198ba4ab17b16f3f5 : SUCCESS in 9m 47s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/5d4ace275a1d402a9ebe5bd72cbb3d45 : SUCCESS in 25m 55s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/cdf1363c79b84182a5467c08142fbbce : FAILURE in 8m 10s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/01f3d9053d3344d0904b10f9adc083fc : FAILURE in 26m 09s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/6d14984c51c4489ebca0140ef0b32873 : SUCCESS in 46m 52s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/b2b97cc5d7394203a3274ad61259fd64 : SUCCESS in 40m 03s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/167fd22739b84f6e876a5d4eca478988 : FAILURE in 39m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffef1c868cef4175bec6912cd9e1aab6 : SUCCESS in 30m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/bb75f66126694469873b51e799eb385e : SUCCESS in 21m 45s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f5952a25a4a34738a4e8f1d6dbf048e2 : SUCCESS in 6m 13s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":1},{"id":"8f5c0052329ab07ba8bd877ca447a1a96c0dd840","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-09 15:27:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f948efbcd55e8a54a72099830715b2b5184bbb3d","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2026-09-10 13:38:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"949f225bfdbc0afa156d2f51e83780212bff88f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:39:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"980420726270bf1fcef69f413d10c3a68ff699a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:33:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaf8ed9c93d74876a2e6580b904dcdfb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6c2b516e4632469e8be5ed20c3424fac : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f536e9059e448f19e0ee9c443598b43 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b406c35048a44e486314b76c13c8a76 : SUCCESS in 11m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84730020a8904631a8d145a0584795e8 : SUCCESS in 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be609d2b5b2e4a74bf5e2c1b9e61a2ae : SUCCESS in 14m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fb4ff9fa682c4ba483b2bb53f85fdb12 : SUCCESS in 4m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/3dce434877c6462fa67bd6f5172a1fe8 : SUCCESS in 1h 07m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/777bf7e9cd3e4ba38c8dfcfd475950ba : SUCCESS in 1h 48m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffff19bd4d8945d38c0e69fdf72b4f66 : SUCCESS in 5m 20s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/5d343b52fabd483d8c7dc5e7b7ec768d : SUCCESS in 42m 29s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/d21166d0234a4e73a65df30f0309df5a : SUCCESS in 33m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12538df7b666464cba51642da57c735c : SUCCESS in 26m 27s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e457b7707f9c485fb12affcaf5aba517 : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7c7cdcbe233692bc2672fcb2b378f56b90b582bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:33:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f961a4d78193c2c905e321eb81cc772f70c4fd07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 15:35:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b33a460f85cb47a698f2b06801ddf1bf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7916ebe0cfa646c3a0faee448a06bd20 : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19094f69afb14dce913f6b2b1fabf119 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb226d0fda7ba1f1decd134fbfe7cff0df607f3f","revisions":{"cb226d0fda7ba1f1decd134fbfe7cff0df607f3f":{"kind":"REWORK","_number":1,"created":"2026-09-05 13:30:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/1004178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/1004178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/1004178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/1004178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/1004178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/1004178/1"}}},"commit":{"parents":[{"commit":"16afc813b7e6de727d8a91e065d7824b06e32925","subject":"Merge \"Abandon server-side paged search cursor when sizelimit is reached\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/16afc813b7e6de727d8a91e065d7824b06e32925"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 13:28:31.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 13:29:47.000000000","tz":540},"subject":"Remove Python 3.10 support","message":"Remove Python 3.10 support\n\nPython 3.10 was removed from tested runtimes during this cycle, and\nis no longer tested.\n\nChange-Id: Ia2ff0b477fd1a0089b25173cbf54713f04b6bd0e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cb226d0fda7ba1f1decd134fbfe7cff0df607f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cb226d0fda7ba1f1decd134fbfe7cff0df607f3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
