)]}'
{"id":"openstack%2Fkeystone~1004205","triplet_id":"openstack%2Fkeystone~master~I704c7e390c6ee040173120a30fbf9f7abd034e90","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-09 18:28:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I704c7e390c6ee040173120a30fbf9f7abd034e90","subject":"tox: Drop redundant passenv","status":"MERGED","created":"2026-09-05 18:00:15.000000000","updated":"2026-09-09 18:30:03.000000000","submitted":"2026-09-09 18:28:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004205","meta_rev_id":"1e4328c18e49ab857115ebfe90253362c7c7a8f8","_number":1004205,"virtual_id_number":1004205,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-09 18:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2026-09-09 15:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2026-09-09 15:40:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-05 18:45:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-05 19:50:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-09 15:28:51.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"f08778a0f1eb060557696ab51611ebba307b80fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:00:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec1cf35670cd2d78435c076aba4046b308566244","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:11:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7b8c4ebd04d5ade1c98da90282d96d57a6c2a0b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-05 18:14:11.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b8c1ee99a5fd6d3840aff35527060763f8958c78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 18:45:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e3f19586c3f47a6b16a8e91c092f9dc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/83332ad265ff4653b01c0d563b54089e : SUCCESS in 21m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4927ac2a9eac428ca7ccc28e35966ef8 : SUCCESS in 17m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/029817c02d10411f9c0812bc0483e580 : SUCCESS in 21m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2e1228696d7253c036366ecd40895c77d2f24334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 19:50:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01b734113447499a9ad70ceea3a818bd\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b74834f68a054d119c42e6740bec8a72 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78292be7aef74c5bbb01126e91b9d81d : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8bbbbd4fba54d43b87694ed011ebe24 : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4abfe196dec849c4a606f66b16278217 : SUCCESS in 9m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45b5c9a8d2cc4d8a8f5f53920800da75 : SUCCESS in 13m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1e2f701a914e4ebabff09863a7b04118 : SUCCESS in 16m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59bc3af1e82e40a6afff97b6d2f230ec : SUCCESS in 11m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/2015955284f94657a2ef0867e6c6fe9f : SUCCESS in 1h 06m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2d99e88d92d4cb2aa9ac5efe764ea5e : SUCCESS in 1h 34m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bfb4afc5f7540ea8fa037bad1a3bf82 : SUCCESS in 14m 28s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/266bb930d06f44a38ee767e02f442faf : SUCCESS in 9m 53s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/6481b7e005814e6a8dde9648736e2855 : SUCCESS in 47m 25s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/0693975790194fb0b56666a7e769b422 : FAILURE in 18m 53s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/2338c6377cdb40358f3c26012de4536c : FAILURE in 28m 06s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/a323ab83c0cc41fdaa611f03d506cf1c : SUCCESS in 50m 53s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/17a9993c53b346a9a9a60bbd7530cc1e : SUCCESS in 21m 26s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8fc2a01bdebc48828da8f3bc20ddae7f : SUCCESS in 29m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2c9e7b1bb374135bc8822a0a48889b2 : SUCCESS in 52m 10s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c2fcbfce22ba417cb0fdd8ce34ecb136 : SUCCESS in 35m 33s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a8efa9339ad04bd6b39058770d3b6aac : SUCCESS in 7m 59s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"f5f24ee7fd3d0d5facb47418a5f0358e3513ed90","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-09 15:28:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"610786faecd9178b4f33a67969752b425ecdb0b7","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-09 15:40:45.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"21f42ff355fe463d2008d90c26de0a22a71b183f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 15:41:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aae0ae80cfd6d24868617b7d94993ba2493f8d6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:28:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d27ad5b1f15a45e585244554695338c7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f51b47187b204d10959bf1e881e6acf0 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f030ce30b44ae2859deb290a7df04e : SUCCESS in 8m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b7077722c584317a06bd7caf21a83dc : SUCCESS in 15m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2867330b5e3349b8a84b594d1cc1b6fc : SUCCESS in 10m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1033dabc5a942b49d083f98fa240485 : SUCCESS in 17m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/6e2c2218d7d24ad8a64c4050e421c22e : SUCCESS in 1h 06m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/937e1b39da91423289a9bac9275cd5f7 : SUCCESS in 1h 30m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00185b9791ea4492910e5ed3adbbfc19 : SUCCESS in 15m 13s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/7ac405e29817482b97d23b5f38ea20f1 : SUCCESS in 54m 24s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/90bbaf7362d64c00adf4ef0177a8fb73 : SUCCESS in 33m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/472c6177d48b461b8bad82f251a2b4c7 : SUCCESS in 43m 10s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f23c5f88f448482ba5226e801e2fe193 : SUCCESS in 43m 29s","accounts_in_message":[],"_revision_number":3},{"id":"4cb05d9972f1159d61003ca88294897f60e45028","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:28:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1e4328c18e49ab857115ebfe90253362c7c7a8f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 18:30:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/969bc5e0ac694da7866c34d68a49588e\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/70222ee90cd2401f9ec7d6c8ec0a1b4d : SUCCESS in 54s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae849d0bf594c888ce73cd392e8fa44 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a4312950ad94d17b02a5405983427bc : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d6a5cee02626a7e026dc3ccb9a4fceec041245a2","revisions":{"1626afba512a092f87522b1c055b0df3ca1f638f":{"kind":"REWORK","_number":1,"created":"2026-09-05 18:00:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/1004205/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/05/1004205/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/1004205/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/1004205/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/1004205/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/05/1004205/1"}}},"commit":{"parents":[{"commit":"16afc813b7e6de727d8a91e065d7824b06e32925","subject":"Merge \"Abandon server-side paged search cursor when sizelimit is reached\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/16afc813b7e6de727d8a91e065d7824b06e32925"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 17:59:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 17:59:35.000000000","tz":540},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\nAlso PBR_VERSION variable seems to be unused.\n\ncf. https: //tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I704c7e390c6ee040173120a30fbf9f7abd034e90\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1626afba512a092f87522b1c055b0df3ca1f638f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1626afba512a092f87522b1c055b0df3ca1f638f"}]},"branch":"refs/heads/master"},"c07b12454efb38fa0822eee97038b93bf38f7958":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-09-05 18:11:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/1004205/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/05/1004205/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/1004205/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/1004205/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/1004205/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/05/1004205/2"}}},"commit":{"parents":[{"commit":"16afc813b7e6de727d8a91e065d7824b06e32925","subject":"Merge \"Abandon server-side paged search cursor when sizelimit is reached\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/16afc813b7e6de727d8a91e065d7824b06e32925"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 17:59:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:11:45.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have add these explicitly.\nAlso PBR_VERSION variable seems to be unused.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I704c7e390c6ee040173120a30fbf9f7abd034e90\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c07b12454efb38fa0822eee97038b93bf38f7958"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c07b12454efb38fa0822eee97038b93bf38f7958"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d6a5cee02626a7e026dc3ccb9a4fceec041245a2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-05 18:14:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/05/1004205/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/05/1004205/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/1004205/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/1004205/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/1004205/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/05/1004205/3"}}},"commit":{"parents":[{"commit":"16afc813b7e6de727d8a91e065d7824b06e32925","subject":"Merge \"Abandon server-side paged search cursor when sizelimit is reached\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/16afc813b7e6de727d8a91e065d7824b06e32925"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 17:59:35.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-05 18:14:11.000000000","tz":0},"subject":"tox: Drop redundant passenv","message":"tox: Drop redundant passenv\n\nSince 4.0.0[1], tox passes basic environment variables automatically\nand we don\u0027t have to add these explicitly.\nAlso PBR_VERSION variable seems to be unused.\n\ncf. https://tox.wiki/en/latest/reference/config.html#passenv\n\nSome variables are always passed through to ensure the basic\nfunctionality of standard library functions or tooling like pip.\nThis is also not case sensitive on all platforms except Windows.\n\n[1] https://github.com/tox-dev/tox/commit/afb3b395dd\n\nChange-Id: I704c7e390c6ee040173120a30fbf9f7abd034e90\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d6a5cee02626a7e026dc3ccb9a4fceec041245a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d6a5cee02626a7e026dc3ccb9a4fceec041245a2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
