)]}'
{"id":"openstack%2Fkeystone~1004354","triplet_id":"openstack%2Fkeystone~master~I30714a6176784e48ac48f8647180eda995dac937","project":"openstack/keystone","branch":"master","attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-09-06 22:38:01.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I30714a6176784e48ac48f8647180eda995dac937","subject":"Fix typos","status":"NEW","created":"2026-09-06 22:37:03.000000000","updated":"2026-09-07 00:48:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3864ed42d454230e8b792bade143b6df13e3fbf7","_number":1004354,"virtual_id_number":1004354,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-07 00:48:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-06 22:38:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-09-06 23:03:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-09-07 00:48:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dbcbec2f2e8625f8a3de324a2c4aa54481e09f83","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:37:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"577d2d779050ceb82ff35d40bdd41af70e95b0fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:37:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0f71490c51eb1e449098520b046e48a4252de7de","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 22:38:01.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"5d8c4b11ff46f60ce0e5cfea276f38c59f6d0242","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 23:03:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a4280a60a85480fac59d3617926fe84\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1707523af6874e41a8b02547e2cddc07 : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b948ab21d2014dd28f10074f0e45c82b : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1ab776d5ffc94ed7bc4d1fbb7a8da900 : SUCCESS in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15044ee1b07c26f99a0bb9a37929a083be910b62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:04:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"df0df732341831ba6d1a65764d76935fc2197641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:05:10.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"0fc03f58f51a3f86c742e546075c8caefaae3553","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 23:37:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/037bcfb52a5d48acace90c98e12eae00\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc079f639dc045a39634d649289a5580 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/62ed44a81eba413e9882c28787e9d059 : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6fd7531597b146bf949425274c4187b9 : SUCCESS in 30m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3536b0bf4d9a3032ae6a64698802e0cbd1497c39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-06 23:38:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"6bd7be15b81ebaf45efb9f9d25b105a91e0e47e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 00:10:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8366c7ae269f48d68eb1d183a73757c6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3f11fe0f43f0406289c2b34454618433 : SUCCESS in 20m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1acb8b94709648de83526feefaa340de : SUCCESS in 19m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e0baf77051e44a889c26e08e4566a69c : SUCCESS in 30m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3864ed42d454230e8b792bade143b6df13e3fbf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 00:48:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0b4c866543949688a748aebcf62385d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2f030274b4e4bc8bd24678f80147968 : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c2211ca0c194b82be17b491118702a3 : SUCCESS in 7m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87adfcd3fe9d4af9b4361851f3ee5e64 : SUCCESS in 7m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c848621335744b3190a81d2bd9cdd6a2 : SUCCESS in 8m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/324677efc3774d1f8c01d27e54d4dc63 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f63554facfe4345a92dae1f7bce90ea : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/110a995ef5184bf598f1f52d3996551b : SUCCESS in 1h 05m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a89bbc5e11d4bdea3479d182f614dd6 : SUCCESS in 1h 08m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de9e688a218844848e1b172663a9be76 : SUCCESS in 14m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c5534e3b66ac47ccbbf4d58ef7cb0e1a : SUCCESS in 15m 36s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/9dc4619ac1524c55ab58800f7822b9c0 : SUCCESS in 49m 34s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/f031616581ac456a87da0d1bb6b818ae : FAILURE in 10m 17s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/8cbeb3f600c94ec78e08c7983b81ee36 : FAILURE in 13m 42s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/045b6ad429be49968858d1fa17ddc920 : SUCCESS in 50m 28s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/5e971128a2aa4cf5b8d8221d0ff61e91 : SUCCESS in 49m 06s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4aa6fccb298a48e19c8ea916ca821522 : SUCCESS in 22m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35e90fa7a7544c1a8addf706f1b9e67a : SUCCESS in 32m 00s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/44a920daa6ed4482bcc2235022d4296d : SUCCESS in 43m 19s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/99d851953b5e4952ba9ff18b7b57a2cf : SUCCESS in 10m 23s (non-voting)\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job keystone-tempest-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5d94554a41ef245d86a6dce1bea6dde6eb606c7a","revisions":{"e45083b01d20f81b633980dc0d15c61e168732c1":{"kind":"REWORK","_number":1,"created":"2026-09-06 22:37:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/54/1004354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/54/1004354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/54/1004354/1"}}},"commit":{"parents":[{"commit":"5f1d13f38f758810ee32d49d22bb3703ea0d2adf","subject":"Merge \"Use a %% to escape the % sign\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f1d13f38f758810ee32d49d22bb3703ea0d2adf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:37:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:37:03.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I30714a6176784e48ac48f8647180eda995dac937\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e45083b01d20f81b633980dc0d15c61e168732c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e45083b01d20f81b633980dc0d15c61e168732c1"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"5959d42ddbbad8043fa543c5a163ba151396a89f":{"kind":"REWORK","_number":2,"created":"2026-09-06 22:37:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/54/1004354/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/54/1004354/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/54/1004354/2"}}},"commit":{"parents":[{"commit":"5f1d13f38f758810ee32d49d22bb3703ea0d2adf","subject":"Merge \"Use a %% to escape the % sign\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f1d13f38f758810ee32d49d22bb3703ea0d2adf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:37:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:37:54.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: I30714a6176784e48ac48f8647180eda995dac937\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5959d42ddbbad8043fa543c5a163ba151396a89f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5959d42ddbbad8043fa543c5a163ba151396a89f"}]},"branch":"refs/heads/master"},"a6edc060f997cc111492d8dceef67fd5b8592e43":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-06 23:04:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/54/1004354/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/54/1004354/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/54/1004354/3"}}},"commit":{"parents":[{"commit":"5f1d13f38f758810ee32d49d22bb3703ea0d2adf","subject":"Merge \"Use a %% to escape the % sign\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f1d13f38f758810ee32d49d22bb3703ea0d2adf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:37:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:04:20.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I30714a6176784e48ac48f8647180eda995dac937\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a6edc060f997cc111492d8dceef67fd5b8592e43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a6edc060f997cc111492d8dceef67fd5b8592e43"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b662d248b658871a1a1e731501e1b7ef4b729960":{"kind":"REWORK","_number":4,"created":"2026-09-06 23:05:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/54/1004354/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/54/1004354/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/54/1004354/4"}}},"commit":{"parents":[{"commit":"5f1d13f38f758810ee32d49d22bb3703ea0d2adf","subject":"Merge \"Use a %% to escape the % sign\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f1d13f38f758810ee32d49d22bb3703ea0d2adf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:37:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:05:10.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I30714a6176784e48ac48f8647180eda995dac937\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b662d248b658871a1a1e731501e1b7ef4b729960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b662d248b658871a1a1e731501e1b7ef4b729960"}]},"branch":"refs/heads/master"},"5d94554a41ef245d86a6dce1bea6dde6eb606c7a":{"kind":"REWORK","_number":5,"created":"2026-09-06 23:38:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/54/1004354/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/54/1004354/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/1004354/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/54/1004354/5"}}},"commit":{"parents":[{"commit":"5f1d13f38f758810ee32d49d22bb3703ea0d2adf","subject":"Merge \"Use a %% to escape the % sign\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5f1d13f38f758810ee32d49d22bb3703ea0d2adf"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 22:37:03.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-06 23:38:00.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: I30714a6176784e48ac48f8647180eda995dac937\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5d94554a41ef245d86a6dce1bea6dde6eb606c7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5d94554a41ef245d86a6dce1bea6dde6eb606c7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
