)]}'
{"id":"openstack%2Fkeystone~173643","triplet_id":"openstack%2Fkeystone~master~I2cd261f2377da3738e1b0befe62166dac5c49a2f","project":"openstack/keystone","branch":"master","topic":"websso_doc_typo","hashtags":[],"change_id":"I2cd261f2377da3738e1b0befe62166dac5c49a2f","subject":"Fix incorrect setting in WebSSO documentation","status":"MERGED","created":"2015-04-15 01:53:36.000000000","updated":"2015-04-15 05:21:21.000000000","submitted":"2015-04-15 05:21:20.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0204a5985b8748615cfe482318ca624f37c09c06","_number":173643,"virtual_id_number":173643,"owner":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":2,"date":"2015-04-15 05:21:20.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":9101,"name":"wanghong","email":"reviewwanghong@126.com","username":"wanghong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"all":[{"value":2,"date":"2015-04-15 02:46:08.000000000","_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-04-15 02:55:33.000000000","_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":1,"date":"2015-04-15 02:36:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9101,"name":"wanghong","email":"reviewwanghong@126.com","username":"wanghong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"all":[{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-04-15 02:55:33.000000000","_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":9101,"name":"wanghong","email":"reviewwanghong@126.com","username":"wanghong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"_account_id":9101,"name":"wanghong","email":"reviewwanghong@126.com","username":"wanghong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-04-15 02:36:08.000000000","updated_by":{"_account_id":9101,"name":"wanghong","email":"reviewwanghong@126.com","username":"wanghong"},"reviewer":{"_account_id":9101,"name":"wanghong","email":"reviewwanghong@126.com","username":"wanghong"},"state":"REVIEWER"},{"updated":"2015-04-15 02:46:08.000000000","updated_by":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"reviewer":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"state":"REVIEWER"},{"updated":"2015-04-15 02:55:33.000000000","updated_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"reviewer":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"state":"REVIEWER"},{"updated":"2015-04-15 05:21:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"96ba2e4b38f0e8a1bc94484a09341f661604440a","author":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"date":"2015-04-15 01:53:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8866e636f28186b3aeb0f77aa96bd3e3e6b5ecf3","author":{"_account_id":9101,"name":"wanghong","email":"reviewwanghong@126.com","username":"wanghong"},"date":"2015-04-15 02:36:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"af7fe9350927f3eaa27d0a840f06d974c7201da6","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2015-04-15 02:46:08.000000000","message":"Patch Set 1: Code-Review+2\n\nwhoops, thanks","accounts_in_message":[],"_revision_number":1},{"id":"7409a9b84b0f10bb73125b6a2ab430835f312b2a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-15 02:55:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-pep8 http://logs.openstack.org/43/173643/1/check/gate-keystone-pep8/4c17812/ : SUCCESS in 2m 56s\n- gate-keystone-docs http://docs-draft.openstack.org/43/173643/1/check/gate-keystone-docs/d93809c//doc/build/html/ : SUCCESS in 6m 32s\n- gate-keystone-python27 http://logs.openstack.org/43/173643/1/check/gate-keystone-python27/8e93968/ : SUCCESS in 10m 31s\n- check-tempest-dsvm-full http://logs.openstack.org/43/173643/1/check/check-tempest-dsvm-full/2cdd0e7/ : SUCCESS in 45m 38s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/43/173643/1/check/check-tempest-dsvm-postgres-full/40a6523/ : SUCCESS in 46m 20s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/43/173643/1/check/check-tempest-dsvm-neutron-full/33df94a/ : SUCCESS in 57m 47s\n- check-grenade-dsvm http://logs.openstack.org/43/173643/1/check/check-grenade-dsvm/b510db1/ : SUCCESS in 44m 18s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/173643/1/check/gate-tempest-dsvm-large-ops/53b1c95/ : SUCCESS in 23m 02s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/173643/1/check/gate-tempest-dsvm-neutron-large-ops/c25fe45/ : SUCCESS in 24m 00s\n- check-swift-dsvm-functional http://logs.openstack.org/43/173643/1/check/check-swift-dsvm-functional/b5496a2/ : SUCCESS in 21m 52s\n- gate-rally-dsvm-keystone http://logs.openstack.org/43/173643/1/check/gate-rally-dsvm-keystone/f8a35e7/ : SUCCESS in 29m 41s (non-voting)\n- gate-keystone-tox-bandit http://logs.openstack.org/43/173643/1/check/gate-keystone-tox-bandit/69699ed/ : SUCCESS in 1m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a42d482c0aa8394006941e417ae51179b92ac21f","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2015-04-15 02:55:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b4bc4b23a9e1cd437d1b56a01df083a0e65c5d0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-15 02:55:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"8a6b877f2d836ee9519e030e41f54725cd4ed471","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-15 05:21:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-keystone-docs http://docs-draft.openstack.org/43/173643/1/gate/gate-keystone-docs/bc33ee1//doc/build/html/ : SUCCESS in 5m 07s\n- gate-keystone-pep8 http://logs.openstack.org/43/173643/1/gate/gate-keystone-pep8/e643bcb/ : SUCCESS in 3m 05s\n- gate-keystone-python27 http://logs.openstack.org/43/173643/1/gate/gate-keystone-python27/39049f5/ : SUCCESS in 10m 13s\n- gate-tempest-dsvm-full http://logs.openstack.org/43/173643/1/gate/gate-tempest-dsvm-full/531b455/ : SUCCESS in 44m 51s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/43/173643/1/gate/gate-tempest-dsvm-postgres-full/34e3809/ : SUCCESS in 41m 15s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/43/173643/1/gate/gate-tempest-dsvm-neutron-full/d8fc238/ : SUCCESS in 59m 35s\n- gate-grenade-dsvm http://logs.openstack.org/43/173643/1/gate/gate-grenade-dsvm/31874aa/ : SUCCESS in 43m 28s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/43/173643/1/gate/gate-tempest-dsvm-large-ops/ef23c9d/ : SUCCESS in 26m 22s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/43/173643/1/gate/gate-tempest-dsvm-neutron-large-ops/5eab5f3/ : SUCCESS in 21m 27s\n- gate-swift-dsvm-functional http://logs.openstack.org/43/173643/1/gate/gate-swift-dsvm-functional/8eeac52/ : SUCCESS in 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"332e6e12584b5fc26c1bc507345f274e32d25871","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-04-15 05:21:21.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2713c8adedbe00baa6e5750bf981a641c0a7a65","revisions":{"c2713c8adedbe00baa6e5750bf981a641c0a7a65":{"kind":"REWORK","_number":1,"created":"2015-04-15 01:53:36.000000000","uploader":{"_account_id":9098,"name":"Nathan Kinder","email":"nkinder@redhat.com","username":"nkinder"},"ref":"refs/changes/43/173643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/43/173643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/173643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/173643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/173643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/43/173643/1"}}},"commit":{"parents":[{"commit":"2298b589f95857d4417ddca713871ad6ffcdf5bf","subject":"Merge \"Redundant events on group grant revocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2298b589f95857d4417ddca713871ad6ffcdf5bf"}]}],"author":{"name":"Nathan Kinder","email":"nkinder@redhat.com","date":"2015-04-15 01:52:03.000000000","tz":-420},"committer":{"name":"Nathan Kinder","email":"nkinder@redhat.com","date":"2015-04-15 01:52:03.000000000","tz":-420},"subject":"Fix incorrect setting in WebSSO documentation","message":"Fix incorrect setting in WebSSO documentation\n\nThe WebSSO documentation describes the WEBSSO_INITIAL_CHOICE setting,\nbut it then shows an example with the incorrect configuration setting\nname (WEBSSO_CHOICES).  This corrects the example.\n\nChange-Id: I2cd261f2377da3738e1b0befe62166dac5c49a2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c2713c8adedbe00baa6e5750bf981a641c0a7a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c2713c8adedbe00baa6e5750bf981a641c0a7a65"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2298b589f95857d4417ddca713871ad6ffcdf5bf","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
