)]}'
{"id":"openstack%2Fkeystone~247017","triplet_id":"openstack%2Fkeystone~master~I7ec3335f6166b4ec211bbae7cf9281efd4b0a982","project":"openstack/keystone","branch":"master","topic":"bug/1410029","hashtags":[],"change_id":"I7ec3335f6166b4ec211bbae7cf9281efd4b0a982","subject":"Removed conflict wrapper on delete_project","status":"ABANDONED","created":"2015-11-18 16:27:12.000000000","updated":"2016-11-14 21:28:47.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2981f269791dc2ad0091759b9bd97f571d6e5134","_number":247017,"virtual_id_number":247017,"owner":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":-1,"date":"2016-01-29 05:59:28.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":17123,"name":"Tom Cocozzello (tjcocozz)","email":"tjcocozz@us.ibm.com","username":"tcocozzello"},{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},{"date":"2015-11-23 14:15:03.000000000","_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},{"date":"2015-11-23 19:05:39.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"date":"2015-11-23 14:18:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2015-11-20 21:44:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17123,"name":"Tom Cocozzello (tjcocozz)","email":"tjcocozz@us.ibm.com","username":"tcocozzello"},{"value":-1,"date":"2016-06-22 22:19:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},{"value":-1,"date":"2016-01-21 02:59:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":-1,"date":"2016-05-10 22:52:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":17123,"name":"Tom Cocozzello (tjcocozz)","email":"tjcocozz@us.ibm.com","username":"tcocozzello"},{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},{"_account_id":17123,"name":"Tom Cocozzello (tjcocozz)","email":"tjcocozz@us.ibm.com","username":"tcocozzello"},{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-11-18 16:29:35.000000000","updated_by":{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},"reviewer":{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},"state":"REVIEWER"},{"updated":"2015-11-20 21:44:49.000000000","updated_by":{"_account_id":17123,"name":"Tom Cocozzello (tjcocozz)","email":"tjcocozz@us.ibm.com","username":"tcocozzello"},"reviewer":{"_account_id":17123,"name":"Tom Cocozzello (tjcocozz)","email":"tjcocozz@us.ibm.com","username":"tcocozzello"},"state":"REVIEWER"},{"updated":"2015-11-23 14:18:47.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"state":"REVIEWER"},{"updated":"2015-11-23 19:05:39.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2016-01-29 05:59:28.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2016-05-10 22:52:21.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2016-06-22 22:19:38.000000000","updated_by":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"reviewer":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"state":"REVIEWER"}],"messages":[{"id":"9294352a34eeba003853cf6e39816ae07fc8b83f","author":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"date":"2015-11-18 16:27:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c71fc60f5306d5b8d7ae7c10f48bdb2a6edc647","author":{"_account_id":10046,"name":"Henrique Truta","email":"henrique@lsd.ufcg.edu.br","username":"henriquetruta"},"date":"2015-11-18 16:29:35.000000000","message":"Patch Set 1: Code-Review-1\n\nYou should explain in the commit message why it is being done, even we have it at the bug description. Also, capitalize the \u0027removed\u0027.","accounts_in_message":[],"_revision_number":1},{"id":"80f5e440c55693f7b84768820d157403f4576334","author":{"_account_id":17123,"name":"Tom Cocozzello (tjcocozz)","email":"tjcocozz@us.ibm.com","username":"tcocozzello"},"date":"2015-11-18 16:30:40.000000000","message":"Patch Set 1: Code-Review-1\n\nhttps://github.com/openstack/keystone/blob/ab063791d3e516644996a5261977b73e92dc5ac7/keystone/assignment/controllers.py#L451\n\nThe bug references removing a different wrappers as well","accounts_in_message":[],"_revision_number":1},{"id":"09ae73e440048f88ec6587113478bf2100b60e1e","author":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"date":"2015-11-18 16:50:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4f0aa28efc3aa0b5d7fb0b46cf71e7de6033bd40","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-18 18:36:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-pep8 http://logs.openstack.org/17/247017/2/check/gate-keystone-pep8/0c516c7/ : SUCCESS in 3m 26s\n- gate-keystone-docs http://docs-draft.openstack.org/17/247017/2/check/gate-keystone-docs/9746116//doc/build/html/ : SUCCESS in 4m 52s\n- gate-keystone-python27 http://logs.openstack.org/17/247017/2/check/gate-keystone-python27/39f1a53/ : SUCCESS in 7m 28s\n- gate-keystone-python34 http://logs.openstack.org/17/247017/2/check/gate-keystone-python34/2da2205/ : SUCCESS in 7m 31s\n- gate-tempest-dsvm-full http://logs.openstack.org/17/247017/2/check/gate-tempest-dsvm-full/fab93b9/ : SUCCESS in 41m 58s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/17/247017/2/check/gate-tempest-dsvm-postgres-full/3f17cb6/ : SUCCESS in 1h 04m 25s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/17/247017/2/check/gate-tempest-dsvm-neutron-full/9d55d81/ : SUCCESS in 1h 16m 57s\n- gate-grenade-dsvm http://logs.openstack.org/17/247017/2/check/gate-grenade-dsvm/216cab6/ : SUCCESS in 45m 31s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/17/247017/2/check/gate-tempest-dsvm-large-ops/eab33e8/ : SUCCESS in 42m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/17/247017/2/check/gate-tempest-dsvm-neutron-large-ops/53a1bed/ : SUCCESS in 26m 30s\n- gate-rally-dsvm-keystone http://logs.openstack.org/17/247017/2/check/gate-rally-dsvm-keystone/02084dc/ : SUCCESS in 31m 37s (non-voting)\n- gate-keystone-tox-bandit http://logs.openstack.org/17/247017/2/check/gate-keystone-tox-bandit/1bf3aae/ : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2},{"id":"d4e59adea7d991de55c31fdcf9bdb113bff93b45","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2015-11-20 20:27:58.000000000","message":"Patch Set 2: Code-Review-1\n\nI really think that we have to add some tests to show that decorator was useless.","accounts_in_message":[],"_revision_number":2},{"id":"e43bea930a9c3aea7a5adc4857796240b44ac0be","author":{"_account_id":17123,"name":"Tom Cocozzello (tjcocozz)","email":"tjcocozz@us.ibm.com","username":"tcocozzello"},"date":"2015-11-20 21:44:49.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nthe bug mentions \"we don\u0027t use it for any other delete_entity() method other than project\"  just an fyi you also use it for delete_trust() vvv\n\nhttp://git.openstack.org/cgit/openstack/keystone/tree/keystone/trust/backends/sql.py#n175\n\n\nI think this is good to go.  A test would be impossible since you can\u0027t get that error.  Good catch Ajaya.  And the tests currently should be testing the full functionality of the code.  Is there any edge case where this could happen? Maybe a core can clear up some concerns here.","accounts_in_message":[],"_revision_number":2},{"id":"6e120b78d5f1b688bb1981a6c72b03c8fe8883f6","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2015-11-23 12:50:21.000000000","message":"Patch Set 2:\n\nHi Tom, I see your point and I agree in parts with you, but I still thinking that we can add some more tests. \n\nI\u0027m not asking for create a test trying delete a duplicate entry, but if He was saying that we cannot get the errors, I believe that we can add tests showing that we cannot create or update duplicate projects ( I searched any tests on backend and backend_sql related to this but I didn\u0027t found). In addition, there is another case, when a Ops add this duplicate project manually, so maybe this wrapper can be useful.","accounts_in_message":[],"_revision_number":2},{"id":"ccb6a4e9d7180009d314360e0695e9377672048f","author":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"date":"2015-11-23 14:15:03.000000000","message":"Patch Set 2:\n\nHi Raildo,\n\nAppreciate your comments. \n\nIf you say there are no unit tests written for creating a duplicate project, maybe we can file a different bug and create a new patch for it.\n\nIdeally an ops should never have to do such a thing and even if he tries to do so, the constraints on the DB table should not allow an operator to add a duplicate project.","accounts_in_message":[],"_revision_number":2},{"id":"d6e5033bb77080d63ba522d122f8ca65cc46d919","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2015-11-23 14:18:47.000000000","message":"Patch Set 2: -Code-Review\n\nI will not hold the patch since you don\u0027t agree with the comment but I don\u0027t think that is a good practice :)","accounts_in_message":[],"_revision_number":2},{"id":"2584c4ea6b950a643e7d359f059efc7748abf1b5","author":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"date":"2015-11-23 14:38:31.000000000","message":"Patch Set 2:\n\nI am more than happy to work on a patch with test which tries to create a duplicate project.\n\nhttps://bugs.launchpad.net/keystone/+bug/1518997\n\nPatch is being worked on as we speak. :)","accounts_in_message":[],"_revision_number":2},{"id":"aaac80278bfd9f057cf75f37582e6225730173ef","author":{"_account_id":17123,"name":"Tom Cocozzello (tjcocozz)","email":"tjcocozz@us.ibm.com","username":"tcocozzello"},"date":"2015-11-23 14:48:35.000000000","message":"Patch Set 2:\n\nAjaya, you might want to assign that bug to yourself.","accounts_in_message":[],"_revision_number":2},{"id":"9ae3536929f74cc72369b3d56512089066d2725e","author":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"date":"2015-11-23 15:36:56.000000000","message":"Patch Set 2:\n\nHi Raildo, Tom\n\nThere are already tests for checking duplicate project creation.\n\nPlease see:\nhttps://github.com/openstack/keystone/blob/master/keystone/tests/unit/test_backend.py#L750-L768\n\nThe bug has been marked invalid by Steve. So no need to worry about the bug also. FYI, according to Steve bugs are reserved for issues which the end customers face instead of the ones the developers face. :)","accounts_in_message":[],"_revision_number":2},{"id":"cad2ac004b068f7c70b5440bd813ccff0c85a4d7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2015-11-23 19:05:39.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nSince we\u0027re removing these statements; maybe we can include https://github.com/openstack/keystone/blob/f468f32d8f5fa3f01799e5ad3022eb8e065dc32a/keystone/trust/backends/sql.py#L175 \n\n\nAs far as testing goes. A test could be written to add two entities with conflicting IDs/names to the backend (not through the API). Just to document the behavior of the schema.","accounts_in_message":[],"_revision_number":2},{"id":"c74d0c7be2ca141d6afbe4c8e6af828456bae79c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2016-01-21 02:59:19.000000000","message":"Patch Set 2: Code-Review-1\n\n-1 just to unstick from my review queue and to raise awareness of Samuel\u0027s comment in the bug report - https://bugs.launchpad.net/keystone/+bug/1410029","accounts_in_message":[],"_revision_number":2},{"id":"6880e96a8eccd28f300d2aca09234f81cfb2c6c9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-01-29 05:59:28.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"72f5e762364dff22d2c0e936656638571fec7edf","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2016-05-10 22:52:21.000000000","message":"Patch Set 2: Code-Review-1\n\nPlease rebase and bring back.","accounts_in_message":[],"_revision_number":2},{"id":"df4136b933ecf1f0a9dea3867928ef5f482070af","author":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"date":"2016-06-22 22:19:38.000000000","message":"Patch Set 2: Code-Review-1\n\nPlease see my comments on the bug report.","accounts_in_message":[],"_revision_number":2},{"id":"15f355ebf2b6f586f22a7845d614d03d6388bcb0","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-11-14 21:28:47.000000000","message":"Abandoned\n\nno action in 5 months and bug is marked as invalid, auto-abandoning","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"67297b9a78979eaa524d32adaf3703c416b4405e","revisions":{"426de1d921e3c9dcc0aa95a13b545420184b9141":{"kind":"REWORK","_number":1,"created":"2015-11-18 16:27:12.000000000","uploader":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"ref":"refs/changes/17/247017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/247017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/247017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/247017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/247017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/247017/1"}}},"commit":{"parents":[{"commit":"fb10b816c49843052c2b6788308cb980ec83e4f3","subject":"Merge \"test_backend_sql work with python34\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fb10b816c49843052c2b6788308cb980ec83e4f3"}]}],"author":{"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","date":"2015-11-18 16:25:24.000000000","tz":330},"committer":{"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","date":"2015-11-18 16:26:21.000000000","tz":330},"subject":"removed conflict wrapper on delete_project","message":"removed conflict wrapper on delete_project\n\nChange-Id: I7ec3335f6166b4ec211bbae7cf9281efd4b0a982\nCloses-Bug: #1410029\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/426de1d921e3c9dcc0aa95a13b545420184b9141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/426de1d921e3c9dcc0aa95a13b545420184b9141"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fb10b816c49843052c2b6788308cb980ec83e4f3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"67297b9a78979eaa524d32adaf3703c416b4405e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2015-11-18 16:50:27.000000000","uploader":{"_account_id":11428,"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","username":"ajayaa"},"ref":"refs/changes/17/247017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/247017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/247017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/247017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/247017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/247017/2"}}},"commit":{"parents":[{"commit":"fb10b816c49843052c2b6788308cb980ec83e4f3","subject":"Merge \"test_backend_sql work with python34\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fb10b816c49843052c2b6788308cb980ec83e4f3"}]}],"author":{"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","date":"2015-11-18 16:25:24.000000000","tz":330},"committer":{"name":"Ajaya Agrawal","email":"ajku.agr@gmail.com","date":"2015-11-18 16:40:54.000000000","tz":330},"subject":"Removed conflict wrapper on delete_project","message":"Removed conflict wrapper on delete_project\n\nSince the wrapper checks for intergrity error\nand duplicate entry error and it is impossible\nto get these errors while deleting a project.\n\nChange-Id: I7ec3335f6166b4ec211bbae7cf9281efd4b0a982\nCloses-Bug: #1410029\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/67297b9a78979eaa524d32adaf3703c416b4405e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/67297b9a78979eaa524d32adaf3703c416b4405e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fb10b816c49843052c2b6788308cb980ec83e4f3","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
