)]}'
{"id":"openstack%2Fkeystone~280747","triplet_id":"openstack%2Fkeystone~master~Ifb42829988e12e506fc16585ea500976f573f29d","project":"openstack/keystone","branch":"master","topic":"extclean1","hashtags":[],"change_id":"Ifb42829988e12e506fc16585ea500976f573f29d","subject":"Tidy up configuration documentation for inherited assignments","status":"MERGED","created":"2016-02-16 14:59:43.000000000","updated":"2016-02-22 18:05:56.000000000","submitted":"2016-02-22 18:05:56.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4e1e13c25e1e923b24caec5af4adb796e8a69053","_number":280747,"virtual_id_number":280747,"owner":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":2,"date":"2016-02-22 18:05:56.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"all":[{"value":2,"date":"2016-02-16 16:16:15.000000000","_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-02-16 20:49:15.000000000","_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"all":[{"value":1,"date":"2016-02-22 16:54:35.000000000","_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-02-16 20:49:15.000000000","updated_by":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"reviewer":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"state":"REVIEWER"},{"updated":"2016-02-22 16:54:35.000000000","updated_by":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"reviewer":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"state":"REVIEWER"},{"updated":"2016-02-22 18:05:56.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"eef98a52582677ef73e46862a41804c8383f14f5","author":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"date":"2016-02-16 14:59:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7291c2d3aefdce3c96b96a5ac4c31eb49da9ec5c","author":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"date":"2016-02-16 15:01:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"254fa7f5c4285dfbf623f489b88e2eac819e09aa","author":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"date":"2016-02-16 15:03:07.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3718ee708278d6dab77256c5ae75609705785702","author":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"date":"2016-02-16 15:11:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"383c491b80221472200660eadb0c29bf406311a2","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-02-16 16:03:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bcca79a03e1a019ddfa0862205c993596385574a","author":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"date":"2016-02-16 16:12:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5cf1714df3fcae5f1b6293d7d11ce8aa92fad50b","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-02-16 16:16:15.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nthanks for keeping me honest!","accounts_in_message":[],"_revision_number":4},{"id":"d57939007f1f051b81984f8d5791b9ca186f8085","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-16 16:49:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-pep8 http://logs.openstack.org/47/280747/4/check/gate-keystone-pep8/a4637b7/ : SUCCESS in 2m 04s\n- gate-keystone-docs http://docs-draft.openstack.org/47/280747/4/check/gate-keystone-docs/9289214//doc/build/html/ : SUCCESS in 2m 52s\n- gate-keystone-python27 http://logs.openstack.org/47/280747/4/check/gate-keystone-python27/9b43915/ : SUCCESS in 8m 42s\n- gate-keystone-python34 http://logs.openstack.org/47/280747/4/check/gate-keystone-python34/d07f8ab/ : SUCCESS in 6m 46s\n- gate-tempest-dsvm-full http://logs.openstack.org/47/280747/4/check/gate-tempest-dsvm-full/79884ca/ : SUCCESS in 54m 10s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/47/280747/4/check/gate-tempest-dsvm-postgres-full/41afb63/ : SUCCESS in 52m 35s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/47/280747/4/check/gate-tempest-dsvm-neutron-full/79ae35a/ : SUCCESS in 55m 26s\n- gate-grenade-dsvm http://logs.openstack.org/47/280747/4/check/gate-grenade-dsvm/d9d211b/ : FAILURE in 40m 44s\n- gate-keystone-tox-legacy_drivers http://logs.openstack.org/47/280747/4/check/gate-keystone-tox-legacy_drivers/f34c22f/ : SUCCESS in 3m 33s\n- gate-tempest-dsvm-keystone-eventlet-full http://logs.openstack.org/47/280747/4/check/gate-tempest-dsvm-keystone-eventlet-full/ee23c44/ : SUCCESS in 1h 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"399c44bcfbe75329ab81b8fa2f78f063bb77a83d","author":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"date":"2016-02-16 17:03:20.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"e2b160b0263754efd126a5e3a7b63f9f24404d45","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-16 19:35:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-pep8 http://logs.openstack.org/47/280747/4/check/gate-keystone-pep8/28c6221/ : SUCCESS in 1m 59s\n- gate-keystone-docs http://docs-draft.openstack.org/47/280747/4/check/gate-keystone-docs/4e65793//doc/build/html/ : SUCCESS in 4m 31s\n- gate-keystone-python27 http://logs.openstack.org/47/280747/4/check/gate-keystone-python27/2bab778/ : SUCCESS in 7m 16s\n- gate-keystone-python34 http://logs.openstack.org/47/280747/4/check/gate-keystone-python34/0fe2629/ : SUCCESS in 6m 10s\n- gate-tempest-dsvm-full http://logs.openstack.org/47/280747/4/check/gate-tempest-dsvm-full/2bac9ca/ : SUCCESS in 57m 51s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/47/280747/4/check/gate-tempest-dsvm-postgres-full/fb4f785/ : SUCCESS in 1h 01m 45s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/47/280747/4/check/gate-tempest-dsvm-neutron-full/1db7b0b/ : SUCCESS in 45m 33s\n- gate-grenade-dsvm http://logs.openstack.org/47/280747/4/check/gate-grenade-dsvm/aa446ad/ : SUCCESS in 41m 50s\n- gate-keystone-tox-legacy_drivers http://logs.openstack.org/47/280747/4/check/gate-keystone-tox-legacy_drivers/fab3fb8/ : SUCCESS in 4m 51s\n- gate-tempest-dsvm-keystone-eventlet-full http://logs.openstack.org/47/280747/4/check/gate-tempest-dsvm-keystone-eventlet-full/407ecfd/ : SUCCESS in 1h 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7a682cf99dfbd1d7d963f50bba4bd64cb738117a","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2016-02-16 20:49:15.000000000","message":"Patch Set 4: Code-Review+2\n\nNot really core if you can disable it.","accounts_in_message":[],"_revision_number":4},{"id":"e5f674d9abd166b62265b561f221e90663f82cb1","author":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"date":"2016-02-16 21:09:21.000000000","message":"Patch Set 4:\n\nthe disable ability is deprecated, and scheduled for removal (but not in M)","accounts_in_message":[],"_revision_number":4},{"id":"34a462bff331e220e69150793b8dbc8c8458791c","author":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"date":"2016-02-22 13:31:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e0cbcc97a0936bb857ff1a66fcdfca848f0c20bd","author":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"date":"2016-02-22 13:55:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"92687d9e256f39d78d30e6b611bd56d4be097d44","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-02-22 16:54:35.000000000","message":"Patch Set 4: Workflow+1\n\ni\u0027m okay with this, we can make adjustments as we go, but this is correct","accounts_in_message":[],"_revision_number":4},{"id":"c011954106036f7d70ee6e550491acbf5a3298aa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-22 16:54:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"d97b5c049e5db56ab94d36384b715cde8d5ca9eb","date":"2016-02-22 18:05:56.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":4},{"id":"ee5df1643118ffb7e7fe0fb74e28b5de94fd2f4d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-02-22 18:05:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-keystone-docs http://docs-draft.openstack.org/47/280747/4/gate/gate-keystone-docs/2fb44cc//doc/build/html/ : SUCCESS in 4m 29s\n- gate-keystone-pep8 http://logs.openstack.org/47/280747/4/gate/gate-keystone-pep8/be06259/ : SUCCESS in 2m 26s\n- gate-keystone-python27 http://logs.openstack.org/47/280747/4/gate/gate-keystone-python27/0de120c/ : SUCCESS in 8m 38s\n- gate-keystone-python34 http://logs.openstack.org/47/280747/4/gate/gate-keystone-python34/469fc45/ : SUCCESS in 7m 08s\n- gate-tempest-dsvm-full http://logs.openstack.org/47/280747/4/gate/gate-tempest-dsvm-full/609fba0/ : SUCCESS in 44m 18s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/47/280747/4/gate/gate-tempest-dsvm-postgres-full/8202343/ : SUCCESS in 37m 57s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/47/280747/4/gate/gate-tempest-dsvm-neutron-full/2f9a392/ : SUCCESS in 1h 02m 25s\n- gate-grenade-dsvm http://logs.openstack.org/47/280747/4/gate/gate-grenade-dsvm/99ffa16/ : SUCCESS in 44m 31s\n- gate-keystone-tox-legacy_drivers http://logs.openstack.org/47/280747/4/gate/gate-keystone-tox-legacy_drivers/54e1db0/ : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d39bf80e34b3ce5e3b20f2b149765bbe3032a9ce","revisions":{"e77e1923940ab7a2134ffb384fccf81a7121ac80":{"kind":"REWORK","_number":1,"created":"2016-02-16 14:59:43.000000000","uploader":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"ref":"refs/changes/47/280747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/47/280747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/47/280747/1"}}},"commit":{"parents":[{"commit":"a320eaa90323c04082204de9b9e630fec96661aa","subject":"Updating sample configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a320eaa90323c04082204de9b9e630fec96661aa"}]}],"author":{"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","date":"2016-02-16 14:57:22.000000000","tz":0},"committer":{"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","date":"2016-02-16 14:57:22.000000000","tz":0},"subject":"Tidy up configuration documentation for inherited assignments","message":"Tidy up configuration documentation for inherited assignments\n\nRemove the \"extension\" nomenclature, state that this is enabled\nby default and remove the details of earlier experimental\nphases (since they are now more than 3 released ago).\n\nChange-Id: Ifb42829988e12e506fc16585ea500976f573f29d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e77e1923940ab7a2134ffb384fccf81a7121ac80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e77e1923940ab7a2134ffb384fccf81a7121ac80"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a320eaa90323c04082204de9b9e630fec96661aa","is_merged_in_target_branch":true,"change_id":"I974491f8fa556110affeb6ba683d178cf5621f8b","change_number":280545,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"dd8f79a2c94f165efbe16f3e077db7118714ab60":{"kind":"REWORK","_number":2,"created":"2016-02-16 15:01:36.000000000","uploader":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"ref":"refs/changes/47/280747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/47/280747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/47/280747/2"}}},"commit":{"parents":[{"commit":"a320eaa90323c04082204de9b9e630fec96661aa","subject":"Updating sample configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a320eaa90323c04082204de9b9e630fec96661aa"}]}],"author":{"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","date":"2016-02-16 14:57:22.000000000","tz":0},"committer":{"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","date":"2016-02-16 15:01:21.000000000","tz":0},"subject":"Tidy up configuration documentation for inherited assignments","message":"Tidy up configuration documentation for inherited assignments\n\nRemove the \"extension\" nomenclature, state that this is enabled\nby default and remove the details of earlier experimental\nphases (since they are now more than 3 released ago).\n\nChange-Id: Ifb42829988e12e506fc16585ea500976f573f29d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/dd8f79a2c94f165efbe16f3e077db7118714ab60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/dd8f79a2c94f165efbe16f3e077db7118714ab60"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a320eaa90323c04082204de9b9e630fec96661aa","is_merged_in_target_branch":true,"change_id":"I974491f8fa556110affeb6ba683d178cf5621f8b","change_number":280545,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"680bde2791095184538b0ffb3b6b650b74b2af93":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2016-02-16 15:03:07.000000000","uploader":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"ref":"refs/changes/47/280747/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/47/280747/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/47/280747/3"}}},"commit":{"parents":[{"commit":"a320eaa90323c04082204de9b9e630fec96661aa","subject":"Updating sample configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a320eaa90323c04082204de9b9e630fec96661aa"}]}],"author":{"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","date":"2016-02-16 14:57:22.000000000","tz":0},"committer":{"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","date":"2016-02-16 15:02:46.000000000","tz":0},"subject":"Tidy up configuration documentation for inherited assignments","message":"Tidy up configuration documentation for inherited assignments\n\nRemove the \"extension\" nomenclature, state that this is enabled\nby default and remove the details of earlier experimental\nphases (since they are now more than 3 releases ago).\n\nChange-Id: Ifb42829988e12e506fc16585ea500976f573f29d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/680bde2791095184538b0ffb3b6b650b74b2af93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/680bde2791095184538b0ffb3b6b650b74b2af93"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a320eaa90323c04082204de9b9e630fec96661aa","is_merged_in_target_branch":true,"change_id":"I974491f8fa556110affeb6ba683d178cf5621f8b","change_number":280545,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d39bf80e34b3ce5e3b20f2b149765bbe3032a9ce":{"kind":"REWORK","_number":4,"created":"2016-02-16 15:11:50.000000000","uploader":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"ref":"refs/changes/47/280747/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/47/280747/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/47/280747/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/47/280747/4"}}},"commit":{"parents":[{"commit":"a320eaa90323c04082204de9b9e630fec96661aa","subject":"Updating sample configuration file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a320eaa90323c04082204de9b9e630fec96661aa"}]}],"author":{"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","date":"2016-02-16 14:57:22.000000000","tz":0},"committer":{"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","date":"2016-02-16 15:11:37.000000000","tz":0},"subject":"Tidy up configuration documentation for inherited assignments","message":"Tidy up configuration documentation for inherited assignments\n\nRemove the \"extension\" nomenclature, state that this is enabled\nby default and remove the details of earlier experimental\nphases (since they are now more than 3 releases ago).\n\nChange-Id: Ifb42829988e12e506fc16585ea500976f573f29d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d39bf80e34b3ce5e3b20f2b149765bbe3032a9ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d39bf80e34b3ce5e3b20f2b149765bbe3032a9ce"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a320eaa90323c04082204de9b9e630fec96661aa","is_merged_in_target_branch":true,"change_id":"I974491f8fa556110affeb6ba683d178cf5621f8b","change_number":280545,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
