)]}'
{"id":"openstack%2Fkeystone~28338","triplet_id":"openstack%2Fkeystone~master~Ib7a5a5f0dcac82229b7b7fb3df4b1ef141586c5f","project":"openstack/keystone","branch":"master","topic":"bug/1176204","hashtags":[],"change_id":"Ib7a5a5f0dcac82229b7b7fb3df4b1ef141586c5f","subject":"Skip IPv6 tests for eventlet dns","status":"MERGED","created":"2013-05-06 19:09:15.000000000","updated":"2013-05-07 03:52:21.000000000","submitted":"2013-05-07 03:52:20.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b839f8fe6d66da1ca338ef72dba8474eab068641","_number":28338,"virtual_id_number":28338,"owner":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":2,"date":"2013-05-07 03:52:20.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":1,"date":"2013-05-07 02:51:22.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},{"value":0,"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":494,"name":"termie","email":"termie@openstack.org","username":"termie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"all":[{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2013-05-07 03:06:06.000000000","_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},{"value":0,"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":494,"name":"termie","email":"termie@openstack.org","username":"termie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"all":[{"value":0,"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":1,"date":"2013-05-07 03:06:06.000000000","_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},{"value":0,"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":494,"name":"termie","email":"termie@openstack.org","username":"termie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":494,"name":"termie","email":"termie@openstack.org","username":"termie"},{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-06 19:49:06.000000000","updated_by":{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},"reviewer":{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},"state":"REVIEWER"},{"updated":"2013-05-06 20:25:00.000000000","updated_by":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"reviewer":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"state":"REVIEWER"},{"updated":"2013-05-06 20:34:20.000000000","updated_by":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"reviewer":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"state":"REVIEWER"},{"updated":"2013-05-06 21:10:17.000000000","updated_by":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"reviewer":{"_account_id":5707,"name":"Henry Nash","email":"henryn@linux.vnet.ibm.com","username":"henry-nash"},"state":"REVIEWER"},{"updated":"2013-05-06 21:10:52.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2013-05-06 23:32:59.000000000","updated_by":{"_account_id":494,"name":"termie","email":"termie@openstack.org","username":"termie"},"reviewer":{"_account_id":494,"name":"termie","email":"termie@openstack.org","username":"termie"},"state":"REVIEWER"},{"updated":"2013-05-07 02:01:57.000000000","updated_by":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"reviewer":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"state":"REVIEWER"},{"updated":"2013-05-07 02:02:32.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2013-05-07 02:51:22.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-05-07 03:06:06.000000000","updated_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"reviewer":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"state":"REVIEWER"},{"updated":"2013-05-07 03:52:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"3bfa3791ce3a0acb9e3c836014cd67f7def43c91","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-05-06 19:10:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"28b0b560ae649a81be12cf084cff08b13ade3b86","author":{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},"date":"2013-05-06 19:35:46.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nsmall nit. see inline comment","accounts_in_message":[],"_revision_number":2},{"id":"b664761cb81bb14b0f7aadddc481321504f636bc","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-05-06 19:37:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a938c7350d9a10e4919064b719fda219a8b41bdb","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-05-06 19:37:19.000000000","message":"Patch Set 3:\n\nNew patchset patch-id matches previous patchset, but whitespace content has changed.","accounts_in_message":[],"_revision_number":3},{"id":"8528033117adcb37bc75e77b43ddb379725389f6","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-05-06 19:38:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f1984576cbf72559dc4f7f6d4cb91757f3f46a6d","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-05-06 19:38:19.000000000","message":"Patch Set 4:\n\nNew patchset patch-id matches previous patchset, but whitespace content has changed.","accounts_in_message":[],"_revision_number":4},{"id":"6a35844d698888f6b8dd3850aedbdaa9b600c1f0","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-05-06 19:38:24.000000000","message":"Patch Set 2: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"58c86a711dbc2a15b0542dafdd27fa2b0b2fd147","author":{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},"date":"2013-05-06 19:49:06.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\nGreat job tracking this down!!!","accounts_in_message":[],"_revision_number":4},{"id":"cc699a1a93b164103ae58990e4929cd75f0300ec","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-06 20:06:33.000000000","message":"Patch Set 4: Works for me\n\nSmokeStack Results (patch set 4):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/85661\n","accounts_in_message":[],"_revision_number":4},{"id":"2b1bf3f767a3b3b8062e37b449afb62ac0701be4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-06 20:23:26.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/28338/4/check/gate-keystone-pep8/4908 : SUCCESS in 00:01:07\n- http://docs-draft.openstack.org/28338/4/check/gate-keystone-docs/3282/doc/build/html/ : SUCCESS in 00:02:02\n- http://logs.openstack.org/28338/4/check/gate-keystone-python26/3802 : SUCCESS in 00:07:59\n- http://logs.openstack.org/28338/4/check/gate-keystone-python26-centos6/46 : FAILURE in 00:06:56 (non-voting)\n- http://logs.openstack.org/28338/4/check/gate-keystone-python27/3814 : SUCCESS in 00:05:27\n- http://logs.openstack.org/28338/4/check/gate-tempest-devstack-vm-full/14720 : SUCCESS in 00:44:21\n- http://logs.openstack.org/28338/4/check/gate-tempest-devstack-vm-postgres-full/14435 : FAILURE in 00:44:59 (non-voting)\n- http://logs.openstack.org/28338/4/check/gate-tempest-devstack-vm-quantum/20903 : SUCCESS in 00:21:35\n- http://logs.openstack.org/28338/4/check/gate-grenade-devstack-vm/2515 : SUCCESS in 00:14:44 (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"6eae360496ba7f19884f56f226a0aa3710017469","author":{"_account_id":5280,"name":"Nicolas Simonds","email":"nic@metacloud.com","username":"nicolas.simonds"},"date":"2013-05-06 20:34:20.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":4},{"id":"4732620dd245f3e2a0bac2e6bdfb025ee3fa6c10","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-05-06 22:02:27.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\neverything is failing gate because of this.  IPv6 is broken across the board, skipping tests is the least of it.","accounts_in_message":[],"_revision_number":4},{"id":"8ee7802c4b24e8191a3766051237fd306058e23d","author":{"_account_id":494,"name":"termie","email":"termie@openstack.org","username":"termie"},"date":"2013-05-06 23:32:59.000000000","message":"Patch Set 4: Looks good to me (core reviewer)\n\nlooks alright, sucks to be skipping more tests, though","accounts_in_message":[],"_revision_number":4},{"id":"9119c81e1f19dd0983b498e9142fb23bc58330f1","author":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"date":"2013-05-06 23:41:56.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":4},{"id":"4f8ab34c584cdc03f3983e3bd6675c2f1b1a66b4","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-05-07 00:05:27.000000000","message":"Patch Set 4: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":4},{"id":"d4ba3685b3db16949ce81104fc169b35fdfa4a8d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-07 00:13:52.000000000","message":"Patch Set 4: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"bd19e8a2ee4fddc8a2d31c4abb4241b099be6fa1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-07 00:59:02.000000000","message":"Patch Set 4: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/28338/4/gate/gate-keystone-docs/3284 : FAILURE in 00:00:46\n- http://logs.openstack.org/28338/4/gate/gate-keystone-pep8/4910 : FAILURE in 00:00:45\n- http://logs.openstack.org/28338/4/gate/gate-keystone-python26/3804 : FAILURE in 00:00:55\n- http://logs.openstack.org/28338/4/gate/gate-keystone-python27/3816 : FAILURE in 00:00:53\n- http://logs.openstack.org/28338/4/gate/gate-tempest-devstack-vm-full/14752 : SUCCESS in 00:42:03\n- http://logs.openstack.org/28338/4/gate/gate-tempest-devstack-vm-postgres-full/14467 : FAILURE in 00:44:58 (non-voting)\n- http://logs.openstack.org/28338/4/gate/gate-tempest-devstack-vm-quantum/20944 : SUCCESS in 00:23:20\n- http://logs.openstack.org/28338/4/gate/gate-grenade-devstack-vm/2539 : SUCCESS in 00:14:59 (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"86a90007a772a28a41652423e16837d62abdedb7","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-05-07 02:00:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"57effae0b795fd704fb02ebb15f0166fb6bab1ae","author":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"date":"2013-05-07 02:01:57.000000000","message":"Patch Set 5: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":5},{"id":"e3ffbaf772d5fe379bee46d8c55128ffff3ff7c5","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2013-05-07 02:02:32.000000000","message":"Patch Set 5: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":5},{"id":"464873736d3b7a373c9298c5ea1f372e7ccacbf1","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-07 02:22:51.000000000","message":"Patch Set 5: Works for me\n\nSmokeStack Results (patch set 5):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/85685\n","accounts_in_message":[],"_revision_number":5},{"id":"c9f3cf9827c5aed207008cd208f673ba6530f78f","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-05-07 02:23:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7db1c0feb80da7207685e1bda3d9126dbacc68e6","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-05-07 02:25:02.000000000","message":"Patch Set 6:\n\nThis is the original patch, with the memcached change removed. I made a minor change to the commit message in order to get gerrit to accept it.","accounts_in_message":[],"_revision_number":6},{"id":"868a4f351fe7470eb6d77c7de590dddb21eebdab","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-07 02:51:22.000000000","message":"Patch Set 6: Works for me\n\nSmokeStack Results (patch set 6):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/85692\n","accounts_in_message":[],"_revision_number":6},{"id":"c8471ee5c6ba75b2285fefb81fb4084a1a530432","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-05-07 03:06:06.000000000","message":"Patch Set 6: Looks good to me (core reviewer); Approved\n\nShould now merge","accounts_in_message":[],"_revision_number":6},{"id":"fdd21343ed0c7e50ff566a61b0da180da7642bd6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-07 03:06:07.000000000","message":"Patch Set 6:\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"3351683409d0a42f5ee2cb733f4ccc5829af4475","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-07 03:08:56.000000000","message":"Patch Set 6: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/28338/6/check/gate-keystone-pep8/4914 : FAILURE in 00:00:18\n- http://logs.openstack.org/28338/6/check/gate-keystone-docs/3288 : FAILURE in 00:00:22\n- http://logs.openstack.org/28338/6/check/gate-keystone-python26/3808 : FAILURE in 00:00:30\n- http://logs.openstack.org/28338/6/check/gate-keystone-python26-centos6/51 : FAILURE in 00:00:31 (non-voting)\n- http://logs.openstack.org/28338/6/check/gate-keystone-python27/3820 : FAILURE in 00:00:22\n- http://logs.openstack.org/28338/6/check/gate-tempest-devstack-vm-full/14771 : SUCCESS in 00:44:45\n- http://logs.openstack.org/28338/6/check/gate-tempest-devstack-vm-postgres-full/14486 : FAILURE in 00:43:49 (non-voting)\n- http://logs.openstack.org/28338/6/check/gate-tempest-devstack-vm-quantum/20963 : SUCCESS in 00:23:08\n- http://logs.openstack.org/28338/6/check/gate-grenade-devstack-vm/2557 : FAILURE in 00:17:12 (non-voting)\n","accounts_in_message":[],"_revision_number":6},{"id":"be5594f7d1228de421f93999d426d26a5cc1500e","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-05-07 03:31:49.000000000","message":"Patch Set 6:\n\nreverify no bug","accounts_in_message":[],"_revision_number":6},{"id":"d12292d57a6cac4634e2e4d6a5fcb42f39a26df1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-07 03:52:20.000000000","message":"Patch Set 6: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/28338/6/gate/gate-keystone-docs/3289/doc/build/html/ : SUCCESS in 00:01:55\n- http://logs.openstack.org/28338/6/gate/gate-keystone-pep8/4915 : SUCCESS in 00:01:07\n- http://logs.openstack.org/28338/6/gate/gate-keystone-python26/3809 : SUCCESS in 00:07:59\n- http://logs.openstack.org/28338/6/gate/gate-keystone-python27/3821 : SUCCESS in 00:05:41\n- http://logs.openstack.org/28338/6/gate/gate-tempest-devstack-vm-full/14774 : SUCCESS in 00:44:19\n- http://logs.openstack.org/28338/6/gate/gate-tempest-devstack-vm-postgres-full/14489 : FAILURE in 00:46:03 (non-voting)\n- http://logs.openstack.org/28338/6/gate/gate-tempest-devstack-vm-quantum/20966 : SUCCESS in 00:21:21\n- http://logs.openstack.org/28338/6/gate/gate-grenade-devstack-vm/2559 : SUCCESS in 00:15:06 (non-voting)\n","accounts_in_message":[],"_revision_number":6},{"id":"25cf1493047468d7f814e9fe8a1f2f68946d55f4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-07 03:52:21.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b55620d7ee0f87b0339285b16dd6fd68e5516b10","revisions":{"433998db4dca65f6c94bc7431acd24a2c56e48eb":{"kind":"REWORK","_number":1,"created":"2013-05-06 19:09:15.000000000","uploader":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"ref":"refs/changes/38/28338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/28338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/28338/1"}}},"commit":{"parents":[{"commit":"22d96b270b6794f15471761073a4d5e1065f35b0","subject":"Merge \"HACKING LDAP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/22d96b270b6794f15471761073a4d5e1065f35b0"}]}],"author":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 18:42:26.000000000","tz":-300},"committer":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 19:08:18.000000000","tz":-300},"subject":"Skip IPv6 tests for eventlet dns","message":"Skip IPv6 tests for eventlet dns\n\nEventlet address resolving doesn\u0027t support IPv6. This is causing the\ngate to fail.\n\nThe failure is somewhat erratic in that it requires that the dnspython\npackage is installed in the environment. When dnspython is there then\neventlet takes over getaddrinfo, and eventlet\u0027s getaddrinfo doesn\u0027t\nsupport IPv6. If dnspython isn\u0027t there, then the eventlet patching\ndoesn\u0027t happen and IPv6 address resolution works.\n\nThe fix is to skip the 3 tests that force the server to listen\non \"::1\".\n\nFixes bug 1176204\n\nOnce eventlet is updated, we can add the tests back in. Here\u0027s the\neventlet code in question:\nhttps://bitbucket.org/eventlet/eventlet/src/e0f578180d7d82d2ed3d8a96d520103503c524ec/eventlet/support/greendns.py?at\u003d0.12#cl-166\n\nChange-Id: Ib7a5a5f0dcac82229b7b7fb3df4b1ef141586c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/433998db4dca65f6c94bc7431acd24a2c56e48eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/433998db4dca65f6c94bc7431acd24a2c56e48eb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"22d96b270b6794f15471761073a4d5e1065f35b0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e428f8239e515a214af6ae883b0b4c27f17622d4":{"kind":"REWORK","_number":2,"created":"2013-05-06 19:10:37.000000000","uploader":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"ref":"refs/changes/38/28338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/28338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/28338/2"}}},"commit":{"parents":[{"commit":"22d96b270b6794f15471761073a4d5e1065f35b0","subject":"Merge \"HACKING LDAP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/22d96b270b6794f15471761073a4d5e1065f35b0"}]}],"author":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 18:42:26.000000000","tz":-300},"committer":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 19:10:39.000000000","tz":-300},"subject":"Skip IPv6 tests for eventlet dns","message":"Skip IPv6 tests for eventlet dns\n\nEventlet address resolving doesn\u0027t support IPv6. This is causing the\ngate to fail.\n\nThe failure is somewhat erratic in that it requires that the dnspython\npackage is installed in the environment. When dnspython is there then\neventlet takes over getaddrinfo, and eventlet\u0027s getaddrinfo doesn\u0027t\nsupport IPv6. If dnspython isn\u0027t there, then the eventlet patching\ndoesn\u0027t happen and IPv6 address resolution works.\n\nThe fix is to skip the 3 tests that force the server to listen\non \"::1\".\n\nFixes bug 1176204\n\nOnce eventlet is updated, we can add the tests back in. Here\u0027s the\neventlet code in question:\nhttps://bitbucket.org/eventlet/eventlet/src/e0f578180d7d82d2ed3d8a96d520103503c524ec/eventlet/support/greendns.py?at\u003d0.12#cl-166\n\nChange-Id: Ib7a5a5f0dcac82229b7b7fb3df4b1ef141586c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e428f8239e515a214af6ae883b0b4c27f17622d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e428f8239e515a214af6ae883b0b4c27f17622d4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"22d96b270b6794f15471761073a4d5e1065f35b0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"d9f944215d836636e4e5209924c9cb1f01bec9f0":{"kind":"REWORK","_number":3,"created":"2013-05-06 19:37:12.000000000","uploader":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"ref":"refs/changes/38/28338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/28338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/28338/3"}}},"commit":{"parents":[{"commit":"22d96b270b6794f15471761073a4d5e1065f35b0","subject":"Merge \"HACKING LDAP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/22d96b270b6794f15471761073a4d5e1065f35b0"}]}],"author":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 18:42:26.000000000","tz":-300},"committer":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 19:35:33.000000000","tz":-300},"subject":"Skip IPv6 tests for eventlet dns","message":"Skip IPv6 tests for eventlet dns\n\nEventlet address resolving doesn\u0027t support IPv6. This is causing the\ngate to fail.\n\nThe failure happens when the dnspython package is installed in the\nenvironment. When dnspython is there then eventlet takes over\ngetaddrinfo, and eventlet\u0027s getaddrinfo doesn\u0027t support IPv6. If\ndnspython isn\u0027t there, then the eventlet patching doesn\u0027t happen\nand IPv6 address resolution works.\n\nIt appears that the pip dependencies have recently been updated\nso that dnspython gets installed now.\n\nThe fix is to skip the 3 tests that force the server to listen\non \"::1\".\n\nFixes bug 1176204\n\nOnce eventlet is updated, we can add the tests back in. Here\u0027s the\neventlet code in question:\nhttps://bitbucket.org/eventlet/eventlet/src/e0f578180d7d82d2ed3d8a96d520103503c524ec/eventlet/support/greendns.py?at\u003d0.12#cl-166\n\nChange-Id: Ib7a5a5f0dcac82229b7b7fb3df4b1ef141586c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d9f944215d836636e4e5209924c9cb1f01bec9f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d9f944215d836636e4e5209924c9cb1f01bec9f0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"22d96b270b6794f15471761073a4d5e1065f35b0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b5ccae96dbdebae48d252c4489c10ed0f06024e8":{"kind":"REWORK","_number":4,"created":"2013-05-06 19:38:11.000000000","uploader":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"ref":"refs/changes/38/28338/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/28338/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/28338/4"}}},"commit":{"parents":[{"commit":"22d96b270b6794f15471761073a4d5e1065f35b0","subject":"Merge \"HACKING LDAP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/22d96b270b6794f15471761073a4d5e1065f35b0"}]}],"author":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 18:42:26.000000000","tz":-300},"committer":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 19:38:11.000000000","tz":-300},"subject":"Skip IPv6 tests for eventlet dns","message":"Skip IPv6 tests for eventlet dns\n\nEventlet address resolving doesn\u0027t support IPv6. This is causing the\ngate to fail.\n\nThe failure happens when the dnspython package is installed in the\nenvironment. When dnspython is there then eventlet takes over\ngetaddrinfo, and eventlet\u0027s getaddrinfo doesn\u0027t support IPv6. If\ndnspython isn\u0027t there, then the eventlet patching doesn\u0027t happen\nand IPv6 address resolution works.\n\nIt appears that the pip dependencies have recently been updated\nso that dnspython gets installed now.\n\nThe fix is to skip the 3 tests that force the server to listen\non \"::1\".\n\nFixes bug 1176204\n\nOnce eventlet is updated, we can add the tests back in. Here\u0027s the\neventlet code in question:\nhttps://bitbucket.org/eventlet/eventlet/src/e0f578180d7d82d2ed3d8a96d520103503c524ec/eventlet/support/greendns.py?at\u003d0.12#cl-166\n\nChange-Id: Ib7a5a5f0dcac82229b7b7fb3df4b1ef141586c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b5ccae96dbdebae48d252c4489c10ed0f06024e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b5ccae96dbdebae48d252c4489c10ed0f06024e8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"22d96b270b6794f15471761073a4d5e1065f35b0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b611f1f2019e29ba7744f27ce1fc303fd017c331":{"kind":"REWORK","_number":5,"created":"2013-05-07 02:00:18.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/38/28338/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/28338/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/28338/5"}}},"commit":{"parents":[{"commit":"22d96b270b6794f15471761073a4d5e1065f35b0","subject":"Merge \"HACKING LDAP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/22d96b270b6794f15471761073a4d5e1065f35b0"}]}],"author":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 18:42:26.000000000","tz":-300},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2013-05-07 01:59:29.000000000","tz":-240},"subject":"Two essential bug fixes.","message":"Two essential bug fixes.\n\nSkip IPv6 tests for eventlet dns\n\nEventlet address resolving doesn\u0027t support IPv6. This is causing the\ngate to fail.\n\nThe failure happens when the dnspython package is installed in the\nenvironment. When dnspython is there then eventlet takes over\ngetaddrinfo, and eventlet\u0027s getaddrinfo doesn\u0027t support IPv6. If\ndnspython isn\u0027t there, then the eventlet patching doesn\u0027t happen\nand IPv6 address resolution works.\n\nIt appears that the pip dependencies have recently been updated\nso that dnspython gets installed now.\n\nThe fix is to skip the 3 tests that force the server to listen\non \"::1\".\n\nFixes bug 1176204\n\nOnce eventlet is updated, we can add the tests back in. Here\u0027s the\neventlet code in question:\nhttps://bitbucket.org/eventlet/eventlet/src/e0f578180d7d82d2ed3d8a96d520103503c524ec/eventlet/support/greendns.py?at\u003d0.12#cl-166\n\nBlacklist python-memcache 1.50, it was a busted release\n\nFixes: Bug 1177136\nChange-Id: Ib7a5a5f0dcac82229b7b7fb3df4b1ef141586c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b611f1f2019e29ba7744f27ce1fc303fd017c331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b611f1f2019e29ba7744f27ce1fc303fd017c331"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"22d96b270b6794f15471761073a4d5e1065f35b0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b55620d7ee0f87b0339285b16dd6fd68e5516b10":{"kind":"REWORK","_number":6,"created":"2013-05-07 02:23:57.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/38/28338/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/38/28338/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/38/28338/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/38/28338/6"}}},"commit":{"parents":[{"commit":"22d96b270b6794f15471761073a4d5e1065f35b0","subject":"Merge \"HACKING LDAP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/22d96b270b6794f15471761073a4d5e1065f35b0"}]}],"author":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-05-06 18:42:26.000000000","tz":-300},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2013-05-07 02:23:27.000000000","tz":-240},"subject":"Skip IPv6 tests for eventlet dns","message":"Skip IPv6 tests for eventlet dns\n\nEventlet address resolving doesn\u0027t support IPv6. This is causing the\ngate to fail.\n\nThe failure happens when the dnspython package is installed in the\nenvironment. When dnspython is there then eventlet takes over\ngetaddrinfo, and eventlet\u0027s getaddrinfo doesn\u0027t support IPv6. If\ndnspython isn\u0027t there, then the eventlet patching doesn\u0027t happen\nand IPv6 address resolution works.\n\nIt appears that the pip dependencies have recently been updated\nso that dnspython gets installed now.\n\nThe fix is to skip the 3 tests that force the server to listen\non \"::1\".\n\nFixes bug 1176204\n\nOnce eventlet is updated, we can add the tests back in. Here\u0027s the\neventlet code in question:\n\nhttps://bitbucket.org/eventlet/eventlet/src/e0f578180d7d82d2ed3d8a96d520103503c524ec/eventlet/support/greendns.py?at\u003d0.12#cl-166\n\nChange-Id: Ib7a5a5f0dcac82229b7b7fb3df4b1ef141586c5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b55620d7ee0f87b0339285b16dd6fd68e5516b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b55620d7ee0f87b0339285b16dd6fd68e5516b10"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"22d96b270b6794f15471761073a4d5e1065f35b0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
