)]}'
{"id":"openstack%2Fkeystone~30128","triplet_id":"openstack%2Fkeystone~master~If0fbd3ad484128432e0564e37ae9a144cda822b5","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"If0fbd3ad484128432e0564e37ae9a144cda822b5","subject":"Remove useless private method","status":"MERGED","created":"2013-05-22 16:19:27.000000000","updated":"2013-05-24 16:21:57.000000000","submitted":"2013-05-24 16:21:56.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"74efbd52f7a03da5b421506fb377f87212c8a0af","_number":30128,"virtual_id_number":30128,"owner":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-05-24 16:21:56.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5956,"name":"Kristy Siu","email":"kwss@kent.ac.uk","username":"kwss"},{"value":1,"date":"2013-05-22 17:11:03.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":0,"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"recommended":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-05-24 13:04:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5956,"name":"Kristy Siu","email":"kwss@kent.ac.uk","username":"kwss"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":1,"date":"2013-05-22 17:21:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2013-05-22 20:31:48.000000000","_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":1,"date":"2013-05-22 22:47:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},{"value":2,"date":"2013-05-24 15:35:40.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5956,"name":"Kristy Siu","email":"kwss@kent.ac.uk","username":"kwss"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2013-05-22 20:31:48.000000000","_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":0,"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},{"value":1,"date":"2013-05-24 15:35:40.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":5956,"name":"Kristy Siu","email":"kwss@kent.ac.uk","username":"kwss"},{"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-05-22 17:11:03.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-05-22 17:21:48.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2013-05-22 20:31:48.000000000","updated_by":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"reviewer":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"state":"REVIEWER"},{"updated":"2013-05-22 22:47:23.000000000","updated_by":{"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},"reviewer":{"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},"state":"REVIEWER"},{"updated":"2013-05-24 13:04:32.000000000","updated_by":{"_account_id":5956,"name":"Kristy Siu","email":"kwss@kent.ac.uk","username":"kwss"},"reviewer":{"_account_id":5956,"name":"Kristy Siu","email":"kwss@kent.ac.uk","username":"kwss"},"state":"REVIEWER"},{"updated":"2013-05-24 15:35:40.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2013-05-24 16:21:56.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"30e38b822b285a092218255586a16fc2df07d4bc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-22 17:10:26.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/30128/1/check/gate-keystone-pep8/5170 : SUCCESS in 1m 28s\n- http://docs-draft.openstack.org/30128/1/check/gate-keystone-docs/3543/doc/build/html/ : SUCCESS in 2m 12s\n- http://logs.openstack.org/30128/1/check/gate-keystone-python26/4065 : SUCCESS in 6m 33s\n- http://logs.openstack.org/30128/1/check/gate-keystone-python27/4076 : SUCCESS in 5m 50s\n- http://logs.openstack.org/30128/1/check/gate-tempest-devstack-vm-full/17758 : SUCCESS in 44m 50s\n- http://logs.openstack.org/30128/1/check/gate-tempest-devstack-vm-postgres-full/17468 : FAILURE in 50m 41s (non-voting)\n- http://logs.openstack.org/30128/1/check/gate-tempest-devstack-vm-quantum/24469 : SUCCESS in 22m 23s\n- http://logs.openstack.org/30128/1/check/gate-grenade-devstack-vm/5265 : FAILURE in 11m 19s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"70636ec1b39ab1c21791f705e9aa91834de1b9c7","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-05-22 17:11:03.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/87429\n","accounts_in_message":[],"_revision_number":1},{"id":"0ad1e4aa5f9070cdde5cf2cf272591ea9f6d92f6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2013-05-22 17:21:48.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"7cd3b988b91497dc6da021a36574fb965a9bdea4","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-05-22 20:31:48.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"48e7a6ebe2f38c33be3526cd7ca6bcae3406b814","author":{"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},"date":"2013-05-22 22:47:23.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"cdef196e89c277dd693220bd3c1ce0dd826633dc","author":{"_account_id":5956,"name":"Kristy Siu","email":"kwss@kent.ac.uk","username":"kwss"},"date":"2013-05-24 13:04:32.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"03c654663a9e2a170d4f6694088c5b48fdc70456","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2013-05-24 15:35:40.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":1},{"id":"378d045d1377da4d4255b145138d338c7c571037","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-24 15:35:42.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"fc756eee416bff5ce9d2ed9ca126c3dcb5ecdd87","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-24 16:21:56.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/30128/1/gate/gate-keystone-docs/3634/doc/build/html/ : SUCCESS in 1m 21s\n- http://logs.openstack.org/30128/1/gate/gate-keystone-pep8/5261 : SUCCESS in 1m 22s\n- http://logs.openstack.org/30128/1/gate/gate-keystone-python26/4156 : SUCCESS in 8m 01s\n- http://logs.openstack.org/30128/1/gate/gate-keystone-python27/4167 : SUCCESS in 6m 49s\n- http://logs.openstack.org/30128/1/gate/gate-tempest-devstack-vm-full/18353 : SUCCESS in 44m 29s\n- http://logs.openstack.org/30128/1/gate/gate-tempest-devstack-vm-postgres-full/18063 : FAILURE in 46m 03s (non-voting)\n- http://logs.openstack.org/30128/1/gate/gate-tempest-devstack-vm-quantum/25122 : SUCCESS in 19m 56s\n- http://logs.openstack.org/30128/1/gate/gate-grenade-devstack-vm/5810 : SUCCESS in 16m 15s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"74efbd52f7a03da5b421506fb377f87212c8a0af","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-05-24 16:21:57.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a212ce3ce7723228c111bb92098a6848c9c398c","revisions":{"3a212ce3ce7723228c111bb92098a6848c9c398c":{"kind":"REWORK","_number":1,"created":"2013-05-22 16:19:27.000000000","uploader":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"ref":"refs/changes/28/30128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/28/30128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/30128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/30128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/28/30128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/28/30128/1"}}},"commit":{"parents":[{"commit":"7db13c58bc8fe0aac35b68ac7296cabee2b5004d","subject":"Merge \"Read-only default domain for LDAP (bug 1168726)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7db13c58bc8fe0aac35b68ac7296cabee2b5004d"}]}],"author":{"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","date":"2013-05-22 16:18:54.000000000","tz":-300},"committer":{"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","date":"2013-05-22 16:18:54.000000000","tz":-300},"subject":"Remove useless private method","message":"Remove useless private method\n\nChange-Id: If0fbd3ad484128432e0564e37ae9a144cda822b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3a212ce3ce7723228c111bb92098a6848c9c398c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3a212ce3ce7723228c111bb92098a6848c9c398c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7db13c58bc8fe0aac35b68ac7296cabee2b5004d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
