)]}'
{"id":"openstack%2Fkeystone~32932","triplet_id":"openstack%2Fkeystone~master~I10b138f319b309db7c2747920ab5bd9e727a4557","project":"openstack/keystone","branch":"master","topic":"bug/1190647","hashtags":[],"change_id":"I10b138f319b309db7c2747920ab5bd9e727a4557","subject":"Correct LDAP configuration doc","status":"MERGED","created":"2013-06-13 17:52:51.000000000","updated":"2013-06-13 22:38:35.000000000","submitted":"2013-06-13 22:38:34.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5bb70cf93bb8f4c9788cefbc66c01ca86f9b61ce","_number":32932,"virtual_id_number":32932,"owner":{"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-06-13 22:38:34.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":1,"date":"2013-06-13 18:13:45.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-06-13 21:05:50.000000000","_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":2,"date":"2013-06-13 18:15:16.000000000","_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2013-06-13 21:05:50.000000000","_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"date":"2013-06-13 18:15:16.000000000","_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-06-13 18:13:45.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-06-13 18:15:16.000000000","updated_by":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"reviewer":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"state":"REVIEWER"},{"updated":"2013-06-13 21:05:50.000000000","updated_by":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"reviewer":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"state":"REVIEWER"},{"updated":"2013-06-13 22:38:34.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"78c34e3870636f47fd2b719dd2e44be8838db898","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-13 18:13:45.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/91097\n","accounts_in_message":[],"_revision_number":1},{"id":"69bb1d0b87d3468752a82ebbbe999e670356b7ab","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-06-13 18:15:16.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"eeaac9cef75070717fc34d368caade62058d7a44","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-13 18:35:17.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/32932/1/check/gate-keystone-pep8/5696 : SUCCESS in 1m 26s\n- http://docs-draft.openstack.org/32932/1/check/gate-keystone-docs/4069/doc/build/html/ : SUCCESS in 1m 23s\n- http://logs.openstack.org/32932/1/check/gate-keystone-python26/4590 : SUCCESS in 12m 38s\n- http://logs.openstack.org/32932/1/check/gate-keystone-python27/4602 : SUCCESS in 7m 07s\n- http://logs.openstack.org/32932/1/check/gate-tempest-devstack-vm-full/23072 : SUCCESS in 42m 09s\n- http://logs.openstack.org/32932/1/check/gate-tempest-devstack-vm-postgres-full/22779 : SUCCESS in 39m 43s\n- http://logs.openstack.org/32932/1/check/gate-tempest-devstack-vm-quantum/30392 : SUCCESS in 16m 18s\n- http://logs.openstack.org/32932/1/check/gate-grenade-devstack-vm/10153 : SUCCESS in 13m 52s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"f64efda739fc0f8fdf7dc92479e2146f5f5c7eee","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-06-13 21:05:50.000000000","message":"Patch Set 1: Looks good to me (core reviewer); Approved\n\nthanks!","accounts_in_message":[],"_revision_number":1},{"id":"e23cd29c632db51ba8cfc12140bfe797bb1e82a1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-13 21:05:53.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"895f3920c5a387abcae8ab6795e7524249b57e2e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-13 22:38:34.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/32932/1/gate/gate-keystone-docs/4071/doc/build/html/ : SUCCESS in 2m 18s\n- http://logs.openstack.org/32932/1/gate/gate-keystone-pep8/5698 : SUCCESS in 2m 05s\n- http://logs.openstack.org/32932/1/gate/gate-keystone-python26/4592 : SUCCESS in 15m 07s\n- http://logs.openstack.org/32932/1/gate/gate-keystone-python27/4604 : SUCCESS in 8m 52s\n- http://logs.openstack.org/32932/1/gate/gate-tempest-devstack-vm-full/23153 : SUCCESS in 41m 01s\n- http://logs.openstack.org/32932/1/gate/gate-tempest-devstack-vm-postgres-full/22860 : SUCCESS in 51m 27s\n- http://logs.openstack.org/32932/1/gate/gate-tempest-devstack-vm-quantum/30487 : SUCCESS in 23m 06s\n- http://logs.openstack.org/32932/1/gate/gate-grenade-devstack-vm/10220 : FAILURE in 33m 45s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"524ae2c6fa4a4da83ec439faa9383f8c9abea4d4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-13 22:38:35.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9633837649ed1883c84a75d526ee50659bcc30e","revisions":{"c9633837649ed1883c84a75d526ee50659bcc30e":{"kind":"REWORK","_number":1,"created":"2013-06-13 17:52:51.000000000","uploader":{"_account_id":6456,"name":"Sahdev Zala","email":"spzala@us.ibm.com","username":"spzala"},"ref":"refs/changes/32/32932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/32/32932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/32/32932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/32/32932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/32/32932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/32/32932/1"}}},"commit":{"parents":[{"commit":"6fc4c4e78a6c281505e4c5f542542c8c2cb1f66a","subject":"Force simple Bind for authentication","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6fc4c4e78a6c281505e4c5f542542c8c2cb1f66a"}]}],"author":{"name":"Sahdev Zala","email":"spzala@us.ibm.com","date":"2013-06-13 17:40:36.000000000","tz":-300},"committer":{"name":"Sahdev Zala","email":"spzala@us.ibm.com","date":"2013-06-13 17:52:47.000000000","tz":-300},"subject":"Correct LDAP configuration doc","message":"Correct LDAP configuration doc\n\nCorrect the wrong naming used for domain controller in the configuration doc.\n\nFixes Bug1190647\n\nChange-Id: I10b138f319b309db7c2747920ab5bd9e727a4557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c9633837649ed1883c84a75d526ee50659bcc30e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c9633837649ed1883c84a75d526ee50659bcc30e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6fc4c4e78a6c281505e4c5f542542c8c2cb1f66a","is_merged_in_target_branch":true,"change_id":"Id0c19f09d615446927db1ba074561b129329b5c8","change_number":32896,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
