)]}'
{"id":"openstack%2Fkeystone~341884","triplet_id":"openstack%2Fkeystone~master~I0043518bdd09d2b2db5be48a942ae91ffc74a524","project":"openstack/keystone","branch":"master","topic":"warnerrors","hashtags":[],"change_id":"I0043518bdd09d2b2db5be48a942ae91ffc74a524","subject":"Disable warnerrors in setup.cfg temporarily","status":"MERGED","created":"2016-07-14 00:46:16.000000000","updated":"2016-07-14 05:10:17.000000000","submitted":"2016-07-14 05:10:16.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f35ce373a0f7c48fc99238e0b9cb93aa2dc78b9","_number":341884,"virtual_id_number":341884,"owner":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":2,"date":"2016-07-14 05:10:16.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"all":[{"value":2,"date":"2016-07-14 04:08:33.000000000","_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"all":[{"value":1,"date":"2016-07-14 04:08:33.000000000","_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-07-14 04:08:33.000000000","updated_by":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"reviewer":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"state":"REVIEWER"},{"updated":"2016-07-14 05:10:16.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"344bd39f57ab4d4c3e173545ab7f63ebabc8020c","author":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"date":"2016-07-14 00:46:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69600c5384ddc7697a0430287cab7225d4bd7083","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-14 01:51:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-docs http://docs-draft.openstack.org/84/341884/1/check/gate-keystone-docs/5dbe629//doc/build/html/ : SUCCESS in 4m 34s\n- gate-keystone-pep8 http://logs.openstack.org/84/341884/1/check/gate-keystone-pep8/f1225f0/ : SUCCESS in 2m 24s\n- gate-keystone-python27-db http://logs.openstack.org/84/341884/1/check/gate-keystone-python27-db/c677e90/ : SUCCESS in 10m 51s\n- gate-keystone-python34-db http://logs.openstack.org/84/341884/1/check/gate-keystone-python34-db/71606cf/ : SUCCESS in 10m 21s\n- gate-keystone-python35-db-nv http://logs.openstack.org/84/341884/1/check/gate-keystone-python35-db-nv/bf618fe/ : FAILURE in 15m 19s (non-voting)\n- gate-tempest-dsvm-full http://logs.openstack.org/84/341884/1/check/gate-tempest-dsvm-full/90190c1/ : SUCCESS in 45m 05s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/84/341884/1/check/gate-tempest-dsvm-postgres-full/073702c/ : SUCCESS in 41m 03s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/84/341884/1/check/gate-tempest-dsvm-neutron-full/d350f43/ : SUCCESS in 51m 46s\n- gate-grenade-dsvm http://logs.openstack.org/84/341884/1/check/gate-grenade-dsvm/da4e4ee/ : SUCCESS in 27m 12s\n- gate-keystone-tox-db-legacy_drivers http://logs.openstack.org/84/341884/1/check/gate-keystone-tox-db-legacy_drivers/93c20c7/ : SUCCESS in 4m 36s\n- gate-keystone-dsvm-functional http://logs.openstack.org/84/341884/1/check/gate-keystone-dsvm-functional/ca11346/ : SUCCESS in 13m 09s\n- gate-keystone-dsvm-functional-v3-only-nv http://logs.openstack.org/84/341884/1/check/gate-keystone-dsvm-functional-v3-only-nv/7d09879/ : SUCCESS in 17m 08s (non-voting)\n- gate-tempest-dsvm-keystone-uwsgi-full-nv http://logs.openstack.org/84/341884/1/check/gate-tempest-dsvm-keystone-uwsgi-full-nv/b3ab25f/ : SUCCESS in 1h 04m 19s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-nv http://logs.openstack.org/84/341884/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-nv/bb9c9f2/ : SUCCESS in 45m 41s (non-voting)\n- keystone-coverage-db http://logs.openstack.org/84/341884/1/check/keystone-coverage-db/2b1afb7/ : SUCCESS in 20m 22s","accounts_in_message":[],"_revision_number":1},{"id":"122b3a6981932f9cb39f0c7c2866d07407ece9d7","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2016-07-14 04:08:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthis is essential or we risk breaking our gate","accounts_in_message":[],"_revision_number":1},{"id":"41e2317fe4fde79b96b5e476196b2bcc61743215","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-14 04:08:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"e1776b5556650a9b006ece1b46a563f2c6a438ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-07-14 05:10:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-keystone-docs http://docs-draft.openstack.org/84/341884/1/gate/gate-keystone-docs/e4e45fa//doc/build/html/ : SUCCESS in 4m 50s\n- gate-keystone-pep8 http://logs.openstack.org/84/341884/1/gate/gate-keystone-pep8/313d34e/ : SUCCESS in 3m 24s\n- gate-keystone-python27-db http://logs.openstack.org/84/341884/1/gate/gate-keystone-python27-db/a256f41/ : SUCCESS in 11m 43s\n- gate-keystone-python34-db http://logs.openstack.org/84/341884/1/gate/gate-keystone-python34-db/74c9463/ : SUCCESS in 11m 34s\n- gate-tempest-dsvm-full http://logs.openstack.org/84/341884/1/gate/gate-tempest-dsvm-full/8cdafc3/ : SUCCESS in 55m 25s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/84/341884/1/gate/gate-tempest-dsvm-postgres-full/aa69be9/ : SUCCESS in 52m 27s\n- gate-tempest-dsvm-neutron-full http://logs.openstack.org/84/341884/1/gate/gate-tempest-dsvm-neutron-full/3fb9b11/ : SUCCESS in 55m 41s\n- gate-grenade-dsvm http://logs.openstack.org/84/341884/1/gate/gate-grenade-dsvm/0d4749a/ : SUCCESS in 46m 05s\n- gate-keystone-dsvm-functional http://logs.openstack.org/84/341884/1/gate/gate-keystone-dsvm-functional/7d999e2/ : SUCCESS in 27m 42s\n- gate-keystone-tox-db-legacy_drivers http://logs.openstack.org/84/341884/1/gate/gate-keystone-tox-db-legacy_drivers/2a0133e/ : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e03efe08ae36a276a50621e52b180d35b8c56189","date":"2016-07-14 05:10:17.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36394a6417d03e212876d43f18d2867f58c71502","revisions":{"36394a6417d03e212876d43f18d2867f58c71502":{"kind":"REWORK","_number":1,"created":"2016-07-14 00:46:16.000000000","uploader":{"_account_id":7191,"name":"Jamie Lennox","email":"jamielennox@gmail.com","username":"jamielennox"},"ref":"refs/changes/84/341884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/84/341884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/84/341884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/84/341884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/84/341884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/84/341884/1"}}},"commit":{"parents":[{"commit":"86656bc991589507fdde54da5ece69043f920d7e","subject":"Merge \"Correct normal response codes in v2.0 token docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/86656bc991589507fdde54da5ece69043f920d7e"}]}],"author":{"name":"Jamie Lennox","email":"jamielennox@gmail.com","date":"2016-07-14 00:43:40.000000000","tz":600},"committer":{"name":"Jamie Lennox","email":"jamielennox@gmail.com","date":"2016-07-14 00:43:40.000000000","tz":600},"subject":"Disable warnerrors in setup.cfg temporarily","message":"Disable warnerrors in setup.cfg temporarily\n\nPBR\u0027s warnerrors is supposed to fail a build job when warnings are\npresent in the documentation. Unfortunately that hasn\u0027t been working for\na while.\n\nWith an upcoming release of PBR this flag will be fixed and any\nwarnings will fail docs jobs again. There is therefore a coordinated\neffort to disable this flag until the PBR release has been completed and\nrequirements updated. After which time we can reenable the flag and\nensure docs build successfully.\n\nhttp://lists.openstack.org/pipermail/openstack-dev/2016-June/097849.html\n\nChange-Id: I0043518bdd09d2b2db5be48a942ae91ffc74a524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/36394a6417d03e212876d43f18d2867f58c71502"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/36394a6417d03e212876d43f18d2867f58c71502"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"86656bc991589507fdde54da5ece69043f920d7e","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
