)]}'
{"id":"openstack%2Fkeystone~34541","triplet_id":"openstack%2Fkeystone~master~Ia0fc0b3d4a92416dcd65ddb49b3b0bf9a0777363","project":"openstack/keystone","branch":"master","topic":"bp/refactor-ldap-driver","hashtags":[],"change_id":"Ia0fc0b3d4a92416dcd65ddb49b3b0bf9a0777363","subject":"Abstract out attribute_ignore assigning in LDAP driver","status":"MERGED","created":"2013-06-26 12:40:56.000000000","updated":"2013-08-10 00:57:59.000000000","submitted":"2013-08-10 00:57:58.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f32a3e4d667c2bd96b672f79999f7b9848b9e1fc","_number":34541,"virtual_id_number":34541,"owner":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-08-10 00:57:58.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":1,"date":"2013-08-06 09:28:31.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},{"value":0,"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":2,"date":"2013-08-07 22:20:22.000000000","_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":2,"date":"2013-08-10 00:12:58.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":1,"date":"2013-08-09 21:06:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},{"value":0,"date":"2013-08-06 09:08:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"date":"2013-08-07 22:20:22.000000000","_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":1,"date":"2013-08-10 00:12:58.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},{"value":0,"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-07-08 18:49:47.000000000","updated_by":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"reviewer":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"state":"REVIEWER"},{"updated":"2013-08-06 09:28:31.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-08-07 22:20:22.000000000","updated_by":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"reviewer":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"state":"REVIEWER"},{"updated":"2013-08-09 21:06:19.000000000","updated_by":{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},"reviewer":{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},"state":"REVIEWER"},{"updated":"2013-08-10 00:12:58.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2013-08-10 00:57:58.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"470a23a60200d4972c2a8d470f364451d4972d86","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-26 13:02:55.000000000","message":"Patch Set 1: Works for me\n\nSmokeStack Results (patch set 1):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/93975\n","accounts_in_message":[],"_revision_number":1},{"id":"80970e1b0496448e99f524b192844a1e05b76497","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-26 13:27:02.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34541/1/check/gate-keystone-pep8/5880 : SUCCESS in 3m 06s\n- http://docs-draft.openstack.org/34541/1/check/gate-keystone-docs/4253/doc/build/html/ : SUCCESS in 1m 51s\n- http://logs.openstack.org/34541/1/check/gate-keystone-python26/4773 : SUCCESS in 13m 01s\n- http://logs.openstack.org/34541/1/check/gate-keystone-python27/4786 : SUCCESS in 7m 32s\n- http://logs.openstack.org/34541/1/check/gate-tempest-devstack-vm-full/26409 : SUCCESS in 45m 48s\n- http://logs.openstack.org/34541/1/check/gate-tempest-devstack-vm-postgres-full/26115 : SUCCESS in 44m 35s\n- http://logs.openstack.org/34541/1/check/gate-tempest-devstack-vm-quantum/34051 : SUCCESS in 21m 37s\n- http://logs.openstack.org/34541/1/check/gate-grenade-devstack-vm/12988 : FAILURE in 28m 21s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"2b9c76d5c9ba746a7ee1c9b10b782c5a646dcd63","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-06-27 20:04:57.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"7fbf7159b6dbd11a682ea455ea4725fe68a566c4","author":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"date":"2013-06-28 13:27:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0d56bccfa38efa8f5f04bb4df285316985abd971","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-06-28 14:05:31.000000000","message":"Patch Set 2: Works for me\n\nSmokeStack Results (patch set 2):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/94774\n","accounts_in_message":[],"_revision_number":2},{"id":"351cc34e6916d2eb2324a17eb0f522c7cd1f0e15","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-28 14:14:32.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34541/2/check/gate-keystone-pep8/5944 : SUCCESS in 2m 00s\n- http://docs-draft.openstack.org/34541/2/check/gate-keystone-docs/4317/doc/build/html/ : SUCCESS in 2m 01s\n- http://logs.openstack.org/34541/2/check/gate-keystone-python26/4837 : SUCCESS in 13m 05s\n- http://logs.openstack.org/34541/2/check/gate-keystone-python27/4850 : SUCCESS in 7m 09s\n- http://logs.openstack.org/34541/2/check/gate-tempest-devstack-vm-full/27108 : SUCCESS in 46m 49s\n- http://logs.openstack.org/34541/2/check/gate-tempest-devstack-vm-postgres-full/26814 : SUCCESS in 46m 23s\n- http://logs.openstack.org/34541/2/check/gate-tempest-devstack-vm-quantum/34834 : SUCCESS in 23m 25s\n- http://logs.openstack.org/34541/2/check/gate-grenade-devstack-vm/13594 : FAILURE in 24m 12s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"166ccac7a42f8ab4673c98b5db0c069e1bbd609e","author":{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},"date":"2013-06-28 21:28:25.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\nIs there a bug or blueprint associated with this change? There is no documented context for why you are doing what you are doing. Also there is no test case that has been added for this.","accounts_in_message":[],"_revision_number":2},{"id":"787aee51cbe9bd5fc97d12827908421864eff9f5","author":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"date":"2013-07-01 13:14:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bbca4a5ccf932924e077a94afe9b013664a19dee","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-07-01 13:14:26.000000000","message":"Patch Set 3:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":3},{"id":"a9471919f24ff671b6efc334a1c515b955b527db","author":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"date":"2013-07-01 13:40:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"94469fb64a19b48cdbb9ac2e66a5ec9849177824","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-07-01 13:40:21.000000000","message":"Patch Set 4:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":4},{"id":"0a729216c5a19a9069db16de33015f9233e7b4be","author":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"date":"2013-07-01 13:49:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"47ecf43c7ee4ced5cc60171872ae9bf1bb06ff2b","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-07-01 13:49:29.000000000","message":"Patch Set 5:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":5},{"id":"42cbb331391e1bd510481a5e7d599eefcb117496","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-01 14:29:06.000000000","message":"Patch Set 5: Works for me\n\nSmokeStack Results (patch set 5):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/95337\n","accounts_in_message":[],"_revision_number":5},{"id":"40bd4406dd903f909f05635d754e2f27cafebaf9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-01 14:41:42.000000000","message":"Patch Set 5: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/34541/5/check/gate-keystone-pep8/5986 : SUCCESS in 2m 17s\n- http://docs-draft.openstack.org/34541/5/check/gate-keystone-docs/4359/doc/build/html/ : SUCCESS in 2m 07s\n- http://logs.openstack.org/34541/5/check/gate-keystone-python26/4879 : FAILURE in 14m 15s\n- http://logs.openstack.org/34541/5/check/gate-keystone-python27/4892 : SUCCESS in 7m 54s\n- http://logs.openstack.org/34541/5/check/gate-tempest-devstack-vm-full/27548 : SUCCESS in 52m 06s\n- http://logs.openstack.org/34541/5/check/gate-tempest-devstack-vm-postgres-full/27254 : SUCCESS in 46m 27s\n- http://logs.openstack.org/34541/5/check/gate-tempest-devstack-vm-quantum/35356 : SUCCESS in 23m 59s\n- http://logs.openstack.org/34541/5/check/gate-grenade-devstack-vm/13984 : FAILURE in 23m 35s (non-voting)\n","accounts_in_message":[],"_revision_number":5},{"id":"931f5f5910b01c72b930597c8ae9d1d74cfb0d09","author":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"date":"2013-07-02 09:00:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0b9b7bc5cea829ada7fe45224af6d14cd0854e80","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-07-02 09:00:28.000000000","message":"Patch Set 6:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":6},{"id":"2c8e0632ea499ff8908dc68336d212dd09211f4d","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-07-02 09:24:49.000000000","message":"Patch Set 6: Works for me\n\nSmokeStack Results (patch set 6):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/95541\n","accounts_in_message":[],"_revision_number":6},{"id":"992bff6517e7dcded9ebb04a4ef92d0c9031933b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-02 09:56:37.000000000","message":"Patch Set 6: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/34541/6/check/gate-keystone-pep8/6003 : SUCCESS in 2m 06s\n- http://docs-draft.openstack.org/34541/6/check/gate-keystone-docs/4376/doc/build/html/ : SUCCESS in 2m 03s\n- http://logs.openstack.org/34541/6/check/gate-keystone-python26/4896 : SUCCESS in 11m 50s\n- http://logs.openstack.org/34541/6/check/gate-keystone-python27/4909 : SUCCESS in 7m 32s\n- http://logs.openstack.org/34541/6/check/gate-tempest-devstack-vm-full/27752 : SUCCESS in 43m 45s\n- http://logs.openstack.org/34541/6/check/gate-tempest-devstack-vm-postgres-full/27458 : SUCCESS in 56m 04s\n- http://logs.openstack.org/34541/6/check/gate-tempest-devstack-vm-quantum/35590 : FAILURE in 19m 23s\n- http://logs.openstack.org/34541/6/check/gate-grenade-devstack-vm/14168 : FAILURE in 23m 48s (non-voting)\n","accounts_in_message":[],"_revision_number":6},{"id":"c1f68d1024270af096aa729c9b8f03a711a9aec2","author":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"date":"2013-07-02 11:16:55.000000000","message":"Patch Set 6:\n\nrecheck bug 1194026","accounts_in_message":[],"_revision_number":6},{"id":"e372d9ef0b5b50517f406024e2932199745a03f1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-07-02 12:03:09.000000000","message":"Patch Set 6: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34541/6/check/gate-keystone-pep8/6007 : SUCCESS in 1m 29s\n- http://docs-draft.openstack.org/34541/6/check/gate-keystone-docs/4380/doc/build/html/ : SUCCESS in 1m 28s\n- http://logs.openstack.org/34541/6/check/gate-keystone-python26/4900 : SUCCESS in 12m 26s\n- http://logs.openstack.org/34541/6/check/gate-keystone-python27/4913 : SUCCESS in 7m 03s\n- http://logs.openstack.org/34541/6/check/gate-tempest-devstack-vm-full/27787 : SUCCESS in 43m 19s\n- http://logs.openstack.org/34541/6/check/gate-tempest-devstack-vm-postgres-full/27493 : SUCCESS in 46m 04s\n- http://logs.openstack.org/34541/6/check/gate-tempest-devstack-vm-quantum/35629 : SUCCESS in 21m 03s\n- http://logs.openstack.org/34541/6/check/gate-grenade-devstack-vm/14201 : FAILURE in 17m 20s (non-voting)\n","accounts_in_message":[],"_revision_number":6},{"id":"9066841524f871ee34128d58b2e83b28d88f04f6","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-07-03 19:19:57.000000000","message":"Patch Set 6: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":6},{"id":"3cb12a6d68119e6142df2c0e9fa049a12184589c","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-07-08 18:49:47.000000000","message":"Patch Set 6: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":6},{"id":"755468921e2df3b6db314b3af780f5ee9b144cbb","author":{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},"date":"2013-07-15 19:27:02.000000000","message":"Patch Set 6: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":6},{"id":"9568acea1108b2330d1283bb8328a4405a702400","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2013-07-16 01:28:20.000000000","message":"Patch Set 6:\n\nreverify no bug","accounts_in_message":[],"_revision_number":6},{"id":"f3667c4ca8fc654c9f613613fb14a7c740d45697","author":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"date":"2013-08-06 09:07:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"76ac35367efda18537d90c0e31f4aced001ce31a","author":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"date":"2013-08-06 09:08:23.000000000","message":"Patch Set 7:\n\nRebased. Moved some code the assignment driver that was in identity before.","accounts_in_message":[],"_revision_number":7},{"id":"6d313d63c192b313e3b951499feac13c65d3b0b1","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-08-06 09:28:31.000000000","message":"Patch Set 7: Works for me\n\nSmokeStack Results (patch set 7):\n\tLibvirt (Fedora 18) Success: http://smokestack.openstack.org/?go\u003d/jobs/109529\n","accounts_in_message":[],"_revision_number":7},{"id":"3630719acebc5f56671f34c428db862f50b88f05","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-06 10:05:59.000000000","message":"Patch Set 7: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/41/34541/7/check/gate-keystone-pep8/fd17aa5 : SUCCESS in 2m 38s\n- http://docs-draft.openstack.org/41/34541/7/check/gate-keystone-docs/99657c1/doc/build/html/ : SUCCESS in 1m 23s\n- http://logs.openstack.org/41/34541/7/check/gate-keystone-python26/d671674 : SUCCESS in 13m 28s\n- http://logs.openstack.org/41/34541/7/check/gate-keystone-python27/ad06094 : SUCCESS in 7m 19s\n- http://logs.openstack.org/41/34541/7/check/gate-tempest-devstack-vm-full/1d2114c : SUCCESS in 46m 39s\n- http://logs.openstack.org/41/34541/7/check/gate-tempest-devstack-vm-cells-full/8559683 : FAILURE in 13m 36s (non-voting)\n- http://logs.openstack.org/41/34541/7/check/gate-tempest-devstack-vm-testr-full/6813b7a : FAILURE in 29m 54s (non-voting)\n- http://logs.openstack.org/41/34541/7/check/gate-tempest-devstack-vm-postgres-full/b9b7688 : SUCCESS in 58m 06s\n- http://logs.openstack.org/41/34541/7/check/gate-tempest-devstack-vm-neutron/212de3e : SUCCESS in 19m 14s\n- http://logs.openstack.org/41/34541/7/check/gate-grenade-devstack-vm/3487170 : SUCCESS in 24m 47s\n","accounts_in_message":[],"_revision_number":7},{"id":"80332bf925be5ec09f56a1dcd10cff6c2b962100","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-08-07 22:20:22.000000000","message":"Patch Set 7: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":7},{"id":"dee2dd9be4df2d6e47f0049d50fe2f0ac48caf7f","author":{"_account_id":6460,"name":"Brad Topol","email":"btopol@us.ibm.com","username":"btopol"},"date":"2013-08-09 21:06:19.000000000","message":"Patch Set 7: Looks good to me, but someone else must approve\n\nVery nice refactoring!!! Great job!!!","accounts_in_message":[],"_revision_number":7},{"id":"c0e282ec9f634418ab36f43232f6a1f13b9a6322","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2013-08-10 00:12:58.000000000","message":"Patch Set 7: Looks good to me (core reviewer); Approved\n\nI agree with Brant, very nice refactor!","accounts_in_message":[],"_revision_number":7},{"id":"c3be2ac1a318d4c2bf92a2853d6beb78da29c388","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-10 00:13:01.000000000","message":"Patch Set 7: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"5b34485a7330ad165b278099b249c205e4f30e5d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-10 00:57:58.000000000","message":"Patch Set 7: Verified\n\nBuild succeeded.\n\n- http://docs-draft.openstack.org/41/34541/7/gate/gate-keystone-docs/0aa11bb/doc/build/html/ : SUCCESS in 1m 28s\n- http://logs.openstack.org/41/34541/7/gate/gate-keystone-pep8/bcf22a2 : SUCCESS in 1m 24s\n- http://logs.openstack.org/41/34541/7/gate/gate-keystone-python26/cc65f68 : SUCCESS in 14m 46s\n- http://logs.openstack.org/41/34541/7/gate/gate-keystone-python27/9135950 : SUCCESS in 8m 54s\n- http://logs.openstack.org/41/34541/7/gate/gate-tempest-devstack-vm-full/cade0fc : SUCCESS in 44m 55s\n- http://logs.openstack.org/41/34541/7/gate/gate-tempest-devstack-vm-postgres-full/2041a1d : SUCCESS in 42m 01s\n- http://logs.openstack.org/41/34541/7/gate/gate-tempest-devstack-vm-neutron/c5882f4 : SUCCESS in 17m 54s\n- http://logs.openstack.org/41/34541/7/gate/gate-grenade-devstack-vm/96efe9a : SUCCESS in 21m 21s\n","accounts_in_message":[],"_revision_number":7},{"id":"f652c770c5733d6e733b921d5db8c33658c28e23","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-10 00:57:59.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6f954d2c2982a6096c13053dafd96ebddc6017bb","revisions":{"5376e227d061f9e05e79e90941ad6b56e9416eda":{"kind":"REWORK","_number":1,"created":"2013-06-26 12:40:56.000000000","uploader":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"ref":"refs/changes/41/34541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/41/34541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/41/34541/1"}}},"commit":{"parents":[{"commit":"b70ece72123632b8370afdbd1b44d737d7e291c1","subject":"Abstract out attribute_mapping filling in LDAP driver.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b70ece72123632b8370afdbd1b44d737d7e291c1"}]}],"author":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-04-19 14:11:37.000000000","tz":240},"committer":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-06-26 11:58:29.000000000","tz":240},"subject":"Abstract out attribute_ignore assigning in LDAP driver","message":"Abstract out attribute_ignore assigning in LDAP driver\n\nChange-Id: Ia0fc0b3d4a92416dcd65ddb49b3b0bf9a0777363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5376e227d061f9e05e79e90941ad6b56e9416eda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5376e227d061f9e05e79e90941ad6b56e9416eda"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b70ece72123632b8370afdbd1b44d737d7e291c1","is_merged_in_target_branch":false,"change_id":"I4b3bb2900b54f046b05d68f15fb6e35b324ca9f7","change_number":34540,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1accfccceb43b32138cab3b915910984e3ae0a43":{"kind":"REWORK","_number":2,"created":"2013-06-28 13:27:09.000000000","uploader":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"ref":"refs/changes/41/34541/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/41/34541/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/41/34541/2"}}},"commit":{"parents":[{"commit":"b70ece72123632b8370afdbd1b44d737d7e291c1","subject":"Abstract out attribute_mapping filling in LDAP driver.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b70ece72123632b8370afdbd1b44d737d7e291c1"}]}],"author":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-04-19 14:11:37.000000000","tz":240},"committer":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-06-28 13:20:07.000000000","tz":240},"subject":"Abstract out attribute_ignore assigning in LDAP driver","message":"Abstract out attribute_ignore assigning in LDAP driver\n\nChange-Id: Ia0fc0b3d4a92416dcd65ddb49b3b0bf9a0777363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1accfccceb43b32138cab3b915910984e3ae0a43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1accfccceb43b32138cab3b915910984e3ae0a43"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b70ece72123632b8370afdbd1b44d737d7e291c1","is_merged_in_target_branch":false,"change_id":"I4b3bb2900b54f046b05d68f15fb6e35b324ca9f7","change_number":34540,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f188fd31f6c2c7a7a401e9bac6d2e96de58ab8f8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2013-07-01 13:14:05.000000000","uploader":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"ref":"refs/changes/41/34541/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/41/34541/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/41/34541/3"}}},"commit":{"parents":[{"commit":"b70ece72123632b8370afdbd1b44d737d7e291c1","subject":"Abstract out attribute_mapping filling in LDAP driver.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b70ece72123632b8370afdbd1b44d737d7e291c1"}]}],"author":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-04-19 14:11:37.000000000","tz":240},"committer":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-07-01 12:56:00.000000000","tz":240},"subject":"Abstract out attribute_ignore assigning in LDAP driver This is used to avoid code duplication in keystone/identity/backends/ldap.py. All changes are fully covered by test_*option_name*_attribute_ignore() in test_backend_ldap.py. There are no blueprints/bugs associated with this change.","message":"Abstract out attribute_ignore assigning in LDAP driver\nThis is used to avoid code duplication in keystone/identity/backends/ldap.py. All changes are fully covered by test_*option_name*_attribute_ignore() in test_backend_ldap.py. There are no blueprints/bugs associated with this change.\n\nChange-Id: Ia0fc0b3d4a92416dcd65ddb49b3b0bf9a0777363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f188fd31f6c2c7a7a401e9bac6d2e96de58ab8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f188fd31f6c2c7a7a401e9bac6d2e96de58ab8f8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b70ece72123632b8370afdbd1b44d737d7e291c1","is_merged_in_target_branch":false,"change_id":"I4b3bb2900b54f046b05d68f15fb6e35b324ca9f7","change_number":34540,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a850694210ebc7e3d0755a1924dc7741593906fc":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2013-07-01 13:40:00.000000000","uploader":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"ref":"refs/changes/41/34541/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/41/34541/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/41/34541/4"}}},"commit":{"parents":[{"commit":"b70ece72123632b8370afdbd1b44d737d7e291c1","subject":"Abstract out attribute_mapping filling in LDAP driver.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b70ece72123632b8370afdbd1b44d737d7e291c1"}]}],"author":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-04-19 14:11:37.000000000","tz":240},"committer":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-07-01 13:34:31.000000000","tz":240},"subject":"Abstract out attribute_ignore assigning in LDAP driver","message":"Abstract out attribute_ignore assigning in LDAP driver\n\nThis is used to avoid code duplication in keystone/identity/backends/ldap.py. All changes are fully covered by test_*option_name*_attribute_ignore() in test_backend_ldap.py.\n\nThere are no blueprints/bugs associated with this change.\n\nChange-Id: Ia0fc0b3d4a92416dcd65ddb49b3b0bf9a0777363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a850694210ebc7e3d0755a1924dc7741593906fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a850694210ebc7e3d0755a1924dc7741593906fc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b70ece72123632b8370afdbd1b44d737d7e291c1","is_merged_in_target_branch":false,"change_id":"I4b3bb2900b54f046b05d68f15fb6e35b324ca9f7","change_number":34540,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b503fc4599f948695bee770217ea2ee97904c359":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2013-07-01 13:49:08.000000000","uploader":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"ref":"refs/changes/41/34541/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/41/34541/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/41/34541/5"}}},"commit":{"parents":[{"commit":"b70ece72123632b8370afdbd1b44d737d7e291c1","subject":"Abstract out attribute_mapping filling in LDAP driver.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b70ece72123632b8370afdbd1b44d737d7e291c1"}]}],"author":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-04-19 14:11:37.000000000","tz":240},"committer":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-07-01 13:41:18.000000000","tz":240},"subject":"Abstract out attribute_ignore assigning in LDAP driver","message":"Abstract out attribute_ignore assigning in LDAP driver\n\nThis is used to avoid code duplication in keystone/identity/\nbackends/ldap.py. All changes are fully covered by test_*option_\nname*_attribute_ignore() in test_backend_ldap.py.\n\nThere are no blueprints/bugs associated with this change.\n\nChange-Id: Ia0fc0b3d4a92416dcd65ddb49b3b0bf9a0777363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b503fc4599f948695bee770217ea2ee97904c359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b503fc4599f948695bee770217ea2ee97904c359"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b70ece72123632b8370afdbd1b44d737d7e291c1","is_merged_in_target_branch":false,"change_id":"I4b3bb2900b54f046b05d68f15fb6e35b324ca9f7","change_number":34540,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f409a271f2dc5ebfb41e79935c4154f977fbec93":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2013-07-02 09:00:06.000000000","uploader":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"ref":"refs/changes/41/34541/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/41/34541/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/41/34541/6"}}},"commit":{"parents":[{"commit":"1b5671a57dda5af5e8252c0608744d92c96a8285","subject":"Abstract out attribute_mapping filling in LDAP driver.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1b5671a57dda5af5e8252c0608744d92c96a8285"}]}],"author":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-04-19 14:11:37.000000000","tz":240},"committer":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-07-02 08:58:29.000000000","tz":240},"subject":"Abstract out attribute_ignore assigning in LDAP driver","message":"Abstract out attribute_ignore assigning in LDAP driver\n\nThis is used to avoid code duplication in keystone/identity/\nbackends/ldap.py. All changes are fully covered by test_*option_\nname*_attribute_ignore() in test_backend_ldap.py.\n\nbp refactor-ldap-driver\n\nChange-Id: Ia0fc0b3d4a92416dcd65ddb49b3b0bf9a0777363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f409a271f2dc5ebfb41e79935c4154f977fbec93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f409a271f2dc5ebfb41e79935c4154f977fbec93"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1b5671a57dda5af5e8252c0608744d92c96a8285","is_merged_in_target_branch":false,"change_id":"I4b3bb2900b54f046b05d68f15fb6e35b324ca9f7","change_number":34540,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6f954d2c2982a6096c13053dafd96ebddc6017bb":{"kind":"REWORK","_number":7,"created":"2013-08-06 09:07:03.000000000","uploader":{"_account_id":7293,"name":"Elena Ezhova","email":"ezhovaelena19@gmail.com","username":"elena"},"ref":"refs/changes/41/34541/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/41/34541/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/41/34541/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/41/34541/7"}}},"commit":{"parents":[{"commit":"1289bbf12e8cf6e7792df0de44a3bc9bf6e9c505","subject":"Abstract out attribute_mapping filling in LDAP driver.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1289bbf12e8cf6e7792df0de44a3bc9bf6e9c505"}]}],"author":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-04-19 14:11:37.000000000","tz":240},"committer":{"name":"Elena Ezhova","email":"eezhova@mirantis.com","date":"2013-08-06 08:54:09.000000000","tz":240},"subject":"Abstract out attribute_ignore assigning in LDAP driver","message":"Abstract out attribute_ignore assigning in LDAP driver\n\nThis is used to avoid code duplication in keystone/identity/\nbackends/ldap.py. All changes are fully covered by test_*option_\nname*_attribute_ignore() in test_backend_ldap.py.\n\nbp refactor-ldap-driver\n\nChange-Id: Ia0fc0b3d4a92416dcd65ddb49b3b0bf9a0777363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6f954d2c2982a6096c13053dafd96ebddc6017bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6f954d2c2982a6096c13053dafd96ebddc6017bb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1289bbf12e8cf6e7792df0de44a3bc9bf6e9c505","is_merged_in_target_branch":true,"change_id":"I4b3bb2900b54f046b05d68f15fb6e35b324ca9f7","change_number":34540,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
