)]}'
{"id":"openstack%2Fkeystone~34622","triplet_id":"openstack%2Fkeystone~stable%2Fgrizzly~Ib5e6fab1d3a20420eb05d6d50860f5040e8d6a15","project":"openstack/keystone","branch":"stable/grizzly","topic":"bug/1170186","hashtags":[],"change_id":"Ib5e6fab1d3a20420eb05d6d50860f5040e8d6a15","subject":"Maintain tokens after role assignments","status":"ABANDONED","created":"2013-06-26 21:44:58.000000000","updated":"2013-08-10 09:23:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee67eb5c9e5a43ec69bfc956cd56ee8051353e75","_number":34622,"virtual_id_number":34622,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":1,"date":"2013-08-03 10:23:20.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":2340,"name":"Ionut Artarisi","email":"iartarisi@suse.cz","username":"mapleoin"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":-2,"date":"2013-08-04 09:36:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-08-03 21:04:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2340,"name":"Ionut Artarisi","email":"iartarisi@suse.cz","username":"mapleoin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":-1,"date":"2013-08-05 19:06:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"date":"2013-08-04 09:36:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2013-08-03 21:04:06.000000000","_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":2340,"name":"Ionut Artarisi","email":"iartarisi@suse.cz","username":"mapleoin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":2340,"name":"Ionut Artarisi","email":"iartarisi@suse.cz","username":"mapleoin"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-06-27 04:58:08.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2013-06-28 15:09:33.000000000","updated_by":{"_account_id":2340,"name":"Ionut Artarisi","email":"iartarisi@suse.cz","username":"mapleoin"},"reviewer":{"_account_id":2340,"name":"Ionut Artarisi","email":"iartarisi@suse.cz","username":"mapleoin"},"state":"REVIEWER"},{"updated":"2013-08-03 10:23:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2013-08-03 21:04:06.000000000","updated_by":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"reviewer":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"state":"REVIEWER"},{"updated":"2013-08-04 09:36:15.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"REVIEWER"},{"updated":"2013-08-05 19:06:57.000000000","updated_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"reviewer":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"state":"REVIEWER"}],"messages":[{"id":"34f88e47d0ea6e261318a8d329af4b1ab85c0a89","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2013-06-26 21:45:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a827a3b049df42ec21c20fc0e7df0db6695b1aa8","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-06-26 21:45:59.000000000","message":"Patch Set 2:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":2},{"id":"ce0c01f062472b0726aa08e690c055a0f2d2e226","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-26 22:19:53.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34622/2/check/gate-keystone-pep8/5894 : SUCCESS in 1m 25s\n- http://docs-draft.openstack.org/34622/2/check/gate-keystone-docs/4267/doc/build/html/ : SUCCESS in 2m 06s\n- http://logs.openstack.org/34622/2/check/gate-keystone-python26/4787 : SUCCESS in 12m 56s\n- http://logs.openstack.org/34622/2/check/gate-keystone-python27/4800 : SUCCESS in 8m 16s\n- http://logs.openstack.org/34622/2/check/gate-tempest-devstack-vm-full/26588 : SUCCESS in 33m 13s\n- http://logs.openstack.org/34622/2/check/gate-tempest-devstack-vm-postgres-full/26294 : SUCCESS in 33m 53s\n- http://logs.openstack.org/34622/2/check/gate-tempest-devstack-vm-quantum/34248 : SUCCESS in 17m 53s\n- http://logs.openstack.org/34622/2/check/gate-grenade-devstack-vm/13141 : SUCCESS in 17m 01s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"6a2acd45f46635b8dd5584b1d241d09df806b3d7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2013-06-27 04:58:08.000000000","message":"Patch Set 2: I would prefer that you didn\u0027t merge this\n\nWas this cherry picked from master? Would it be possibly to cherry pick with the -x command to include the hash of the commit that was the original?","accounts_in_message":[],"_revision_number":2},{"id":"bf9795ab6644a5a0ae963172528401e4e28a2a9b","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2013-06-27 06:29:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"290e2a311ea57a1acc9e1d40f451101db02b489b","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-06-27 06:29:50.000000000","message":"Patch Set 3:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":3},{"id":"201019e547532526e96b5b48e20e24d454d8ad79","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-27 07:04:21.000000000","message":"Patch Set 3: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGithub#Test_Failures\n\n- http://logs.openstack.org/34622/3/check/gate-keystone-pep8/5905 : SUCCESS in 1m 17s\n- http://docs-draft.openstack.org/34622/3/check/gate-keystone-docs/4278/doc/build/html/ : SUCCESS in 2m 48s\n- http://logs.openstack.org/34622/3/check/gate-keystone-python26/4798 : SUCCESS in 13m 20s\n- http://logs.openstack.org/34622/3/check/gate-keystone-python27/4811 : SUCCESS in 8m 23s\n- http://logs.openstack.org/34622/3/check/gate-tempest-devstack-vm-full/26687 : FAILURE in 4m 33s\n- http://logs.openstack.org/34622/3/check/gate-tempest-devstack-vm-postgres-full/26393 : SUCCESS in 34m 30s\n- http://logs.openstack.org/34622/3/check/gate-tempest-devstack-vm-quantum/34356 : SUCCESS in 19m 36s\n- http://logs.openstack.org/34622/3/check/gate-grenade-devstack-vm/13213 : SUCCESS in 19m 05s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"8872043eb10a6130b88a614a3cf652ade09f433c","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2013-06-27 09:54:56.000000000","message":"Patch Set 3:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":3},{"id":"898aad1dfa051480891d4b0efe69021a3169cc4a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-06-27 10:29:34.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/34622/3/check/gate-keystone-pep8/5911 : SUCCESS in 1m 18s\n- http://docs-draft.openstack.org/34622/3/check/gate-keystone-docs/4284/doc/build/html/ : SUCCESS in 1m 58s\n- http://logs.openstack.org/34622/3/check/gate-keystone-python26/4804 : SUCCESS in 15m 45s\n- http://logs.openstack.org/34622/3/check/gate-keystone-python27/4817 : SUCCESS in 8m 07s\n- http://logs.openstack.org/34622/3/check/gate-tempest-devstack-vm-full/26742 : SUCCESS in 34m 15s\n- http://logs.openstack.org/34622/3/check/gate-tempest-devstack-vm-postgres-full/26448 : SUCCESS in 34m 31s\n- http://logs.openstack.org/34622/3/check/gate-tempest-devstack-vm-quantum/34422 : SUCCESS in 17m 05s\n- http://logs.openstack.org/34622/3/check/gate-grenade-devstack-vm/13268 : SUCCESS in 19m 07s (non-voting)\n","accounts_in_message":[],"_revision_number":3},{"id":"daf5592380ed9b0cd1b5f35556f7ba2c88683862","author":{"_account_id":2340,"name":"Ionut Artarisi","email":"iartarisi@suse.cz","username":"mapleoin"},"date":"2013-06-28 15:09:33.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\nI\u0027ve tested this and it fixes the bug. I think it\u0027s worthwhile to backport this as it\u0027s a very user-visible change.","accounts_in_message":[],"_revision_number":3},{"id":"8aa2346855df2c03304ca232d009c1fa5369d117","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-07-08 15:40:30.000000000","message":"Patch Set 3: I would prefer that you didn\u0027t merge this\n\nwhat was wrong with the commit message in patchset 1? the new summary line is useless","accounts_in_message":[],"_revision_number":3},{"id":"c516812dd07d7cef73c7631010b15f8b2657cf0b","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-07-11 01:37:07.000000000","message":"Patch Set 3: I would prefer that you didn\u0027t merge this\n\nredo commit message as dolph recommends","accounts_in_message":[],"_revision_number":3},{"id":"75666cf5ab0f81b3edde2d021aa18803d5665907","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-07-18 06:03:03.000000000","message":"Patch Set 3: Abandoned\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":3},{"id":"ba95bd7cabb64f653471828e44aca6aead716800","author":{"_account_id":9,"name":"LaunchpadSync","username":"launchpadsync"},"date":"2013-07-18 06:03:03.000000000","message":"Patch Set 3:\n\ncode review expired after 1 week of no activity after a negative review, it can be restored using the `Restore Change` button under the Patch Set on the web interface","accounts_in_message":[],"_revision_number":3},{"id":"14a27085d1039630cebbea7dc2e19198fbf632b5","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2013-08-03 09:47:18.000000000","message":"Patch Set 3: Restored\n\nnew commit message","accounts_in_message":[],"_revision_number":3},{"id":"5186e2699d0d39e5828b6b13a3e6d63699cd1719","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2013-08-03 09:48:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0f1ee2356d22627cd7a195f9eff29acfd0b70592","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2013-08-03 09:48:48.000000000","message":"Patch Set 4:\n\nNew patchset patch-id matches previous patchset, but commit message has changed.","accounts_in_message":[],"_revision_number":4},{"id":"08bd7ede009f70a296896bcfcc6511f6d42a0eef","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-08-03 10:23:20.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- http://logs.openstack.org/22/34622/4/check/gate-keystone-pep8/678862f : SUCCESS in 1m 25s\n- http://docs-draft.openstack.org/22/34622/4/check/gate-keystone-docs/e237a63/doc/build/html/ : SUCCESS in 2m 35s\n- http://logs.openstack.org/22/34622/4/check/gate-keystone-python26/fa344a0 : SUCCESS in 15m 34s\n- http://logs.openstack.org/22/34622/4/check/gate-keystone-python27/048e75c : SUCCESS in 8m 14s\n- http://logs.openstack.org/22/34622/4/check/gate-tempest-devstack-vm-full/0c41b53 : SUCCESS in 34m 33s\n- http://logs.openstack.org/22/34622/4/check/gate-tempest-devstack-vm-cells-full/800090c : SUCCESS in 34m 23s (non-voting)\n- http://logs.openstack.org/22/34622/4/check/gate-tempest-devstack-vm-postgres-full/3ea8e22 : SUCCESS in 34m 28s\n- http://logs.openstack.org/22/34622/4/check/gate-tempest-devstack-vm-neutron/2dd8f7c : SUCCESS in 18m 11s (non-voting)\n- http://logs.openstack.org/22/34622/4/check/gate-grenade-devstack-vm-stable/0dee702 : FAILURE in 16m 15s (non-voting)\n","accounts_in_message":[],"_revision_number":4},{"id":"1ec99cdf57afa8b03cac94b41fe4b54960408b84","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-08-03 21:04:06.000000000","message":"Patch Set 4: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"07d8f16dadecc2ff0ad8da727732e519f980073f","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2013-08-04 09:36:15.000000000","message":"Patch Set 4: Do not merge\n\nPlease keep master Change-Id: Iacd2d9e09be4ab3d6a3c5acf4074e4af7e300602\nhttps://wiki.openstack.org/wiki/StableBranch#Change-Ids\n\nDolph, is this worth freeze exception?\nIf yes, let\u0027s discuss it on the openstack-stable-maint before approval, for example http://lists.openstack.org/pipermail/openstack-stable-maint/2013-August/thread.html#1066","accounts_in_message":[],"_revision_number":4},{"id":"c64d0b0f15b2636985fa16dbc813f5f299af5eea","author":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"date":"2013-08-05 17:13:39.000000000","message":"Patch Set 4:\n\ncherry pick with original change-id https://review.openstack.org/40249","accounts_in_message":[],"_revision_number":4},{"id":"460a4eb99a8a9433bcb2816b895c6ef2afeb1669","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-08-05 19:06:45.000000000","message":"Patch Set 4:\n\nPlease mark this review as abandoned, as the other one with the same change Id was used instead.","accounts_in_message":[],"_revision_number":4},{"id":"6e14a7d9b28aef3d3154f7f4eeb9a8095a2bc300","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2013-08-05 19:06:57.000000000","message":"Patch Set 4: I would prefer that you didn\u0027t merge this\n\n","accounts_in_message":[],"_revision_number":4},{"id":"f6220292d0106160dcb51bf4702fa1591172dabe","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2013-08-10 09:23:48.000000000","message":"Patch Set 4: Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ea0c141e94da0abf706b0751b81c00d8872f28d8","revisions":{"77fe95cd5b06323488b6b36d8af21ada103def5d":{"kind":"REWORK","_number":1,"created":"2013-06-26 21:44:58.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/34622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/22/34622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/22/34622/1"}}},"commit":{"parents":[{"commit":"76a94c68d621c893941a817d36f8ddbe9559dbde","subject":"Merge \"Ignore conflict on v2 auto role assignment (bug 1161963)\" into stable/grizzly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/76a94c68d621c893941a817d36f8ddbe9559dbde"}]}],"author":{"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","date":"2013-06-03 19:21:36.000000000","tz":-300},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2013-06-26 20:40:38.000000000","tz":120},"subject":"Maintain tokens after role assignments (bug 1170186)","message":"Maintain tokens after role assignments (bug 1170186)\n\nChange-Id: Ib5e6fab1d3a20420eb05d6d50860f5040e8d6a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/77fe95cd5b06323488b6b36d8af21ada103def5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/77fe95cd5b06323488b6b36d8af21ada103def5d"}]},"parents_data":[{"branch_name":"refs/heads/stable/grizzly","commit_id":"76a94c68d621c893941a817d36f8ddbe9559dbde","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/grizzly"},"832a5a22b7b777356d91a7c1a2c4a50d7573a28e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2013-06-26 21:45:51.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/34622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/22/34622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/22/34622/2"}}},"commit":{"parents":[{"commit":"76a94c68d621c893941a817d36f8ddbe9559dbde","subject":"Merge \"Ignore conflict on v2 auto role assignment (bug 1161963)\" into stable/grizzly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/76a94c68d621c893941a817d36f8ddbe9559dbde"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2013-06-26 21:45:10.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2013-06-26 21:45:10.000000000","tz":120},"subject":"Backport from master:","message":"Backport from master:\n\n    Author: Dolph Mathews \u003cdolph.mathews@gmail.com\u003e\n    Date:   Mon Jun 3 14:21:36 2013 -0500\n\n    Maintain tokens after role assignments (bug 1170186)\n\nChange-Id: Ib5e6fab1d3a20420eb05d6d50860f5040e8d6a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/832a5a22b7b777356d91a7c1a2c4a50d7573a28e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/832a5a22b7b777356d91a7c1a2c4a50d7573a28e"}]},"parents_data":[{"branch_name":"refs/heads/stable/grizzly","commit_id":"76a94c68d621c893941a817d36f8ddbe9559dbde","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/grizzly"},"825160d26b3efecd3a39cadaabe0880a84f12dfc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2013-06-27 06:29:42.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/34622/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/22/34622/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/22/34622/3"}}},"commit":{"parents":[{"commit":"76a94c68d621c893941a817d36f8ddbe9559dbde","subject":"Merge \"Ignore conflict on v2 auto role assignment (bug 1161963)\" into stable/grizzly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/76a94c68d621c893941a817d36f8ddbe9559dbde"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2013-06-26 21:45:10.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2013-06-27 06:29:13.000000000","tz":120},"subject":"Backport from master:","message":"Backport from master:\n\n    Author: Dolph Mathews \u003cdolph.mathews@gmail.com\u003e\n    Date:   Mon Jun 3 14:21:36 2013 -0500\n\n    Maintain tokens after role assignments (bug 1170186)\n\n    (cherry picked from commit 132ff6d85e02acdce7483c2848686676cb4f14a0)\n\nChange-Id: Ib5e6fab1d3a20420eb05d6d50860f5040e8d6a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/825160d26b3efecd3a39cadaabe0880a84f12dfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/825160d26b3efecd3a39cadaabe0880a84f12dfc"}]},"parents_data":[{"branch_name":"refs/heads/stable/grizzly","commit_id":"76a94c68d621c893941a817d36f8ddbe9559dbde","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/grizzly"},"ea0c141e94da0abf706b0751b81c00d8872f28d8":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2013-08-03 09:48:40.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/22/34622/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/22/34622/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/22/34622/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/22/34622/4"}}},"commit":{"parents":[{"commit":"27a5b42dbbdcb1f10138542cfa2fc5584470bace","subject":"Fix issue with v3 tokens and group membership roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/27a5b42dbbdcb1f10138542cfa2fc5584470bace"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2013-06-26 21:45:10.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2013-08-03 09:47:58.000000000","tz":120},"subject":"Maintain tokens after role assignments","message":"Maintain tokens after role assignments\n\nFixes bug 1170186.\n\nBackport from master:\n    Author: Dolph Mathews \u003cdolph.mathews@gmail.com\u003e\n    Date:   Mon Jun 3 14:21:36 2013 -0500\n\n    Maintain tokens after role assignments (bug 1170186)\n\n    (cherry picked from commit 132ff6d85e02acdce7483c2848686676cb4f14a0)\n\nChange-Id: Ib5e6fab1d3a20420eb05d6d50860f5040e8d6a15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea0c141e94da0abf706b0751b81c00d8872f28d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea0c141e94da0abf706b0751b81c00d8872f28d8"}]},"parents_data":[{"branch_name":"refs/heads/stable/grizzly","commit_id":"27a5b42dbbdcb1f10138542cfa2fc5584470bace","is_merged_in_target_branch":true,"change_id":"I48aaf79241c87377c6940ab6193fc3acd4006c94","change_number":38484,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/stable/grizzly"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
