)]}'
{"id":"openstack%2Fkeystone~387154","triplet_id":"openstack%2Fkeystone~master~Iaf63a76660dc8d4674af6f24e516db44d00ef705","project":"openstack/keystone","branch":"master","topic":"refactor-auth","hashtags":[],"change_id":"Iaf63a76660dc8d4674af6f24e516db44d00ef705","subject":"Refactor authorization move filterprotected to authorization.py","status":"ABANDONED","created":"2016-10-17 04:59:20.000000000","updated":"2016-10-18 01:59:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b77d03078de6ce1223551768263f7193fcc9760b","_number":387154,"virtual_id_number":387154,"owner":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":-1,"date":"2016-10-17 05:52:49.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":-1,"date":"2016-10-17 05:05:18.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-10-17 05:52:49.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"53b0fe927a3b1768cd359dc77a9649dd68cadbf8","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-10-17 04:59:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ae825efda67e50a80c9b40c2de0c249ebb45937","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-10-17 05:05:18.000000000","message":"Patch Set 1: Workflow-1\n\nfor reference only","accounts_in_message":[],"_revision_number":1},{"id":"4d321710bc1707cb781962a218676793fd4d2035","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-10-17 05:52:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-docs-ubuntu-xenial http://docs-draft.openstack.org/54/387154/1/check/gate-keystone-docs-ubuntu-xenial/14108c0//doc/build/html/ : SUCCESS in 2m 55s\n- gate-keystone-pep8-ubuntu-xenial http://logs.openstack.org/54/387154/1/check/gate-keystone-pep8-ubuntu-xenial/44f9946/ : FAILURE in 4m 26s\n- gate-keystone-python27-db-ubuntu-xenial http://logs.openstack.org/54/387154/1/check/gate-keystone-python27-db-ubuntu-xenial/3987288/ : SUCCESS in 8m 43s\n- gate-keystone-python34-db http://logs.openstack.org/54/387154/1/check/gate-keystone-python34-db/f64b56e/ : SUCCESS in 12m 06s\n- gate-keystone-python35-db http://logs.openstack.org/54/387154/1/check/gate-keystone-python35-db/27bcdc1/ : SUCCESS in 8m 27s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/54/387154/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/a759e75/ : SUCCESS in 52m 03s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/54/387154/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/e6459b8/ : SUCCESS in 43m 13s\n- gate-keystone-dsvm-functional-ubuntu-xenial http://logs.openstack.org/54/387154/1/check/gate-keystone-dsvm-functional-ubuntu-xenial/0ac305d/ : SUCCESS in 14m 00s\n- gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/54/387154/1/check/gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv/46f7483/ : SUCCESS in 14m 01s (non-voting)\n- gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv http://logs.openstack.org/54/387154/1/check/gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv/3d9e219/ : SUCCESS in 40m 56s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/54/387154/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/a64c621/ : SUCCESS in 43m 35s (non-voting)\n- keystone-coverage-db-ubuntu-xenial http://logs.openstack.org/54/387154/1/check/keystone-coverage-db-ubuntu-xenial/0b29029/ : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":1},{"id":"25d1f2c2acd56220864f5c6b8920ff6649c60eb8","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2016-10-18 01:59:19.000000000","message":"Abandoned\n\nWas submitted for illustration purposes.  This and other reviews squashed into a single patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ecdf34c76941dec62e2fa09b6350c64dd5b3dd8","revisions":{"7ecdf34c76941dec62e2fa09b6350c64dd5b3dd8":{"kind":"REWORK","_number":1,"created":"2016-10-17 04:59:20.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/54/387154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/54/387154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/387154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/387154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/54/387154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/54/387154/1"}}},"commit":{"parents":[{"commit":"149a9d48e5c128cf27cc06b06a10e40fd7f85283","subject":"Move more authorization functions to authorization.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/149a9d48e5c128cf27cc06b06a10e40fd7f85283"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-14 21:09:33.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2016-10-14 21:09:33.000000000","tz":-240},"subject":"Refactor authorization move filterprotected to authorization.py","message":"Refactor authorization move filterprotected to authorization.py\n\nChange-Id: Iaf63a76660dc8d4674af6f24e516db44d00ef705\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7ecdf34c76941dec62e2fa09b6350c64dd5b3dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7ecdf34c76941dec62e2fa09b6350c64dd5b3dd8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"149a9d48e5c128cf27cc06b06a10e40fd7f85283","is_merged_in_target_branch":false,"change_id":"Ifebdfb8dbeae313061a6f22589da0f38b2d5e5b6","change_number":387153,"patch_set_number":1,"change_status":"ABANDONED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
