)]}'
{"id":"openstack%2Fkeystone~437402","triplet_id":"openstack%2Fkeystone~master~I4efe4b1b90c93110509cd599f9dd047c313dade3","project":"openstack/keystone","branch":"master","topic":"bug/1662762","hashtags":[],"change_id":"I4efe4b1b90c93110509cd599f9dd047c313dade3","subject":"Fix MFA rule checks for LDAP auth","status":"MERGED","created":"2017-02-23 14:32:28.000000000","updated":"2017-02-24 15:41:31.000000000","submitted":"2017-02-24 15:27:18.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aec2687c1b9bbc2fe9e50e509223cee5f562da38","_number":437402,"virtual_id_number":437402,"owner":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"date":"2017-02-24 15:41:31.000000000","post_submit":true,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":2,"date":"2017-02-24 15:27:18.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":2,"date":"2017-02-24 14:21:00.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},{"value":1,"date":"2017-02-24 14:21:00.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"date":"2017-02-24 06:25:43.000000000","_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-02-23 14:50:11.000000000","updated_by":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"reviewer":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"state":"REVIEWER"},{"updated":"2017-02-24 06:25:43.000000000","updated_by":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"reviewer":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"state":"REVIEWER"},{"updated":"2017-02-24 14:21:00.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2017-02-24 15:27:18.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-02-24 15:41:31.000000000","updated_by":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"reviewer":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"state":"REVIEWER"}],"messages":[{"id":"effa53dba03a09ced0c5b2d918e4b0a4bf664bae","author":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"date":"2017-02-23 14:32:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71a1d132c587117827f7acf462664738d8ca19e2","author":{"_account_id":13637,"name":"Divya K Konoor","email":"dikonoor@in.ibm.com","username":"dikonoor"},"date":"2017-02-23 14:50:11.000000000","message":"Patch Set 1:\n\nI had had a discussion on this fix with the community and dstanek was not in agreement with this fix. He was of the opinion that an empty options attribute as an attribute for the user. I don\u0027t recollect the reason for this though.","accounts_in_message":[],"_revision_number":1},{"id":"d0496df95babcb2fe134975f4168c53d48be798c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-23 15:53:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-docs-ubuntu-xenial http://docs-draft.openstack.org/02/437402/1/check/gate-keystone-docs-ubuntu-xenial/485750c//doc/build/html/ : SUCCESS in 5m 14s\n- gate-keystone-pep8-ubuntu-xenial http://logs.openstack.org/02/437402/1/check/gate-keystone-pep8-ubuntu-xenial/14e5e79/ : SUCCESS in 4m 52s\n- gate-keystone-python27-ubuntu-xenial http://logs.openstack.org/02/437402/1/check/gate-keystone-python27-ubuntu-xenial/0ff9845/ : SUCCESS in 8m 58s\n- gate-keystone-python35 http://logs.openstack.org/02/437402/1/check/gate-keystone-python35/c268485/ : SUCCESS in 9m 18s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/02/437402/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/d6e8745/ : SUCCESS in 1h 19m 34s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/02/437402/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/61a91ee/ : SUCCESS in 55m 56s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/02/437402/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/054c9e4/ : SUCCESS in 53m 17s\n- gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv http://logs.openstack.org/02/437402/1/check/gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv/5b8effe/ : SUCCESS in 30m 58s (non-voting)\n- gate-keystone-dsvm-functional-ubuntu-xenial http://logs.openstack.org/02/437402/1/check/gate-keystone-dsvm-functional-ubuntu-xenial/e2f7a3a/ : SUCCESS in 17m 44s\n- gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/02/437402/1/check/gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv/ee82fef/ : SUCCESS in 31m 40s (non-voting)\n- gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/02/437402/1/check/gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv/3c83e4f/ : SUCCESS in 32m 57s (non-voting)\n- gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv http://logs.openstack.org/02/437402/1/check/gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv/1b9d269/ : FAILURE in 1h 01m 07s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/02/437402/1/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/4f23b6c/ : SUCCESS in 1h 05m 40s (non-voting)\n- keystone-coverage-ubuntu-xenial http://logs.openstack.org/02/437402/1/check/keystone-coverage-ubuntu-xenial/042e80f/ : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bbeb78cc56007520882872957f52c5f009fec53d","author":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"date":"2017-02-23 18:17:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f9f12028e22d284091359101d97cc02cc739651","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-02-23 18:58:39.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)\n\nPlease provide a test and/or comments in the code to ensure we do not regress in the future here.","accounts_in_message":[],"_revision_number":2},{"id":"1e43aa113db0e3537aa252d73e677f5093a28173","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-02-23 18:58:48.000000000","message":"Patch Set 2: Code-Review-1 -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"05cd1ba9a819277aaae45346ba2c6372fde56b1b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-23 19:29:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-docs-ubuntu-xenial http://docs-draft.openstack.org/02/437402/2/check/gate-keystone-docs-ubuntu-xenial/dd784c5//doc/build/html/ : SUCCESS in 10m 27s\n- gate-keystone-pep8-ubuntu-xenial http://logs.openstack.org/02/437402/2/check/gate-keystone-pep8-ubuntu-xenial/b66636a/ : SUCCESS in 3m 39s\n- gate-keystone-python27-ubuntu-xenial http://logs.openstack.org/02/437402/2/check/gate-keystone-python27-ubuntu-xenial/9467fe1/ : FAILURE in 7m 33s\n- gate-keystone-python35 http://logs.openstack.org/02/437402/2/check/gate-keystone-python35/6223ff9/ : FAILURE in 7m 59s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/02/437402/2/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/f3e3f2f/ : FAILURE in 1h 10m 22s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/02/437402/2/check/gate-grenade-dsvm-neutron-ubuntu-xenial/7d866ab/ : SUCCESS in 36m 00s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/02/437402/2/check/gate-tempest-dsvm-py35-ubuntu-xenial/9b24309/ : SUCCESS in 50m 36s\n- gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv http://logs.openstack.org/02/437402/2/check/gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv/fca5baf/ : SUCCESS in 16m 16s (non-voting)\n- gate-keystone-dsvm-functional-ubuntu-xenial http://logs.openstack.org/02/437402/2/check/gate-keystone-dsvm-functional-ubuntu-xenial/c3f5310/ : SUCCESS in 17m 14s\n- gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/02/437402/2/check/gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv/a1a4f1f/ : SUCCESS in 25m 00s (non-voting)\n- gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/02/437402/2/check/gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv/614ac40/ : SUCCESS in 16m 55s (non-voting)\n- gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv http://logs.openstack.org/02/437402/2/check/gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv/598d437/ : FAILURE in 35m 47s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/02/437402/2/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/f254914/ : FAILURE in 1h 00m 49s (non-voting)\n- keystone-coverage-ubuntu-xenial http://logs.openstack.org/02/437402/2/check/keystone-coverage-ubuntu-xenial/28130b1/ : FAILURE in 8m 39s","accounts_in_message":[],"_revision_number":2},{"id":"43cf63e1008c348e7d64d492ce895f5b4bf2bf41","author":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"date":"2017-02-23 22:56:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"97b242d11f202f311709b874a63eff60246ce165","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-24 00:32:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-docs-ubuntu-xenial http://docs-draft.openstack.org/02/437402/3/check/gate-keystone-docs-ubuntu-xenial/93e1e97//doc/build/html/ : SUCCESS in 3m 53s\n- gate-keystone-pep8-ubuntu-xenial http://logs.openstack.org/02/437402/3/check/gate-keystone-pep8-ubuntu-xenial/1d3a5c1/ : SUCCESS in 3m 28s\n- gate-keystone-python27-ubuntu-xenial http://logs.openstack.org/02/437402/3/check/gate-keystone-python27-ubuntu-xenial/73db9f8/ : FAILURE in 7m 46s\n- gate-keystone-python35 http://logs.openstack.org/02/437402/3/check/gate-keystone-python35/301c149/ : FAILURE in 7m 40s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/02/437402/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/8204105/ : SUCCESS in 1h 34m 47s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/02/437402/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/427b1e3/ : SUCCESS in 35m 52s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/02/437402/3/check/gate-tempest-dsvm-py35-ubuntu-xenial/6cd0bdb/ : SUCCESS in 49m 47s\n- gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv http://logs.openstack.org/02/437402/3/check/gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv/a0f3f06/ : SUCCESS in 16m 56s (non-voting)\n- gate-keystone-dsvm-functional-ubuntu-xenial http://logs.openstack.org/02/437402/3/check/gate-keystone-dsvm-functional-ubuntu-xenial/45e2bb2/ : SUCCESS in 17m 21s\n- gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/02/437402/3/check/gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv/1c7b480/ : SUCCESS in 24m 25s (non-voting)\n- gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/02/437402/3/check/gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv/58a8b96/ : SUCCESS in 25m 20s (non-voting)\n- gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv http://logs.openstack.org/02/437402/3/check/gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv/f84583a/ : FAILURE in 36m 47s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/02/437402/3/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/1e14c46/ : SUCCESS in 1h 08m 34s (non-voting)\n- keystone-coverage-ubuntu-xenial http://logs.openstack.org/02/437402/3/check/keystone-coverage-ubuntu-xenial/8824b31/ : FAILURE in 13m 16s","accounts_in_message":[],"_revision_number":3},{"id":"595f50665db807f5b656f27f9f8e067594b44f2f","author":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"date":"2017-02-24 06:20:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5e49e2db5a5982077755f0e337ea602f562e1baf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-24 07:51:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-keystone-docs-ubuntu-xenial http://docs-draft.openstack.org/02/437402/4/check/gate-keystone-docs-ubuntu-xenial/0f5db5e//doc/build/html/ : SUCCESS in 3m 29s\n- gate-keystone-pep8-ubuntu-xenial http://logs.openstack.org/02/437402/4/check/gate-keystone-pep8-ubuntu-xenial/83e9dcd/ : SUCCESS in 3m 09s\n- gate-keystone-python27-ubuntu-xenial http://logs.openstack.org/02/437402/4/check/gate-keystone-python27-ubuntu-xenial/224560b/ : SUCCESS in 8m 23s\n- gate-keystone-python35 http://logs.openstack.org/02/437402/4/check/gate-keystone-python35/455cddf/ : SUCCESS in 11m 24s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/02/437402/4/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/4dad80e/ : SUCCESS in 1h 02m 50s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/02/437402/4/check/gate-grenade-dsvm-neutron-ubuntu-xenial/4758eb8/ : SUCCESS in 35m 05s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/02/437402/4/check/gate-tempest-dsvm-py35-ubuntu-xenial/14c1a84/ : SUCCESS in 48m 38s\n- gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv http://logs.openstack.org/02/437402/4/check/gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv/08e2aca/ : SUCCESS in 16m 16s (non-voting)\n- gate-keystone-dsvm-functional-ubuntu-xenial http://logs.openstack.org/02/437402/4/check/gate-keystone-dsvm-functional-ubuntu-xenial/7a9c516/ : SUCCESS in 17m 48s\n- gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/02/437402/4/check/gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv/c40c848/ : SUCCESS in 23m 58s (non-voting)\n- gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/02/437402/4/check/gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv/7047746/ : SUCCESS in 37m 33s (non-voting)\n- gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv http://logs.openstack.org/02/437402/4/check/gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv/9811c0c/ : FAILURE in 35m 17s (non-voting)\n- gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv http://logs.openstack.org/02/437402/4/check/gate-tempest-dsvm-neutron-identity-v3-only-full-ubuntu-xenial-nv/011a7ac/ : SUCCESS in 1h 29m 23s (non-voting)\n- keystone-coverage-ubuntu-xenial http://logs.openstack.org/02/437402/4/check/keystone-coverage-ubuntu-xenial/af19b0f/ : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":4},{"id":"863773f351d5248da2afb43852e13aeefdaccc90","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-02-24 14:21:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThis looks correct. Because we don\u0027t lean on shadow users enough at the moment to ensure that the things like options never hit the backend (pending fixes as what was agreeed upon at the PTG) this is the most correct fix.","accounts_in_message":[],"_revision_number":4},{"id":"b6c31b796c658c64d6963b5b19de1c3748bea369","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-24 14:21:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"936edb77e5eb07204448e29f0bef937cd3850509","date":"2017-02-24 15:27:18.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":4},{"id":"6be6fa2cc30041d59c25fd4944af1ae052cad6f8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-24 15:27:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-keystone-docs-ubuntu-xenial http://docs-draft.openstack.org/02/437402/4/gate/gate-keystone-docs-ubuntu-xenial/dee3523//doc/build/html/ : SUCCESS in 5m 58s\n- gate-keystone-pep8-ubuntu-xenial http://logs.openstack.org/02/437402/4/gate/gate-keystone-pep8-ubuntu-xenial/9cc5fc6/ : SUCCESS in 3m 12s\n- gate-keystone-python27-ubuntu-xenial http://logs.openstack.org/02/437402/4/gate/gate-keystone-python27-ubuntu-xenial/b1d8d46/ : SUCCESS in 8m 15s\n- gate-keystone-python35 http://logs.openstack.org/02/437402/4/gate/gate-keystone-python35/3702267/ : SUCCESS in 10m 25s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/02/437402/4/gate/gate-tempest-dsvm-neutron-full-ubuntu-xenial/06e895e/ : SUCCESS in 1h 05m 07s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/02/437402/4/gate/gate-grenade-dsvm-neutron-ubuntu-xenial/ed38633/ : SUCCESS in 39m 17s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/02/437402/4/gate/gate-tempest-dsvm-py35-ubuntu-xenial/0b53147/ : SUCCESS in 56m 20s\n- gate-keystone-dsvm-functional-ubuntu-xenial http://logs.openstack.org/02/437402/4/gate/gate-keystone-dsvm-functional-ubuntu-xenial/15a848b/ : SUCCESS in 18m 42s","accounts_in_message":[],"_revision_number":4},{"id":"b6a390e52956b82ea6547cefca6079e0501406f7","author":{"_account_id":6482,"name":"Steve Martinelli","email":"s.martinelli@gmail.com","username":"stevemar"},"date":"2017-02-24 15:41:31.000000000","message":"Patch Set 4:\n\nbackport?","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4e0029455ab45e3b9a15fe9fc151c14c502b7bdd","revisions":{"b794683c4b1b22e8531fc5787f6e501b138c6a33":{"kind":"REWORK","_number":1,"created":"2017-02-23 14:32:28.000000000","uploader":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"ref":"refs/changes/02/437402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/02/437402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/02/437402/1"}}},"commit":{"parents":[{"commit":"1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5","subject":"Merge \"Rename protocol cascade delete migration file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5"}]}],"author":{"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","date":"2017-02-23 12:36:49.000000000","tz":-300},"committer":{"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","date":"2017-02-23 14:32:21.000000000","tz":-300},"subject":"Fix MFA rule checks for LDAP auth","message":"Fix MFA rule checks for LDAP auth\n\nLDAP authentication was broken by the addition of MFA rule checking,\nThis patch fixes that.\n\nChange-Id: I4efe4b1b90c93110509cd599f9dd047c313dade3\nCloses-Bug: #1662762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b794683c4b1b22e8531fc5787f6e501b138c6a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b794683c4b1b22e8531fc5787f6e501b138c6a33"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"9d3d56de9e356fb19f6837b0e56bd7c7945db638":{"kind":"REWORK","_number":2,"created":"2017-02-23 18:17:43.000000000","uploader":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"ref":"refs/changes/02/437402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/02/437402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/02/437402/2"}}},"commit":{"parents":[{"commit":"1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5","subject":"Merge \"Rename protocol cascade delete migration file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5"}]}],"author":{"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","date":"2017-02-23 18:15:58.000000000","tz":-300},"committer":{"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","date":"2017-02-23 18:15:58.000000000","tz":-300},"subject":"Fix MFA rule checks for LDAP auth","message":"Fix MFA rule checks for LDAP auth\n\nLDAP authentication was broken by the addition of MFA rule checking.\nThis patch fixes that.\n\nChange-Id: I4efe4b1b90c93110509cd599f9dd047c313dade3\nCloses-Bug: #1662762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9d3d56de9e356fb19f6837b0e56bd7c7945db638"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9d3d56de9e356fb19f6837b0e56bd7c7945db638"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7493f70fbc2e7212d95c5cd9be77cf896ab4fbc7":{"kind":"REWORK","_number":3,"created":"2017-02-23 22:56:20.000000000","uploader":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"ref":"refs/changes/02/437402/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/02/437402/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/02/437402/3"}}},"commit":{"parents":[{"commit":"1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5","subject":"Merge \"Rename protocol cascade delete migration file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5"}]}],"author":{"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","date":"2017-02-23 18:15:58.000000000","tz":-300},"committer":{"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","date":"2017-02-23 22:56:04.000000000","tz":-300},"subject":"Fix MFA rule checks for LDAP auth","message":"Fix MFA rule checks for LDAP auth\n\nLDAP authentication was broken by the addition of MFA rule checking.\nThis patch fixes that.\n\nChange-Id: I4efe4b1b90c93110509cd599f9dd047c313dade3\nCloses-Bug: #1662762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7493f70fbc2e7212d95c5cd9be77cf896ab4fbc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7493f70fbc2e7212d95c5cd9be77cf896ab4fbc7"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4e0029455ab45e3b9a15fe9fc151c14c502b7bdd":{"kind":"REWORK","_number":4,"created":"2017-02-24 06:20:01.000000000","uploader":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"ref":"refs/changes/02/437402/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/02/437402/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/02/437402/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/02/437402/4"}}},"commit":{"parents":[{"commit":"1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5","subject":"Merge \"Rename protocol cascade delete migration file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5"}]}],"author":{"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","date":"2017-02-24 05:41:11.000000000","tz":-300},"committer":{"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","date":"2017-02-24 05:55:32.000000000","tz":-300},"subject":"Fix MFA rule checks for LDAP auth","message":"Fix MFA rule checks for LDAP auth\n\nLDAP authentication was broken by the addition of MFA rule checking.\nThis patch fixes that.\n\nChange-Id: I4efe4b1b90c93110509cd599f9dd047c313dade3\nCloses-Bug: #1662762\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4e0029455ab45e3b9a15fe9fc151c14c502b7bdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4e0029455ab45e3b9a15fe9fc151c14c502b7bdd"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1d4c72cd6fdad2c75e6b16ecf089852c604b8cd5","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
