)]}'
{"id":"openstack%2Fkeystone~478243","triplet_id":"openstack%2Fkeystone~master~I59c71c2479a9b9f2c9fda4e9c469d970af6bc779","project":"openstack/keystone","branch":"master","topic":"bp/removed-as-of-pike","hashtags":[],"change_id":"I59c71c2479a9b9f2c9fda4e9c469d970af6bc779","subject":"Remove support for direct import of drivers","status":"ABANDONED","created":"2017-06-27 19:13:02.000000000","updated":"2018-09-08 01:03:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8a06a5f6367aad9a95e73aee4c3a3afc508d9dc7","_number":478243,"virtual_id_number":478243,"owner":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":-1,"date":"2017-12-28 18:38:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":-1,"date":"2017-12-28 17:13:42.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-12-28 17:13:42.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2017-12-28 18:38:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"46a577ad036a6fe86e745f2beec3e88ea1a5ca46","author":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"date":"2017-06-27 19:13:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a04651207d854503489145d336d2f8b5a595977","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-27 20:35:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-docs-ubuntu-xenial http://docs-draft.openstack.org/43/478243/1/check/gate-keystone-docs-ubuntu-xenial/2a3635f//doc/build/html/ : SUCCESS in 4m 58s\n- gate-keystone-pep8-ubuntu-xenial http://logs.openstack.org/43/478243/1/check/gate-keystone-pep8-ubuntu-xenial/dc12f94/ : SUCCESS in 3m 15s\n- gate-keystone-python27-ubuntu-xenial http://logs.openstack.org/43/478243/1/check/gate-keystone-python27-ubuntu-xenial/3f58b7e/ : FAILURE in 11m 37s\n- gate-keystone-python35 http://logs.openstack.org/43/478243/1/check/gate-keystone-python35/d30fe11/ : FAILURE in 9m 57s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/478243/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/86ed57e/ : SUCCESS in 1h 21m 52s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/478243/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/6f3b89e/ : SUCCESS in 56m 02s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/43/478243/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/6b88921/ : SUCCESS in 1h 10m 01s\n- gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv http://logs.openstack.org/43/478243/1/check/gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv/a290066/ : SUCCESS in 28m 45s (non-voting)\n- gate-keystone-dsvm-functional-ubuntu-xenial http://logs.openstack.org/43/478243/1/check/gate-keystone-dsvm-functional-ubuntu-xenial/600d336/ : SUCCESS in 15m 40s\n- gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/43/478243/1/check/gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv/541bdcc/ : FAILURE in 17m 12s (non-voting)\n- gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/43/478243/1/check/gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv/bd7e77e/ : FAILURE in 23m 24s (non-voting)\n- gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv http://logs.openstack.org/43/478243/1/check/gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv/a94dbc5/ : SUCCESS in 1h 06m 38s (non-voting)\n- keystone-coverage-ubuntu-xenial http://logs.openstack.org/43/478243/1/check/keystone-coverage-ubuntu-xenial/80b9d12/ : FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ca42c962039e3b69e902a49e55499728eacd3cab","author":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"date":"2017-06-28 00:59:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5b0afcc20b252cb6a3b24b9c8464606bec79db1","author":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"date":"2017-06-28 01:27:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c9a85504157c981e85feb2bac6cc872fb469a081","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-28 02:53:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-docs-ubuntu-xenial http://docs-draft.openstack.org/43/478243/3/check/gate-keystone-docs-ubuntu-xenial/c48046e//doc/build/html/ : SUCCESS in 3m 12s\n- gate-keystone-pep8-ubuntu-xenial http://logs.openstack.org/43/478243/3/check/gate-keystone-pep8-ubuntu-xenial/5b27e47/ : SUCCESS in 3m 11s\n- gate-keystone-python27-ubuntu-xenial http://logs.openstack.org/43/478243/3/check/gate-keystone-python27-ubuntu-xenial/266a534/ : FAILURE in 9m 36s\n- gate-keystone-python35 http://logs.openstack.org/43/478243/3/check/gate-keystone-python35/32f139e/ : FAILURE in 8m 58s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/43/478243/3/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/5cf8089/ : SUCCESS in 1h 25m 37s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/43/478243/3/check/gate-grenade-dsvm-neutron-ubuntu-xenial/45448f4/ : SUCCESS in 52m 17s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/43/478243/3/check/gate-tempest-dsvm-py35-ubuntu-xenial/7883959/ : SUCCESS in 1h 16m 48s\n- gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv http://logs.openstack.org/43/478243/3/check/gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv/2870e9c/ : SUCCESS in 26m 44s (non-voting)\n- gate-keystone-dsvm-functional-ubuntu-xenial http://logs.openstack.org/43/478243/3/check/gate-keystone-dsvm-functional-ubuntu-xenial/d16049d/ : SUCCESS in 16m 34s\n- gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/43/478243/3/check/gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv/2f29e7e/ : FAILURE in 28m 33s (non-voting)\n- gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/43/478243/3/check/gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv/3c1088b/ : FAILURE in 26m 37s (non-voting)\n- gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv http://logs.openstack.org/43/478243/3/check/gate-tempest-dsvm-keystone-uwsgi-full-ubuntu-xenial-nv/cdc985a/ : SUCCESS in 1h 07m 51s (non-voting)\n- keystone-coverage-ubuntu-xenial http://logs.openstack.org/43/478243/3/check/keystone-coverage-ubuntu-xenial/8b29276/ : FAILURE in 10m 38s","accounts_in_message":[],"_revision_number":3},{"id":"b231f12b739615ae89b24604e8730623d451c8b8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 02:58:11.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":3},{"id":"4860e61ac155518c633d1c5dd98bdd1da059702f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-12-28 17:12:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6bbc51278bd3de9ac0d6feedb6de1d7238a89fe1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-12-28 17:13:42.000000000","message":"Patch Set 4: Workflow-1\n\nThis still needs a little help with one test case:\n\nhttps://github.com/openstack/keystone/blob/master/keystone/tests/unit/test_auth_plugin.py#L69-L111","accounts_in_message":[],"_revision_number":4},{"id":"69f08bb762b0c77d835a79ccc8da135772b4c495","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-28 18:38:51.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/43/478243/4/check/openstack-tox-pep8/dd06e5a/ : SUCCESS in 4m 46s\n- openstack-tox-py27 http://logs.openstack.org/43/478243/4/check/openstack-tox-py27/9e847e4/ : FAILURE in 10m 39s\n- openstack-tox-py35 http://logs.openstack.org/43/478243/4/check/openstack-tox-py35/fd7effd/ : FAILURE in 10m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/43/478243/4/check/build-openstack-sphinx-docs/db17f18/html/ : SUCCESS in 6m 49s\n- tempest-full http://logs.openstack.org/43/478243/4/check/tempest-full/07484bc/ : SUCCESS in 1h 15m 30s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/43/478243/4/check/legacy-grenade-dsvm-neutron/e9b1340/ : SUCCESS in 50m 27s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/43/478243/4/check/legacy-tempest-dsvm-py35/9b80532/ : POST_FAILURE in 4m 40s\n- build-openstack-releasenotes http://logs.openstack.org/43/478243/4/check/build-openstack-releasenotes/f18962f/html/ : SUCCESS in 5m 36s\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/43/478243/4/check/legacy-keystoneclient-dsvm-functional/a543d8f/ : SUCCESS in 25m 09s (non-voting)\n- legacy-keystone-dsvm-functional http://logs.openstack.org/43/478243/4/check/legacy-keystone-dsvm-functional/47113cd/ : SUCCESS in 23m 16s\n- legacy-keystone-dsvm-functional-v3-only http://logs.openstack.org/43/478243/4/check/legacy-keystone-dsvm-functional-v3-only/3d9c1f7/ : FAILURE in 30m 08s (non-voting)\n- legacy-keystone-dsvm-py35-functional-v3-only http://logs.openstack.org/43/478243/4/check/legacy-keystone-dsvm-py35-functional-v3-only/7036598/ : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/43/478243/4/check/openstack-tox-cover/4e0e6e9/ : POST_FAILURE in 11m 15s\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/43/478243/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/27f7ab1/ : SUCCESS in 1h 18m 30s (non-voting)\n- openstack-ansible-keystone-rolling-upgrade http://logs.openstack.org/43/478243/4/check/openstack-ansible-keystone-rolling-upgrade/1260d1c/ : FAILURE in 24m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bb34bdcb9f4be6c802987cbb0b45e021fa5c29db","author":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"date":"2018-09-08 01:03:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a27966721d70de1c8d5a18ce24664febc3fa8372","revisions":{"49298d09147b812cd10c9a010bb052ab425e26a0":{"kind":"REWORK","_number":1,"created":"2017-06-27 19:13:02.000000000","uploader":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"ref":"refs/changes/43/478243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/43/478243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/43/478243/1"}}},"commit":{"parents":[{"commit":"a1976aa2c9b70de30ea6f646b430bd49f82dbcc1","subject":"Merge \"Remove note about kvs from admin-guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a1976aa2c9b70de30ea6f646b430bd49f82dbcc1"}]}],"author":{"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","date":"2017-06-27 20:14:55.000000000","tz":-240},"committer":{"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","date":"2017-06-27 20:14:55.000000000","tz":-240},"subject":"Remove support for direct import of drivers","message":"Remove support for direct import of drivers\n\nChange-Id: I59c71c2479a9b9f2c9fda4e9c469d970af6bc779\nPartially-Implements: bp removed-as-of-pike\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/49298d09147b812cd10c9a010bb052ab425e26a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/49298d09147b812cd10c9a010bb052ab425e26a0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a1976aa2c9b70de30ea6f646b430bd49f82dbcc1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"67e1b6d2320807399fc6a2fa628066d27ef1c863":{"kind":"REWORK","_number":2,"created":"2017-06-28 00:59:19.000000000","uploader":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"ref":"refs/changes/43/478243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/43/478243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/43/478243/2"}}},"commit":{"parents":[{"commit":"a1976aa2c9b70de30ea6f646b430bd49f82dbcc1","subject":"Merge \"Remove note about kvs from admin-guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a1976aa2c9b70de30ea6f646b430bd49f82dbcc1"}]}],"author":{"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","date":"2017-06-27 20:14:55.000000000","tz":-240},"committer":{"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","date":"2017-06-28 02:01:31.000000000","tz":-240},"subject":"Remove support for direct import of drivers","message":"Remove support for direct import of drivers\n\nPartially-Implements: bp removed-as-of-pike\nChange-Id: I59c71c2479a9b9f2c9fda4e9c469d970af6bc779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/67e1b6d2320807399fc6a2fa628066d27ef1c863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/67e1b6d2320807399fc6a2fa628066d27ef1c863"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a1976aa2c9b70de30ea6f646b430bd49f82dbcc1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3d5b9d340ece809f7fd85192efd034c5b85beba0":{"kind":"REWORK","_number":3,"created":"2017-06-28 01:27:18.000000000","uploader":{"_account_id":17860,"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","username":"samueldmq"},"ref":"refs/changes/43/478243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/43/478243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/43/478243/3"}}},"commit":{"parents":[{"commit":"a1976aa2c9b70de30ea6f646b430bd49f82dbcc1","subject":"Merge \"Remove note about kvs from admin-guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a1976aa2c9b70de30ea6f646b430bd49f82dbcc1"}]}],"author":{"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","date":"2017-06-27 20:14:55.000000000","tz":-240},"committer":{"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","date":"2017-06-28 02:29:40.000000000","tz":-240},"subject":"Remove support for direct import of drivers","message":"Remove support for direct import of drivers\n\nPartially-Implements: bp removed-as-of-pike\nChange-Id: I59c71c2479a9b9f2c9fda4e9c469d970af6bc779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3d5b9d340ece809f7fd85192efd034c5b85beba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3d5b9d340ece809f7fd85192efd034c5b85beba0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a1976aa2c9b70de30ea6f646b430bd49f82dbcc1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a27966721d70de1c8d5a18ce24664febc3fa8372":{"kind":"REWORK","_number":4,"created":"2017-12-28 17:12:57.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/43/478243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/43/478243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/43/478243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/43/478243/4"}}},"commit":{"parents":[{"commit":"1e21c52f3b35c8108fadb4c14c2059adf7b53935","subject":"Merge \"modify LOG.error tip message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1e21c52f3b35c8108fadb4c14c2059adf7b53935"}]}],"author":{"name":"Samuel de Medeiros Queiroz","email":"samueldmq@gmail.com","date":"2017-06-27 20:14:55.000000000","tz":-240},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2017-12-28 16:51:13.000000000","tz":0},"subject":"Remove support for direct import of drivers","message":"Remove support for direct import of drivers\n\nPartially-Implements: bp removed-as-of-queens\nChange-Id: I59c71c2479a9b9f2c9fda4e9c469d970af6bc779\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a27966721d70de1c8d5a18ce24664febc3fa8372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a27966721d70de1c8d5a18ce24664febc3fa8372"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1e21c52f3b35c8108fadb4c14c2059adf7b53935","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
