)]}'
{"id":"openstack%2Fkeystone~499703","triplet_id":"openstack%2Fkeystone~master~I4ba17855efd797c0db9f4824936b49e4bff54b6a","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"I4ba17855efd797c0db9f4824936b49e4bff54b6a","subject":"Remove Dependency Injection","status":"MERGED","created":"2017-08-31 16:23:31.000000000","updated":"2017-12-15 02:33:16.000000000","submitted":"2017-12-15 02:33:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"499703-1513305196582-217fbfef","meta_rev_id":"d735d6f7fb85104988b684e50e3a75aa46bb1a42","_number":499703,"virtual_id_number":499703,"owner":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"date":"2017-12-14 19:15:36.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":2,"date":"2017-12-15 02:33:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-12-14 16:15:02.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2017-12-14 22:48:40.000000000","_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":1,"date":"2017-12-14 22:48:40.000000000","_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-31 17:45:16.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-12-14 16:15:02.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2017-12-14 22:48:40.000000000","updated_by":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"reviewer":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"state":"REVIEWER"},{"updated":"2017-12-15 02:33:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be09b299049292280a52d52e151cfde08c7a9a88","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-08-31 16:23:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e2de01ccf441104d45a1b14faf642af7386957c","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-08-31 16:25:16.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0ffaed0e172bca799a72ccb6fae5b8c3cc6d0c12","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-31 17:45:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-keystone-docs-ubuntu-xenial http://logs.openstack.org/03/499703/1/check/gate-keystone-docs-ubuntu-xenial/132913f/ : FAILURE in 3m 15s\n- gate-keystone-pep8-ubuntu-xenial http://logs.openstack.org/03/499703/1/check/gate-keystone-pep8-ubuntu-xenial/f7774bc/ : FAILURE in 5m 48s\n- gate-keystone-python27-ubuntu-xenial http://logs.openstack.org/03/499703/1/check/gate-keystone-python27-ubuntu-xenial/e554ef1/ : FAILURE in 4m 15s\n- gate-keystone-python35 http://logs.openstack.org/03/499703/1/check/gate-keystone-python35/afdedc0/ : FAILURE in 2m 48s\n- gate-tempest-dsvm-neutron-full-ubuntu-xenial http://logs.openstack.org/03/499703/1/check/gate-tempest-dsvm-neutron-full-ubuntu-xenial/52c209b/ : FAILURE in 14m 59s\n- gate-grenade-dsvm-neutron-ubuntu-xenial http://logs.openstack.org/03/499703/1/check/gate-grenade-dsvm-neutron-ubuntu-xenial/3e78c73/ : FAILURE in 35m 56s\n- gate-tempest-dsvm-py35-ubuntu-xenial http://logs.openstack.org/03/499703/1/check/gate-tempest-dsvm-py35-ubuntu-xenial/3df01f8/ : FAILURE in 12m 02s\n- gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv http://logs.openstack.org/03/499703/1/check/gate-keystoneclient-dsvm-functional-ubuntu-xenial-nv/57ebce5/ : FAILURE in 11m 55s (non-voting)\n- gate-keystone-dsvm-functional-ubuntu-xenial http://logs.openstack.org/03/499703/1/check/gate-keystone-dsvm-functional-ubuntu-xenial/3880ca5/ : FAILURE in 15m 05s\n- gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/03/499703/1/check/gate-keystone-dsvm-functional-v3-only-ubuntu-xenial-nv/3bfd867/ : FAILURE in 14m 28s (non-voting)\n- gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv http://logs.openstack.org/03/499703/1/check/gate-keystone-dsvm-py35-functional-v3-only-ubuntu-xenial-nv/f7b3d3a/ : FAILURE in 14m 56s (non-voting)\n- keystone-coverage-ubuntu-xenial http://logs.openstack.org/03/499703/1/check/keystone-coverage-ubuntu-xenial/b76c10d/ : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b604a56ea394c64e3df9bc58f6662eeb2a639008","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-04 23:50:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5014e21de992d0a6c8dec8e844fea2ce0235eeb","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-05 00:04:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b70610bf3b699297c592d32c8fed7e19d5c74f5f","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-05 00:06:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c4095e65aeb482da14bcede8443810803e108b05","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-05 00:32:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dc7fbba9e8045174055d53a8898eb57d947464e7","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-05 00:33:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f44a68b761023f6ffefe598443e89fe661f789f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-05 01:52:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/499703/4/check/openstack-tox-pep8/d1b2e9d/ : SUCCESS in 4m 16s\n- openstack-tox-py27 http://logs.openstack.org/03/499703/4/check/openstack-tox-py27/bd1d113/ : SUCCESS in 11m 08s\n- openstack-tox-py35 http://logs.openstack.org/03/499703/4/check/openstack-tox-py35/ac33bc0/ : SUCCESS in 10m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/499703/4/check/build-openstack-sphinx-docs/eebdaf9/html/ : SUCCESS in 6m 06s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/03/499703/4/check/legacy-tempest-dsvm-py35/5e3b8bc/ : SUCCESS in 1h 14m 43s\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/03/499703/4/check/legacy-keystoneclient-dsvm-functional/4491870/ : SUCCESS in 25m 53s (non-voting)\n- legacy-keystone-dsvm-functional http://logs.openstack.org/03/499703/4/check/legacy-keystone-dsvm-functional/66766f6/ : SUCCESS in 31m 34s\n- legacy-keystone-dsvm-functional-v3-only http://logs.openstack.org/03/499703/4/check/legacy-keystone-dsvm-functional-v3-only/53385bd/ : FAILURE in 25m 40s (non-voting)\n- legacy-keystone-dsvm-py35-functional-v3-only http://logs.openstack.org/03/499703/4/check/legacy-keystone-dsvm-py35-functional-v3-only/9e6a7c9/ : SUCCESS in 25m 10s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/03/499703/4/check/openstack-tox-cover/47f95d8/cover/ : SUCCESS in 15m 02s\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/03/499703/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/9b2bb24/ : SUCCESS in 1h 17m 18s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/03/499703/4/check/legacy-tempest-dsvm-neutron-full/fcfaaf2/ : SUCCESS in 1h 13m 46s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/03/499703/4/check/legacy-grenade-dsvm-neutron/1f1e955/ : SUCCESS in 48m 46s\n- openstack-ansible-keystone-rolling-upgrade http://logs.openstack.org/03/499703/4/check/openstack-ansible-keystone-rolling-upgrade/573e453/ : FAILURE in 25m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5509cc2f6b503a3d377e4e46ca4550d12af8dabc","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-12-07 20:12:42.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)\n\n+1 on the overall direction. This moves us away from having to deal with dependency management for each manager individually. In the future, it would be cool to try and do something with registration and metaclasses and see if that helps get rid of special ordering during setup.\n\nhttps://review.openstack.org/#/c/492621/","accounts_in_message":[],"_revision_number":4},{"id":"19227969f11cf1e4fcc6d256bc839c11db6fff3e","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-08 15:22:53.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e33c92730fc26067594c58a41186e59eac9fd670","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-13 18:56:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"68ad3ac497c1a4c27de47db7c73e10d38494e489","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-13 19:00:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1fe68f1f1851313c7e112f2bd4ff43081d3248e7","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-13 20:18:16.000000000","message":"Patch Set 6:\n\nThe next step is to stop using self.XXX_api and reference provider_api.ProviderAPIs.XXX_api directly.","accounts_in_message":[],"_revision_number":6},{"id":"ddfa1ca6f604ae9bc9b46052ca2bb94c4d0a568b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-13 20:23:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/499703/6/check/openstack-tox-pep8/e64045e/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/03/499703/6/check/openstack-tox-py27/61b18e9/ : SUCCESS in 11m 17s\n- openstack-tox-py35 http://logs.openstack.org/03/499703/6/check/openstack-tox-py35/fc134b7/ : SUCCESS in 14m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/499703/6/check/build-openstack-sphinx-docs/c045fb2/html/ : SUCCESS in 5m 41s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/03/499703/6/check/legacy-tempest-dsvm-py35/e500ae2/ : SUCCESS in 1h 17m 44s\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/03/499703/6/check/legacy-keystoneclient-dsvm-functional/d32b54a/ : SUCCESS in 31m 40s (non-voting)\n- legacy-keystone-dsvm-functional http://logs.openstack.org/03/499703/6/check/legacy-keystone-dsvm-functional/fb2f4e5/ : SUCCESS in 23m 53s\n- legacy-keystone-dsvm-functional-v3-only http://logs.openstack.org/03/499703/6/check/legacy-keystone-dsvm-functional-v3-only/3cde238/ : SUCCESS in 26m 09s (non-voting)\n- legacy-keystone-dsvm-py35-functional-v3-only http://logs.openstack.org/03/499703/6/check/legacy-keystone-dsvm-py35-functional-v3-only/1a21c26/ : FAILURE in 23m 37s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/03/499703/6/check/openstack-tox-cover/6abb6f5/cover/ : SUCCESS in 11m 08s\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/03/499703/6/check/legacy-tempest-dsvm-ldap-domain-specific-driver/872f068/ : SUCCESS in 1h 18m 40s (non-voting)\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/03/499703/6/check/legacy-tempest-dsvm-neutron-full/fc9094c/ : SUCCESS in 1h 19m 22s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/03/499703/6/check/legacy-grenade-dsvm-neutron/6108e96/ : SUCCESS in 46m 27s\n- openstack-ansible-keystone-rolling-upgrade http://logs.openstack.org/03/499703/6/check/openstack-ansible-keystone-rolling-upgrade/e0292d5/ : FAILURE in 54m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bf943afa5179ff8b0441ea0a06bf84a77c816897","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-12-14 15:54:13.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bb1576afd4e344ba004b8f4257cc9ada9edbb88a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2017-12-14 16:15:02.000000000","message":"Patch Set 6: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"41c271a064d6d16848b1e433fffd5ed0713eb66e","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2017-12-14 19:15:36.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c61914b126ca7e37c19851971b39fe9d98ed1912","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2017-12-14 22:48:40.000000000","message":"Patch Set 6: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"5efabe1bf558a07bdbceb9325e8fce37aa9e5a6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 22:49:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3ae12d6dfe6b048f541edb4738e962450864a06c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 00:43:23.000000000","message":"Patch Set 6:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"bfab17304ac6ed22b5e081738f30f99f7fb6dc30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 02:33:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"d735d6f7fb85104988b684e50e3a75aa46bb1a42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-15 02:33:16.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/03/499703/6/gate/openstack-tox-pep8/29c26d0/ : SUCCESS in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/03/499703/6/gate/openstack-tox-py27/9d3d5d0/ : SUCCESS in 15m 27s\n- openstack-tox-py35 http://logs.openstack.org/03/499703/6/gate/openstack-tox-py35/2134429/ : SUCCESS in 9m 56s\n- build-openstack-sphinx-docs http://logs.openstack.org/03/499703/6/gate/build-openstack-sphinx-docs/d8e21aa/html/ : SUCCESS in 6m 20s\n- legacy-tempest-dsvm-py35 http://logs.openstack.org/03/499703/6/gate/legacy-tempest-dsvm-py35/b750359/ : SUCCESS in 1h 11m 35s\n- legacy-keystone-dsvm-functional http://logs.openstack.org/03/499703/6/gate/legacy-keystone-dsvm-functional/518f8e0/ : SUCCESS in 23m 10s\n- legacy-tempest-dsvm-neutron-full http://logs.openstack.org/03/499703/6/gate/legacy-tempest-dsvm-neutron-full/6ee83f8/ : SUCCESS in 1h 25m 23s\n- legacy-grenade-dsvm-neutron http://logs.openstack.org/03/499703/6/gate/legacy-grenade-dsvm-neutron/0067046/ : SUCCESS in 42m 28s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"81f9fe6fed62ec629804c9367fbb9ebfd584388c","revisions":{"94c73a3238fc2f690a1fb8c58f99abc9fce128e3":{"kind":"REWORK","_number":1,"created":"2017-08-31 16:23:31.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/03/499703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/499703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/499703/1"}}},"commit":{"parents":[{"commit":"746a2ee2fa0cb41dc3027237660d7990d9a9c88d","subject":"Merge \"Delete redundant code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/746a2ee2fa0cb41dc3027237660d7990d9a9c88d"}]}],"author":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-08-31 16:23:00.000000000","tz":-420},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-08-31 16:23:00.000000000","tz":-420},"subject":"WIP: Remove dependency injection","message":"WIP: Remove dependency injection\n\nChange-Id: I4ba17855efd797c0db9f4824936b49e4bff54b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/94c73a3238fc2f690a1fb8c58f99abc9fce128e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/94c73a3238fc2f690a1fb8c58f99abc9fce128e3"}]},"branch":"refs/heads/master"},"6fb8ab615c785767aad2951cd79cdc8deed8dc0c":{"kind":"REWORK","_number":2,"created":"2017-12-04 23:50:15.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/03/499703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/499703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/499703/2"}}},"commit":{"parents":[{"commit":"51d5b63a083450468cec474b9b6400df5d977091","subject":"Merge \"Assert default project id is not domain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51d5b63a083450468cec474b9b6400df5d977091"}]}],"author":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-04 23:49:48.000000000","tz":-480},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-04 23:49:48.000000000","tz":-480},"subject":"WIP: Remove Dependency Injection","message":"WIP: Remove Dependency Injection\n\nChange-Id: I4ba17855efd797c0db9f4824936b49e4bff54b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6fb8ab615c785767aad2951cd79cdc8deed8dc0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6fb8ab615c785767aad2951cd79cdc8deed8dc0c"}]},"branch":"refs/heads/master"},"a45845dfa45e7b148c97b05071bad39b206b1351":{"kind":"REWORK","_number":3,"created":"2017-12-05 00:04:57.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/03/499703/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/499703/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/499703/3"}}},"commit":{"parents":[{"commit":"51d5b63a083450468cec474b9b6400df5d977091","subject":"Merge \"Assert default project id is not domain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51d5b63a083450468cec474b9b6400df5d977091"}]}],"author":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-04 23:49:48.000000000","tz":-480},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-05 00:04:49.000000000","tz":-480},"subject":"WIP: Remove Dependency Injection","message":"WIP: Remove Dependency Injection\n\nChange-Id: I4ba17855efd797c0db9f4824936b49e4bff54b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a45845dfa45e7b148c97b05071bad39b206b1351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a45845dfa45e7b148c97b05071bad39b206b1351"}]},"branch":"refs/heads/master"},"7d6e685311424f49acf7f5194fbfc1b70da082a1":{"kind":"REWORK","_number":4,"created":"2017-12-05 00:32:15.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/03/499703/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/499703/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/499703/4"}}},"commit":{"parents":[{"commit":"51d5b63a083450468cec474b9b6400df5d977091","subject":"Merge \"Assert default project id is not domain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51d5b63a083450468cec474b9b6400df5d977091"}]}],"author":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-04 23:49:48.000000000","tz":-480},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-05 00:25:46.000000000","tz":-480},"subject":"WIP: Remove Dependency Injection","message":"WIP: Remove Dependency Injection\n\nRefactors all of keystone\u0027s dependency injection to maintain a\nsingle centralized repository of instantiated objects. This\nmeans that we are no longer having to resolve order. All\nobjects that need to reference the various manager APIs simply\ndo so via the __getattr__ built into the Manager common object\nor the ProviderAPIMixin object.\n\nThis is also the first step towards correcting our tests to\nwhere they cannot run \"load_backends\" multiple times.\n\nThis forces any/all managers to properly run super()\nas the way to register the api is via __init__.\n\nThis eliminates all use of the @dependency.requires and\n@dependency.provides decorators, simplifying the objects\nall around.\n\nAny instantiations of a Manager after keystone is running\nwill now generate an error, ensuring everything for keystone\nis running before handling requests. An exception is for\nCLI and CLI tests, as the CLI may directly instantiate\nmanagers and will not lock the registry.\n\nChange-Id: I4ba17855efd797c0db9f4824936b49e4bff54b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7d6e685311424f49acf7f5194fbfc1b70da082a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7d6e685311424f49acf7f5194fbfc1b70da082a1"}]},"branch":"refs/heads/master"},"333c3c6e27d9cc21caeb05bfb4cdca57bdefd6db":{"kind":"REWORK","_number":5,"created":"2017-12-13 18:56:19.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/03/499703/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/499703/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/499703/5"}}},"commit":{"parents":[{"commit":"3cc3986a89723c483ee55d6c383b54cb973336f2","subject":"Remove extra parameter for token auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3cc3986a89723c483ee55d6c383b54cb973336f2"}]}],"author":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-04 23:49:48.000000000","tz":-480},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-13 18:56:06.000000000","tz":-480},"subject":"Remove Dependency Injection","message":"Remove Dependency Injection\n\nRefactors all of keystone\u0027s dependency injection to maintain a\nsingle centralized repository of instantiated objects. This\nmeans that we are no longer having to resolve order. All\nobjects that need to reference the various manager APIs simply\ndo so via the __getattr__ built into the Manager common object\nor the ProviderAPIMixin object.\n\nThis is also the first step towards correcting our tests to\nwhere they cannot run \"load_backends\" multiple times.\n\nThis forces any/all managers to properly run super()\nas the way to register the api is via __init__.\n\nThis eliminates all use of the @dependency.requires and\n@dependency.provides decorators, simplifying the objects\nall around.\n\nAny instantiations of a Manager after keystone is running\nwill now generate an error, ensuring everything for keystone\nis running before handling requests. An exception is for\nCLI and CLI tests, as the CLI may directly instantiate\nmanagers and will not lock the registry.\n\nChange-Id: I4ba17855efd797c0db9f4824936b49e4bff54b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/333c3c6e27d9cc21caeb05bfb4cdca57bdefd6db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/333c3c6e27d9cc21caeb05bfb4cdca57bdefd6db"}]},"branch":"refs/heads/master"},"81f9fe6fed62ec629804c9367fbb9ebfd584388c":{"kind":"REWORK","_number":6,"created":"2017-12-13 19:00:28.000000000","uploader":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"ref":"refs/changes/03/499703/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/499703/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/499703/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/499703/6"}}},"commit":{"parents":[{"commit":"3cc3986a89723c483ee55d6c383b54cb973336f2","subject":"Remove extra parameter for token auth","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3cc3986a89723c483ee55d6c383b54cb973336f2"}]}],"author":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-04 23:49:48.000000000","tz":-480},"committer":{"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","date":"2017-12-13 18:59:39.000000000","tz":-480},"subject":"Remove Dependency Injection","message":"Remove Dependency Injection\n\nRefactors all of keystone\u0027s dependency injection to maintain a\nsingle centralized repository of instantiated objects. This\nmeans that we are no longer having to resolve order. All\nobjects that need to reference the various manager APIs simply\ndo so via the __getattr__ built into the Manager common object\nor the ProviderAPIMixin object.\n\nThis is also the first step towards correcting our tests to\nwhere they cannot run \"load_backends\" multiple times.\n\nThis forces any/all managers to properly run super()\nas the way to register the api is via __init__.\n\nThis eliminates all use of the @dependency.requires and\n@dependency.provides decorators, simplifying the objects\nall around.\n\nAny instantiations of a Manager after keystone is running\nwill now generate an error, ensuring everything for keystone\nis running before handling requests. An exception is for\nCLI and CLI tests, as the CLI may directly instantiate\nmanagers and will not lock the registry.\n\nChange-Id: I4ba17855efd797c0db9f4824936b49e4bff54b6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/81f9fe6fed62ec629804c9367fbb9ebfd584388c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/81f9fe6fed62ec629804c9367fbb9ebfd584388c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
