)]}'
{"id":"openstack%2Fkeystone~50076","triplet_id":"openstack%2Fkeystone~master~If66b9bd3ebcfc3a59151b52a46407a36d2a9352b","project":"openstack/keystone","branch":"master","topic":"use-abc-for-plugins","hashtags":[],"change_id":"If66b9bd3ebcfc3a59151b52a46407a36d2a9352b","subject":"Use abstract base class for catalog driver","status":"MERGED","created":"2013-10-07 13:51:51.000000000","updated":"2013-10-16 23:11:57.000000000","submitted":"2013-10-16 23:11:56.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cf52fde57213c957e791539ae0438a53a1fc1b8e","_number":50076,"virtual_id_number":50076,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2013-10-16 23:11:56.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2013-10-15 02:19:19.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-10-08 12:26:08.000000000","_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":1,"date":"2013-10-08 12:26:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"date":"2013-10-16 22:27:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":2,"date":"2013-10-15 23:10:52.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2013-10-08 12:26:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"date":"2013-10-16 22:27:06.000000000","_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":1,"date":"2013-10-15 23:10:52.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-10-08 12:26:08.000000000","updated_by":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"reviewer":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"state":"REVIEWER"},{"updated":"2013-10-08 12:26:08.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2013-10-08 12:26:08.000000000","updated_by":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"reviewer":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"state":"REVIEWER"},{"updated":"2013-10-15 02:19:19.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-10-15 23:10:52.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2013-10-16 22:27:06.000000000","updated_by":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"reviewer":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"state":"REVIEWER"},{"updated":"2013-10-16 23:11:56.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5c3332c4ff31b51cedef06121b9fad91e687a9c2","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2013-10-07 14:16:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d0129eb18bcfc66c73425c73a54c300f4418190a","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-10-07 14:24:12.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"de4f5ab89271792deea5bd4952175caedd3aa19a","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2013-10-07 14:38:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a89c76ef10407f5e1b8845977237c948b3a47e6","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-10-07 14:40:00.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":3},{"id":"9bf197308e1fa9f1625147be4469f9324f401ddd","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2013-10-07 15:14:03.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"23be66a95bf114fec110c57cbe3fb41836a3ccae","author":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"date":"2013-10-07 15:17:18.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"a1d74d2de7a26a85cb531608c9bd06b51908ffe0","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-07 15:20:59.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- gate-keystone-pep8 http://logs.openstack.org/76/50076/3/check/gate-keystone-pep8/e42e040 : SUCCESS in 1m 44s\n- gate-keystone-docs http://docs-draft.openstack.org/76/50076/3/check/gate-keystone-docs/dcff3fe/doc/build/html/ : SUCCESS in 1m 33s\n- gate-keystone-python26 http://logs.openstack.org/76/50076/3/check/gate-keystone-python26/60739a9 : SUCCESS in 20m 01s\n- gate-keystone-python27 http://logs.openstack.org/76/50076/3/check/gate-keystone-python27/a1e9671 : SUCCESS in 14m 15s\n- check-tempest-devstack-vm-full http://logs.openstack.org/76/50076/3/check/check-tempest-devstack-vm-full/7984184 : SUCCESS in 36m 30s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/76/50076/3/check/check-tempest-devstack-vm-postgres-full/ea58f33 : SUCCESS in 39m 05s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/76/50076/3/check/check-tempest-devstack-vm-neutron/08ffd66 : SUCCESS in 19m 35s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/76/50076/3/check/gate-tempest-devstack-vm-large-ops/e8c0f43 : SUCCESS in 13m 42s\n- check-grenade-devstack-vm http://logs.openstack.org/76/50076/3/check/check-grenade-devstack-vm/d391b7e : SUCCESS in 25m 01s\n- check-swift-devstack-vm-functional http://logs.openstack.org/76/50076/3/check/check-swift-devstack-vm-functional/e946282 : SUCCESS in 14m 15s\n","accounts_in_message":[],"_revision_number":3},{"id":"003f04839bddce2dd07f8e030589d185f40fa2ce","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2013-10-08 12:24:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b570fc17eb9eb2d40a19ffcfb9a222530bb50140","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-10-08 12:26:08.000000000","message":"Patch Set 4: Looks good to me (core reviewer)\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":4},{"id":"f3b86e85e666d337ce9ea02db2b654e5dac7a970","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2013-10-08 12:26:08.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":4},{"id":"bef965216f2648d818c5583d310cd50e0beb9711","author":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"date":"2013-10-08 12:26:08.000000000","message":"Patch Set 4: Looks good to me, but someone else must approve\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":4},{"id":"a30fa5863938d1d80ecbdcadf1813f624b8731f9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-08 12:59:24.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- gate-keystone-pep8 http://logs.openstack.org/76/50076/4/check/gate-keystone-pep8/740de84 : SUCCESS in 1m 52s\n- gate-keystone-docs http://docs-draft.openstack.org/76/50076/4/check/gate-keystone-docs/f887967/doc/build/html/ : SUCCESS in 2m 08s\n- gate-keystone-python26 http://logs.openstack.org/76/50076/4/check/gate-keystone-python26/fa9bd7c : SUCCESS in 20m 18s\n- gate-keystone-python27 http://logs.openstack.org/76/50076/4/check/gate-keystone-python27/d9cbb0e : SUCCESS in 15m 30s\n- check-tempest-devstack-vm-full http://logs.openstack.org/76/50076/4/check/check-tempest-devstack-vm-full/a7cb185 : SUCCESS in 29m 49s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/76/50076/4/check/check-tempest-devstack-vm-postgres-full/5ea6fa6 : SUCCESS in 31m 17s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/76/50076/4/check/check-tempest-devstack-vm-neutron/889c469 : SUCCESS in 26m 37s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/76/50076/4/check/gate-tempest-devstack-vm-large-ops/cedd0ec : SUCCESS in 12m 56s\n- check-grenade-devstack-vm http://logs.openstack.org/76/50076/4/check/check-grenade-devstack-vm/146ab6b : SUCCESS in 26m 09s\n- check-swift-devstack-vm-functional http://logs.openstack.org/76/50076/4/check/check-swift-devstack-vm-functional/b5878cc : SUCCESS in 13m 43s\n","accounts_in_message":[],"_revision_number":4},{"id":"01f60e3d71101c623f0e5b22601a17e5e6bab512","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-10-15 02:19:19.000000000","message":"Patch Set 4: Works for me\n\nSmokeStack Results (patch set 4):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/145396 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/145397 : SUCCESS \n","accounts_in_message":[],"_revision_number":4},{"id":"867be890dda0cff7dd3c1f4a9f30d32fc5f554ba","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2013-10-15 23:10:52.000000000","message":"Patch Set 4: Looks good to me (core reviewer); Approved\n\n(2 inline comments)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"4f4435310800f5d4725ac066eeeff14f77273e9c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-15 23:10:54.000000000","message":"Patch Set 4: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"4d730259d8dd50b3c8a2e110031a2f27bad084b3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-16 00:21:16.000000000","message":"Patch Set 4: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-keystone-docs http://docs-draft.openstack.org/76/50076/4/gate/gate-keystone-docs/3461aa2/doc/build/html/ : SUCCESS in 1m 36s\n- gate-keystone-pep8 http://logs.openstack.org/76/50076/4/gate/gate-keystone-pep8/766e515 : SUCCESS in 1m 45s\n- gate-keystone-python26 http://logs.openstack.org/76/50076/4/gate/gate-keystone-python26/2e9162d : SUCCESS in 20m 58s\n- gate-keystone-python27 http://logs.openstack.org/76/50076/4/gate/gate-keystone-python27/7a02e27 : SUCCESS in 13m 45s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-full/78189ad : FAILURE in 45m 30s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-postgres-full/cd20b94 : SUCCESS in 46m 47s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-neutron/d1865f7 : SUCCESS in 17m 56s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-large-ops/42612de : SUCCESS in 14m 14s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-neutron-large-ops/8ef19b9 : SUCCESS in 16m 29s (non-voting)\n- gate-grenade-devstack-vm http://logs.openstack.org/76/50076/4/gate/gate-grenade-devstack-vm/baa7e37 : SUCCESS in 28m 12s\n- gate-swift-devstack-vm-functional http://logs.openstack.org/76/50076/4/gate/gate-swift-devstack-vm-functional/5322c2a : SUCCESS in 12m 57s\n","accounts_in_message":[],"_revision_number":4},{"id":"99d76fe67255f11936fa0cb3dbf8790fac5f40a8","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2013-10-16 00:21:29.000000000","message":"Patch Set 4:\n\nI noticed tempest failed, I think you hit bugs: https://bugs.launchpad.net/bugs/1218391","accounts_in_message":[],"_revision_number":4},{"id":"b82156f5bbfff03f57fc94a6e9b6d7ae3ccf4faa","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2013-10-16 01:12:03.000000000","message":"Patch Set 4:\n\nreverify bug 1218391","accounts_in_message":[],"_revision_number":4},{"id":"b079a53acdf8fe821b9177830041f2882ff6857a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-16 01:12:05.000000000","message":"Patch Set 4: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"cc33facebdb47bf7f12d73431facf880c1a0a7b7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-16 03:03:46.000000000","message":"Patch Set 4: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-keystone-docs http://docs-draft.openstack.org/76/50076/4/gate/gate-keystone-docs/0e014d0/doc/build/html/ : SUCCESS in 1m 50s\n- gate-keystone-pep8 http://logs.openstack.org/76/50076/4/gate/gate-keystone-pep8/0c568ca : SUCCESS in 1m 37s\n- gate-keystone-python26 http://logs.openstack.org/76/50076/4/gate/gate-keystone-python26/222512d : SUCCESS in 23m 31s\n- gate-keystone-python27 http://logs.openstack.org/76/50076/4/gate/gate-keystone-python27/f1d9561 : SUCCESS in 15m 25s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-full/fbe2dda : SUCCESS in 45m 17s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-postgres-full/a1acc69 : SUCCESS in 45m 31s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-neutron/63d675b : SUCCESS in 19m 29s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-large-ops/33a2ffd : SUCCESS in 15m 50s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-neutron-large-ops/5cbbfb6 : SUCCESS in 17m 41s (non-voting)\n- gate-grenade-devstack-vm http://logs.openstack.org/76/50076/4/gate/gate-grenade-devstack-vm/36cbeba : FAILURE in 26m 18s\n- gate-swift-devstack-vm-functional http://logs.openstack.org/76/50076/4/gate/gate-swift-devstack-vm-functional/a54c565 : SUCCESS in 13m 00s\n","accounts_in_message":[],"_revision_number":4},{"id":"307332888baca5fd92a1c9bf406472c0e1accca6","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-10-16 22:27:06.000000000","message":"Patch Set 4:\n\nreverify bug 1218391","accounts_in_message":[],"_revision_number":4},{"id":"715ed3970c16181f1f644d8c39d79dad33c8d760","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-16 22:27:08.000000000","message":"Patch Set 4: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"fd0bf70b37157c4d1543792a84645b0c950cd579","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-16 23:11:56.000000000","message":"Patch Set 4: Verified\n\nBuild succeeded.\n\n- gate-keystone-docs http://docs-draft.openstack.org/76/50076/4/gate/gate-keystone-docs/39f51a7/doc/build/html/ : SUCCESS in 2m 05s\n- gate-keystone-pep8 http://logs.openstack.org/76/50076/4/gate/gate-keystone-pep8/daa1fbd : SUCCESS in 1m 33s\n- gate-keystone-python26 http://logs.openstack.org/76/50076/4/gate/gate-keystone-python26/c30b38b : SUCCESS in 19m 26s\n- gate-keystone-python27 http://logs.openstack.org/76/50076/4/gate/gate-keystone-python27/33683b5 : SUCCESS in 11m 07s\n- gate-tempest-devstack-vm-full http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-full/7f4221e : SUCCESS in 42m 27s\n- gate-tempest-devstack-vm-postgres-full http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-postgres-full/b2ec6c7 : SUCCESS in 44m 38s\n- gate-tempest-devstack-vm-neutron http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-neutron/b4e1032 : SUCCESS in 21m 18s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-large-ops/ceba025 : SUCCESS in 16m 36s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/76/50076/4/gate/gate-tempest-devstack-vm-neutron-large-ops/122dd8e : SUCCESS in 16m 03s (non-voting)\n- gate-grenade-devstack-vm http://logs.openstack.org/76/50076/4/gate/gate-grenade-devstack-vm/103d2d0 : SUCCESS in 25m 22s\n- gate-swift-devstack-vm-functional http://logs.openstack.org/76/50076/4/gate/gate-swift-devstack-vm-functional/7bb1f08 : SUCCESS in 12m 01s\n","accounts_in_message":[],"_revision_number":4},{"id":"9d6f9f4b7013628c70c576571fb2dc09eeb71ede","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-16 23:11:57.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8c1899cc8f1cdfd0781388a5450595468ec7de08","revisions":{"c0c2169e5d8c73fad41b80578aa4eb883306df19":{"kind":"REWORK","_number":1,"created":"2013-10-07 13:51:51.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/76/50076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/76/50076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/76/50076/1"}}},"commit":{"parents":[{"commit":"6279c2a17e0d65c719e43377ebd75d281e835746","subject":"Use abstract base class for credential driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6279c2a17e0d65c719e43377ebd75d281e835746"}]}],"author":{"name":"Doug Hellmann","email":"doug.hellmann@dreamhost.com","date":"2013-10-06 16:46:44.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug.hellmann@dreamhost.com","date":"2013-10-07 13:42:31.000000000","tz":-240},"subject":"Use abstract base class for catalog driver","message":"Use abstract base class for catalog driver\n\nUse the abc module for managing abstract base classes in the driver\nlayer and add missing methods to concrete classes.\n\nbp abstract-base-class-drivers\n\nChange-Id: If66b9bd3ebcfc3a59151b52a46407a36d2a9352b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c0c2169e5d8c73fad41b80578aa4eb883306df19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c0c2169e5d8c73fad41b80578aa4eb883306df19"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6279c2a17e0d65c719e43377ebd75d281e835746","is_merged_in_target_branch":false,"change_id":"Iae65d406bdb65f5ae0fc08f0c1d4b48890eca889","change_number":50075,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d57202690e1fa74591c18afc21145e9450aeccf8":{"kind":"REWORK","_number":2,"created":"2013-10-07 14:16:59.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/76/50076/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/76/50076/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/76/50076/2"}}},"commit":{"parents":[{"commit":"e4925b0d6828b246eb4eb952626f8bf83cef2a3d","subject":"Use abstract base class for credential driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e4925b0d6828b246eb4eb952626f8bf83cef2a3d"}]}],"author":{"name":"Doug Hellmann","email":"doug.hellmann@dreamhost.com","date":"2013-10-06 16:46:44.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug.hellmann@dreamhost.com","date":"2013-10-07 14:09:22.000000000","tz":-240},"subject":"Use abstract base class for catalog driver","message":"Use abstract base class for catalog driver\n\nUse the abc module for managing abstract base classes in the driver\nlayer and add missing methods to concrete classes.\n\nbp abstract-base-class-drivers\n\nChange-Id: If66b9bd3ebcfc3a59151b52a46407a36d2a9352b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d57202690e1fa74591c18afc21145e9450aeccf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d57202690e1fa74591c18afc21145e9450aeccf8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e4925b0d6828b246eb4eb952626f8bf83cef2a3d","is_merged_in_target_branch":false,"change_id":"Iae65d406bdb65f5ae0fc08f0c1d4b48890eca889","change_number":50075,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"f1334e27c4ceee0ced765a724dd9a6855a0b8db0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2013-10-07 14:38:34.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/76/50076/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/76/50076/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/76/50076/3"}}},"commit":{"parents":[{"commit":"0308f1bca66309d1937e495e33089216870af9fe","subject":"Use abstract base class for credential driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0308f1bca66309d1937e495e33089216870af9fe"}]}],"author":{"name":"Doug Hellmann","email":"doug.hellmann@dreamhost.com","date":"2013-10-06 16:46:44.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug.hellmann@dreamhost.com","date":"2013-10-07 14:36:52.000000000","tz":-240},"subject":"Use abstract base class for catalog driver","message":"Use abstract base class for catalog driver\n\nUse the abc module for managing abstract base classes in the driver\nlayer and add missing methods to concrete classes.\n\nbp abstract-base-class-drivers\n\nChange-Id: If66b9bd3ebcfc3a59151b52a46407a36d2a9352b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f1334e27c4ceee0ced765a724dd9a6855a0b8db0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f1334e27c4ceee0ced765a724dd9a6855a0b8db0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"0308f1bca66309d1937e495e33089216870af9fe","is_merged_in_target_branch":false,"change_id":"Iae65d406bdb65f5ae0fc08f0c1d4b48890eca889","change_number":50075,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"8c1899cc8f1cdfd0781388a5450595468ec7de08":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2013-10-08 12:24:42.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/76/50076/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/76/50076/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/50076/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/76/50076/4"}}},"commit":{"parents":[{"commit":"c0f6653a13be16707ade16e0f4e6d076aaf7f373","subject":"Use abstract base class for credential driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c0f6653a13be16707ade16e0f4e6d076aaf7f373"}]}],"author":{"name":"Doug Hellmann","email":"doug.hellmann@dreamhost.com","date":"2013-10-06 16:46:44.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug.hellmann@dreamhost.com","date":"2013-10-08 12:24:10.000000000","tz":-240},"subject":"Use abstract base class for catalog driver","message":"Use abstract base class for catalog driver\n\nUse the abc module for managing abstract base classes in the driver\nlayer and add missing methods to concrete classes.\n\nbp abstract-base-class-drivers\n\nChange-Id: If66b9bd3ebcfc3a59151b52a46407a36d2a9352b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8c1899cc8f1cdfd0781388a5450595468ec7de08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8c1899cc8f1cdfd0781388a5450595468ec7de08"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c0f6653a13be16707ade16e0f4e6d076aaf7f373","is_merged_in_target_branch":true,"change_id":"Iae65d406bdb65f5ae0fc08f0c1d4b48890eca889","change_number":50075,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
