)]}'
{"id":"openstack%2Fkeystone~53486","triplet_id":"openstack%2Fkeystone~master~I8f274cdd7f498fb5a9f407ad76e1195346acca26","project":"openstack/keystone","branch":"master","topic":"(detached","hashtags":[],"change_id":"I8f274cdd7f498fb5a9f407ad76e1195346acca26","subject":"Generate autodocs again","status":"ABANDONED","created":"2013-10-24 01:49:24.000000000","updated":"2013-10-24 19:42:17.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d35805a6c3207a0e22db2d58922b7e5d9f7efbce","_number":53486,"virtual_id_number":53486,"owner":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2013-10-24 11:45:54.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2013-10-24 11:23:33.000000000","_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"disliked":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2013-10-24 10:54:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"value":1,"date":"2013-10-24 15:52:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"value":-1,"date":"2013-10-24 17:56:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2013-10-24 10:54:02.000000000","_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2013-10-24 10:54:02.000000000","updated_by":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"reviewer":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"state":"REVIEWER"},{"updated":"2013-10-24 11:23:33.000000000","updated_by":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"reviewer":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2013-10-24 11:45:54.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2013-10-24 15:52:56.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2013-10-24 17:56:07.000000000","updated_by":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"reviewer":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"state":"REVIEWER"}],"messages":[{"id":"907c3fb469fb3fd55a50f1573cc27c0ffcb2b9b4","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-10-24 02:03:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b752b56dcaf323209b032da9fa1826ce754d2a5f","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-10-24 02:29:05.000000000","message":"Patch Set 2: Works for me\n\nSmokeStack Results (patch set 2):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/149672 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/149673 : SUCCESS \n","accounts_in_message":[],"_revision_number":2},{"id":"fa7310d36bbc0e3517289d0af692d7aa079b0bb7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-24 02:35:31.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- gate-keystone-pep8 http://logs.openstack.org/86/53486/2/check/gate-keystone-pep8/25eb66d : SUCCESS in 1m 37s\n- gate-keystone-docs http://docs-draft.openstack.org/86/53486/2/check/gate-keystone-docs/64270ff/doc/build/html/ : SUCCESS in 3m 29s\n- gate-keystone-python26 http://logs.openstack.org/86/53486/2/check/gate-keystone-python26/5bfbdcc : SUCCESS in 15m 39s\n- gate-keystone-python27 http://logs.openstack.org/86/53486/2/check/gate-keystone-python27/fc02d35 : SUCCESS in 15m 53s\n- check-tempest-devstack-vm-full http://logs.openstack.org/86/53486/2/check/check-tempest-devstack-vm-full/d73a624 : SUCCESS in 29m 26s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/86/53486/2/check/check-tempest-devstack-vm-postgres-full/a750a3a : SUCCESS in 31m 55s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/86/53486/2/check/check-tempest-devstack-vm-neutron/caa495e : SUCCESS in 25m 20s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/86/53486/2/check/gate-tempest-devstack-vm-large-ops/c510ff8 : SUCCESS in 14m 17s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/86/53486/2/check/gate-tempest-devstack-vm-neutron-large-ops/ff9347b : SUCCESS in 16m 18s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/86/53486/2/check/check-grenade-devstack-vm/e1fb340 : SUCCESS in 24m 56s\n- check-swift-devstack-vm-functional http://logs.openstack.org/86/53486/2/check/check-swift-devstack-vm-functional/3bce960 : SUCCESS in 13m 56s\n","accounts_in_message":[],"_revision_number":2},{"id":"7afd6484b7923f1de3b0729e16a6080afd468840","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-10-24 10:54:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9d22d88444b4635f78685bfa26998e30d3d439a5","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-10-24 10:54:10.000000000","message":"Patch Set 2: (4 inline comments)\n\n","accounts_in_message":[],"_revision_number":2},{"id":"e23e6ef5d315c4422310249b3d6e62431234114b","author":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"date":"2013-10-24 10:54:25.000000000","message":"Patch Set 3: Looks good to me (core reviewer)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"11858ea3c2215790279e2b0f391db23a71de3482","author":{"_account_id":2166,"name":"SmokeStack CI","username":"smokestack","tags":["SERVICE_USER"]},"date":"2013-10-24 11:23:33.000000000","message":"Patch Set 3: Works for me\n\nSmokeStack Results (patch set 3):\n- Libvirt (Fedora 19) https://smokestack.openstack.org/?go\u003d/jobs/149881 : SUCCESS \n- Libvirt (Centos 6.4) https://smokestack.openstack.org/?go\u003d/jobs/149882 : SUCCESS \n","accounts_in_message":[],"_revision_number":3},{"id":"a429afc2003f6a156dd3e134aabb45905bc31911","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2013-10-24 11:45:54.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- gate-keystone-pep8 http://logs.openstack.org/86/53486/3/check/gate-keystone-pep8/a3be73b : SUCCESS in 2m 52s\n- gate-keystone-docs http://docs-draft.openstack.org/86/53486/3/check/gate-keystone-docs/a044fdc/doc/build/html/ : SUCCESS in 3m 26s\n- gate-keystone-python26 http://logs.openstack.org/86/53486/3/check/gate-keystone-python26/009a298 : SUCCESS in 16m 01s\n- gate-keystone-python27 http://logs.openstack.org/86/53486/3/check/gate-keystone-python27/0d1a395 : SUCCESS in 15m 39s\n- check-tempest-devstack-vm-full http://logs.openstack.org/86/53486/3/check/check-tempest-devstack-vm-full/e074e84 : SUCCESS in 29m 26s\n- check-tempest-devstack-vm-postgres-full http://logs.openstack.org/86/53486/3/check/check-tempest-devstack-vm-postgres-full/d7a7af3 : SUCCESS in 30m 15s\n- check-tempest-devstack-vm-neutron http://logs.openstack.org/86/53486/3/check/check-tempest-devstack-vm-neutron/58250c8 : SUCCESS in 24m 42s\n- gate-tempest-devstack-vm-large-ops http://logs.openstack.org/86/53486/3/check/gate-tempest-devstack-vm-large-ops/a04d233 : SUCCESS in 16m 13s\n- gate-tempest-devstack-vm-neutron-large-ops http://logs.openstack.org/86/53486/3/check/gate-tempest-devstack-vm-neutron-large-ops/0db813b : SUCCESS in 16m 53s (non-voting)\n- check-grenade-devstack-vm http://logs.openstack.org/86/53486/3/check/check-grenade-devstack-vm/fe25f0a : SUCCESS in 51m 45s\n- check-swift-devstack-vm-functional http://logs.openstack.org/86/53486/3/check/check-swift-devstack-vm-functional/154fe5d : SUCCESS in 20m 59s\n","accounts_in_message":[],"_revision_number":3},{"id":"57584b263d28380547e40ec0ae29827006738c35","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2013-10-24 15:52:56.000000000","message":"Patch Set 3: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":3},{"id":"818399df9c09762f64aea1db05e079acf122693f","author":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"date":"2013-10-24 17:56:07.000000000","message":"Patch Set 3: I would prefer that you didn\u0027t merge this\n\nThe docs generated from this are really ugly.  I have another review where I\u0027m trying to fix up the docs: https://review.openstack.org/#/c/52929/","accounts_in_message":[],"_revision_number":3},{"id":"2731dc56811c3fe1ca4732c08fe62d2ffbed3836","author":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"date":"2013-10-24 19:42:17.000000000","message":"Patch Set 3: Abandoned\n\nLooks like https://review.openstack.org/#/c/52929/ will fix this problem, too.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e013dfc2c5a609cf1195d88a126739e4a5e7594e","revisions":{"345ebcd273f5524167368f65208243a9c80304ed":{"kind":"REWORK","_number":1,"created":"2013-10-24 01:49:24.000000000","uploader":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"ref":"refs/changes/86/53486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/86/53486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/53486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/53486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/53486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/86/53486/1"}}},"commit":{"parents":[{"commit":"9448cd81d0105043ca19f45a49d0e2de5eab9118","subject":"Merge \"Adds decorator to deprecate functions and methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9448cd81d0105043ca19f45a49d0e2de5eab9118"}]}],"author":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-10-24 01:47:24.000000000","tz":-300},"committer":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-10-24 01:48:38.000000000","tz":-300},"subject":"Generate autodocs again","message":"Generate autodocs again\n\nThe generated documentation here:\nhttp://docs.openstack.org/developer/keystone/#code-documentation\n\nDidn\u0027t include the generated documentation. The generated docs\nused to be in the \"Code Documentation\" section.\n\nYou can generate the docs locally with:\n  tox -evenv python setup.py build_sphinx\n\nChange-Id: I8f274cdd7f498fb5a9f407ad76e1195346acca26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/345ebcd273f5524167368f65208243a9c80304ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/345ebcd273f5524167368f65208243a9c80304ed"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9448cd81d0105043ca19f45a49d0e2de5eab9118","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"1d54f6c948602180840ce387bbba8536204d3959":{"kind":"REWORK","_number":2,"created":"2013-10-24 02:03:30.000000000","uploader":{"_account_id":6486,"name":"Brant Knudson","email":"blk@acm.org","username":"blk-u"},"ref":"refs/changes/86/53486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/86/53486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/53486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/53486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/53486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/86/53486/2"}}},"commit":{"parents":[{"commit":"9448cd81d0105043ca19f45a49d0e2de5eab9118","subject":"Merge \"Adds decorator to deprecate functions and methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9448cd81d0105043ca19f45a49d0e2de5eab9118"}]}],"author":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-10-24 01:47:24.000000000","tz":-300},"committer":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-10-24 02:03:12.000000000","tz":-300},"subject":"Generate autodocs again","message":"Generate autodocs again\n\nThe generated documentation here:\nhttp://docs.openstack.org/developer/keystone/#code-documentation\n\nDidn\u0027t include the generated documentation. The generated docs\nused to be in the \"Code Documentation\" section.\n\nYou can generate the docs locally with:\n  tox -evenv python setup.py build_sphinx\n\nChange-Id: I8f274cdd7f498fb5a9f407ad76e1195346acca26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d54f6c948602180840ce387bbba8536204d3959"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d54f6c948602180840ce387bbba8536204d3959"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9448cd81d0105043ca19f45a49d0e2de5eab9118","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e013dfc2c5a609cf1195d88a126739e4a5e7594e":{"kind":"REWORK","_number":3,"created":"2013-10-24 10:54:02.000000000","uploader":{"_account_id":4,"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","username":"dolph"},"ref":"refs/changes/86/53486/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/86/53486/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/53486/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/53486/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/86/53486/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/86/53486/3"}}},"commit":{"parents":[{"commit":"9448cd81d0105043ca19f45a49d0e2de5eab9118","subject":"Merge \"Adds decorator to deprecate functions and methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9448cd81d0105043ca19f45a49d0e2de5eab9118"}]}],"author":{"name":"Brant Knudson","email":"bknudson@us.ibm.com","date":"2013-10-24 01:47:24.000000000","tz":-300},"committer":{"name":"Dolph Mathews","email":"dolph.mathews@gmail.com","date":"2013-10-24 10:53:56.000000000","tz":-300},"subject":"Generate autodocs again","message":"Generate autodocs again\n\nThe generated documentation here:\nhttp://docs.openstack.org/developer/keystone/#code-documentation\n\nDidn\u0027t include the generated documentation. The generated docs\nused to be in the \"Code Documentation\" section.\n\nYou can generate the docs locally with:\n  tox -evenv python setup.py build_sphinx\n\nChange-Id: I8f274cdd7f498fb5a9f407ad76e1195346acca26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e013dfc2c5a609cf1195d88a126739e4a5e7594e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e013dfc2c5a609cf1195d88a126739e4a5e7594e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9448cd81d0105043ca19f45a49d0e2de5eab9118","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
