)]}'
{"id":"openstack%2Fkeystone~585417","triplet_id":"openstack%2Fkeystone~master~If2872ef64eaae262c1a9ce2eb3f773ac2e922f4e","project":"openstack/keystone","branch":"master","topic":"bp/mutable-config","hashtags":[],"change_id":"If2872ef64eaae262c1a9ce2eb3f773ac2e922f4e","subject":"Enable mutable config","status":"ABANDONED","created":"2018-07-24 16:07:38.000000000","updated":"2022-01-21 20:57:44.000000000","total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3f585faf090c6f71bccd6ab23a614467305aff39","_number":585417,"virtual_id_number":585417,"owner":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2018-07-25 21:05:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2018-07-26 03:11:29.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":-1,"date":"2018-10-11 10:48:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"date":"2018-10-16 18:42:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-25 21:05:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-07-26 03:11:29.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2018-10-11 10:48:55.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2018-10-16 18:42:18.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"}],"messages":[{"id":"ff4ed8da449b546db9de9029a42cc0aaa025199a","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2018-07-24 16:07:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b714d1e6a7812510bc86e123549c744cf59a6aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-24 19:59:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/17/585417/1/check/openstack-tox-pep8/4a22e1e/ : SUCCESS in 6m 38s\n- openstack-tox-py27 http://logs.openstack.org/17/585417/1/check/openstack-tox-py27/804cc38/ : SUCCESS in 12m 33s\n- openstack-tox-py35 http://logs.openstack.org/17/585417/1/check/openstack-tox-py35/01ed11f/ : SUCCESS in 12m 52s\n- build-openstack-sphinx-docs http://logs.openstack.org/17/585417/1/check/build-openstack-sphinx-docs/16fa02d/html/ : SUCCESS in 8m 03s\n- tempest-full http://logs.openstack.org/17/585417/1/check/tempest-full/5292ffe/ : SUCCESS in 1h 28m 00s\n- neutron-grenade http://logs.openstack.org/17/585417/1/check/neutron-grenade/6d8951b/ : SUCCESS in 53m 32s\n- tempest-full-py3 http://logs.openstack.org/17/585417/1/check/tempest-full-py3/6d31540/ : SUCCESS in 1h 40m 17s\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/17/585417/1/check/legacy-keystoneclient-dsvm-functional/d1e472a/ : SUCCESS in 30m 31s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/17/585417/1/check/openstack-tox-cover/356ee2f/ : TIMED_OUT in 51m 28s\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/17/585417/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/c73f3b6/ : SUCCESS in 1h 47m 59s (non-voting)\n- keystone-dsvm-functional http://logs.openstack.org/17/585417/1/check/keystone-dsvm-functional/9739453/ : SUCCESS in 27m 49s\n- keystone-dsvm-functional-v3-only http://logs.openstack.org/17/585417/1/check/keystone-dsvm-functional-v3-only/1f3b7d9/ : SUCCESS in 28m 08s (non-voting)\n- keystone-dsvm-py35-functional-v3-only http://logs.openstack.org/17/585417/1/check/keystone-dsvm-py35-functional-v3-only/37e9bc8/ : SUCCESS in 27m 36s (non-voting)\n- openstack-ansible-keystone-rolling-upgrade http://logs.openstack.org/17/585417/1/check/openstack-ansible-keystone-rolling-upgrade/038dbb5/ : FAILURE in 16m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/17/585417/1/check/openstack-tox-lower-constraints/9bb0402/ : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":1},{"id":"fa972dc3f07bd0bfdd9ccf4345439c062369dd56","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2018-07-25 18:30:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fbece1cffe906f8880a9ec1c53c6f99b9c99d594","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-25 21:05:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/17/585417/2/check/openstack-tox-pep8/9d19dd0/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/17/585417/2/check/openstack-tox-py27/21e3581/ : SUCCESS in 13m 41s\n- openstack-tox-py35 http://logs.openstack.org/17/585417/2/check/openstack-tox-py35/337665d/ : SUCCESS in 14m 26s\n- build-openstack-sphinx-docs http://logs.openstack.org/17/585417/2/check/build-openstack-sphinx-docs/561360a/html/ : SUCCESS in 8m 15s\n- tempest-full http://logs.openstack.org/17/585417/2/check/tempest-full/ec48dd8/ : SUCCESS in 1h 35m 44s\n- neutron-grenade http://logs.openstack.org/17/585417/2/check/neutron-grenade/94fcf72/ : SUCCESS in 48m 03s\n- tempest-full-py3 http://logs.openstack.org/17/585417/2/check/tempest-full-py3/2869595/ : SUCCESS in 1h 26m 44s\n- build-openstack-releasenotes http://logs.openstack.org/17/585417/2/check/build-openstack-releasenotes/cab6c73/html/ : SUCCESS in 10m 55s\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/17/585417/2/check/legacy-keystoneclient-dsvm-functional/4c8f954/ : SUCCESS in 32m 45s (non-voting)\n- openstack-tox-cover http://logs.openstack.org/17/585417/2/check/openstack-tox-cover/40e0ddb/cover/ : SUCCESS in 16m 16s\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/17/585417/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/e41611a/ : SUCCESS in 1h 33m 23s (non-voting)\n- keystone-dsvm-functional http://logs.openstack.org/17/585417/2/check/keystone-dsvm-functional/f0ca7c1/ : SUCCESS in 34m 43s\n- keystone-dsvm-functional-v3-only http://logs.openstack.org/17/585417/2/check/keystone-dsvm-functional-v3-only/94cdd9a/ : SUCCESS in 27m 25s (non-voting)\n- keystone-dsvm-py35-functional-v3-only http://logs.openstack.org/17/585417/2/check/keystone-dsvm-py35-functional-v3-only/d4306f1/ : SUCCESS in 35m 36s (non-voting)\n- openstack-ansible-keystone-rolling-upgrade http://logs.openstack.org/17/585417/2/check/openstack-ansible-keystone-rolling-upgrade/968c4e9/ : FAILURE in 17m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/17/585417/2/check/openstack-tox-lower-constraints/e2d9ad0/ : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":2},{"id":"9b86b053511a7357a5a04cf43d9f6ab916494a0a","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-07-26 03:11:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de5ffcb71cce187776fc83fba1adb6733a3eaf88","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-07-27 15:07:39.000000000","message":"Patch Set 2: Code-Review-2\n\nThe -2 here is strictly procedural since it is recommended that we wait until Stein. The deadline for implementing community goals was rocky-3, which we barely missed.\n\nUnless we can deem this approach extremely low-risk, then we might have to wait until Stein opens for development.","accounts_in_message":[],"_revision_number":2},{"id":"98351c46152b65d1080fd1adcb9d2712207f7466","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-07-27 15:20:02.000000000","message":"Patch Set 2:\n\n\u003e The -2 here is strictly procedural since it is recommended that we\n \u003e wait until Stein. The deadline for implementing community goals was\n \u003e rocky-3, which we barely missed.\n \u003e \n \u003e Unless we can deem this approach extremely low-risk, then we might\n \u003e have to wait until Stein opens for development.\n\nTC related discussion - http://eavesdrop.openstack.org/irclogs/%23openstack-tc/%23openstack-tc.2018-07-27.log.html#t2018-07-27T15:03:49","accounts_in_message":[],"_revision_number":2},{"id":"02b4938e69d5a817f63906ac66c425194c3e68d6","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-08-27 16:30:38.000000000","message":"Patch Set 2: -Code-Review\n\nUnblocking this since we\u0027re technically open for Stein development.","accounts_in_message":[],"_revision_number":2},{"id":"6f32d242c61672abf8b470648a7ec460a035a1fa","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-08-27 16:38:27.000000000","message":"Patch Set 2: Code-Review+1\n\n(4 comments)\n\nRelatively minor comments, otherwise this looks good.","accounts_in_message":[],"_revision_number":2},{"id":"e38194bea8f37ca32451478fdd6621d6b3560db7","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2018-10-11 10:48:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nNot a strong -1 but marking -1 to make sure we get answers to my, wxy\u0027s and lbragstad\u0027s questions","accounts_in_message":[],"_revision_number":2},{"id":"0455fd9b4fcc088ad639350c60d9de77d687699d","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-10-16 16:07:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0bf1e9013b2cb39380af8d5da72e70f0ef7c609e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-10-16 18:42:18.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"f2f719442052542bd4633d34202cf72ff08ac14f","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2020-01-28 17:18:05.000000000","message":"Removed reviewer Morgan Fainberg with the following votes:\n\n* Code-Review-1 by Morgan Fainberg \u003cmorgan.fainberg@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"3f585faf090c6f71bccd6ab23a614467305aff39","tag":"autogenerated:gerrit:abandon","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-21 20:57:44.000000000","message":"Abandoned\n\nAbandoning since there hasn\u0027t been any recent activity, if anyone wants to continue this work, please feel free to restore this or create a new change.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"168dc144741ec3635a8d117a7355789ca2ac3a79","revisions":{"851d44f7d39adcc546055027f109702000bbe31e":{"kind":"REWORK","_number":1,"created":"2018-07-24 16:07:38.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/17/585417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/585417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/585417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/585417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/585417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/585417/1"}}},"commit":{"parents":[{"commit":"05c6a6584650f312a49ebe1b2917d780b943c380","subject":"Merge \"Add pycadf initiator for flask resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/05c6a6584650f312a49ebe1b2917d780b943c380"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2018-07-24 16:05:34.000000000","tz":-240},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2018-07-24 16:07:19.000000000","tz":-240},"subject":"WIP - Enable mutable config","message":"WIP - Enable mutable config\n\nblueprint mutable-config\n\nChange-Id: If2872ef64eaae262c1a9ce2eb3f773ac2e922f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/851d44f7d39adcc546055027f109702000bbe31e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/851d44f7d39adcc546055027f109702000bbe31e"}]},"branch":"refs/heads/master"},"168dc144741ec3635a8d117a7355789ca2ac3a79":{"kind":"REWORK","_number":2,"created":"2018-07-25 18:30:23.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/17/585417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/17/585417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/585417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/585417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/17/585417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/17/585417/2"}}},"commit":{"parents":[{"commit":"05c6a6584650f312a49ebe1b2917d780b943c380","subject":"Merge \"Add pycadf initiator for flask resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/05c6a6584650f312a49ebe1b2917d780b943c380"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2018-07-24 16:05:34.000000000","tz":-240},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2018-07-25 18:27:28.000000000","tz":-240},"subject":"Enable mutable config","message":"Enable mutable config\n\nThis enables support for reloading config files without restarting\nthe service. This is accomplished by checking the mtime of\nconfig files every time an API request hits the keystone server.\n\nA global before_request function with the flask framework,\ntherefore only APIs that have migrated to flask will trigger it\nto fire off.\n\nblueprint mutable-config\n\nChange-Id: If2872ef64eaae262c1a9ce2eb3f773ac2e922f4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/168dc144741ec3635a8d117a7355789ca2ac3a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/168dc144741ec3635a8d117a7355789ca2ac3a79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
