)]}'
{"id":"openstack%2Fkeystone~605235","triplet_id":"openstack%2Fkeystone~master~I100bca162e71a9d394ed5787b976b13b1e57987f","project":"openstack/keystone","branch":"master","topic":"explicit-domain-id","hashtags":[],"change_id":"I100bca162e71a9d394ed5787b976b13b1e57987f","subject":"Allow an explicit_domain_id parameter when creating a domain","status":"MERGED","created":"2018-09-25 21:15:31.000000000","updated":"2019-04-09 22:36:45.000000000","submitted":"2019-04-09 22:36:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"605235-1554849405934-bedb62c6","meta_rev_id":"ea5bf6828cc8996928a209c11dd70ff478a9918a","_number":605235,"virtual_id_number":605235,"owner":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":2,"date":"2019-04-09 22:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},{"value":2,"date":"2019-04-09 16:32:48.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-04-09 20:17:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":2,"date":"2019-04-09 20:44:00.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"all":[{"value":0,"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"value":0,"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"value":1,"date":"2019-04-09 20:44:00.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-01 03:39:16.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2018-10-01 03:39:16.000000000","updated_by":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"reviewer":{"_account_id":7725,"name":"David Stanek","email":"dstanek@dstanek.com","username":"dstanek"},"state":"REVIEWER"},{"updated":"2018-10-01 03:39:16.000000000","updated_by":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"reviewer":{"_account_id":13063,"name":"WEI CHEN","email":"dave.jungler@gmail.com","username":"wei.d.chen"},"state":"REVIEWER"},{"updated":"2018-10-01 03:39:16.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2018-10-01 03:39:16.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2018-10-02 14:57:48.000000000","updated_by":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"reviewer":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"state":"REVIEWER"},{"updated":"2018-10-03 04:10:50.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2018-11-06 02:26:25.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2018-11-06 06:36:40.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-02-01 14:42:13.000000000","updated_by":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"reviewer":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"state":"REVIEWER"},{"updated":"2019-04-09 16:32:48.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-04-09 20:17:45.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"state":"REVIEWER"},{"updated":"2019-04-09 20:44:00.000000000","updated_by":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"reviewer":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"state":"REVIEWER"},{"updated":"2019-04-09 22:36:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62aa6bc566c940a7871af0135f1cd64eb005f6a5","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2018-09-25 21:15:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70f7ba46b3cdefb5a7e05a90d89364d983c89668","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-26 04:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/1/check/openstack-tox-cover/b0482d8/cover/ : SUCCESS in 15m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/1/check/openstack-tox-lower-constraints/4ec7a68/ : SUCCESS in 14m 34s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/1/check/openstack-tox-pep8/003274b/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/1/check/openstack-tox-py27/70c6217/ : SUCCESS in 10m 02s\n- openstack-tox-py35 http://logs.openstack.org/35/605235/1/check/openstack-tox-py35/51b7ac0/ : SUCCESS in 12m 29s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/1/check/openstack-tox-py36/2c69ecb/ : SUCCESS in 11m 18s\n- openstack-tox-docs http://logs.openstack.org/35/605235/1/check/openstack-tox-docs/a7edd69/html/ : SUCCESS in 5m 38s\n- tempest-full http://logs.openstack.org/35/605235/1/check/tempest-full/47a3d10/ : SUCCESS in 1h 32m 58s\n- neutron-grenade http://logs.openstack.org/35/605235/1/check/neutron-grenade/bea686f/ : SUCCESS in 55m 00s\n- tempest-full-py3 http://logs.openstack.org/35/605235/1/check/tempest-full-py3/a5060dd/ : SUCCESS in 1h 34m 35s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/1/check/keystone-dsvm-functional/2cd8195/ : SUCCESS in 27m 05s\n- keystone-dsvm-functional-federation http://logs.openstack.org/35/605235/1/check/keystone-dsvm-functional-federation/2f9afcc/ : SUCCESS in 36m 22s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/35/605235/1/check/keystone-dsvm-py35-functional-federation/1f641f9/ : FAILURE in 26m 31s (non-voting)\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/35/605235/1/check/legacy-keystoneclient-dsvm-functional/2426155/ : SUCCESS in 28m 28s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/59b0f05/ : SUCCESS in 1h 20m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88083915096036a01a4c61d58238c47a8a949bb1","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-09-26 06:23:58.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for the patch. I think a release note is also required for this change.","accounts_in_message":[],"_revision_number":1},{"id":"ca3015368dfbf3550e672b042ce4e7ab6db42064","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2018-09-26 14:15:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c61c3bc42fb1f216ccd1527da8cd21a39a3cacba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-26 22:51:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/2/check/openstack-tox-cover/4e14bf8/cover/ : SUCCESS in 11m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/2/check/openstack-tox-lower-constraints/fd4d309/ : SUCCESS in 24m 25s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/2/check/openstack-tox-pep8/2b66b4f/ : SUCCESS in 6m 42s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/2/check/openstack-tox-py27/beca610/ : SUCCESS in 13m 19s\n- openstack-tox-py35 http://logs.openstack.org/35/605235/2/check/openstack-tox-py35/022e2c9/ : SUCCESS in 11m 54s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/2/check/openstack-tox-py36/a74487d/ : SUCCESS in 10m 41s\n- openstack-tox-docs http://logs.openstack.org/35/605235/2/check/openstack-tox-docs/e3b34dc/html/ : SUCCESS in 5m 34s\n- tempest-full http://logs.openstack.org/35/605235/2/check/tempest-full/46ac655/ : SUCCESS in 1h 36m 30s\n- neutron-grenade http://logs.openstack.org/35/605235/2/check/neutron-grenade/5f489d9/ : SUCCESS in 58m 04s\n- tempest-full-py3 http://logs.openstack.org/35/605235/2/check/tempest-full-py3/0b0e1c0/ : SUCCESS in 1h 27m 42s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/2/check/build-openstack-releasenotes/33e6f5e/html/ : SUCCESS in 11m 28s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/2/check/keystone-dsvm-functional/2fd8854/ : SUCCESS in 30m 53s\n- keystone-dsvm-functional-federation http://logs.openstack.org/35/605235/2/check/keystone-dsvm-functional-federation/da4fc71/ : SUCCESS in 30m 35s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/35/605235/2/check/keystone-dsvm-py35-functional-federation/e6443b6/ : FAILURE in 33m 27s (non-voting)\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/35/605235/2/check/legacy-keystoneclient-dsvm-functional/1f67237/ : SUCCESS in 26m 53s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/dee0e1b/ : SUCCESS in 1h 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d191cc1622beb5a22479a04fae04d6438376fc6e","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-09-27 08:50:22.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"66cd2cca0fe046103a6e30d77c121765ff99cbf0","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2018-10-01 13:38:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fba464a78537d16cca475c6086eb9965698d9696","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-01 17:32:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/3/check/openstack-tox-cover/d531887/cover/ : SUCCESS in 10m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/3/check/openstack-tox-lower-constraints/aa405a3/ : SUCCESS in 14m 31s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/3/check/openstack-tox-pep8/7555e0d/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/3/check/openstack-tox-py27/efc1f83/ : SUCCESS in 10m 08s\n- openstack-tox-py35 http://logs.openstack.org/35/605235/3/check/openstack-tox-py35/493b099/ : SUCCESS in 9m 17s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/3/check/openstack-tox-py36/6574f44/ : SUCCESS in 9m 15s\n- openstack-tox-docs http://logs.openstack.org/35/605235/3/check/openstack-tox-docs/d2c2568/html/ : SUCCESS in 5m 59s\n- tempest-full http://logs.openstack.org/35/605235/3/check/tempest-full/0cd2af9/ : SUCCESS in 1h 30m 19s\n- neutron-grenade http://logs.openstack.org/35/605235/3/check/neutron-grenade/eb11380/ : SUCCESS in 46m 44s\n- tempest-full-py3 http://logs.openstack.org/35/605235/3/check/tempest-full-py3/0d0906a/ : SUCCESS in 1h 19m 25s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/3/check/build-openstack-releasenotes/0537b48/html/ : SUCCESS in 8m 21s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/3/check/keystone-dsvm-functional/0ef43a5/ : SUCCESS in 25m 08s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/3/check/keystone-dsvm-py3-functional/b188630/ : SUCCESS in 28m 55s\n- keystone-dsvm-functional-federation http://logs.openstack.org/35/605235/3/check/keystone-dsvm-functional-federation/1ce7692/ : SUCCESS in 27m 09s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/35/605235/3/check/keystone-dsvm-py35-functional-federation/b41ac14/ : SUCCESS in 26m 06s (non-voting)\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/35/605235/3/check/legacy-keystoneclient-dsvm-functional/20a600c/ : SUCCESS in 27m 26s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/3/check/legacy-tempest-dsvm-ldap-domain-specific-driver/e2ff322/ : SUCCESS in 1h 55m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"980bc9bb007e4a7786b7bebfebe37c8c9cf3818d","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2018-10-02 14:57:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d8dbdb75e6648ccea30aa754327d339491e8bf17","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-10-02 17:27:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3fe9d99046f2be520e6a16b5d365361354eb2595","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2018-10-02 21:28:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dced4f63e3013640dde4be90303c6c5c6e6225ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-03 02:16:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/4/check/openstack-tox-cover/b91664f/cover/ : SUCCESS in 11m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/4/check/openstack-tox-lower-constraints/f2f376c/ : SUCCESS in 16m 09s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/4/check/openstack-tox-pep8/5c5a471/ : SUCCESS in 7m 26s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/4/check/openstack-tox-py27/c2887a5/ : SUCCESS in 12m 05s\n- openstack-tox-py35 http://logs.openstack.org/35/605235/4/check/openstack-tox-py35/e922a17/ : SUCCESS in 13m 21s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/4/check/openstack-tox-py36/dc37480/ : SUCCESS in 11m 44s\n- openstack-tox-docs http://logs.openstack.org/35/605235/4/check/openstack-tox-docs/ca56e58/html/ : SUCCESS in 6m 24s\n- tempest-full http://logs.openstack.org/35/605235/4/check/tempest-full/bd49688/ : SUCCESS in 1h 34m 10s\n- neutron-grenade http://logs.openstack.org/35/605235/4/check/neutron-grenade/211b4e3/ : SUCCESS in 54m 08s\n- tempest-full-py3 http://logs.openstack.org/35/605235/4/check/tempest-full-py3/648392c/ : SUCCESS in 1h 55m 07s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/4/check/build-openstack-releasenotes/bd6e58c/html/ : SUCCESS in 12m 27s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/4/check/keystone-dsvm-functional/ddf7c4d/ : SUCCESS in 31m 12s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/4/check/keystone-dsvm-py3-functional/d542063/ : SUCCESS in 32m 23s\n- keystone-dsvm-functional-federation http://logs.openstack.org/35/605235/4/check/keystone-dsvm-functional-federation/a3e845b/ : SUCCESS in 36m 20s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/35/605235/4/check/keystone-dsvm-py35-functional-federation/44e86b8/ : SUCCESS in 40m 43s (non-voting)\n- legacy-keystoneclient-dsvm-functional http://logs.openstack.org/35/605235/4/check/legacy-keystoneclient-dsvm-functional/2e45db1/ : SUCCESS in 27m 20s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/2d87544/ : SUCCESS in 1h 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f7cda0dfef3f83e55dcf29c63ed0bc28384c3198","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-10-03 04:10:50.000000000","message":"Patch Set 4: Code-Review+1\n\nFirst pass looks good. Will review more in depth soon. This generally works for me.","accounts_in_message":[],"_revision_number":4},{"id":"44f3bd8c267df94dd30b2e15b21bcf76f8266944","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-11-06 02:26:25.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"acc82ac1af249aa66f2e50e23f94677b1e331c80","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-11-06 06:36:40.000000000","message":"Patch Set 4: Code-Review-1\n\nThe 409 Conflict Error message[1] should be updated as well. Now it always says the name is duplicated even using different name but the same id.\n\n[1]:https://github.com/openstack/keystone/blob/master/keystone/resource/core.py#L225","accounts_in_message":[],"_revision_number":4},{"id":"831f2a05138c571fe2807d143814592b57331a00","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-01-22 16:04:16.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"9bb03384db87361cd9027f115763e65a93ea61fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-22 19:40:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/5/check/openstack-tox-cover/8cb4f45/cover/ : SUCCESS in 13m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/5/check/openstack-tox-lower-constraints/adc2a37/ : SUCCESS in 16m 50s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/5/check/openstack-tox-pep8/986e71d/ : SUCCESS in 6m 11s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/5/check/openstack-tox-py27/e84804e/ : SUCCESS in 12m 07s\n- openstack-tox-py35 http://logs.openstack.org/35/605235/5/check/openstack-tox-py35/02eb0b8/ : SUCCESS in 12m 25s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/5/check/openstack-tox-py36/fd5b6f6/ : SUCCESS in 10m 52s\n- openstack-tox-docs http://logs.openstack.org/35/605235/5/check/openstack-tox-docs/272487c/html/ : SUCCESS in 7m 13s\n- tempest-full http://logs.openstack.org/35/605235/5/check/tempest-full/f1589df/ : SUCCESS in 1h 43m 53s\n- neutron-grenade http://logs.openstack.org/35/605235/5/check/neutron-grenade/53571c2/ : SUCCESS in 1h 00m 12s\n- grenade-py3 http://logs.openstack.org/35/605235/5/check/grenade-py3/9deec4e/ : SUCCESS in 1h 08m 20s\n- tempest-full-py3 http://logs.openstack.org/35/605235/5/check/tempest-full-py3/3fafffe/ : SUCCESS in 1h 48m 42s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/5/check/build-openstack-releasenotes/67f8e82/html/ : SUCCESS in 9m 25s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/5/check/keystone-dsvm-functional/dbe1ea6/ : SUCCESS in 28m 30s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/5/check/keystone-dsvm-py3-functional/60d48d3/ : SUCCESS in 28m 56s\n- keystone-dsvm-functional-federation http://logs.openstack.org/35/605235/5/check/keystone-dsvm-functional-federation/a1c56e3/ : SUCCESS in 27m 52s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/35/605235/5/check/keystone-dsvm-py35-functional-federation/ea22f50/ : SUCCESS in 36m 23s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/35/605235/5/check/keystoneclient-devstack-functional/37fafa7/ : SUCCESS in 17m 51s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/5/check/legacy-tempest-dsvm-ldap-domain-specific-driver/d13bd7c/ : SUCCESS in 2h 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"babe331f638f6af0c3a1c0d500a5fb2f3ecf649d","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-01-26 01:06:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"76bf929628791be76a6c6392718136c3ae4e70bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-26 02:51:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/6/check/openstack-tox-cover/e68a236/ : FAILURE in 11m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/6/check/openstack-tox-lower-constraints/4c6df40/ : FAILURE in 19m 03s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/6/check/openstack-tox-pep8/b38d1d3/ : SUCCESS in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/6/check/openstack-tox-py27/0be472e/ : FAILURE in 12m 11s\n- openstack-tox-py35 http://logs.openstack.org/35/605235/6/check/openstack-tox-py35/337f3a8/ : FAILURE in 12m 52s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/6/check/openstack-tox-py36/bae063a/ : FAILURE in 11m 55s\n- openstack-tox-docs http://logs.openstack.org/35/605235/6/check/openstack-tox-docs/06f0900/html/ : SUCCESS in 7m 33s\n- tempest-full http://logs.openstack.org/35/605235/6/check/tempest-full/6d1ed41/ : FAILURE in 1h 21m 50s\n- neutron-grenade http://logs.openstack.org/35/605235/6/check/neutron-grenade/b575a3c/ : SUCCESS in 56m 03s\n- grenade-py3 http://logs.openstack.org/35/605235/6/check/grenade-py3/b7347a8/ : SUCCESS in 1h 04m 35s\n- tempest-full-py3 http://logs.openstack.org/35/605235/6/check/tempest-full-py3/aa98471/ : FAILURE in 1h 03m 55s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/6/check/build-openstack-releasenotes/e21c043/html/ : SUCCESS in 11m 47s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/6/check/keystone-dsvm-functional/5c2229d/ : SUCCESS in 29m 14s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/6/check/keystone-dsvm-py3-functional/10c9358/ : SUCCESS in 29m 27s\n- keystone-dsvm-functional-federation http://logs.openstack.org/35/605235/6/check/keystone-dsvm-functional-federation/8c6c297/ : SUCCESS in 24m 13s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/35/605235/6/check/keystone-dsvm-py35-functional-federation/3233e6a/ : SUCCESS in 31m 25s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/35/605235/6/check/keystoneclient-devstack-functional/becb3f1/ : SUCCESS in 17m 11s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/6/check/legacy-tempest-dsvm-ldap-domain-specific-driver/49b625e/ : FAILURE in 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2cbb42e3dd4312153f7e54013bcd3df7644682a0","author":{"_account_id":28032,"name":"Oleksiy Petrenko","email":"opetrenko@mirantis.com","username":"enacero"},"date":"2019-02-01 14:42:13.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3518c1566d9bc44fcd11ab42095c4b3a7c9c9c98","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-02 03:22:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e90f53bd598c7467eaa1fd163ee7e00592cb2c28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 04:52:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/7/check/openstack-tox-cover/15680eb/ : FAILURE in 35m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/7/check/openstack-tox-lower-constraints/593a25f/ : FAILURE in 35m 37s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/7/check/openstack-tox-pep8/9e1540b/ : SUCCESS in 7m 02s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/7/check/openstack-tox-py27/ed3120a/ : FAILURE in 24m 05s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/7/check/openstack-tox-py36/0d104cc/ : FAILURE in 25m 48s\n- openstack-tox-py37 http://logs.openstack.org/35/605235/7/check/openstack-tox-py37/3104bb7/ : FAILURE in 23m 08s\n- openstack-tox-docs http://logs.openstack.org/35/605235/7/check/openstack-tox-docs/5d54947/html/ : SUCCESS in 7m 29s\n- tempest-full http://logs.openstack.org/35/605235/7/check/tempest-full/6e4215f/ : FAILURE in 1h 24m 46s\n- neutron-grenade http://logs.openstack.org/35/605235/7/check/neutron-grenade/58b5997/ : SUCCESS in 51m 36s\n- grenade-py3 http://logs.openstack.org/35/605235/7/check/grenade-py3/76a2816/ : SUCCESS in 1h 03m 02s\n- tempest-full-py3 http://logs.openstack.org/35/605235/7/check/tempest-full-py3/d1e4cbe/ : FAILURE in 1h 01m 38s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/7/check/build-openstack-releasenotes/8a763af/html/ : SUCCESS in 9m 58s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/7/check/keystone-dsvm-functional/ec880a9/ : SUCCESS in 25m 51s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/7/check/keystone-dsvm-py3-functional/f610fe8/ : SUCCESS in 36m 04s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/35/605235/7/check/keystone-dsvm-functional-federation-opensuse15/3ced23c/ : SUCCESS in 39m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/35/605235/7/check/keystone-dsvm-py3-functional-federation-opensuse15/037d83b/ : SUCCESS in 37m 56s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/35/605235/7/check/keystoneclient-devstack-functional/e69fa6e/ : SUCCESS in 13m 24s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/7/check/legacy-tempest-dsvm-ldap-domain-specific-driver/77bc519/ : FAILURE in 23m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e88bfc3bdf611e3206c196c71b702d97bebbb9c1","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-02 17:04:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9b8ca44ff66f210e89e3afe1ea4dbcb42dd12e43","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-04-02 18:06:33.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"dc3b405bd9550528d2bcd2f0754d75cbdcb0d5a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 20:11:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/8/check/openstack-tox-cover/573cb9e/cover/ : SUCCESS in 33m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/8/check/openstack-tox-lower-constraints/beef65d/ : TIMED_OUT in 41m 35s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/8/check/openstack-tox-pep8/e5de768/ : SUCCESS in 8m 27s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/8/check/openstack-tox-py27/f998e79/ : SUCCESS in 18m 59s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/8/check/openstack-tox-py36/559a94b/ : SUCCESS in 25m 26s\n- openstack-tox-py37 http://logs.openstack.org/35/605235/8/check/openstack-tox-py37/728352a/ : SUCCESS in 24m 11s\n- openstack-tox-docs http://logs.openstack.org/35/605235/8/check/openstack-tox-docs/8ada9b3/html/ : SUCCESS in 8m 23s\n- tempest-full http://logs.openstack.org/35/605235/8/check/tempest-full/e041b09/ : SUCCESS in 1h 48m 15s\n- neutron-grenade http://logs.openstack.org/35/605235/8/check/neutron-grenade/61875e9/ : SUCCESS in 1h 01m 11s\n- grenade-py3 http://logs.openstack.org/35/605235/8/check/grenade-py3/41b8c68/ : SUCCESS in 50m 33s\n- tempest-full-py3 http://logs.openstack.org/35/605235/8/check/tempest-full-py3/eb7f7ef/ : SUCCESS in 1h 34m 15s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/8/check/build-openstack-releasenotes/a54bd78/html/ : SUCCESS in 13m 24s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/8/check/keystone-dsvm-functional/6bf4cb5/ : SUCCESS in 35m 07s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/8/check/keystone-dsvm-py3-functional/8d4e905/ : SUCCESS in 37m 48s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/35/605235/8/check/keystone-dsvm-functional-federation-opensuse15/2168c58/ : SUCCESS in 39m 08s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/35/605235/8/check/keystone-dsvm-py3-functional-federation-opensuse15/7831eb7/ : SUCCESS in 40m 31s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/35/605235/8/check/keystoneclient-devstack-functional/27b520b/ : SUCCESS in 21m 34s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/8/check/legacy-tempest-dsvm-ldap-domain-specific-driver/25b1200/ : SUCCESS in 1h 30m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"634976461e872dbbd8d756b88141e16b75e50a03","author":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"date":"2019-04-06 16:41:42.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ee5918ac2e29bb4930e44bd2dc1715e93dd7834c","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-06 23:22:05.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"239f630a9dc388b0b39779230c17822b5fc8d2f2","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-06 23:22:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b5e90f31e67b58f8c1074dba4cbf1ef268a58430","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-07 01:02:34.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/9/check/openstack-tox-cover/b56c05e/cover/ : SUCCESS in 28m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/9/check/openstack-tox-lower-constraints/9d7b098/ : SUCCESS in 26m 49s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/9/check/openstack-tox-pep8/a66696b/ : SUCCESS in 5m 36s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/9/check/openstack-tox-py27/d72924b/ : SUCCESS in 21m 28s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/9/check/openstack-tox-py36/6d5a882/ : SUCCESS in 27m 03s\n- openstack-tox-py37 http://logs.openstack.org/35/605235/9/check/openstack-tox-py37/f7a9f61/ : SUCCESS in 23m 51s\n- openstack-tox-docs http://logs.openstack.org/35/605235/9/check/openstack-tox-docs/ae6de04/html/ : SUCCESS in 7m 29s\n- tempest-full http://logs.openstack.org/35/605235/9/check/tempest-full/918645a/ : FAILURE in 1h 33m 59s\n- neutron-grenade http://logs.openstack.org/35/605235/9/check/neutron-grenade/e95ce04/ : SUCCESS in 56m 38s (non-voting)\n- grenade-py3 http://logs.openstack.org/35/605235/9/check/grenade-py3/573937b/ : SUCCESS in 55m 26s\n- tempest-full-py3 http://logs.openstack.org/35/605235/9/check/tempest-full-py3/2f79d5b/ : SUCCESS in 1h 33m 50s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/9/check/build-openstack-releasenotes/730ef0d/html/ : SUCCESS in 11m 15s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/9/check/keystone-dsvm-functional/9e46fe5/ : SUCCESS in 34m 04s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/9/check/keystone-dsvm-py3-functional/1c204ca/ : SUCCESS in 33m 17s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/35/605235/9/check/keystone-dsvm-functional-federation-opensuse15/e44a764/ : SUCCESS in 39m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/35/605235/9/check/keystone-dsvm-py3-functional-federation-opensuse15/d7fa35c/ : FAILURE in 43m 28s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/35/605235/9/check/keystoneclient-devstack-functional/ffff71d/ : SUCCESS in 17m 26s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/9/check/legacy-tempest-dsvm-ldap-domain-specific-driver/959c815/ : SUCCESS in 1h 36m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"aa4c1f0763a5309b75ec6127d816aa21078bc954","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-07 01:03:45.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"994c8eab8a3c7b063cff6af722a9873807aa1ef9","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-07 13:03:17.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"a0c9cdd1cc9a0a0f65c19ebf26e6cc48590c09c1","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-07 13:03:37.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"00c76a166e4e6c755e70e22ff7445388d2885f75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-07 14:50:26.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/10/check/openstack-tox-cover/3820474/cover/ : SUCCESS in 28m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/10/check/openstack-tox-lower-constraints/7a2a7b3/ : SUCCESS in 31m 03s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/10/check/openstack-tox-pep8/2ef0344/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/10/check/openstack-tox-py27/7754306/ : SUCCESS in 20m 04s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/10/check/openstack-tox-py36/8f0f57e/ : SUCCESS in 22m 46s\n- openstack-tox-py37 http://logs.openstack.org/35/605235/10/check/openstack-tox-py37/ffeb222/ : SUCCESS in 19m 17s\n- openstack-tox-docs http://logs.openstack.org/35/605235/10/check/openstack-tox-docs/11591bb/html/ : SUCCESS in 7m 19s\n- tempest-full http://logs.openstack.org/35/605235/10/check/tempest-full/41394d3/ : SUCCESS in 1h 45m 37s\n- neutron-grenade http://logs.openstack.org/35/605235/10/check/neutron-grenade/962b48c/ : SUCCESS in 1h 08m 26s (non-voting)\n- grenade-py3 http://logs.openstack.org/35/605235/10/check/grenade-py3/b6f9c93/ : SUCCESS in 54m 59s\n- tempest-full-py3 http://logs.openstack.org/35/605235/10/check/tempest-full-py3/d05c14c/ : SUCCESS in 1h 40m 09s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/10/check/build-openstack-releasenotes/8834539/html/ : SUCCESS in 9m 36s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/10/check/keystone-dsvm-functional/ea41500/ : SUCCESS in 31m 37s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/10/check/keystone-dsvm-py3-functional/bdec55e/ : SUCCESS in 33m 10s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/35/605235/10/check/keystone-dsvm-functional-federation-opensuse15/8474b2a/ : SUCCESS in 29m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/35/605235/10/check/keystone-dsvm-py3-functional-federation-opensuse15/8dd696a/ : SUCCESS in 37m 41s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/35/605235/10/check/keystoneclient-devstack-functional/a3c2ebe/ : SUCCESS in 12m 42s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/10/check/legacy-tempest-dsvm-ldap-domain-specific-driver/dcc8b30/ : SUCCESS in 1h 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"241a08e2284d252d62f38618c9c5fa8078880ef0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-04-08 21:17:12.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7cbf02453f6c74433468d0dd39558e9ccddc7245","author":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"date":"2019-04-08 21:35:49.000000000","message":"Patch Set 10: Code-Review+1\n\nLGTM once Colleen\u0027s comments are addressed.","accounts_in_message":[],"_revision_number":10},{"id":"1db8d13489e966a9b898036337b230273463296b","author":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"date":"2019-04-09 16:29:58.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"0bc877ce05ac8a50d81b12b1318bf41f644d226b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-04-09 16:32:48.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"b3e3fe141e40cb2349eaf6cd1ceda7bb15d74d3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 20:02:37.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/35/605235/11/check/openstack-tox-cover/ceee643/cover/ : SUCCESS in 30m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/11/check/openstack-tox-lower-constraints/21d3415/ : SUCCESS in 31m 58s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/11/check/openstack-tox-pep8/8a99366/ : SUCCESS in 7m 13s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/11/check/openstack-tox-py27/fa50aca/ : SUCCESS in 20m 38s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/11/check/openstack-tox-py36/1844c33/ : SUCCESS in 21m 41s\n- openstack-tox-py37 http://logs.openstack.org/35/605235/11/check/openstack-tox-py37/109b407/ : SUCCESS in 25m 20s\n- openstack-tox-docs http://logs.openstack.org/35/605235/11/check/openstack-tox-docs/d047f21/html/ : SUCCESS in 10m 12s\n- tempest-full http://logs.openstack.org/35/605235/11/check/tempest-full/b7f141d/ : SUCCESS in 1h 37m 09s\n- neutron-grenade http://logs.openstack.org/35/605235/11/check/neutron-grenade/2e4fd23/ : SUCCESS in 52m 11s (non-voting)\n- grenade-py3 http://logs.openstack.org/35/605235/11/check/grenade-py3/302130b/ : SUCCESS in 1h 02m 33s\n- tempest-full-py3 http://logs.openstack.org/35/605235/11/check/tempest-full-py3/2074b47/ : SUCCESS in 1h 46m 32s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/11/check/build-openstack-releasenotes/1b90e20/html/ : SUCCESS in 10m 04s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/11/check/keystone-dsvm-functional/50bedf3/ : SUCCESS in 36m 10s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/11/check/keystone-dsvm-py3-functional/8383a5e/ : SUCCESS in 29m 35s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/35/605235/11/check/keystone-dsvm-functional-federation-opensuse15/b938761/ : SUCCESS in 33m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/35/605235/11/check/keystone-dsvm-py3-functional-federation-opensuse15/6421a27/ : FAILURE in 34m 31s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/35/605235/11/check/keystoneclient-devstack-functional/28641f0/ : SUCCESS in 14m 25s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/35/605235/11/check/legacy-tempest-dsvm-ldap-domain-specific-driver/2d79ffa/ : SUCCESS in 1h 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"980f338ca33b39bd8fd8ec7a6070b29438c1b002","author":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"date":"2019-04-09 20:17:45.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"b4920e701bbb82ffc857104f3b5dfb0b4fea9cc2","author":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"date":"2019-04-09 20:44:00.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"bde373f8f63fca5fd1b93f71ca57cd528c156221","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 20:44:25.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"9a7c8a0fc410635e505af48ace650cd87657d432","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 22:36:45.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/605235/11/gate/openstack-tox-lower-constraints/b23bf09/ : SUCCESS in 30m 21s\n- openstack-tox-pep8 http://logs.openstack.org/35/605235/11/gate/openstack-tox-pep8/e84ea1c/ : SUCCESS in 7m 34s\n- openstack-tox-py27 http://logs.openstack.org/35/605235/11/gate/openstack-tox-py27/682d2bf/ : SUCCESS in 22m 11s\n- openstack-tox-py36 http://logs.openstack.org/35/605235/11/gate/openstack-tox-py36/a5b2564/ : SUCCESS in 28m 49s\n- openstack-tox-py37 http://logs.openstack.org/35/605235/11/gate/openstack-tox-py37/9730e37/ : SUCCESS in 23m 17s\n- openstack-tox-docs http://logs.openstack.org/35/605235/11/gate/openstack-tox-docs/b15366b/html/ : SUCCESS in 8m 09s\n- tempest-full http://logs.openstack.org/35/605235/11/gate/tempest-full/5910034/ : SUCCESS in 1h 39m 29s\n- neutron-grenade http://logs.openstack.org/35/605235/11/gate/neutron-grenade/5e7d0cd/ : SUCCESS in 58m 25s (non-voting)\n- grenade-py3 http://logs.openstack.org/35/605235/11/gate/grenade-py3/6cf4c45/ : SUCCESS in 1h 13m 44s\n- tempest-full-py3 http://logs.openstack.org/35/605235/11/gate/tempest-full-py3/6082e54/ : SUCCESS in 1h 49m 12s\n- build-openstack-releasenotes http://logs.openstack.org/35/605235/11/gate/build-openstack-releasenotes/ef79eab/html/ : SUCCESS in 11m 59s\n- keystone-dsvm-functional http://logs.openstack.org/35/605235/11/gate/keystone-dsvm-functional/446330c/ : SUCCESS in 28m 29s\n- keystone-dsvm-py3-functional http://logs.openstack.org/35/605235/11/gate/keystone-dsvm-py3-functional/541d044/ : SUCCESS in 31m 12s","accounts_in_message":[],"_revision_number":11},{"id":"bc67b82299259134c2590cbd475701a631f4f327","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 22:36:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"4cd99e7197f3abbdf800fa8ef392c24148d4c390","revisions":{"48dcc08dba29ba69e6349e950e2ad8d9b8d0011f":{"kind":"REWORK","_number":1,"created":"2018-09-25 21:15:31.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/1"}}},"commit":{"parents":[{"commit":"50b2d6aa0e7b9cd9f43daf67db94e49882a57448","subject":"Merge \"Implement Trust Flush via keystone-manage.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/50b2d6aa0e7b9cd9f43daf67db94e49882a57448"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:15:13.000000000","tz":-240},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/48dcc08dba29ba69e6349e950e2ad8d9b8d0011f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/48dcc08dba29ba69e6349e950e2ad8d9b8d0011f"}]},"branch":"refs/heads/master"},"f770b732f74e9dd8273c740be3ef74cd4a89411d":{"kind":"REWORK","_number":2,"created":"2018-09-26 14:15:28.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/2"}}},"commit":{"parents":[{"commit":"50b2d6aa0e7b9cd9f43daf67db94e49882a57448","subject":"Merge \"Implement Trust Flush via keystone-manage.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/50b2d6aa0e7b9cd9f43daf67db94e49882a57448"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-26 14:15:09.000000000","tz":-240},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f770b732f74e9dd8273c740be3ef74cd4a89411d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f770b732f74e9dd8273c740be3ef74cd4a89411d"}]},"branch":"refs/heads/master"},"661f0b8a3f410c3cbac7ee90786f08944a47df1d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-10-01 13:38:14.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/3"}}},"commit":{"parents":[{"commit":"9c91aa74daba42145c07d5dcc75592812a631701","subject":"Merge \"Add build_target arguement to enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9c91aa74daba42145c07d5dcc75592812a631701"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"ayoung","email":"ayoung@redhat.com","date":"2018-10-01 13:38:14.000000000","tz":0},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/661f0b8a3f410c3cbac7ee90786f08944a47df1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/661f0b8a3f410c3cbac7ee90786f08944a47df1d"}]},"branch":"refs/heads/master"},"ccf0c8a6dd8fe2edb6a673b12eb467cf26295e0d":{"kind":"REWORK","_number":4,"created":"2018-10-02 21:28:50.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/4"}}},"commit":{"parents":[{"commit":"9c91aa74daba42145c07d5dcc75592812a631701","subject":"Merge \"Add build_target arguement to enforcer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9c91aa74daba42145c07d5dcc75592812a631701"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-10-02 21:28:40.000000000","tz":-240},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ccf0c8a6dd8fe2edb6a673b12eb467cf26295e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ccf0c8a6dd8fe2edb6a673b12eb467cf26295e0d"}]},"branch":"refs/heads/master"},"96c288b44de2a0abccd1315b13421d282dd81fa4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-01-22 16:04:16.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/5"}}},"commit":{"parents":[{"commit":"2c47e935fa866ddea25ec3954bbe18af4258f812","subject":"Merge \"Remove registered limit policies from policy.v3cloudsample.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2c47e935fa866ddea25ec3954bbe18af4258f812"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"ayoung","email":"ayoung@redhat.com","date":"2019-01-22 16:04:16.000000000","tz":0},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/96c288b44de2a0abccd1315b13421d282dd81fa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/96c288b44de2a0abccd1315b13421d282dd81fa4"}]},"branch":"refs/heads/master"},"9b86b0be25c69cf63dafc76d92c8bc02e206ed34":{"kind":"REWORK","_number":6,"created":"2019-01-26 01:06:58.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/6"}}},"commit":{"parents":[{"commit":"2c47e935fa866ddea25ec3954bbe18af4258f812","subject":"Merge \"Remove registered limit policies from policy.v3cloudsample.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2c47e935fa866ddea25ec3954bbe18af4258f812"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2019-01-22 17:46:48.000000000","tz":-300},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9b86b0be25c69cf63dafc76d92c8bc02e206ed34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9b86b0be25c69cf63dafc76d92c8bc02e206ed34"}]},"branch":"refs/heads/master"},"f30701592b05412856bce8a5ff5f5ee5895d897f":{"kind":"REWORK","_number":7,"created":"2019-04-02 03:22:03.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/7"}}},"commit":{"parents":[{"commit":"af4ec20f83424270974223439ccb591f88fdbcc2","subject":"Merge \"Remove redundant policies from v3cloudsample\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/af4ec20f83424270974223439ccb591f88fdbcc2"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2019-04-02 03:21:55.000000000","tz":-240},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f30701592b05412856bce8a5ff5f5ee5895d897f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f30701592b05412856bce8a5ff5f5ee5895d897f"}]},"branch":"refs/heads/master"},"44aebf7613aaf2665338fce120c1e56e6b9dcf6a":{"kind":"REWORK","_number":8,"created":"2019-04-02 17:04:48.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/8"}}},"commit":{"parents":[{"commit":"af4ec20f83424270974223439ccb591f88fdbcc2","subject":"Merge \"Remove redundant policies from v3cloudsample\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/af4ec20f83424270974223439ccb591f88fdbcc2"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2019-04-02 17:04:38.000000000","tz":-240},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/44aebf7613aaf2665338fce120c1e56e6b9dcf6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/44aebf7613aaf2665338fce120c1e56e6b9dcf6a"}]},"branch":"refs/heads/master"},"9a338ae5c23cb4b43072ee784725dd3371798bc8":{"kind":"REWORK","_number":9,"created":"2019-04-06 23:22:23.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/9"}}},"commit":{"parents":[{"commit":"af4ec20f83424270974223439ccb591f88fdbcc2","subject":"Merge \"Remove redundant policies from v3cloudsample\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/af4ec20f83424270974223439ccb591f88fdbcc2"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2019-04-06 23:22:16.000000000","tz":-240},"subject":"llow an explicit_domain_id parameter when creating a domain","message":"llow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9a338ae5c23cb4b43072ee784725dd3371798bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9a338ae5c23cb4b43072ee784725dd3371798bc8"}]},"branch":"refs/heads/master"},"63e22ef3b39c38da7012d815c4a0215a19252c54":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2019-04-07 13:03:17.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/10"}}},"commit":{"parents":[{"commit":"af4ec20f83424270974223439ccb591f88fdbcc2","subject":"Merge \"Remove redundant policies from v3cloudsample\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/af4ec20f83424270974223439ccb591f88fdbcc2"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"ayoung","email":"ayoung@redhat.com","date":"2019-04-07 13:03:11.000000000","tz":0},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/63e22ef3b39c38da7012d815c4a0215a19252c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/63e22ef3b39c38da7012d815c4a0215a19252c54"}]},"branch":"refs/heads/master"},"4cd99e7197f3abbdf800fa8ef392c24148d4c390":{"kind":"REWORK","_number":11,"created":"2019-04-09 16:29:58.000000000","uploader":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"ref":"refs/changes/35/605235/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/605235/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/605235/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/605235/11"}}},"commit":{"parents":[{"commit":"af4ec20f83424270974223439ccb591f88fdbcc2","subject":"Merge \"Remove redundant policies from v3cloudsample\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/af4ec20f83424270974223439ccb591f88fdbcc2"}]}],"author":{"name":"Adam Young","email":"ayoung@redhat.com","date":"2018-09-25 21:09:02.000000000","tz":-240},"committer":{"name":"ayoung","email":"ayoung@redhat.com","date":"2019-04-09 16:29:52.000000000","tz":0},"subject":"Allow an explicit_domain_id parameter when creating a domain","message":"Allow an explicit_domain_id parameter when creating a domain\n\nThis allows domain_ids to match across distinct Keystone\ndeployments The domain_id is used to create unique\nidentifiers with the mapping backend.  When this\noption is used, mapped user identifiers can be\nconsistant across different Keystone servers.\n\ncloses-bug: 1794527\n\nChange-Id: I100bca162e71a9d394ed5787b976b13b1e57987f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4cd99e7197f3abbdf800fa8ef392c24148d4c390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4cd99e7197f3abbdf800fa8ef392c24148d4c390"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
