)]}'
{"id":"openstack%2Fkeystone~607820","triplet_id":"openstack%2Fkeystone~master~I090e2470726d22b2670a2cca89025063419f5262","project":"openstack/keystone","branch":"master","topic":"bug/1750678","hashtags":[],"change_id":"I090e2470726d22b2670a2cca89025063419f5262","subject":"Implement scope type checking for EC2 credentials","status":"MERGED","created":"2018-10-04 07:15:27.000000000","updated":"2019-09-16 21:08:58.000000000","submitted":"2019-09-15 09:19:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"607820-1568539198986-75f62c1c","meta_rev_id":"4df9f8450817d8f0426d73c87e144a223a1cba3b","_number":607820,"virtual_id_number":607820,"owner":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-09-15 09:19:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-09-14 08:59:15.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-09-14 08:59:15.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":1,"date":"2019-09-14 10:08:22.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 13:30:35.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-09-14 10:08:22.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-09-15 09:19:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eaeccf1f7d288ad8f2627904e2f019fffdc12eda","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-04 07:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae65fea89057b935174d2191aec21796dd49554c","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-04 07:17:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"57aaf8f33339147e756200b3246f65495991bb80","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-04 07:17:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7dcba1acc6fb80437946ef80d40402bb689acdc5","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-04 07:18:07.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"741c4a27aa81cef27d1d56e047ea666ee6fb5806","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-04 07:18:12.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"df525e8bdd51f4bc7550eb5b9f72ae2b097858bc","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-04 07:35:27.000000000","message":"Patch Set 3:\n\nIt did not find the place where policies/ec2_credential.py are beinig enforced. ec2 credentials controller call credential controller which enforce the credential\u0027s policies. so ec2 credential\u0027s policy are not being enforced for ec2 credential APIs. \n\nSo the question arises whether ec2 credential policies should be there or be deprecated/removed?","accounts_in_message":[],"_revision_number":3},{"id":"a108c1d0e078bf85323d75c2fba17428ad5323b8","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-04 08:06:32.000000000","message":"Patch Set 3:\n\nAfter confirming with cmurphy on IRC got to know that the ec2 policies are enforced with controller.protected decorator [1].\n\nSo for implementing system scope one approach can be changing the controller.protected decorator or second can be enforcing ec2 policies same as of  credential policies.\n\n[1] http://git.openstack.org/cgit/openstack/keystone/tree/keystone/contrib/ec2/controllers.py#n302","accounts_in_message":[],"_revision_number":3},{"id":"2693a330ce26a610b5bfc3b119ceae1ec5f1e3c0","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-05 06:25:29.000000000","message":"Patch Set 3:\n\nOn Hold for now as will proceed after transitioning of @controller.protected decorator to ENFORCER.enforce_call.Dependent on that.","accounts_in_message":[],"_revision_number":3},{"id":"8eb5ba3eda53f8e463be5dba39ff0ded71d59970","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-11-05 02:41:35.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8ffc9f994ca95d79d5d403bcfa80ab38b2e4bd6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-05 04:04:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/20/607820/4/check/openstack-tox-cover/8fefa07/cover/ : SUCCESS in 12m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/607820/4/check/openstack-tox-lower-constraints/221a4de/ : TIMED_OUT in 40m 42s\n- openstack-tox-pep8 http://logs.openstack.org/20/607820/4/check/openstack-tox-pep8/17ba98f/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/20/607820/4/check/openstack-tox-py27/d1637fd/ : SUCCESS in 8m 59s\n- openstack-tox-py35 http://logs.openstack.org/20/607820/4/check/openstack-tox-py35/13db231/ : SUCCESS in 10m 01s\n- openstack-tox-py36 http://logs.openstack.org/20/607820/4/check/openstack-tox-py36/d2bb166/ : SUCCESS in 9m 17s\n- openstack-tox-docs http://logs.openstack.org/20/607820/4/check/openstack-tox-docs/f0a1432/html/ : SUCCESS in 4m 32s\n- tempest-full http://logs.openstack.org/20/607820/4/check/tempest-full/b25fa94/ : SUCCESS in 1h 21m 55s\n- neutron-grenade http://logs.openstack.org/20/607820/4/check/neutron-grenade/56c3259/ : SUCCESS in 53m 47s\n- tempest-full-py3 http://logs.openstack.org/20/607820/4/check/tempest-full-py3/71eec2c/ : SUCCESS in 1h 11m 24s\n- keystone-dsvm-functional http://logs.openstack.org/20/607820/4/check/keystone-dsvm-functional/f494de3/ : SUCCESS in 32m 34s\n- keystone-dsvm-py3-functional http://logs.openstack.org/20/607820/4/check/keystone-dsvm-py3-functional/34d00cc/ : SUCCESS in 29m 28s\n- keystone-dsvm-functional-federation http://logs.openstack.org/20/607820/4/check/keystone-dsvm-functional-federation/d3bbd56/ : FAILURE in 26m 24s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/20/607820/4/check/keystone-dsvm-py35-functional-federation/399e28c/ : FAILURE in 34m 57s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/20/607820/4/check/keystoneclient-devstack-functional/1aa93bf/ : SUCCESS in 24m 03s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/20/607820/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/d3f31a9/ : SUCCESS in 1h 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d67586ff38cd2a87a7743dc0327a00064cf7ffd4","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-06 22:16:21.000000000","message":"Patch Set 4:\n\nLast updated 10 months ago, is this still in progress?","accounts_in_message":[],"_revision_number":4},{"id":"ffbd5475ad9ec49b57c9be529ed044ca28b64824","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-08 13:08:11.000000000","message":"Patch Set 4:\n\n\u003e Last updated 10 months ago, is this still in progress?\n\nStarting over this.","accounts_in_message":[],"_revision_number":4},{"id":"da0eaf11e60e1a58be707f3b761468596279f913","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-08 15:05:04.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"55ca6d123b955c1dec0b3f2d56a51629f820c4b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-08 16:56:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dcba0c5603645ec8390d87e7ec6d129 : SUCCESS in 39m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/744e2ca1a8794831a65f8be69650794f : SUCCESS in 32m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d48c6f54eab465699a6a5af3de59889 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/681036a0f988480ea2c8b1bd56ef388f : SUCCESS in 35m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd9382784a1446289b496ed6713b7f8c : SUCCESS in 32m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2b157bd0157421ea33189d4edcbc326 : SUCCESS in 30m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c141da6c0734fe59c794e936dd8edad : SUCCESS in 10m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/95af8b40a3f74d43a76d62f58fdcb591 : SUCCESS in 1h 44m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6c5cd7659c244b81837a8fb1879946ec : SUCCESS in 1h 02m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/15456f44045346c7af5e55905b8091a1 : SUCCESS in 1h 05m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/544efa9293aa4c68a331d4fb1e490680 : SUCCESS in 1h 25m 57s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/02ec8855f205422da74093fb27643bf7 : SUCCESS in 39m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/34959d1cbcf646758df02178807c7894 : SUCCESS in 33m 56s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0961adbd48e34bb5bc9946b1cf6bf76e : RETRY_LIMIT in 3m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/32d102e7ef904ea3931d1de3d1998840 : RETRY_LIMIT in 3m 56s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c94b574b0fa64c51a9c673a1efc6e2bc : SUCCESS in 18m 32s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/111a3a0d5db44eb697d2f3d8150614ff : SUCCESS in 1h 42m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3990500b99564609b914121e358bdcd1 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":5},{"id":"430a82703a557ad770e2670d6ac1489b42a9d0b6","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-09 10:59:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"47fae820371c02df669d3b55ae94b9193645aca1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:02:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/019f8e9d685f4c5086c94834c84aee3e : SUCCESS in 34m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b2979dee5e1497692cce01f8abfa8ad : SUCCESS in 36m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e156f447514dbfb6d570b316e16a10 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8602fb9dee2241feb583f7b0edb148b3 : SUCCESS in 30m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e57c1715664f4e38bb7209bd7bb6b050 : SUCCESS in 33m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/628fe2f50c914a599b3ad4dcc18159a8 : SUCCESS in 32m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db1797c763a045e6a7807c878bba3bf3 : SUCCESS in 12m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/56b2a84228674b8b839333d671589d75 : SUCCESS in 1h 44m 18s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f8a7270e5229463aa45c71665dbba74e : SUCCESS in 58m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3f24f580cbb443bfa132112a3c91f939 : SUCCESS in 1h 01m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7aa81e21aa15449da5d8c76425f747d7 : SUCCESS in 1h 16m 58s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/164db4a985a54be8b6923a3390946c4c : SUCCESS in 39m 20s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/46d36de780434cf1bf70afdaa2cb7cd3 : SUCCESS in 33m 05s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/1ff2460895834493adfc2e242eb82645 : SUCCESS in 45m 19s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/eb56ad457f6a4d1e82c36f65f7118b51 : SUCCESS in 41m 07s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0fea5dec218346b5ba9aff234dcd8c67 : SUCCESS in 16m 30s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/01069686655046edbdb1a1e2670a3fb8 : SUCCESS in 1h 54m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87e2ed99db7e420bac3b54699fb34275 : SUCCESS in 1h 15m 14s","accounts_in_message":[],"_revision_number":6},{"id":"765e13902ec343858d2bc67ea0936b1b16350a20","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-09 20:32:20.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ad797030358fdcdef2b00ebcb1b8e00f21c0f25f","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-10 05:54:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f5ccc5c17a356e711d8c6c382c3fd1ebfada110d","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-10 06:39:47.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7e85f9f7eaa75bad8321e536ebc12780de2d82e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 11:44:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1956a95f82f48ffb2ec4a1348d62612 : TIMED_OUT in 1h 00m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20942e95516148808fa13e054ad2320b : TIMED_OUT in 41m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f146890a7964fe790f251724b6b3325 : SUCCESS in 6m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/293a9358213d41459a0819cfc2da722f : FAILURE in 36m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73dd02c78bcb47a2a749e1becc078ebe : FAILURE in 41m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33ed9cfe523243ab8140ffe31cb067ce : FAILURE in 36m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1fd1fa9dcb4e2b9d31651dedc2ada3 : SUCCESS in 9m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c698d27242b446ef9e84b3d6b494704d : SUCCESS in 1h 41m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b1862aba92d046ca88d3435efc02317e : SUCCESS in 59m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ac867fc096a2409b9981a291c602b3c8 : SUCCESS in 1h 18m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd5f19e062d646fbb90fbe1affeb4f83 : SUCCESS in 1h 29m 36s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/1af4fb9cafee4aaf8de7e2059afcb7c8 : SUCCESS in 32m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4cb1a5ad6295477ab7a587757beb21b4 : SUCCESS in 34m 28s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/e9af1186d2e84ec78b4377f9654d67ac : SUCCESS in 32m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/e0ba3deeb54745e680574b586a18ac95 : SUCCESS in 46m 29s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e0f28bc6f0c149ff89ace455715be347 : SUCCESS in 16m 14s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ef17de79040b4c55aaec964a86fe8f52 : SUCCESS in 1h 34m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a8e560755964f30b16a8cd6eb441885 : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":7},{"id":"676c1fc1d248a0ab0f993c7d252853183314466a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-10 16:27:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3b14224193d82f34fd5ded512407138d09ad7a45","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-10 23:14:21.000000000","message":"Patch Set 8: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"99c6a3b45c52ae34c22408372e271673f7bd48ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 00:00:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6645b163292548c887f2b2d891701e53 : SUCCESS in 43m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb5e95145f554a3fa2f95d844a75a297 : SUCCESS in 39m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b9e8af86a084d40a216a75711fb87bb : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30ee4b234f3643d09f4d0bc8552876d9 : SUCCESS in 31m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9548bdea085c431ba86f286a949ed8a5 : SUCCESS in 37m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/054b874f74144ea583fcea4773fc05cd : SUCCESS in 39m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cab06fa1034475eb0abc3d5c9c88d8c : SUCCESS in 12m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b279a8b8d31f4800bbbac08aa945aea8 : SUCCESS in 1h 54m 02s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/c812085fe71d4bd2a6457d2aa9905add : SUCCESS in 58m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/19b90b545fe142b98b419eb8c9b65be5 : SUCCESS in 1h 08m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90e821f4ae1f4f36bc7a19896ab4c67e : SUCCESS in 1h 24m 43s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/76fe85979e2f4138b790fcbca87da70c : SUCCESS in 32m 46s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f5fddff9845b439fab02be15a6d4c565 : SUCCESS in 34m 00s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/f970b94aad4d4958b72c7b963615f799 : SUCCESS in 41m 22s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/2a05cf18271e4bc484a22410aa7a690e : SUCCESS in 41m 45s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/cb713cfa4b324fc2a05b2fe277e1cec5 : SUCCESS in 18m 11s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/e1a9fd9b2a63428893060b87abc97d61 : SUCCESS in 1h 34m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e360c5b7a4d48448ee4fba8823ef477 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":8},{"id":"e34eac776a1fbeb828cafa69a1f6a04c34ed8779","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-11 07:29:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"20642ecb43f306d9ac87802337eaec31f42d7c99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 12:02:42.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2ae25d2932d4503aac4bde7b086914d : FAILURE in 35m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86d663473b5b4c869c7aa430e1afe2c7 : FAILURE in 40m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e82f94898ad47c8828662d474913438 : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d7b52615ec441e2ae60dd2033f67137 : FAILURE in 44m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2620b5c0aec74c779a51f1b1740ad13b : FAILURE in 40m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7178a6bf5644e5688f88187a7e8b80c : FAILURE in 32m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f636a2cfc4246fca0e496dfb8813b1e : SUCCESS in 14m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d733bbe76de24c84a254a52237379f15 : SUCCESS in 1h 40m 56s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8dcd49a86d304716b36246e105cb55dc : SUCCESS in 1h 04m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bf38282c7a824d3196673eee1dc23536 : SUCCESS in 1h 04m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6eae16aa72584659b6dc1f973e4ffb7a : SUCCESS in 1h 24m 40s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c1aede5ee6543729799f795897cb2e2 : SUCCESS in 38m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1210f39c2ea84289b6cec2b8a8215b18 : SUCCESS in 37m 52s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ce62d67821c6438bbea9af991ae24b75 : SUCCESS in 45m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/7c4ddbce6c294821bfa880e555b7cd0f : SUCCESS in 43m 09s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/fc1cdea5cc3644edb4407b69b3e06b4b : SUCCESS in 19m 47s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/21eb256f0c564d57b0529e5f2b1cbc51 : SUCCESS in 1h 45m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6ddb269a52d472e94ce3678ebcb9106 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":9},{"id":"588dde3d304a1e28fe0dc1e2afa34549135ab2ee","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-11 18:59:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"33e55ef88580956810df7ce6c27ecabcd52c7a78","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-11 19:01:28.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"2d264bc140256198bfeb5ec44ffac4bc38ce0464","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 02:25:28.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc1cb12a4ac04f92b586e0b836d691c5 : SUCCESS in 45m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae1b31c63b8e4dba81748adcac7f0d4e : SUCCESS in 43m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597c42756f3048e383a631a4e9b66adf : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2942416a88e44e5e9228711316e1a682 : FAILURE in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbcc85dca8f648fda5e6dd0cbb4083f4 : SUCCESS in 37m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/433acffecee1490f91020a09740c2735 : SUCCESS in 36m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb256813412c480ebe69ae6279a24a5b : SUCCESS in 17m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/926652cb526146a4ab918c9d76eb7b20 : SUCCESS in 1h 44m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/df71c86943a9423d839c0ad6e54dce36 : FAILURE in 50m 00s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c405cfab1988446eb8d94cb2f6e78137 : SUCCESS in 1h 08m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b02731cca90140f691b10bf59bd05cbf : SUCCESS in 1h 35m 57s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/aea75266ccff475faede2dc03d6510c1 : SUCCESS in 36m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/215668996b86430e884ed53598992dc7 : SUCCESS in 35m 37s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/14ae78924b6e4ff5bdfc06ca5069595b : SUCCESS in 39m 34s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c97d30779be0474088bec63f4180c758 : SUCCESS in 42m 56s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/def8eaac1a634787baad3a71f7c3ca78 : SUCCESS in 16m 36s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/055ffde680734c60aa249df4ff257831 : SUCCESS in 1h 38m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4d94f79ddd6423ba6294bb45e2c9757 : SUCCESS in 1h 09m 06s","accounts_in_message":[],"_revision_number":11},{"id":"8dcc53e05646d1e17ca08c0d1ca0524f42b73223","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-12 05:59:53.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"76563a9469bc44026c1b5ec7e6dcc4be9b856635","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 10:29:08.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73be8930b0704af3b2c446b70bf281d6 : SUCCESS in 37m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e25546be8f54296bb723bb545580577 : SUCCESS in 54m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/027ef9389bf74e47a21fb477b348b2e1 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ce3bf64f5814230ad21cdc86349a697 : SUCCESS in 33m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c2ecda6a91c462a9764c436ff341bd7 : SUCCESS in 36m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe2d9d8e4721450684d37987e7665001 : SUCCESS in 32m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467b29c2546e4ae58d153057c7a38cf5 : SUCCESS in 18m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/caf0936e2099467e96a9b5d6a4c2943c : POST_FAILURE in 2h 19m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/6ab36ce37f574fe0a01062817e5304ef : SUCCESS in 56m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1fba5a97c55749ce897e7b256295e3f5 : TIMED_OUT in 3h 01m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40274d0226024acd9cf6c7e70063e49c : SUCCESS in 1h 28m 48s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/da434fee48dc4ed2a2343efd0889e91a : SUCCESS in 44m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9fa65d25cd3f435eb26915ade97fbfa6 : SUCCESS in 43m 58s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c84bcb60db84498588fe3cc08fc1569e : SUCCESS in 44m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/08d62d2d90574678b825984ac4c95ef0 : SUCCESS in 52m 23s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f8eef3c1835048cc975700e2edc7a422 : SUCCESS in 17m 37s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3c0836b0c5c04e8fac58f46ed8f12648 : SUCCESS in 2h 04m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac13b76ea8e5451dbc0fee5ce5bd0c78 : SUCCESS in 1h 05m 54s","accounts_in_message":[],"_revision_number":11},{"id":"19b1243b6073a7cad9d8237ae71f942f8243699c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-12 12:28:35.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"92a28891d9a7329cbd2cd5554db0769ea694150a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-12 14:35:56.000000000","message":"Patch Set 11:\n\nThanks Colleen For working over this.","accounts_in_message":[],"_revision_number":11},{"id":"b8e801544c10671ae5fdc248c00fb9ee7a71f584","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 00:50:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/574b596f21da442f93b698218de571b6 : SUCCESS in 48m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0492a1bbe7c64a01b679f9a53406ecc7 : SUCCESS in 46m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4509e341d905448db9b92e4b024a3a1c : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e6e5bd8e7ed40349f2d25e82a2e7391 : SUCCESS in 38m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb86e241cf4d4c0bb4c3b16a7eaebf41 : SUCCESS in 30m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d182f0204534512a7478a8c1490e4c0 : SUCCESS in 34m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81378d5459e498d89cd82339ceedc06 : SUCCESS in 14m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1638f312a81c4043acc30646837efb96 : SUCCESS in 1h 40m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/37c6aa5af96246b39cd6fe2af35163d7 : SUCCESS in 1h 08m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3f1f6dd169654cd2bbe319c5ad641996 : SUCCESS in 1h 05m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4af0b5319824e1a89326742c16e8490 : SUCCESS in 1h 39m 32s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/10ccfc8747824c3390344dbe096f5c7c : SUCCESS in 33m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/71d916d264904f73b4eebe9833f85eca : SUCCESS in 36m 20s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/20cb4b4288c84d4e917e82dac72c18d9 : SUCCESS in 42m 26s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/37f9ab225b56487ea6a6a37952d31977 : SUCCESS in 41m 46s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f1cec83ce9814205a55a5b42f94acb32 : SUCCESS in 18m 12s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0f0c8e83b4d848dfb63d3d8de70a48f7 : SUCCESS in 1h 32m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a62542bf543a4416b8de5af906ba3ccf : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":11},{"id":"89ccbd184fe52b0abfaf3efeca0f5fc4975524e5","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-13 00:51:02.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"1ce073cf175d236896412173cebf5e4136c6f574","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-13 13:30:35.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"85972e3b5c238beaf13052bc91781a5667feaa24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 13:31:06.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"895eedc09ae5451594636cd1b9e38402903db20e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 15:24:58.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41edac2d00f54541acb3cc88772d434b : SUCCESS in 45m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a5ac995740a4ef9b69441956fcd63e6 : SUCCESS in 6m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a20f72e865814a55a4bc255aceebd01c : SUCCESS in 42m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c494a2c23a94b11be715ccf5620b0b3 : SUCCESS in 38m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25783a0a62fa4faea0b54b197c688bb9 : SUCCESS in 37m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9873ab64a73548b29b74416490f74fe5 : SUCCESS in 13m 39s\n- tempest-full https://zuul.opendev.org/t/openstack/build/865bbf65ea53410b96ae1f53fcc05d72 : SUCCESS in 1h 44m 32s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/79a8b1d58d744e73b18960b43f93610a : SUCCESS in 1h 02m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/73b7eaa9a1d8478b9b55b385853c4459 : SUCCESS in 59m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b4c76bc27a2f435cbd30731bced3d1eb : FAILURE in 1h 12m 16s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/53db828e63654a949eeb14c184de6f6e : SUCCESS in 33m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0a27a596883e4ff487cf21f2d1ddbe8c : SUCCESS in 33m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3f457bd6531441894c3b76a12081345 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":11},{"id":"ec7d1370a50f359f4281e6bd8a3edf6a62ed6a1d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-13 15:32:52.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"5012cb52602116c486b9d21ccc1c5f2a3891fa87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:33:22.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39108410883d47fd8e1337bbc792bffd : SUCCESS in 40m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa7a32418e124de2b6221fc66bbfb5c9 : SUCCESS in 46m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8bc0c7128b84752a848aab6c55e9b86 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf46a86eb8b7404e87ee650841d5baa8 : SUCCESS in 48m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/197697ca4d91442091eea0a2c96e6fb0 : SUCCESS in 36m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24268aebb91a489597c9adc440297fcf : SUCCESS in 38m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d65530f40c44b7cb3e4e4a2fc5f4698 : SUCCESS in 14m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6c54c92011f7444fa4c7f5a8bd9c0120 : SUCCESS in 1h 32m 00s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b4d2c1b21d544ad3b6e140d021d2a704 : SUCCESS in 1h 02m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/06a6ff97cb474c5b96915f6a5e588128 : SUCCESS in 1h 02m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e87c4e3db7f47cc812200c10bea8f2b : SUCCESS in 1h 35m 39s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a2b6f5711db482f9720c3dd295b9203 : SUCCESS in 31m 13s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0fc970f348a8406bbb720488fa0cf9cc : SUCCESS in 38m 05s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/bd70e78497ae485889b40ddb9927faef : SUCCESS in 36m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/81b754a23927473597e59a7b9b2c5aa4 : SUCCESS in 41m 53s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ebb0d2f9f26e4c508e198456b75f08d3 : SUCCESS in 19m 25s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ea524b960d3a4fc9ba845930d5725ce2 : SUCCESS in 1h 43m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12078788f4164030a1e5129c7060e50b : SUCCESS in 1h 08m 47s","accounts_in_message":[],"_revision_number":11},{"id":"5388b627eb86a10bf00ba5f583620bfebd07cb25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:34:13.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"a85a64f1315a76b0b24976cc2993afac629ca946","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 08:10:35.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e734e2867812485782241069bd166697 : FAILURE in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bb9b7ff0f3a4e31add27690c5b4686b : SUCCESS in 8m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02d70c62379f46f9aa1fd15b091e286a : SUCCESS in 51m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcf28e9960f94fc49cd31317cf58904c : SUCCESS in 46m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d0a00f7a8084f74a206396fe7553271 : TIMED_OUT in 1h 30m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66816b5ca85d4629b2f6fb368595284b : RETRY_LIMIT in 10m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a23f37f29a4944f09d3a50d58b46e0ad : SUCCESS in 1h 52m 29s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/40132e12004f467bbd5c68950bf7f986 : SUCCESS in 1h 30m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5e85b1beb1404cf0b5f3cd1093a7adae : FAILURE in 44m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/032cd10bd2e247e8a1b12ab81738bcc9 : SUCCESS in 1h 47m 59s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/f41aa42d43bf46c4b1de5bf9856e2de1 : SUCCESS in 30m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0a11b14447ac4b07a737cde0598e84ec : SUCCESS in 45m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3bc4e82696947cf830659059d13f753 : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":11},{"id":"8149351a952720cf73fe5e3ec7bb0fe449b4056e","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-14 08:30:54.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"335dd8994535ff371367c8c3a902852ccb0f94ee","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-14 08:59:15.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"110ce366933df185c6d37bcfa0496609c2132571","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-14 10:08:22.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"a2863fff017ed1b6e57a4da2ab9a74ee7b77f226","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 11:11:11.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9eb90d541c243ccafa5d3103c05a062 : SUCCESS in 1h 04m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af9b4d5ae3c245b6880e0659c3992d0c : SUCCESS in 52m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d0b94689afc4176989399120449c0c6 : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4d8fc53dc3d148e398fe213dca0c219e : SUCCESS in 51m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e42bddf17133476a8565d4295a216871 : SUCCESS in 33m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c595b6743b74ee58ec3cfce8c7c4224 : SUCCESS in 43m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bee28cbe57a402fb150d9b61bc9bb67 : SUCCESS in 16m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/41bc61bbb8b7464bb73156e8e0abf1dc : SUCCESS in 1h 36m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b46d526d424d47ecae3a27e439d8cdf6 : SUCCESS in 1h 00m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fc1248d7ac76411e865a47b926110d6a : SUCCESS in 1h 03m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e6463dd104b405ebe808c2663566078 : SUCCESS in 1h 29m 50s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e419c3f506346788e10ece1c7f7f3f2 : SUCCESS in 39m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/69d024402cee468ab3b2fcdbb26faf70 : SUCCESS in 37m 05s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/3d40d9c1f277469c86080e06aad95e7e : SUCCESS in 40m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/7ce7761196bc4928b1e74eef18d74c9b : SUCCESS in 48m 57s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2d6cdf1b68904425be41b7c23ed627cf : SUCCESS in 18m 43s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/caf16ced57f64bfcbc730aa0130fad6f : SUCCESS in 1h 44m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af784c04181e421e98ace08790e9a088 : SUCCESS in 1h 06m 27s","accounts_in_message":[],"_revision_number":12},{"id":"854e69b8b42259e57905f2db45302880f1807b0d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:29:10.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"dbbee7efa439b35ff477f39aeae61bcc72b792f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 21:49:06.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/017dbe01c9e640a4a25342c44a3b06f3 : SUCCESS in 53m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b18451b7869a410b886314d485e0cd9f : SUCCESS in 7m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e22d7d1621b54bc186f96fe7744fc132 : SUCCESS in 50m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e737b2eea526421e949f019b81078b1c : SUCCESS in 39m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6aad4a142c874c66a3efdcb45e557ca8 : SUCCESS in 58m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1bf87798892414bb84e6e2ae63658d0 : SUCCESS in 15m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0654bbaa81d3466aa017f4740432e5e8 : POST_FAILURE in 1h 43m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3443640e24a04024b703fc6819b8926d : SUCCESS in 59m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a27c13a59c0e48d6917360a9efbac8c7 : SUCCESS in 1h 07m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34d6dfe9b2844156870eb3441d1bff5a : SUCCESS in 1h 24m 34s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/a4f2461a79a24610b4c83e03b6ad7ace : SUCCESS in 33m 06s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3d08b863bd8a40b79e762f0e1c8aabef : SUCCESS in 40m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b348f434228d4bc79f7bf6bddbf23027 : SUCCESS in 1h 26m 37s","accounts_in_message":[],"_revision_number":12},{"id":"41bacd19b46fc2a266a5894bd6d84bc6eaf06c70","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-14 22:14:51.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"dbde66fb7c1f5dc36796a2f6649542393727febf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 23:58:30.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffa7060b6668447e98e670b1ea3ce3d0 : SUCCESS in 1h 01m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09c2761f4596477dbfa4d855e362a170 : SUCCESS in 1h 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/905ce2a0480e46508c32f089e88e3ad0 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0ee65340d50461aa8b1d3496b017b03 : FAILURE in 53m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47e0f7b15b5c4010bdaa75352ad0f5b6 : TIMED_OUT in 1h 30m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b0a549a8e5245b9a6d5027a027d6920 : TIMED_OUT in 1h 30m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78f7c8528883434787a25ecd467f7972 : SUCCESS in 11m 42s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ed94cf13d63c40cfac0c05d62e5845a8 : SUCCESS in 1h 40m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0a385580a6dd401d8709aabdd596f40f : SUCCESS in 1h 02m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ad3705779bae47329cfc4cf20d8a9a9b : SUCCESS in 1h 06m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19cbf6a30313406894879b7016088bf0 : SUCCESS in 1h 18m 08s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/396f18d99fea45b2b18ef289e9ba7dc8 : SUCCESS in 27m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/79ec8ef6d5314bc78ad8eb949365e54f : SUCCESS in 34m 39s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/fca42f25e8104d859a93553d07fbd5dd : SUCCESS in 41m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/02127d4eeb9b430f9990b1b67b149de5 : SUCCESS in 43m 11s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f916951a220e4fa2ba8683ed9d15da35 : SUCCESS in 17m 42s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/085f77953dd5438a9a1d7519f69b9fa4 : SUCCESS in 1h 37m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cf446d2a18c4316a647ff1bf986b731 : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":12},{"id":"d243818d64b1e4f57c5de75710f4521c3a1b0d17","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-15 00:27:20.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"9c0bc0db240e0b6cbc51fd2d82e992dd484a5222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 02:38:53.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/940baf3772284d3abd02f7473e45cd88 : SUCCESS in 1h 19m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/951cb7ebb9d74fefbf638567ff7b0798 : SUCCESS in 1h 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a37ab19118f4702912ace6fbb0e61f4 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a8909e27ae04721b548ea61bea656fa : SUCCESS in 1h 13m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4d79e53c73b41b38ebb1fe4b9e05d46 : SUCCESS in 1h 10m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61b13381872043608a96303c97110099 : SUCCESS in 1h 13m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26f77188b1254a26b5710c3715960d73 : SUCCESS in 15m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fcc5a5ec811042dabf9c8c7b9bd2ed9d : SUCCESS in 2h 07m 24s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5db57a3f57ff41f3b1d8cef23c0bdee4 : SUCCESS in 1h 03m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a98696ca5f304c088d7ae26d7f797dea : SUCCESS in 1h 04m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6011faa2165740658973cc8d308209de : SUCCESS in 1h 45m 03s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd904394e1d248d59a05e664dc25b628 : SUCCESS in 40m 56s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/27db0d48b27a4e1db8e8abbc6966364f : SUCCESS in 32m 08s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/d2f3a131592b43b992cc58ca3d8246df : SUCCESS in 37m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/c14a4b7fbe364504b821e99219c32211 : SUCCESS in 50m 47s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9a10cf4bf9ed4cd597c3ae9e6b12d98a : SUCCESS in 15m 00s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1485149a16b541d88516c2ba8c9fc4d3 : SUCCESS in 1h 39m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/881d4a77e75f411183821f5608d7ced0 : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":12},{"id":"0425903febf3bb717cdc93db8881bff14b292059","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 06:35:25.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"f0d7e6f5868aec6f6c50ea09b65a80c747a3788a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 09:19:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"baea718dae5a281e9450017f26e7553d5d4feb7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 09:19:58.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/804878bc42ec4de3b0e6436c29b10f7b : SUCCESS in 55m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bccd4e58870e4cc3bdc9328011bbe7e8 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d5abbef4f6b4690b3a4dba2deed191b : SUCCESS in 48m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a0cca853be14e2994db4668b5eac40f : SUCCESS in 45m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96da0a3d06734d50a054d54f84ba5804 : SUCCESS in 40m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f8b0c15bbe48f692fae381f137f1f5 : SUCCESS in 18m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f164df342b424bff8b445bb3b2b06828 : SUCCESS in 1h 51m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/23d6d8fbb0e6447da27d2ef72d6df099 : SUCCESS in 55m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ef1809d339d34ba397db1cd7f9264c0b : SUCCESS in 1h 08m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41e6124ab68a4559b20e588255423f3d : SUCCESS in 1h 37m 34s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/f4ef08fd8f714a278dfbb585cd187ede : SUCCESS in 28m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8bfcb0f698af473586938fe288a264f5 : SUCCESS in 34m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1087e074d7f4f549a250b3f87aaef5c : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":12},{"id":"0abebac1dc4e5e6d2dca2b2e9649208fb7912e06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 09:21:44.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4598358cb7664d2fa8c63636555758b5 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":12},{"id":"f983121c1604bdfa65661b2003766a9b9c7de883","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-16 21:08:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":12,"current_revision":"d009384c9b683008d572f9996e1fe4e5e5d82096","revisions":{"4d8bc6ed6b4a35140e30ed2833c85718980e679c":{"kind":"REWORK","_number":1,"created":"2018-10-04 07:15:27.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/1"}}},"commit":{"parents":[{"commit":"31333131e0fed2482984d1a6a7c54efd90bc59d0","subject":"Merge \"Properly replace flask view args in links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/31333131e0fed2482984d1a6a7c54efd90bc59d0"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"subject":"Implement scope_type checking for ec2 credentials","message":"Implement scope_type checking for ec2 credentials\n\nThis change modifies the policies for the ec2\ncredentials API to be more self-service by properly\nchecking for various scopes.\n\nFurther patch will be updated for test cases for\nthe default roles keystone supports at install time.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nCloses-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4d8bc6ed6b4a35140e30ed2833c85718980e679c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4d8bc6ed6b4a35140e30ed2833c85718980e679c"}]},"branch":"refs/heads/master"},"1a89d9b84b174d7b5920e4659c574d9aa90fc0d4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-10-04 07:17:10.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/2"}}},"commit":{"parents":[{"commit":"31333131e0fed2482984d1a6a7c54efd90bc59d0","subject":"Merge \"Properly replace flask view args in links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/31333131e0fed2482984d1a6a7c54efd90bc59d0"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:17:00.000000000","tz":0},"subject":"[WIP] Implement scope_type checking for ec2 credentials","message":"[WIP] Implement scope_type checking for ec2 credentials\n\nThis change modifies the policies for the ec2\ncredentials API to be more self-service by properly\nchecking for various scopes.\n\nFurther patch will be updated for test cases for\nthe default roles keystone supports at install time.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nCloses-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1a89d9b84b174d7b5920e4659c574d9aa90fc0d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1a89d9b84b174d7b5920e4659c574d9aa90fc0d4"}]},"branch":"refs/heads/master"},"88ab40d388057e49738b60d15f8e53b7adabaa0e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-10-04 07:17:58.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/3"}}},"commit":{"parents":[{"commit":"4001fc6c2a858223d10213f4c87009ac7a66bc1f","subject":"Implement scope_type checking for credentials","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4001fc6c2a858223d10213f4c87009ac7a66bc1f"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:17:58.000000000","tz":0},"subject":"[WIP] Implement scope_type checking for ec2 credentials","message":"[WIP] Implement scope_type checking for ec2 credentials\n\nThis change modifies the policies for the ec2\ncredentials API to be more self-service by properly\nchecking for various scopes.\n\nFurther patch will be updated for test cases for\nthe default roles keystone supports at install time.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nCloses-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/88ab40d388057e49738b60d15f8e53b7adabaa0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/88ab40d388057e49738b60d15f8e53b7adabaa0e"}]},"branch":"refs/heads/master"},"7d681af4b5912dd0cc717351a5c7a1ebcff9578d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-11-05 02:41:35.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/4"}}},"commit":{"parents":[{"commit":"733b37f24d874193b965528bacf1fd56ccffbc79","subject":"Documentation fix - Port number","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/733b37f24d874193b965528bacf1fd56ccffbc79"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-11-05 02:41:35.000000000","tz":0},"subject":"[WIP] Implement scope_type checking for ec2 credentials","message":"[WIP] Implement scope_type checking for ec2 credentials\n\nThis change modifies the policies for the ec2\ncredentials API to be more self-service by properly\nchecking for various scopes.\n\nFurther patch will be updated for test cases for\nthe default roles keystone supports at install time.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nCloses-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7d681af4b5912dd0cc717351a5c7a1ebcff9578d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7d681af4b5912dd0cc717351a5c7a1ebcff9578d"}]},"branch":"refs/heads/master"},"3d5640da8720a0b26366cb1034ccc846476d0a60":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-09-08 15:05:04.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/5"}}},"commit":{"parents":[{"commit":"6aebf179b8b1b48909ea57666f61bbc080bd6888","subject":"Add tests for system member for trusts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6aebf179b8b1b48909ea57666f61bbc080bd6888"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-09-08 15:05:04.000000000","tz":0},"subject":"[WIP] Implement scope_type checking for ec2 credentials","message":"[WIP] Implement scope_type checking for ec2 credentials\n\nThis change modifies the policies for the ec2\ncredentials API to be more self-service by properly\nchecking for various scopes.\n\nFurther patch will be updated for test cases for\nthe default roles keystone supports at install time.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nCloses-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3d5640da8720a0b26366cb1034ccc846476d0a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3d5640da8720a0b26366cb1034ccc846476d0a60"}]},"branch":"refs/heads/master"},"20afd6f85c5ba7518aca12a8fa2594ea13c20a63":{"kind":"REWORK","_number":6,"created":"2019-09-09 10:59:56.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/6"}}},"commit":{"parents":[{"commit":"6aebf179b8b1b48909ea57666f61bbc080bd6888","subject":"Add tests for system member for trusts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6aebf179b8b1b48909ea57666f61bbc080bd6888"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-09-09 10:56:52.000000000","tz":330},"subject":"Implement system reader and member for EC2 credentials","message":"Implement system reader and member for EC2 credentials\n\nThis change updates the EC2 credentials policies to understand\nthe system reader and member role. A follow-on change will add\nsupport for system admin.\nFor the time being, we\u0027re deferring adding support for domain\nand project users to create EC2 credentials but may add it in\nthe future.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nPartial-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/20afd6f85c5ba7518aca12a8fa2594ea13c20a63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/20afd6f85c5ba7518aca12a8fa2594ea13c20a63"}]},"branch":"refs/heads/master"},"2948b72504e759f770c2c406b022cab4d8f1c995":{"kind":"REWORK","_number":7,"created":"2019-09-10 05:54:28.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/7"}}},"commit":{"parents":[{"commit":"6aebf179b8b1b48909ea57666f61bbc080bd6888","subject":"Add tests for system member for trusts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6aebf179b8b1b48909ea57666f61bbc080bd6888"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-09-10 05:32:55.000000000","tz":330},"subject":"Implement scope type checking for EC2 credentials","message":"Implement scope type checking for EC2 credentials\n\nThis change updates the EC2 credentials policies to understand\nthe scope types for EC2 credentials. A follow on patch will\nRemove obsolete credential policies.\n\nTo maintain the compatibility with the old rule the\nequivalent ec2_list_credentials and ec2_get_credentials behaves\ninconsistently. Same for ec2_create_credentials and the\nec2_delete_inconsistently.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nPartial-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2948b72504e759f770c2c406b022cab4d8f1c995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2948b72504e759f770c2c406b022cab4d8f1c995"}]},"branch":"refs/heads/master"},"d9864af2417488edf51ea5b4ca721f54dd289c0d":{"kind":"REWORK","_number":8,"created":"2019-09-10 16:27:02.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/8"}}},"commit":{"parents":[{"commit":"6aebf179b8b1b48909ea57666f61bbc080bd6888","subject":"Add tests for system member for trusts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6aebf179b8b1b48909ea57666f61bbc080bd6888"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-09-10 16:23:25.000000000","tz":330},"subject":"Implement scope type checking for EC2 credentials","message":"Implement scope type checking for EC2 credentials\n\nThis change updates the EC2 credentials policies to understand\nthe scope types for EC2 credentials. A follow on patch will\nRemove obsolete credential policies.\n\nTo maintain the compatibility with the old rule the\nequivalent ec2_list_credentials and ec2_get_credentials behaves\ninconsistently. Same for ec2_create_credentials and the\nec2_delete_inconsistently.\n\nAlso removing the test of ec2 credentials from test_v3_protection\nsince new protection file is being added.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nPartial-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d9864af2417488edf51ea5b4ca721f54dd289c0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d9864af2417488edf51ea5b4ca721f54dd289c0d"}]},"branch":"refs/heads/master"},"6ce03367fe2272bfc89e3f7afe2418c216320a48":{"kind":"REWORK","_number":9,"created":"2019-09-11 07:29:04.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/9"}}},"commit":{"parents":[{"commit":"6aebf179b8b1b48909ea57666f61bbc080bd6888","subject":"Add tests for system member for trusts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6aebf179b8b1b48909ea57666f61bbc080bd6888"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-09-11 07:28:16.000000000","tz":330},"subject":"Implement scope type checking for EC2 credentials","message":"Implement scope type checking for EC2 credentials\n\nThis change updates the EC2 credentials policies to understand\nthe scope types for EC2 credentials. A follow on patch will\nRemove obsolete credential policies.\n\nTo maintain the compatibility with the old rule the\nequivalent ec2_list_credentials and ec2_get_credentials behaves\ninconsistently. Same for ec2_create_credentials and the\nec2_delete_inconsistently.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nPartial-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6ce03367fe2272bfc89e3f7afe2418c216320a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6ce03367fe2272bfc89e3f7afe2418c216320a48"}]},"branch":"refs/heads/master"},"9ad31b2650b37df30db3ccb9e4a9bb173516f6d0":{"kind":"REWORK","_number":10,"created":"2019-09-11 18:59:36.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/20/607820/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/10"}}},"commit":{"parents":[{"commit":"6aebf179b8b1b48909ea57666f61bbc080bd6888","subject":"Add tests for system member for trusts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6aebf179b8b1b48909ea57666f61bbc080bd6888"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-11 18:59:30.000000000","tz":-420},"subject":"Implement scope type checking for EC2 credentials","message":"Implement scope type checking for EC2 credentials\n\nThis change updates the EC2 credentials policies to understand\nthe scope types for EC2 credentials. A follow on patch will\nRemove obsolete credential policies.\n\nTo maintain the compatibility with the old rule the\nequivalent ec2_list_credentials and ec2_get_credentials behaves\ninconsistently. Same for ec2_create_credentials and the\nec2_delete_inconsistently.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nPartial-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ad31b2650b37df30db3ccb9e4a9bb173516f6d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ad31b2650b37df30db3ccb9e4a9bb173516f6d0"}]},"branch":"refs/heads/master"},"cf2b02829d8ea809f3180de62d5e36e189c92df1":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-09-11 19:01:28.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/20/607820/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/11"}}},"commit":{"parents":[{"commit":"a73e057e253e22c6f9ca7adcae6d23c7a102cf15","subject":"Fix timeout Zuul changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a73e057e253e22c6f9ca7adcae6d23c7a102cf15"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-11 19:01:00.000000000","tz":-420},"subject":"Implement scope type checking for EC2 credentials","message":"Implement scope type checking for EC2 credentials\n\nThis change updates the EC2 credentials policies to understand\nthe scope types for EC2 credentials. A follow on patch will\nRemove obsolete credential policies.\n\nTo maintain the compatibility with the old rule the\nequivalent ec2_list_credentials and ec2_get_credentials behaves\ninconsistently. Same for ec2_create_credentials and the\nec2_delete_inconsistently.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nPartial-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cf2b02829d8ea809f3180de62d5e36e189c92df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cf2b02829d8ea809f3180de62d5e36e189c92df1"}]},"branch":"refs/heads/master"},"d009384c9b683008d572f9996e1fe4e5e5d82096":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-09-14 08:59:15.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/20/607820/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/20/607820/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/20/607820/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/20/607820/12"}}},"commit":{"parents":[{"commit":"f3d5dfe323e75d20a38871892769c6f43359ef50","subject":"Merge \"Implement system scope for domain role management\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f3d5dfe323e75d20a38871892769c6f43359ef50"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2018-10-04 07:09:32.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-09-13 08:28:32.000000000","tz":330},"subject":"Implement scope type checking for EC2 credentials","message":"Implement scope type checking for EC2 credentials\n\nThis change updates the EC2 credentials policies to understand\nthe scope types for EC2 credentials. A follow on patch will\nRemove obsolete credential policies.\n\nTo maintain the compatibility with the old rule the\nequivalent ec2_list_credentials and ec2_get_credentials behaves\ninconsistently. Same for ec2_create_credentials and the\nec2_delete_inconsistently.\n\nChange-Id: I090e2470726d22b2670a2cca89025063419f5262\nPartial-Bug: #1750678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d009384c9b683008d572f9996e1fe4e5e5d82096"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d009384c9b683008d572f9996e1fe4e5e5d82096"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
