)]}'
{"id":"openstack%2Fkeystone~612625","triplet_id":"openstack%2Fkeystone~master~I6ccf470ef41c858c9aac90c8ce03dcfdb5468b82","project":"openstack/keystone","branch":"master","topic":"nit-fix","hashtags":[],"change_id":"I6ccf470ef41c858c9aac90c8ce03dcfdb5468b82","subject":"nit: remove some useless code","status":"MERGED","created":"2018-10-23 08:08:40.000000000","updated":"2019-07-09 05:18:55.000000000","submitted":"2019-07-09 05:18:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"612625-1562649535972-83a8d452","meta_rev_id":"3d3aa4fb43f9185618b1698a0f359451c5363847","_number":612625,"virtual_id_number":612625,"owner":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-07-09 05:18:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-07-08 20:39:52.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-07-08 19:41:16.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-08 19:41:16.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":1,"date":"2019-07-08 20:39:52.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-10-23 08:56:32.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2018-10-29 03:35:11.000000000","updated_by":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"reviewer":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"state":"REVIEWER"},{"updated":"2018-11-08 06:23:11.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2018-11-13 15:29:23.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-07-08 20:39:52.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-07-09 05:18:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"176d1511aea8f251d864c3fa84180fe3eaf6179e","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-10-23 08:08:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d0414ec4db079bbf24fdf8b16d07b6bd7bb2154","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-10-23 08:09:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"995656a423baea5d2255486075ff51d149a831c0","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-23 08:56:32.000000000","message":"Patch Set 2: Code-Review-1\n\nCan we also update the docs about removal of these cache backends, in these files also.\n\n[1] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/contributor/caching-layer.rst#n104\n\n[2] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/admin/identity-caching-layer.rst#n82","accounts_in_message":[],"_revision_number":2},{"id":"b2a776f414a326492cc1037db6628fa60ad22633","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-23 08:59:17.000000000","message":"Patch Set 2:\n\n\u003e Can we also remove/clear the docs where these cache backends \n \u003e are mentioned. Like [1] and [2]\n \u003e \n \u003e [1] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/contributor/caching-layer.rst#n104\n \u003e \n \u003e [2] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/admin/identity-caching-layer.rst#n82","accounts_in_message":[],"_revision_number":2},{"id":"e34f51c2c03cf7156e9ac2623ff0b9f72ee091b1","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-10-23 09:13:47.000000000","message":"Patch Set 2:\n\n\u003e Can we also update the docs about removal of these cache backends,\n \u003e in these files also.\n \u003e \n \u003e [1] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/contributor/caching-layer.rst#n104\n \u003e \n \u003e [2] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/admin/identity-caching-layer.rst#n82\n\nGood call.\n\nThe first guide is totally out of date now. I just created a bug for tracing it. https://bugs.launchpad.net/keystone/+bug/1799384\n\nThe second one is correct. Now Keystone use oslo.cache for cache backend setting.","accounts_in_message":[],"_revision_number":2},{"id":"dbc0e22f6481c5d3382dde18a46c2e617bb493a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-23 10:06:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/612625/2/check/openstack-tox-cover/e474716/cover/ : SUCCESS in 11m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/612625/2/check/openstack-tox-lower-constraints/917404f/ : SUCCESS in 12m 55s\n- openstack-tox-pep8 http://logs.openstack.org/25/612625/2/check/openstack-tox-pep8/70c9c3d/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/25/612625/2/check/openstack-tox-py27/8951246/ : SUCCESS in 9m 36s\n- openstack-tox-py35 http://logs.openstack.org/25/612625/2/check/openstack-tox-py35/20d7bfd/ : SUCCESS in 9m 47s\n- openstack-tox-py36 http://logs.openstack.org/25/612625/2/check/openstack-tox-py36/7d1852d/ : SUCCESS in 8m 41s\n- openstack-tox-docs http://logs.openstack.org/25/612625/2/check/openstack-tox-docs/61c4a06/html/ : SUCCESS in 6m 44s\n- tempest-full http://logs.openstack.org/25/612625/2/check/tempest-full/9685ddc/ : SUCCESS in 1h 28m 59s\n- neutron-grenade http://logs.openstack.org/25/612625/2/check/neutron-grenade/c946d08/ : SUCCESS in 56m 36s\n- tempest-full-py3 http://logs.openstack.org/25/612625/2/check/tempest-full-py3/e82fde5/ : SUCCESS in 1h 22m 35s\n- keystone-dsvm-functional http://logs.openstack.org/25/612625/2/check/keystone-dsvm-functional/3889317/ : SUCCESS in 26m 14s\n- keystone-dsvm-py3-functional http://logs.openstack.org/25/612625/2/check/keystone-dsvm-py3-functional/beb1ca1/ : SUCCESS in 33m 26s\n- keystone-dsvm-functional-federation http://logs.openstack.org/25/612625/2/check/keystone-dsvm-functional-federation/84a0359/ : FAILURE in 28m 36s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/25/612625/2/check/keystone-dsvm-py35-functional-federation/40fa84a/ : FAILURE in 28m 44s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/25/612625/2/check/keystoneclient-devstack-functional/480ef2d/ : SUCCESS in 26m 34s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/25/612625/2/check/legacy-tempest-dsvm-ldap-domain-specific-driver/5caa64b/ : SUCCESS in 1h 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"14cef3c30cb8a6f35b97f3099c4ce59334d4cab2","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-10-24 01:33:22.000000000","message":"Patch Set 2:\n\n\u003e \u003e Can we also update the docs about removal of these cache\n \u003e backends,\n \u003e \u003e in these files also.\n \u003e \u003e\n \u003e \u003e [1] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/contributor/caching-layer.rst#n104\n \u003e \u003e\n \u003e \u003e [2] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/admin/identity-caching-layer.rst#n82\n \u003e \n \u003e Good call.\n \u003e \n \u003e The first guide is totally out of date now. I just created a bug\n \u003e for tracing it. https://bugs.launchpad.net/keystone/+bug/1799384\n \u003e \n \u003e The second one is correct. Now Keystone use oslo.cache for cache\n \u003e backend setting.\n\nHi wangxiyuan, In second one there are still some places where the removed backend are still mentioned. Pl have a look\n\nhttp://git.openstack.org/cgit/openstack/keystone/tree/doc/source/admin/identity-caching-layer.rst#n185","accounts_in_message":[],"_revision_number":2},{"id":"49d94514793031b0c8325b0072fd7e07999e1e57","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-10-24 06:44:44.000000000","message":"Patch Set 2:\n\n\u003e \u003e \u003e Can we also update the docs about removal of these cache\n \u003e \u003e backends,\n \u003e \u003e \u003e in these files also.\n \u003e \u003e \u003e\n \u003e \u003e \u003e [1] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/contributor/caching-layer.rst#n104\n \u003e \u003e \u003e\n \u003e \u003e \u003e [2] http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/admin/identity-caching-layer.rst#n82\n \u003e \u003e\n \u003e \u003e Good call.\n \u003e \u003e\n \u003e \u003e The first guide is totally out of date now. I just created a bug\n \u003e \u003e for tracing it. https://bugs.launchpad.net/keystone/+bug/1799384\n \u003e \u003e\n \u003e \u003e The second one is correct. Now Keystone use oslo.cache for cache\n \u003e \u003e backend setting.\n \u003e \n \u003e Hi wangxiyuan, In second one there are still some places where the\n \u003e removed backend are still mentioned. Pl have a look\n \u003e \n \u003e http://git.openstack.org/cgit/openstack/keystone/tree/doc/source/admin/identity-caching-layer.rst#n185\n\nThis is how Keystone working now. These options are from oslo.cache.","accounts_in_message":[],"_revision_number":2},{"id":"06f17c7aa0354a58e9f2e29a67eafd0caa89c5c5","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-10-25 03:37:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"405019ef0fe9ad89e967f92466398503c414428b","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-10-25 03:44:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e3bc56f1aadca5a68adaf61334d22c80dede5a03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 05:52:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/25/612625/4/check/openstack-tox-cover/4bc9297/cover/ : SUCCESS in 11m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/612625/4/check/openstack-tox-lower-constraints/c234d87/ : SUCCESS in 27m 07s\n- openstack-tox-pep8 http://logs.openstack.org/25/612625/4/check/openstack-tox-pep8/07daaa8/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/25/612625/4/check/openstack-tox-py27/326bfc1/ : SUCCESS in 9m 29s\n- openstack-tox-py35 http://logs.openstack.org/25/612625/4/check/openstack-tox-py35/43a2074/ : SUCCESS in 9m 51s\n- openstack-tox-py36 http://logs.openstack.org/25/612625/4/check/openstack-tox-py36/178d318/ : SUCCESS in 8m 13s\n- openstack-tox-docs http://logs.openstack.org/25/612625/4/check/openstack-tox-docs/0d0d4b0/html/ : SUCCESS in 5m 16s\n- tempest-full http://logs.openstack.org/25/612625/4/check/tempest-full/62dc349/ : TIMED_OUT in 2h 07m 03s\n- neutron-grenade http://logs.openstack.org/25/612625/4/check/neutron-grenade/73a0532/ : SUCCESS in 1h 12m 32s\n- tempest-full-py3 http://logs.openstack.org/25/612625/4/check/tempest-full-py3/2c9f246/ : SUCCESS in 1h 29m 28s\n- keystone-dsvm-functional http://logs.openstack.org/25/612625/4/check/keystone-dsvm-functional/39f01f0/ : SUCCESS in 24m 50s\n- keystone-dsvm-py3-functional http://logs.openstack.org/25/612625/4/check/keystone-dsvm-py3-functional/1900f69/ : SUCCESS in 30m 13s\n- keystone-dsvm-functional-federation http://logs.openstack.org/25/612625/4/check/keystone-dsvm-functional-federation/51a6bf3/ : FAILURE in 28m 24s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/25/612625/4/check/keystone-dsvm-py35-functional-federation/5d42e18/ : FAILURE in 32m 33s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/25/612625/4/check/keystoneclient-devstack-functional/f24bfd3/ : SUCCESS in 29m 41s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/25/612625/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/54ba775/ : SUCCESS in 1h 36m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"566096953c655340ff5290fa8b200eddadb8dd30","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-10-25 06:10:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bd55767a4d57c894d67472e8fb38732a4851178b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-25 08:30:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/612625/4/check/openstack-tox-cover/beabb08/cover/ : SUCCESS in 11m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/612625/4/check/openstack-tox-lower-constraints/d2cde05/ : SUCCESS in 15m 08s\n- openstack-tox-pep8 http://logs.openstack.org/25/612625/4/check/openstack-tox-pep8/b485ba5/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/25/612625/4/check/openstack-tox-py27/c7736c3/ : SUCCESS in 11m 41s\n- openstack-tox-py35 http://logs.openstack.org/25/612625/4/check/openstack-tox-py35/c94706e/ : SUCCESS in 10m 46s\n- openstack-tox-py36 http://logs.openstack.org/25/612625/4/check/openstack-tox-py36/8cf5439/ : SUCCESS in 9m 52s\n- openstack-tox-docs http://logs.openstack.org/25/612625/4/check/openstack-tox-docs/e4d7040/html/ : SUCCESS in 6m 24s\n- tempest-full http://logs.openstack.org/25/612625/4/check/tempest-full/31d029f/ : SUCCESS in 1h 41m 01s\n- neutron-grenade http://logs.openstack.org/25/612625/4/check/neutron-grenade/fe60e26/ : SUCCESS in 50m 32s\n- tempest-full-py3 http://logs.openstack.org/25/612625/4/check/tempest-full-py3/4a6f108/ : SUCCESS in 1h 28m 39s\n- keystone-dsvm-functional http://logs.openstack.org/25/612625/4/check/keystone-dsvm-functional/05f0a03/ : SUCCESS in 32m 00s\n- keystone-dsvm-py3-functional http://logs.openstack.org/25/612625/4/check/keystone-dsvm-py3-functional/6d3badd/ : SUCCESS in 30m 06s\n- keystone-dsvm-functional-federation http://logs.openstack.org/25/612625/4/check/keystone-dsvm-functional-federation/134189a/ : FAILURE in 36m 46s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/25/612625/4/check/keystone-dsvm-py35-functional-federation/1cc0471/ : FAILURE in 38m 00s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/25/612625/4/check/keystoneclient-devstack-functional/432e5dc/ : SUCCESS in 32m 00s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/25/612625/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/6bf6aff/ : SUCCESS in 1h 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4d5654bf8e2ae56e3098148b899d30d4bc3f4ac7","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-10-28 10:52:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ec1655dc984cf3a45641b10cfa679e93362253c0","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-10-29 01:14:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"19609e15f903d9fd1559376aefe75797d1e2d9f9","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-10-29 03:35:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0c52750b71adbb9fb6c8bfcfac6c256dbaa2aa5b","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-10-29 11:17:12.000000000","message":"Patch Set 4: Code-Review+2\n\nNo guidance, I remember there were concerns.","accounts_in_message":[],"_revision_number":4},{"id":"6b4b28340805660c3a73545c9269b0b98a35f466","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2018-10-29 17:02:09.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nWe may need to deprecate setup_remote_pydev_debug, see Lance\u0027s comment. If that is not the case then sure.","accounts_in_message":[],"_revision_number":4},{"id":"3bfedd39f9db4fffaef6bf369c53e8cc1a620adf","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-10-30 01:02:33.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n \u003e \n \u003e We may need to deprecate setup_remote_pydev_debug, see Lance\u0027s\n \u003e comment. If that is not the case then sure.\n\nah, yeah. The patch for deprecation: https://review.openstack.org/#/c/568764","accounts_in_message":[],"_revision_number":4},{"id":"4a7b8762e834260dbdcf260a907600ed2450b83e","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-11-08 01:31:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"941885e5f244f7b533870842de2eb0fec6f3075b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-08 03:15:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/612625/5/check/openstack-tox-cover/9b4f690/cover/ : SUCCESS in 16m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/612625/5/check/openstack-tox-lower-constraints/6ede17e/ : SUCCESS in 23m 20s\n- openstack-tox-pep8 http://logs.openstack.org/25/612625/5/check/openstack-tox-pep8/80534c3/ : SUCCESS in 5m 01s\n- openstack-tox-py27 http://logs.openstack.org/25/612625/5/check/openstack-tox-py27/74e12e0/ : SUCCESS in 10m 03s\n- openstack-tox-py35 http://logs.openstack.org/25/612625/5/check/openstack-tox-py35/5083abe/ : SUCCESS in 9m 15s\n- openstack-tox-py36 http://logs.openstack.org/25/612625/5/check/openstack-tox-py36/22a09b6/ : SUCCESS in 9m 33s\n- openstack-tox-docs http://logs.openstack.org/25/612625/5/check/openstack-tox-docs/172031c/html/ : SUCCESS in 5m 10s\n- tempest-full http://logs.openstack.org/25/612625/5/check/tempest-full/6ec2e95/ : SUCCESS in 1h 38m 32s\n- neutron-grenade http://logs.openstack.org/25/612625/5/check/neutron-grenade/c6c20bd/ : SUCCESS in 52m 45s\n- tempest-full-py3 http://logs.openstack.org/25/612625/5/check/tempest-full-py3/15aded5/ : SUCCESS in 1h 16m 04s\n- keystone-dsvm-functional http://logs.openstack.org/25/612625/5/check/keystone-dsvm-functional/8d80364/ : SUCCESS in 27m 30s\n- keystone-dsvm-py3-functional http://logs.openstack.org/25/612625/5/check/keystone-dsvm-py3-functional/9cad05f/ : SUCCESS in 28m 56s\n- keystone-dsvm-functional-federation http://logs.openstack.org/25/612625/5/check/keystone-dsvm-functional-federation/46ecfc4/ : SUCCESS in 24m 33s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/25/612625/5/check/keystone-dsvm-py35-functional-federation/517c33d/ : SUCCESS in 31m 53s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/25/612625/5/check/keystoneclient-devstack-functional/a387db8/ : RETRY_LIMIT in 21m 58s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/25/612625/5/check/legacy-tempest-dsvm-ldap-domain-specific-driver/1f39d7b/ : SUCCESS in 1h 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bc064ab23ea759a7cfa5ef0df4248185fd19bf57","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-11-08 06:23:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1d31fd99ab2c395ecaaafe5285f84baba3aca193","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2018-11-08 14:30:26.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b5c00b585b6cb244863d65cddc9528b8fa6c5864","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-11-09 01:24:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8f67a70b29e72466167e3526a405a8b7f1590b4a","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2018-11-09 20:41:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"edbe34686ca6d5eeed7be099eeae830ae7245af0","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2018-11-13 15:29:23.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"71a33a903aad35eed81df30528ea80587edbb7d2","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-07-08 19:41:16.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"e6341cec29c209ea53411a6b12240cb2abf7d733","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-07-08 20:39:52.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ae8b9e98ff7fbe781727a688ee516a7e75e5312e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 00:36:47.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/612625/6/check/openstack-tox-cover/ef84308/cover/ : SUCCESS in 37m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/612625/6/check/openstack-tox-lower-constraints/6671e11/ : SUCCESS in 36m 08s\n- openstack-tox-pep8 http://logs.openstack.org/25/612625/6/check/openstack-tox-pep8/5bae285/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/25/612625/6/check/openstack-tox-py27/7d601ec/ : SUCCESS in 33m 30s\n- openstack-tox-py36 http://logs.openstack.org/25/612625/6/check/openstack-tox-py36/2926d20/ : SUCCESS in 27m 51s\n- openstack-tox-py37 http://logs.openstack.org/25/612625/6/check/openstack-tox-py37/897c982/ : SUCCESS in 26m 27s\n- openstack-tox-docs http://logs.openstack.org/25/612625/6/check/openstack-tox-docs/78e9f2b/html/ : SUCCESS in 7m 40s\n- tempest-full http://logs.openstack.org/25/612625/6/check/tempest-full/dc3537b/ : SUCCESS in 1h 30m 40s\n- neutron-grenade http://logs.openstack.org/25/612625/6/check/neutron-grenade/6e69c71/ : SUCCESS in 1h 02m 16s\n- grenade-py3 http://logs.openstack.org/25/612625/6/check/grenade-py3/aecbece/ : SUCCESS in 1h 03m 10s\n- tempest-full-py3 http://logs.openstack.org/25/612625/6/check/tempest-full-py3/acc4289/ : SUCCESS in 1h 45m 08s\n- keystone-dsvm-functional http://logs.openstack.org/25/612625/6/check/keystone-dsvm-functional/cd0a0e5/ : SUCCESS in 40m 55s\n- keystone-dsvm-py3-functional http://logs.openstack.org/25/612625/6/check/keystone-dsvm-py3-functional/433a36e/ : SUCCESS in 40m 57s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/25/612625/6/check/keystone-dsvm-functional-federation-opensuse15/eece081/ : SUCCESS in 33m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/25/612625/6/check/keystone-dsvm-py3-functional-federation-opensuse15/9c43bf4/ : SUCCESS in 50m 10s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/25/612625/6/check/keystoneclient-devstack-functional/bf822f4/ : SUCCESS in 18m 44s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/25/612625/6/check/legacy-tempest-dsvm-ldap-domain-specific-driver/f76239e/ : SUCCESS in 1h 39m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c33fba5a0fdf2117700d3bfe92dc7e9de24fd948","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 00:37:15.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"908eb772952a171e5380b91fd615c101095c390a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 05:18:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"528d296c2e14715b1e4ca6fbb74fdca8b85c8dfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 05:18:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/25/612625/6/gate/openstack-tox-lower-constraints/ac0aae4/ : SUCCESS in 35m 09s\n- openstack-tox-pep8 http://logs.openstack.org/25/612625/6/gate/openstack-tox-pep8/3ff0e9a/ : SUCCESS in 6m 21s\n- openstack-tox-py27 http://logs.openstack.org/25/612625/6/gate/openstack-tox-py27/18c498a/ : SUCCESS in 29m 38s\n- openstack-tox-py36 http://logs.openstack.org/25/612625/6/gate/openstack-tox-py36/1ddb0cf/ : SUCCESS in 29m 50s\n- openstack-tox-py37 http://logs.openstack.org/25/612625/6/gate/openstack-tox-py37/c2a61bb/ : SUCCESS in 33m 18s\n- openstack-tox-docs http://logs.openstack.org/25/612625/6/gate/openstack-tox-docs/cea2cb4/html/ : SUCCESS in 8m 58s\n- tempest-full http://logs.openstack.org/25/612625/6/gate/tempest-full/833d69a/ : SUCCESS in 1h 48m 19s\n- neutron-grenade http://logs.openstack.org/25/612625/6/gate/neutron-grenade/4d763bf/ : SUCCESS in 56m 40s\n- grenade-py3 http://logs.openstack.org/25/612625/6/gate/grenade-py3/4513eea/ : SUCCESS in 1h 00m 10s\n- tempest-full-py3 http://logs.openstack.org/25/612625/6/gate/tempest-full-py3/ebc2986/ : SUCCESS in 1h 22m 10s\n- keystone-dsvm-functional http://logs.openstack.org/25/612625/6/gate/keystone-dsvm-functional/01e7308/ : SUCCESS in 33m 07s\n- keystone-dsvm-py3-functional http://logs.openstack.org/25/612625/6/gate/keystone-dsvm-py3-functional/5d80924/ : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"19cd84b8409982a3d226fca25b090ff320a64aad","revisions":{"319252c949bb5b62a2ef7c713706e4b7630c1c85":{"kind":"REWORK","_number":1,"created":"2018-10-23 08:08:40.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/25/612625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/612625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/612625/1"}}},"commit":{"parents":[{"commit":"4b41fa4c8b17b2da510100e9c6cad98d1cd19b0c","subject":"Merge \"Add caching on trust role validation to improve performance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4b41fa4c8b17b2da510100e9c6cad98d1cd19b0c"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-23 08:06:21.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-23 08:07:04.000000000","tz":480},"subject":"nit: remove useless code","message":"nit: remove useless code\n\nThe code for memcache_pool, mongo and noop cache backends\nhave been removed already.\n\nChange-Id: I6ccf470ef41c858c9aac90c8ce03dcfdb5468b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/319252c949bb5b62a2ef7c713706e4b7630c1c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/319252c949bb5b62a2ef7c713706e4b7630c1c85"}]},"branch":"refs/heads/master"},"2cefbf2190aaeaf353502fcbfe30af34c260d453":{"kind":"REWORK","_number":2,"created":"2018-10-23 08:09:28.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/25/612625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/612625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/612625/2"}}},"commit":{"parents":[{"commit":"4b41fa4c8b17b2da510100e9c6cad98d1cd19b0c","subject":"Merge \"Add caching on trust role validation to improve performance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4b41fa4c8b17b2da510100e9c6cad98d1cd19b0c"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-23 08:06:21.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-23 08:09:11.000000000","tz":480},"subject":"nit: remove useless code","message":"nit: remove useless code\n\nThe code for memcache_pool, mongo and noop cache backends\nhave been removed already.\n\nChange-Id: I6ccf470ef41c858c9aac90c8ce03dcfdb5468b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2cefbf2190aaeaf353502fcbfe30af34c260d453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2cefbf2190aaeaf353502fcbfe30af34c260d453"}]},"branch":"refs/heads/master"},"f1cb8bcd012a7f2acde37279182610f8e2b24bf5":{"kind":"REWORK","_number":3,"created":"2018-10-25 03:37:06.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/25/612625/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/612625/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/612625/3"}}},"commit":{"parents":[{"commit":"9f2b84608e7297cc7ed41497ddf45841325a8fd7","subject":"Merge \"Set min and max length for resource_name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9f2b84608e7297cc7ed41497ddf45841325a8fd7"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-23 08:06:21.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-25 03:35:36.000000000","tz":480},"subject":"nit: remove some useless code","message":"nit: remove some useless code\n\nRemove some uneless code\n\nChange-Id: I6ccf470ef41c858c9aac90c8ce03dcfdb5468b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f1cb8bcd012a7f2acde37279182610f8e2b24bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f1cb8bcd012a7f2acde37279182610f8e2b24bf5"}]},"branch":"refs/heads/master"},"2f9dd93a3c3f4b77fb2581393eb48ae1effa41ce":{"kind":"REWORK","_number":4,"created":"2018-10-25 03:44:58.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/25/612625/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/612625/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/612625/4"}}},"commit":{"parents":[{"commit":"9f2b84608e7297cc7ed41497ddf45841325a8fd7","subject":"Merge \"Set min and max length for resource_name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9f2b84608e7297cc7ed41497ddf45841325a8fd7"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-23 08:06:21.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-25 03:44:16.000000000","tz":480},"subject":"nit: remove some useless code","message":"nit: remove some useless code\n\nRemove some uneless code\n\nChange-Id: I6ccf470ef41c858c9aac90c8ce03dcfdb5468b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2f9dd93a3c3f4b77fb2581393eb48ae1effa41ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2f9dd93a3c3f4b77fb2581393eb48ae1effa41ce"}]},"branch":"refs/heads/master"},"01a93949cdfa248eb2acf7362ff85310c74871f0":{"kind":"REWORK","_number":5,"created":"2018-11-08 01:31:10.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/25/612625/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/612625/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/612625/5"}}},"commit":{"parents":[{"commit":"0780297b6325855c6cca9429cb9fbafab3a0d023","subject":"Merge \"Fixing nits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0780297b6325855c6cca9429cb9fbafab3a0d023"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-23 08:06:21.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-11-08 01:30:53.000000000","tz":480},"subject":"nit: remove some useless code","message":"nit: remove some useless code\n\nRemove some uneless code\n\nChange-Id: I6ccf470ef41c858c9aac90c8ce03dcfdb5468b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/01a93949cdfa248eb2acf7362ff85310c74871f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/01a93949cdfa248eb2acf7362ff85310c74871f0"}]},"branch":"refs/heads/master"},"19cd84b8409982a3d226fca25b090ff320a64aad":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-08 19:41:16.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/25/612625/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/612625/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/612625/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/612625/6"}}},"commit":{"parents":[{"commit":"3b13b4e5e7d72c2eaef470d0f84537a279e10e43","subject":"Merge \"Update test cases for os-pki revoke API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3b13b4e5e7d72c2eaef470d0f84537a279e10e43"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan@huawei.com","date":"2018-10-23 08:06:21.000000000","tz":480},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2019-07-08 19:41:16.000000000","tz":0},"subject":"nit: remove some useless code","message":"nit: remove some useless code\n\nRemove some uneless code\n\nChange-Id: I6ccf470ef41c858c9aac90c8ce03dcfdb5468b82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19cd84b8409982a3d226fca25b090ff320a64aad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19cd84b8409982a3d226fca25b090ff320a64aad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
