)]}'
{"id":"openstack%2Fkeystone~621023","triplet_id":"openstack%2Fkeystone~master~I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd","project":"openstack/keystone","branch":"master","topic":"train-rc1-policy-fixes","hashtags":[],"change_id":"I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd","subject":"Allow domain users to access the limit API","status":"MERGED","created":"2018-11-29 21:34:28.000000000","updated":"2019-09-26 07:50:12.000000000","submitted":"2019-09-26 07:48:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"621023-1569484127015-654589fa","meta_rev_id":"232c0d2752d0d3e8604bae349c3a24d083c0bd9b","_number":621023,"virtual_id_number":621023,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 07:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-09-25 02:26:18.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-09-25 05:38:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-09-25 17:16:56.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2019-09-25 17:27:04.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-11 12:00:43.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-12-20 07:18:41.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-03-05 16:19:54.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-09-25 02:26:18.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-09-25 05:38:32.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-09-26 07:48:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"df709f7fe737df2e8912ef1662274e97ae90d06e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-11-29 21:34:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a98c9202048ec16405cfa8ad6c381658ef4e4d1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 01:29:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/1/check/openstack-tox-cover/488e9f0/cover/ : SUCCESS in 14m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/1/check/openstack-tox-lower-constraints/a830324/ : SUCCESS in 18m 13s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/1/check/openstack-tox-pep8/989c9f7/ : SUCCESS in 8m 14s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/1/check/openstack-tox-py27/215e818/ : SUCCESS in 10m 36s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/1/check/openstack-tox-py35/effeae9/ : SUCCESS in 13m 44s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/1/check/openstack-tox-py36/1537aef/ : SUCCESS in 11m 22s\n- openstack-tox-docs http://logs.openstack.org/23/621023/1/check/openstack-tox-docs/48fe93f/html/ : SUCCESS in 7m 07s\n- grenade-py3 http://logs.openstack.org/23/621023/1/check/grenade-py3/18c8665/ : SUCCESS in 1h 03m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3b67db65f29b842610258ae8075c0868881d35f0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-11-30 23:37:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02e2f023799deec7bf56bdd3ce0effda7fcd2154","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-01 04:33:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/2/check/openstack-tox-cover/7898746/cover/ : SUCCESS in 15m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/2/check/openstack-tox-lower-constraints/3bb24d8/ : SUCCESS in 15m 14s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/2/check/openstack-tox-pep8/8767295/ : SUCCESS in 7m 11s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/2/check/openstack-tox-py27/7946f4a/ : SUCCESS in 14m 26s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/2/check/openstack-tox-py35/5d8dea4/ : SUCCESS in 11m 03s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/2/check/openstack-tox-py36/3414201/ : SUCCESS in 10m 26s\n- openstack-tox-docs http://logs.openstack.org/23/621023/2/check/openstack-tox-docs/9b07ba3/html/ : SUCCESS in 6m 48s\n- grenade-py3 http://logs.openstack.org/23/621023/2/check/grenade-py3/d4ed188/ : SUCCESS in 55m 03s","accounts_in_message":[],"_revision_number":2},{"id":"f7218eb9a87f3b0f65f959be8bd6ccf07255d54d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-07 17:54:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1d7bd49e93b6d19a5cdf9d05d5cf280b6f01de31","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-07 18:00:29.000000000","message":"Topic changed from bug/1805880 to implement-default-roles","accounts_in_message":[],"_revision_number":3},{"id":"2075464c1d9549f5a25ae2e583615b32bbad2ad2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-08 00:15:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/3/check/openstack-tox-cover/7fe98d8/cover/ : SUCCESS in 22m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/3/check/openstack-tox-lower-constraints/cfc4c8d/ : SUCCESS in 14m 02s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/3/check/openstack-tox-pep8/a162b55/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/3/check/openstack-tox-py27/5628b11/ : SUCCESS in 9m 56s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/3/check/openstack-tox-py35/84a16df/ : SUCCESS in 11m 07s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/3/check/openstack-tox-py36/3b59f6f/ : SUCCESS in 9m 33s\n- openstack-tox-docs http://logs.openstack.org/23/621023/3/check/openstack-tox-docs/4efc68b/html/ : SUCCESS in 6m 53s\n- tempest-full http://logs.openstack.org/23/621023/3/check/tempest-full/b32f834/ : SUCCESS in 1h 46m 38s\n- neutron-grenade http://logs.openstack.org/23/621023/3/check/neutron-grenade/884c818/ : SUCCESS in 58m 51s\n- grenade-py3 http://logs.openstack.org/23/621023/3/check/grenade-py3/c978cc5/ : SUCCESS in 1h 01m 39s\n- tempest-full-py3 http://logs.openstack.org/23/621023/3/check/tempest-full-py3/06c18ea/ : SUCCESS in 1h 47m 57s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/3/check/keystone-dsvm-functional/dad0aaa/ : SUCCESS in 35m 09s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/3/check/keystone-dsvm-py3-functional/98fcc91/ : SUCCESS in 36m 59s\n- keystone-dsvm-functional-federation http://logs.openstack.org/23/621023/3/check/keystone-dsvm-functional-federation/e02db48/ : SUCCESS in 32m 37s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/23/621023/3/check/keystone-dsvm-py35-functional-federation/ff37460/ : SUCCESS in 33m 41s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/3/check/keystoneclient-devstack-functional/e37f9e2/ : SUCCESS in 12m 42s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/3/check/legacy-tempest-dsvm-ldap-domain-specific-driver/1b980d3/ : SUCCESS in 1h 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de0b597f2dfe6ca75f85deeb5c139ce97dce9dbc","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-11 09:15:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"536b4c73469ffb72fab8511eeed190e806408094","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-11 09:19:25.000000000","message":"Topic changed from clean-up-v3-cloud-sample to implement-default-roles","accounts_in_message":[],"_revision_number":4},{"id":"2e32e28c362ea1b8315f0f605e542a107c362d44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 11:57:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/4/check/openstack-tox-cover/c6c806f/cover/ : SUCCESS in 13m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/4/check/openstack-tox-lower-constraints/4fb9b59/ : SUCCESS in 15m 21s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/4/check/openstack-tox-pep8/2971325/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/4/check/openstack-tox-py27/7fec2d0/ : SUCCESS in 12m 49s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/4/check/openstack-tox-py35/edca123/ : SUCCESS in 13m 13s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/4/check/openstack-tox-py36/bc303be/ : SUCCESS in 10m 59s\n- openstack-tox-docs http://logs.openstack.org/23/621023/4/check/openstack-tox-docs/6b5a0ac/html/ : SUCCESS in 6m 40s\n- tempest-full http://logs.openstack.org/23/621023/4/check/tempest-full/18e2006/ : SUCCESS in 1h 26m 12s\n- neutron-grenade http://logs.openstack.org/23/621023/4/check/neutron-grenade/f7b8f80/ : SUCCESS in 57m 27s\n- grenade-py3 http://logs.openstack.org/23/621023/4/check/grenade-py3/6a3b53e/ : FAILURE in 1h 00m 07s\n- tempest-full-py3 http://logs.openstack.org/23/621023/4/check/tempest-full-py3/de8bf4f/ : SUCCESS in 1h 18m 45s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/4/check/keystone-dsvm-functional/45b6ec3/ : SUCCESS in 30m 09s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/4/check/keystone-dsvm-py3-functional/634b570/ : SUCCESS in 29m 06s\n- keystone-dsvm-functional-federation http://logs.openstack.org/23/621023/4/check/keystone-dsvm-functional-federation/8e80be2/ : SUCCESS in 26m 36s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/23/621023/4/check/keystone-dsvm-py35-functional-federation/cd18475/ : SUCCESS in 27m 41s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/4/check/keystoneclient-devstack-functional/1d2a919/ : SUCCESS in 12m 01s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/db61e06/ : SUCCESS in 1h 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fa383515a66d3e93c285efe7f8c30bb7e9917728","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-12-11 12:00:43.000000000","message":"Patch Set 4:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- grenade-py3: https://bugs.launchpad.net/bugs/1807518\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":4},{"id":"c93253f4c234607fdc9ca0cd112c4084056807ff","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-11 13:10:47.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"62df496ab5960e039bbb69327f8bf2d2fe24ec7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 15:13:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/4/check/openstack-tox-cover/1694470/cover/ : SUCCESS in 14m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/4/check/openstack-tox-lower-constraints/e2a2d11/ : SUCCESS in 19m 00s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/4/check/openstack-tox-pep8/2e4c00d/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/4/check/openstack-tox-py27/add0b0e/ : SUCCESS in 12m 44s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/4/check/openstack-tox-py35/3a0e419/ : SUCCESS in 15m 12s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/4/check/openstack-tox-py36/7615424/ : SUCCESS in 10m 32s\n- openstack-tox-docs http://logs.openstack.org/23/621023/4/check/openstack-tox-docs/b1fea47/html/ : SUCCESS in 6m 51s\n- tempest-full http://logs.openstack.org/23/621023/4/check/tempest-full/bbf26ef/ : SUCCESS in 1h 47m 49s\n- neutron-grenade http://logs.openstack.org/23/621023/4/check/neutron-grenade/969dab4/ : SUCCESS in 58m 05s\n- grenade-py3 http://logs.openstack.org/23/621023/4/check/grenade-py3/f0683cb/ : SUCCESS in 1h 15m 33s\n- tempest-full-py3 http://logs.openstack.org/23/621023/4/check/tempest-full-py3/3b3a151/ : SUCCESS in 1h 47m 56s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/4/check/keystone-dsvm-functional/8282c87/ : SUCCESS in 36m 10s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/4/check/keystone-dsvm-py3-functional/a0b908a/ : SUCCESS in 35m 25s\n- keystone-dsvm-functional-federation http://logs.openstack.org/23/621023/4/check/keystone-dsvm-functional-federation/304c3cd/ : SUCCESS in 29m 33s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/23/621023/4/check/keystone-dsvm-py35-functional-federation/a951c35/ : SUCCESS in 39m 03s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/4/check/keystoneclient-devstack-functional/8c9fadf/ : SUCCESS in 19m 30s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/4/check/legacy-tempest-dsvm-ldap-domain-specific-driver/e1280d0/ : SUCCESS in 1h 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c9b3973dccd8a0c4794f144cc0a6ba5bc24abc1c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-18 22:28:55.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"f0f3772c635870d7a9f1750c3d6f0d769c9d6022","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-19 00:50:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/5/check/openstack-tox-cover/39d9e39/ : FAILURE in 11m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/5/check/openstack-tox-lower-constraints/814b7fd/ : FAILURE in 18m 42s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/5/check/openstack-tox-pep8/4ad87ae/ : SUCCESS in 6m 28s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/5/check/openstack-tox-py27/443b371/ : FAILURE in 11m 45s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/5/check/openstack-tox-py35/466fd7c/ : FAILURE in 10m 26s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/5/check/openstack-tox-py36/0c6c1e3/ : FAILURE in 9m 37s\n- openstack-tox-docs http://logs.openstack.org/23/621023/5/check/openstack-tox-docs/429da1e/html/ : SUCCESS in 7m 56s\n- tempest-full http://logs.openstack.org/23/621023/5/check/tempest-full/1b7d450/ : SUCCESS in 1h 23m 57s\n- neutron-grenade http://logs.openstack.org/23/621023/5/check/neutron-grenade/4e634ba/ : SUCCESS in 48m 45s\n- grenade-py3 http://logs.openstack.org/23/621023/5/check/grenade-py3/7769b10/ : SUCCESS in 1h 02m 49s\n- tempest-full-py3 http://logs.openstack.org/23/621023/5/check/tempest-full-py3/3cd454b/ : SUCCESS in 1h 07m 24s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/5/check/keystone-dsvm-functional/cd1bee3/ : SUCCESS in 30m 53s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/5/check/keystone-dsvm-py3-functional/2febe56/ : SUCCESS in 33m 47s\n- keystone-dsvm-functional-federation http://logs.openstack.org/23/621023/5/check/keystone-dsvm-functional-federation/667869e/ : SUCCESS in 31m 21s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/23/621023/5/check/keystone-dsvm-py35-functional-federation/7048caa/ : SUCCESS in 34m 39s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/5/check/keystoneclient-devstack-functional/fdb5bba/ : SUCCESS in 16m 51s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/5/check/legacy-tempest-dsvm-ldap-domain-specific-driver/29fbaae/ : SUCCESS in 1h 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"79fe757d38b43fc200dce0209b41d9dac046c15f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-19 23:14:26.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"7c087e9b401259a9aa01e1138cf3d3dc12ea15ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-20 00:58:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/6/check/openstack-tox-cover/81f5b3e/cover/ : SUCCESS in 13m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/6/check/openstack-tox-lower-constraints/39a322f/ : SUCCESS in 17m 02s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/6/check/openstack-tox-pep8/ca9acc1/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/6/check/openstack-tox-py27/38e4052/ : SUCCESS in 13m 15s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/6/check/openstack-tox-py35/12b32a6/ : SUCCESS in 13m 09s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/6/check/openstack-tox-py36/e88d846/ : SUCCESS in 9m 03s\n- openstack-tox-docs http://logs.openstack.org/23/621023/6/check/openstack-tox-docs/4ef258c/html/ : SUCCESS in 6m 44s\n- tempest-full http://logs.openstack.org/23/621023/6/check/tempest-full/eed4fbf/ : SUCCESS in 1h 14m 07s\n- neutron-grenade http://logs.openstack.org/23/621023/6/check/neutron-grenade/1fe4a09/ : SUCCESS in 57m 06s\n- grenade-py3 http://logs.openstack.org/23/621023/6/check/grenade-py3/80a22a8/ : SUCCESS in 54m 42s\n- tempest-full-py3 http://logs.openstack.org/23/621023/6/check/tempest-full-py3/416b12f/ : SUCCESS in 1h 06m 35s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/6/check/keystone-dsvm-functional/1b47cf9/ : SUCCESS in 32m 38s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/6/check/keystone-dsvm-py3-functional/133d690/ : SUCCESS in 29m 20s\n- keystone-dsvm-functional-federation http://logs.openstack.org/23/621023/6/check/keystone-dsvm-functional-federation/a1a2ad0/ : SUCCESS in 33m 30s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/23/621023/6/check/keystone-dsvm-py35-functional-federation/478798b/ : SUCCESS in 28m 47s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/6/check/keystoneclient-devstack-functional/538f278/ : SUCCESS in 12m 57s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/6/check/legacy-tempest-dsvm-ldap-domain-specific-driver/4e25302/ : SUCCESS in 1h 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2c9a0076ad5d47c54a860980a411e86f7f6b5787","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-12-20 07:18:41.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"6c1e1101c4af82b24b4835c8b4c5a9df882b43e0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-01-08 18:24:13.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"44e459d31a7af947d62b83bce82c652bd63eeb5f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-01-08 19:23:01.000000000","message":"Topic changed from bug/1805880 to implement-default-roles","accounts_in_message":[],"_revision_number":7},{"id":"59b1d7cd047cafdb232bfa7d1a001364dfa0e5c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 20:12:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/7/check/openstack-tox-cover/f12e6f4/cover/ : SUCCESS in 16m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/7/check/openstack-tox-lower-constraints/9062c16/ : SUCCESS in 15m 08s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/7/check/openstack-tox-pep8/c29a292/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/7/check/openstack-tox-py27/34251a5/ : SUCCESS in 10m 05s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/7/check/openstack-tox-py35/61b770b/ : SUCCESS in 14m 38s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/7/check/openstack-tox-py36/f1c94f5/ : SUCCESS in 10m 55s\n- openstack-tox-docs http://logs.openstack.org/23/621023/7/check/openstack-tox-docs/7a1616f/html/ : SUCCESS in 8m 23s\n- tempest-full http://logs.openstack.org/23/621023/7/check/tempest-full/6cdd01c/ : SUCCESS in 1h 20m 27s\n- neutron-grenade http://logs.openstack.org/23/621023/7/check/neutron-grenade/5fb2306/ : SUCCESS in 47m 36s\n- grenade-py3 http://logs.openstack.org/23/621023/7/check/grenade-py3/f468a34/ : SUCCESS in 55m 26s\n- tempest-full-py3 http://logs.openstack.org/23/621023/7/check/tempest-full-py3/ea6cd57/ : SUCCESS in 1h 18m 58s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/7/check/keystone-dsvm-functional/717af2b/ : SUCCESS in 31m 38s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/7/check/keystone-dsvm-py3-functional/ae64434/ : SUCCESS in 32m 11s\n- keystone-dsvm-functional-federation http://logs.openstack.org/23/621023/7/check/keystone-dsvm-functional-federation/16b7b9a/ : SUCCESS in 30m 18s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/23/621023/7/check/keystone-dsvm-py35-functional-federation/d372428/ : SUCCESS in 32m 01s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/7/check/keystoneclient-devstack-functional/5b333ec/ : SUCCESS in 20m 18s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/7/check/legacy-tempest-dsvm-ldap-domain-specific-driver/562675f/ : SUCCESS in 1h 31m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6de1d0228da8d7b29f03c81f8d355d291fc416bd","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-01-09 14:59:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6c2eb8572f964e6f17dfdc408b9d002af847dba5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 17:08:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/8/check/openstack-tox-cover/4a51c7b/cover/ : SUCCESS in 14m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/8/check/openstack-tox-lower-constraints/d04ea05/ : SUCCESS in 15m 32s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/8/check/openstack-tox-pep8/a55a1fd/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/8/check/openstack-tox-py27/773949e/ : SUCCESS in 14m 46s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/8/check/openstack-tox-py35/f7e8ff7/ : SUCCESS in 11m 08s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/8/check/openstack-tox-py36/029249e/ : SUCCESS in 10m 22s\n- openstack-tox-docs http://logs.openstack.org/23/621023/8/check/openstack-tox-docs/8a6080c/html/ : SUCCESS in 7m 36s\n- tempest-full http://logs.openstack.org/23/621023/8/check/tempest-full/829c3d8/ : SUCCESS in 1h 38m 40s\n- neutron-grenade http://logs.openstack.org/23/621023/8/check/neutron-grenade/eebf83c/ : SUCCESS in 1h 03m 03s\n- grenade-py3 http://logs.openstack.org/23/621023/8/check/grenade-py3/0555de3/ : SUCCESS in 1h 05m 47s\n- tempest-full-py3 http://logs.openstack.org/23/621023/8/check/tempest-full-py3/8b909d3/ : SUCCESS in 1h 26m 09s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/8/check/keystone-dsvm-functional/ef30c27/ : SUCCESS in 26m 55s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/8/check/keystone-dsvm-py3-functional/0b2b5d3/ : SUCCESS in 29m 14s\n- keystone-dsvm-functional-federation http://logs.openstack.org/23/621023/8/check/keystone-dsvm-functional-federation/f4114d1/ : SUCCESS in 29m 04s (non-voting)\n- keystone-dsvm-py35-functional-federation http://logs.openstack.org/23/621023/8/check/keystone-dsvm-py35-functional-federation/92edbd9/ : SUCCESS in 31m 14s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/8/check/keystoneclient-devstack-functional/5a3cd63/ : SUCCESS in 23m 39s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/8/check/legacy-tempest-dsvm-ldap-domain-specific-driver/e76d041/ : SUCCESS in 1h 46m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4edeb14674ec169d8ebdbc1d06c62bd2d0592439","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-01-17 16:28:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2942ce3df779d94de4b2830ed8afb19a51ebca3b","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-01-21 20:58:21.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)\n\nMarking as WIP until we get some more discussion.","accounts_in_message":[],"_revision_number":8},{"id":"70a0cc397edd14ee439b46198848ec7c6dd82a03","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-01-21 21:01:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"157efdf9ea44148d9f4e3915f890f4e212158b31","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-05 15:40:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"6b92742209b3790c2e07a1b2477d8dc43957877c","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-03-05 16:19:54.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)\n\nQuestions in-line.","accounts_in_message":[],"_revision_number":9},{"id":"54d15c54cab8c9ca0a2c5444fc163a5d7f906b9a","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-05 18:28:56.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3f49ad9af85308c6e69d090448b432e9c1aa36e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 18:31:53.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/9/check/openstack-tox-cover/9fcb8c6/cover/ : SUCCESS in 25m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/9/check/openstack-tox-lower-constraints/e835ef4/ : SUCCESS in 23m 12s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/9/check/openstack-tox-pep8/e8b177c/ : FAILURE in 5m 43s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/9/check/openstack-tox-py27/0392ebc/ : SUCCESS in 17m 59s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/9/check/openstack-tox-py35/8d5abfe/ : SUCCESS in 17m 59s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/9/check/openstack-tox-py36/2fbaaf5/ : SUCCESS in 20m 43s\n- openstack-tox-py37 http://logs.openstack.org/23/621023/9/check/openstack-tox-py37/1fd0c87/ : SUCCESS in 18m 37s\n- openstack-tox-docs http://logs.openstack.org/23/621023/9/check/openstack-tox-docs/1ccaac7/html/ : SUCCESS in 7m 38s\n- tempest-full http://logs.openstack.org/23/621023/9/check/tempest-full/7036ebe/ : SUCCESS in 1h 52m 36s\n- neutron-grenade http://logs.openstack.org/23/621023/9/check/neutron-grenade/ef0362a/ : SUCCESS in 50m 21s\n- grenade-py3 http://logs.openstack.org/23/621023/9/check/grenade-py3/301bea0/ : SUCCESS in 58m 17s\n- tempest-full-py3 http://logs.openstack.org/23/621023/9/check/tempest-full-py3/c3105fa/ : SUCCESS in 1h 24m 50s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/9/check/keystone-dsvm-functional/81a9953/ : SUCCESS in 26m 51s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/9/check/keystone-dsvm-py3-functional/2384ec9/ : SUCCESS in 36m 23s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/23/621023/9/check/keystone-dsvm-functional-federation-opensuse15/f4285a6/ : SUCCESS in 34m 16s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/23/621023/9/check/keystone-dsvm-py3-functional-federation-opensuse15/eeba119/ : SUCCESS in 41m 05s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/9/check/keystoneclient-devstack-functional/8dd75c6/ : SUCCESS in 18m 26s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/9/check/legacy-tempest-dsvm-ldap-domain-specific-driver/01ef398/ : SUCCESS in 1h 47m 22s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6425cee89e18a013e5f22aad1baaa4108ab0a7dd","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-05 19:14:17.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"91026025b328c723964c12b7f75a4f2a91848ae2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-05 19:14:45.000000000","message":"Topic changed from bug/1805880 to implement-default-roles","accounts_in_message":[],"_revision_number":10},{"id":"bc2e979839a18aac4525cf93ff1affaad9e6ad6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 21:56:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/10/check/openstack-tox-cover/2bf26e9/cover/ : SUCCESS in 20m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/10/check/openstack-tox-lower-constraints/e67eba3/ : SUCCESS in 22m 03s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/10/check/openstack-tox-pep8/e4ef889/ : SUCCESS in 6m 02s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/10/check/openstack-tox-py27/476aabc/ : SUCCESS in 17m 49s\n- openstack-tox-py35 http://logs.openstack.org/23/621023/10/check/openstack-tox-py35/33eb350/ : SUCCESS in 24m 55s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/10/check/openstack-tox-py36/5ee330c/ : SUCCESS in 17m 28s\n- openstack-tox-py37 http://logs.openstack.org/23/621023/10/check/openstack-tox-py37/98b69ff/ : SUCCESS in 17m 16s\n- openstack-tox-docs http://logs.openstack.org/23/621023/10/check/openstack-tox-docs/71f66f5/html/ : SUCCESS in 8m 45s\n- tempest-full http://logs.openstack.org/23/621023/10/check/tempest-full/3732bf2/ : SUCCESS in 1h 33m 17s\n- neutron-grenade http://logs.openstack.org/23/621023/10/check/neutron-grenade/23cfc54/ : SUCCESS in 59m 30s\n- grenade-py3 http://logs.openstack.org/23/621023/10/check/grenade-py3/4ac1a08/ : SUCCESS in 56m 32s\n- tempest-full-py3 http://logs.openstack.org/23/621023/10/check/tempest-full-py3/930ff05/ : SUCCESS in 1h 20m 48s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/10/check/keystone-dsvm-functional/a5278ab/ : SUCCESS in 25m 22s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/10/check/keystone-dsvm-py3-functional/4d6a4d4/ : SUCCESS in 34m 34s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/23/621023/10/check/keystone-dsvm-functional-federation-opensuse15/39c73e1/ : SUCCESS in 29m 26s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/23/621023/10/check/keystone-dsvm-py3-functional-federation-opensuse15/11f6fdd/ : SUCCESS in 32m 39s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/10/check/keystoneclient-devstack-functional/13f3722/ : SUCCESS in 12m 52s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/10/check/legacy-tempest-dsvm-ldap-domain-specific-driver/5b3b27c/ : SUCCESS in 1h 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fc5a181e0614dbb43e39db13c63bbc2ece55a574","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-06 20:28:05.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"253e34f53ce8839b1bae50e71f406a229df0d58f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 00:45:28.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/11/check/openstack-tox-cover/5aa3c35/cover/ : SUCCESS in 27m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/11/check/openstack-tox-lower-constraints/862b1d4/ : SUCCESS in 23m 51s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/11/check/openstack-tox-pep8/9867f2e/ : SUCCESS in 7m 21s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/11/check/openstack-tox-py27/50d9bed/ : SUCCESS in 15m 31s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/11/check/openstack-tox-py36/3b4cc9e/ : SUCCESS in 17m 13s\n- openstack-tox-py37 http://logs.openstack.org/23/621023/11/check/openstack-tox-py37/0fdb110/ : SUCCESS in 18m 52s\n- openstack-tox-docs http://logs.openstack.org/23/621023/11/check/openstack-tox-docs/a213d4e/html/ : SUCCESS in 7m 36s\n- tempest-full http://logs.openstack.org/23/621023/11/check/tempest-full/b922328/ : TIMED_OUT in 2h 05m 27s\n- neutron-grenade http://logs.openstack.org/23/621023/11/check/neutron-grenade/14e528d/ : SUCCESS in 1h 03m 56s\n- grenade-py3 http://logs.openstack.org/23/621023/11/check/grenade-py3/c641e3a/ : SUCCESS in 1h 04m 53s\n- tempest-full-py3 http://logs.openstack.org/23/621023/11/check/tempest-full-py3/b5ca73c/ : SUCCESS in 1h 21m 17s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/11/check/keystone-dsvm-functional/51967d7/ : SUCCESS in 31m 14s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/11/check/keystone-dsvm-py3-functional/6229d92/ : SUCCESS in 32m 45s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/23/621023/11/check/keystone-dsvm-functional-federation-opensuse15/27fa81e/ : SUCCESS in 36m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/23/621023/11/check/keystone-dsvm-py3-functional-federation-opensuse15/64ac432/ : SUCCESS in 42m 34s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/11/check/keystoneclient-devstack-functional/5c7cb54/ : SUCCESS in 18m 00s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/11/check/legacy-tempest-dsvm-ldap-domain-specific-driver/1a92a62/ : SUCCESS in 1h 57m 55s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"81a1391172ea46a32535fc86d69c70f36af29331","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-03-07 06:22:29.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"741529664d19470749e6a1d019f9cbfea8b1acc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 08:13:13.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/11/check/openstack-tox-cover/72dc717/cover/ : SUCCESS in 23m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/11/check/openstack-tox-lower-constraints/f89ca8b/ : SUCCESS in 22m 16s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/11/check/openstack-tox-pep8/e6b00d4/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/11/check/openstack-tox-py27/ae0d453/ : SUCCESS in 16m 18s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/11/check/openstack-tox-py36/e691095/ : SUCCESS in 18m 30s\n- openstack-tox-py37 http://logs.openstack.org/23/621023/11/check/openstack-tox-py37/511dc7c/ : SUCCESS in 20m 11s\n- openstack-tox-docs http://logs.openstack.org/23/621023/11/check/openstack-tox-docs/19fe9fb/html/ : SUCCESS in 7m 09s\n- tempest-full http://logs.openstack.org/23/621023/11/check/tempest-full/bf2bc11/ : SUCCESS in 1h 38m 19s\n- neutron-grenade http://logs.openstack.org/23/621023/11/check/neutron-grenade/08ba66d/ : SUCCESS in 59m 50s\n- grenade-py3 http://logs.openstack.org/23/621023/11/check/grenade-py3/adb37e8/ : SUCCESS in 59m 24s\n- tempest-full-py3 http://logs.openstack.org/23/621023/11/check/tempest-full-py3/c3c8b79/ : SUCCESS in 1h 28m 57s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/11/check/keystone-dsvm-functional/f5237c5/ : SUCCESS in 31m 28s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/11/check/keystone-dsvm-py3-functional/cfbc9e6/ : SUCCESS in 32m 04s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/23/621023/11/check/keystone-dsvm-functional-federation-opensuse15/c393b8d/ : SUCCESS in 33m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/23/621023/11/check/keystone-dsvm-py3-functional-federation-opensuse15/05f1b19/ : SUCCESS in 33m 58s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/11/check/keystoneclient-devstack-functional/2d60a1b/ : SUCCESS in 17m 37s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/11/check/legacy-tempest-dsvm-ldap-domain-specific-driver/c30959c/ : SUCCESS in 1h 44m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e1de954c4a4bcc860aadfcf31e36fa4f3cd53339","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-03-08 10:55:29.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)\n\nI\u0027m a little surprised at this direction, I\u0027d prefer if we didn\u0027t make these policies completely inconsistent with all our other policies.","accounts_in_message":[],"_revision_number":11},{"id":"a0c6bd8c8b7886be633b96abb1d5bdfd40c3bd8f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-25 19:39:47.000000000","message":"Patch Set 11: Code-Review-2 Workflow-1\n\nWaiting on this until Train opens for development.","accounts_in_message":[],"_revision_number":11},{"id":"4b11e90a0c47c7caf4cdb4e3b1dc65b4fb4316bd","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-27 21:12:33.000000000","message":"Patch Set 11: -Code-Review -Workflow\n\nThis is ready to pick back up now that Train is open for development.","accounts_in_message":[],"_revision_number":11},{"id":"3d236b710df948adb9b8d2aace62ff9a908763f2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-02 14:54:51.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ea33441bb0b0a6da5a334c675282044e94a13535","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-02 15:03:01.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e6d7e81b436b7d26b6c9f740cdd6514ae2b7a943","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 21:21:39.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/23/621023/12/check/openstack-tox-cover/dab9c22/cover/ : SUCCESS in 31m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/621023/12/check/openstack-tox-lower-constraints/d41ca32/ : RETRY_LIMIT in 1m 36s\n- openstack-tox-pep8 http://logs.openstack.org/23/621023/12/check/openstack-tox-pep8/0f21d93/ : RETRY_LIMIT in 2m 32s\n- openstack-tox-py27 http://logs.openstack.org/23/621023/12/check/openstack-tox-py27/f6df457/ : SUCCESS in 24m 36s\n- openstack-tox-py36 http://logs.openstack.org/23/621023/12/check/openstack-tox-py36/2050e52/ : SUCCESS in 30m 37s\n- openstack-tox-py37 http://logs.openstack.org/23/621023/12/check/openstack-tox-py37/f1cd387/ : SUCCESS in 27m 06s\n- openstack-tox-docs http://logs.openstack.org/23/621023/12/check/openstack-tox-docs/3044979/html/ : SUCCESS in 7m 31s\n- tempest-full http://logs.openstack.org/23/621023/12/check/tempest-full/885aa24/ : SUCCESS in 1h 46m 27s\n- neutron-grenade http://logs.openstack.org/23/621023/12/check/neutron-grenade/899e3d6/ : SUCCESS in 57m 39s\n- grenade-py3 http://logs.openstack.org/23/621023/12/check/grenade-py3/00a4857/ : SUCCESS in 55m 52s\n- tempest-full-py3 http://logs.openstack.org/23/621023/12/check/tempest-full-py3/836e906/ : SUCCESS in 1h 41m 29s\n- keystone-dsvm-functional http://logs.openstack.org/23/621023/12/check/keystone-dsvm-functional/c1f6591/ : SUCCESS in 31m 32s\n- keystone-dsvm-py3-functional http://logs.openstack.org/23/621023/12/check/keystone-dsvm-py3-functional/f9a7873/ : SUCCESS in 32m 25s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/23/621023/12/check/keystone-dsvm-functional-federation-opensuse15/1004dc4/ : SUCCESS in 35m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/23/621023/12/check/keystone-dsvm-py3-functional-federation-opensuse15/5b6a479/ : SUCCESS in 38m 15s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/23/621023/12/check/keystoneclient-devstack-functional/7c21b96/ : SUCCESS in 13m 24s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/23/621023/12/check/legacy-tempest-dsvm-ldap-domain-specific-driver/305d2d1/ : RETRY_LIMIT in 3m 49s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"eafb14329ec78e0bbe66363c6e5144ea1060fbd9","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-23 16:06:15.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"4a191e11573ed19cbc02460f1a42ab6534c5780e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-18 14:53:11.000000000","message":"Topic changed from implement-default-roles to train-rc1-policy-fixes","accounts_in_message":[],"_revision_number":12},{"id":"1ddf10ec7c0f0578cefbc28848b44af24a856eaa","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-24 17:30:57.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"70397e20d37a7476b8f780a7390eb63ad814152f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-24 22:36:15.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"d98ca994e69c43dca47565bfdc92663ec550eb11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 01:24:25.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e0e8a2ec8d04b51869fabe885c90e49 : SUCCESS in 15m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d5431e6294d491aa852846da2ad99c1 : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e32fde009c4467a9e2b8849e391072 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/777327d6db234822a583c7e52d326b77 : SUCCESS in 16m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/418fe6124f6d447c869046563c85f840 : SUCCESS in 13m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44ee273a97e04941a4ce5a3acf86c608 : SUCCESS in 17m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/624e0757eed7492aa54694ea7f86a519 : SUCCESS in 13m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/04d140a97e9a481a8fe70b7d2502ea9d : SUCCESS in 1h 57m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8d72356bc05c4d83842e29ee75ea311d : SUCCESS in 1h 16m 13s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/122365e4ce4346efa52052e370843f62 : SUCCESS in 1h 06m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/817b410f1a5748358efae28bef06323a : SUCCESS in 1h 34m 13s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4f57ee66c9a42a4bfc94eab308e5311 : SUCCESS in 34m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/b8d02961fa46445a86570ab0e5a4d094 : SUCCESS in 30m 58s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/0a86b5d0371b4d35898350e303e86e6e : SUCCESS in 38m 05s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/97971701c2c049468a9c6b505a5b817e : SUCCESS in 49m 58s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2ddbe786c8e744b98a6e33b5a7dc4ba2 : SUCCESS in 19m 13s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0fc80b2850874d0893083bdd3c0f6129 : SUCCESS in 1h 24m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e173f4e70304d1abc693f25081a4805 : SUCCESS in 1h 03m 40s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/3b74060c47cb4c6d924842127d34bc9d : SUCCESS in 45m 06s","accounts_in_message":[],"_revision_number":14},{"id":"5878958b6029bf65c59fb7ff8c885b88102cdb67","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 02:13:26.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"748cfeb8992238b745753f405a2cba651f2d04d5","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 02:25:55.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"14c486d694a50e472bf100b32aeb857849ad4696","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 02:26:18.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"68ac40a1afeaa945fecb4c0e48e817669cf21e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 04:09:59.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cdcb668dd6941598256e3c0c61a3be8 : SUCCESS in 14m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7031dc52115144f6b240eb540f527f96 : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43003498b8ca42ea84ccb5cf1181ca45 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/935a90246cab43e6b205e1a3d0db2da4 : SUCCESS in 15m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94016642346549f2a395c98d44aa7885 : SUCCESS in 13m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fd6a51ea0c04fc2b095a5e53ffaf2f4 : SUCCESS in 14m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1333585946d8423aa5e3a65a0ac12abf : SUCCESS in 15m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/db7b59db57c44e9798852aec1fac0c69 : SUCCESS in 1h 29m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3394f423711b453ba7b6d8ab4a28b1ac : SUCCESS in 1h 01m 08s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f37377d697674d9bba992dc982a36cf8 : SUCCESS in 1h 05m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11b82171d6354e5484dd25818e4a95fc : SUCCESS in 1h 27m 05s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/881c935638d04ef08c171995ffe9240d : SUCCESS in 35m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e9183040e36943f18137999e5ff7036c : SUCCESS in 33m 29s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/5bd2bf00bceb4ea9adf22127bbc7be29 : SUCCESS in 42m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/e1f1a708d7c54064b2111361d04756a7 : SUCCESS in 42m 55s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9b7aa182bd814822b1e3f9d853f187a5 : SUCCESS in 17m 50s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/cf14e999d844486398730fff4b16221a : SUCCESS in 1h 37m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1dc16a6fc06498697a4c0be038f05e2 : SUCCESS in 1h 10m 30s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/827da21e65d74d9eaa8e8ce8fff1ca15 : SUCCESS in 46m 59s","accounts_in_message":[],"_revision_number":15},{"id":"6bfa793b3dd1a3ae9b2506530ccf0ed50361a9c8","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-25 05:38:32.000000000","message":"Patch Set 15: Code-Review+1\n\nOk. Only system admins can create, update and delete the limits. Even though the user is domain admin it cannot create limits within their domain. Looks good for the get and list scope to domain users.","accounts_in_message":[],"_revision_number":15},{"id":"970dbcfbb70286f25c61dec88a7f6ead4d5ccf33","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-25 16:30:17.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f3650078836d4abccc8f59d5fd6cce640925e76a","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 16:33:14.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"49b9b96d6accfd639a1b6b9af5ed31549e5a4d53","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-25 17:14:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3d3695dbcf48774de6f0a3ed3b972b1637db464c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-25 17:16:56.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"aef50807d87183dcb18b6fdb683339929146f1f8","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-25 17:27:04.000000000","message":"Patch Set 15: Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"3eb5986e0d085e2dee8beb9d0ae373cd0354d107","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 17:27:38.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"5136bf57f735416475866652ae81b9025a8e480e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 20:30:02.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f88a960e8bb241bd9accb6741fba62fe : SUCCESS in 19m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1995901e0c9e4b619c1c64c081c2c124 : SUCCESS in 8m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9c5cfceff7644f2a9347f0ad8ece7ac : SUCCESS in 20m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cb1bad430be444ba5b154829c61e248 : SUCCESS in 15m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80089f6d14e14f759f0ab77423c21162 : SUCCESS in 16m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f17159f0d2684fe9b3b41a92f3ff571f : SUCCESS in 13m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5f5cefcc7ae34ed79314a58640d38876 : TIMED_OUT in 2h 12m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8191d6dd3a18453cb8e7f3be7945c35c : SUCCESS in 1h 09m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/21ee1e3e21624eee99d95107481fd76f : SUCCESS in 1h 11m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7a43520adc04be0922205501ef6ca14 : SUCCESS in 1h 19m 52s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae83b2b65c5c4216b1af54d5345e7975 : SUCCESS in 32m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3e45f1d501b74079b48d231e94235a75 : SUCCESS in 35m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb0d21278e3b4db7977013488ffe9b98 : SUCCESS in 1h 05m 06s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/27c527f56ed8464b8d2eeb5d41d071d9 : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":15},{"id":"aa4b141e36a65d00a3d7ab0f557212ca62ae15d3","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 20:34:32.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"019a42b3a350150c816acfd19da74d45a8eabf4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 23:25:41.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04ff85c156ca42828a860ed1db512095 : SUCCESS in 16m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff1c59470529489ab22c731d8d72ed42 : TIMED_OUT in 40m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3da2b86be8544eed8f7891d416e92f08 : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/722e103e748c40efa0bf19104ea3b2e9 : SUCCESS in 15m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5612d9a957344344b09b2007570ff0a1 : SUCCESS in 12m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3da4b236ccda4bc187818f05f59c1bb6 : SUCCESS in 12m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0d1048a7364b54942a8bf7fca4311d : SUCCESS in 17m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/2c1e03a6ecfd44fc999278d62b06be80 : SUCCESS in 1h 53m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0b1cce5904bc4d268b188cefbdfc114f : SUCCESS in 1h 04m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7141adb24c5d47b4b9c505fb037c9c21 : SUCCESS in 1h 06m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a092435ae0c7432882ce3d3e8b9d33b8 : SUCCESS in 1h 30m 59s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/2aacd37c1ab64a3bbb86a67c8a939280 : SUCCESS in 43m 55s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/00abe8540cf0421baced2a3747ba2c59 : SUCCESS in 41m 12s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/b55bf87045d74cdbb0487f872f7a7da3 : SUCCESS in 37m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/536f09b9c99d4808bf6294e5f87a8e64 : SUCCESS in 48m 38s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1fc9f4d16e5d436285e39e6ea7912610 : SUCCESS in 20m 18s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/3f30f8f89e45477099ec6abbb219adbc : SUCCESS in 1h 51m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43e9937fede0429aa093beae58da84a0 : SUCCESS in 1h 26m 58s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/0749caf8b76a4cd78569c491c4c69e5d : SUCCESS in 44m 01s","accounts_in_message":[],"_revision_number":15},{"id":"4e03d3a268c278fd6ad8eebd2037a7786b5556e6","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 23:40:36.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"7ecff849880641513331be118da3533fb12f24d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 01:24:32.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f200b77d524541eba90ae04065e36d6b : SUCCESS in 18m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4feff3eba6e4457e83727e15a0a8299e : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88651105dddd467d9106423a1674aa65 : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6a987ad2c03400bb91742ad5d6fdcbc : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0123242bca0421ea48b3b7f82c039cd : SUCCESS in 12m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14cf39ca706c47829d7a8a3f30b31a05 : SUCCESS in 12m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80ae5803d2f0450093556eef468feba5 : SUCCESS in 12m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1c25a998f73f4015933cd794a55728e9 : SUCCESS in 1h 37m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9050d9f3bce04be0b2fedc9115c796d7 : SUCCESS in 48m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b35eecb11d2d4b8582421cd9026d9fb0 : SUCCESS in 1h 05m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40d8faba99444b5e82cf9c41cb9edeae : SUCCESS in 1h 21m 56s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/d3955f8f1a9b4d20a3620b759a93ecee : SUCCESS in 36m 12s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/abbf13c745fa489b8ade6615d497d5ee : SUCCESS in 39m 33s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/ecba4005d86b4373a75ac87918129acc : SUCCESS in 33m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/b73e3bf5a0ca48f8be40d05a134d7d25 : SUCCESS in 39m 27s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4f0c12cc5766417880b00fa4a5add7dc : SUCCESS in 16m 24s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/719dcee407c3409f99c087d142c08794 : SUCCESS in 1h 27m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db8ee1a7bcd247158c9d5275d8ba1f07 : SUCCESS in 1h 08m 42s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/56470c15ffab4e81951b75841a230e96 : SUCCESS in 34m 56s","accounts_in_message":[],"_revision_number":15},{"id":"59dfa07f8428df66fb33eb332c1610951a5d465d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 01:24:48.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"bed4a60056392e18aad51306adea55494e55cc82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 03:12:50.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3423d09f6d00482c922824b3601d1f51 : SUCCESS in 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da171e2bec5b4895b092bfff83bdf437 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b47f1ca4aa9f4f3c9bfde8e71c1e5a71 : FAILURE in 16m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d2d3f8c2987480bbf631d1b3baa330a : SUCCESS in 12m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17728ab87b92402b9a703ba80436a4d2 : SUCCESS in 15m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf0dbc46cf324d849d6defca33823f87 : SUCCESS in 12m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/83e14b39697040d8a87794b7ee3c01f2 : SUCCESS in 1h 42m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/49516a1907954a87b6c1e02f318abc3f : SUCCESS in 1h 10m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c5624ecc1b794bc7981764847be243ac : SUCCESS in 59m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4f2cfcbc06fc4151ade1217ee1b310c8 : SUCCESS in 1h 38m 10s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/075a3c92e24c433ab2be6764040310b8 : SUCCESS in 32m 17s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ee144ef8482d4ee798c6b4c41cb39be3 : SUCCESS in 36m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aea1165030994877ab1743655f8a03bf : SUCCESS in 1h 09m 22s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/07931110aaac4f6881a6b5b1c47646f7 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":15},{"id":"4dfa9a3592abfc76362feef4de1d41943b8b07dd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-26 03:45:42.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"29b03567b73dcecbbdcb61a673221c9f0f874f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 05:41:50.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5121bbfab60d45ca86bca16375e1af4d : SUCCESS in 18m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53dae835f3484c239ae668155809c9f3 : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b16714e9f01d416f8c1996e00c0612d8 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9df30b4213274729ac67736c912764ef : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba869b4404b5477bab868af8d0e24083 : SUCCESS in 12m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94792068187b4237bd5dd9b1f2be1959 : SUCCESS in 15m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285762fc4f49428d8071c29ec4aabbe1 : SUCCESS in 13m 04s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f150954fafd04015aaad1c1f2f5ad171 : SUCCESS in 1h 48m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/071d3bbf0a0140b39291fba5b1b319d8 : SUCCESS in 1h 12m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4db5c18a343f4c9b8943c4d7f9d9d03e : SUCCESS in 1h 00m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2afda616d474a8a866bd6f5cb9f5642 : SUCCESS in 1h 31m 11s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/81635d49e5794b68859c322b014ba25e : SUCCESS in 35m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f0c612e4a3c8426e9e03eb7b9af176ae : SUCCESS in 37m 03s\n- keystone-dsvm-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/bb343efbb5c84d88bd10c0dcbd9b76f4 : SUCCESS in 40m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 https://zuul.opendev.org/t/openstack/build/dcb00567b6d643e19af0d248730093f8 : SUCCESS in 41m 52s (non-voting)\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9dd3a54d3b414fea878b840251acbcd6 : SUCCESS in 14m 56s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/95e4898a37d04679b72fc684f09e4604 : SUCCESS in 1h 33m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fdb57dfc1d14782a8f878910e8feda6 : SUCCESS in 1h 14m 05s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/a92b40d1202a4c5fafe699a435ce8f9d : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":15},{"id":"a9aee1d8a42a9bc02b5a7f158cdfb8a9ea7161f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 05:42:05.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"ba5a1a63bbea6dee80c579f89ab47e79489855ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 07:48:46.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2850128c04845fab5c39cc3336a7d2e : SUCCESS in 15m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f7e0f841715460797a1266c003a4324 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aba17efa9986474f98dea0cc29eb4756 : SUCCESS in 18m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d10cade98ee14dd9b2eda982b28ab69f : SUCCESS in 18m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90226eca100e482bb0bba1daf0163595 : SUCCESS in 17m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b9bf512acae42c9bbe4964abb6401b8 : SUCCESS in 14m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/79ef4bfda9e04cdb9fa8ebbc4cc7ed74 : SUCCESS in 1h 58m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7ecda3b59e9b4a45926aba219315deb5 : SUCCESS in 1h 13m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ab0dc1b87fe843d3926b0ca120efde54 : SUCCESS in 1h 13m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95e578568591451bb6cd3fd249b36d43 : SUCCESS in 1h 27m 57s\n- keystone-dsvm-functional https://zuul.opendev.org/t/openstack/build/17f46986b341490f9288313316f650e9 : SUCCESS in 36m 36s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a426526d977845c7a75a21275d7e8e41 : SUCCESS in 42m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a41143f0abc940a58f442e72c7c03dd5 : SUCCESS in 1h 15m 46s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/a3e2ccb5d582411d859f9df6a3781011 : SUCCESS in 46m 19s","accounts_in_message":[],"_revision_number":15},{"id":"f4a2b9a5ec8d4097fdc9348bb78910123e59467d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 07:48:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"232c0d2752d0d3e8604bae349c3a24d083c0bd9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 07:50:12.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238599b818b2484eb17c56dc7bb49a2a : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"f249c9e2b0f39b688ba356feaca7818adfc9f739","revisions":{"3103b57080101701b6b493638c11aab292138bf6":{"kind":"REWORK","_number":1,"created":"2018-11-29 21:34:28.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/1"}}},"commit":{"parents":[{"commit":"177558ee1c7c26c6761536ae767c8befc6ddd8e6","subject":"Update limit policies for system admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/177558ee1c7c26c6761536ae767c8befc6ddd8e6"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:33:13.000000000","tz":0},"subject":"Add tests for domain users interacting with limits","message":"Add tests for domain users interacting with limits\n\nThis commit introduces some tests that explicitly show how domain\nusers are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3103b57080101701b6b493638c11aab292138bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3103b57080101701b6b493638c11aab292138bf6"}]},"branch":"refs/heads/master"},"6950ecdfd4e726cee763ddb5dba44e5b5cac73d3":{"kind":"REWORK","_number":2,"created":"2018-11-30 23:37:19.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/2"}}},"commit":{"parents":[{"commit":"7984ffb6de85dd2f4ca027acfb714e16558d9809","subject":"Update limit policies for system admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7984ffb6de85dd2f4ca027acfb714e16558d9809"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-30 23:33:04.000000000","tz":0},"subject":"Add tests for domain users interacting with limits","message":"Add tests for domain users interacting with limits\n\nThis commit introduces some tests that explicitly show how domain\nusers are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6950ecdfd4e726cee763ddb5dba44e5b5cac73d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6950ecdfd4e726cee763ddb5dba44e5b5cac73d3"}]},"branch":"refs/heads/master"},"9d9811fe55529932ba3343aa741d1ec24078b54c":{"kind":"REWORK","_number":3,"created":"2018-12-07 17:54:59.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/3"}}},"commit":{"parents":[{"commit":"d3fdfe7bec5a405cfb201e274b60c521623d3821","subject":"Update limit policies for system admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d3fdfe7bec5a405cfb201e274b60c521623d3821"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-12-07 17:49:50.000000000","tz":0},"subject":"Add tests for domain users interacting with limits","message":"Add tests for domain users interacting with limits\n\nThis commit introduces some tests that explicitly show how domain\nusers are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9d9811fe55529932ba3343aa741d1ec24078b54c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9d9811fe55529932ba3343aa741d1ec24078b54c"}]},"branch":"refs/heads/master"},"c393ad78ef8c66b2df49ddae8b649df5a45019a3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-12-11 09:15:47.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/4"}}},"commit":{"parents":[{"commit":"d6378ae22a6c562e0684816c579b553cc4aec1f0","subject":"Update limit policies for system admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d6378ae22a6c562e0684816c579b553cc4aec1f0"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-12-11 09:15:13.000000000","tz":0},"subject":"Add tests for domain users interacting with limits","message":"Add tests for domain users interacting with limits\n\nThis commit introduces some tests that explicitly show how domain\nusers are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c393ad78ef8c66b2df49ddae8b649df5a45019a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c393ad78ef8c66b2df49ddae8b649df5a45019a3"}]},"branch":"refs/heads/master"},"8b303ba13d765aeeaa60deca8fa1f0f9dc2ac611":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2018-12-18 22:28:55.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/5"}}},"commit":{"parents":[{"commit":"afc022ed364b536758f0a3ea19a1d25f98e835d2","subject":"Update limit policies for system admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/afc022ed364b536758f0a3ea19a1d25f98e835d2"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-12-18 22:18:35.000000000","tz":0},"subject":"Add tests for domain users interacting with limits","message":"Add tests for domain users interacting with limits\n\nThis commit introduces some tests that explicitly show how domain\nusers are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8b303ba13d765aeeaa60deca8fa1f0f9dc2ac611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8b303ba13d765aeeaa60deca8fa1f0f9dc2ac611"}]},"branch":"refs/heads/master"},"39b683d0bae0c429a588a49b3dd76ab6ab5b3d6b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2018-12-19 23:14:26.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/6"}}},"commit":{"parents":[{"commit":"c446acc6949b162aee49b04ee4ecbaa4f59cea5e","subject":"Update limit policies for system admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c446acc6949b162aee49b04ee4ecbaa4f59cea5e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-12-19 23:08:54.000000000","tz":0},"subject":"Add tests for domain users interacting with limits","message":"Add tests for domain users interacting with limits\n\nThis commit introduces some tests that explicitly show how domain\nusers are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/39b683d0bae0c429a588a49b3dd76ab6ab5b3d6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/39b683d0bae0c429a588a49b3dd76ab6ab5b3d6b"}]},"branch":"refs/heads/master"},"2777179f3684687c15c00b1b7bae09ada8cfda22":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-01-08 18:24:13.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/7"}}},"commit":{"parents":[{"commit":"4403b9864bb2cc6a29b977b28ee4e811e71ff24b","subject":"Update limit policies for system admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4403b9864bb2cc6a29b977b28ee4e811e71ff24b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-01-08 18:18:16.000000000","tz":0},"subject":"Add tests for domain users interacting with limits","message":"Add tests for domain users interacting with limits\n\nThis commit introduces some tests that explicitly show how domain\nusers are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2777179f3684687c15c00b1b7bae09ada8cfda22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2777179f3684687c15c00b1b7bae09ada8cfda22"}]},"branch":"refs/heads/master"},"e0f8fce136d8b7e5429d7bae7f224fd94219eaf4":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":8,"created":"2019-01-09 14:59:44.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/8"}}},"commit":{"parents":[{"commit":"1d4e40252884f054d692e82e170d3f69228ef7ee","subject":"Update limit policies for system admin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1d4e40252884f054d692e82e170d3f69228ef7ee"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-01-09 14:57:26.000000000","tz":0},"subject":"Allow domain users to access the limit API","message":"Allow domain users to access the limit API\n\nThis commit adds domain-scope to the scope_types for limit policies,\nallowing domain users to access those APIs when enforce_scope is\nenabled. This commit also introduces some tests that explicitly show\nhow domain users are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e0f8fce136d8b7e5429d7bae7f224fd94219eaf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e0f8fce136d8b7e5429d7bae7f224fd94219eaf4"}]},"branch":"refs/heads/master"},"d31b767c131ab4c49ef8d5ef89b2218b6cbfd2dd":{"kind":"REWORK","_number":9,"created":"2019-03-05 15:40:35.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/9"}}},"commit":{"parents":[{"commit":"e6e3857ffcfbd1b82f114e9bf1f0974748f23b5b","subject":"Merge \"Switch federation check jobs to opensuse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e6e3857ffcfbd1b82f114e9bf1f0974748f23b5b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-03-05 15:37:37.000000000","tz":0},"subject":"Allow domain users to access the limit API","message":"Allow domain users to access the limit API\n\nThis commit adds domain-scope to the scope_types for limit policies,\nallowing domain users to access those APIs when enforce_scope is\nenabled. This commit also introduces some tests that explicitly show\nhow domain users are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nThe policy that protects the GET /v3/limits/{limit_id} API was broken\ninto two separate policies. This was done because writing a single\ncheck_str for project-scoped, domain-scoped, and system-scoped tokens\nwhen project_id or domain_id might not be in the limit body exposed a\nFalse positive in the policy engine logic. For example, if a user\ncalls the GET /v3/limits/{limit_id} API with a domain-scoped token and\nthe policy contains `project_id:limit.project.id` to enforce project\nscope, the check will be true with domain limits with domain scoped\ntokens. This is because project_id is None and limit.project.id is\nalso None, resulting in a match.\n\nThis commit also modifies the GET /v3/limit policy to allow project\nusers to filter responses by project_id, which isn\u0027t entirely useful\noutside of just calling the API with a project-scoped token.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d31b767c131ab4c49ef8d5ef89b2218b6cbfd2dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d31b767c131ab4c49ef8d5ef89b2218b6cbfd2dd"}]},"branch":"refs/heads/master"},"fdcb723129865573d646fcf991715e1b650168ce":{"kind":"REWORK","_number":10,"created":"2019-03-05 19:14:17.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/10"}}},"commit":{"parents":[{"commit":"e6e3857ffcfbd1b82f114e9bf1f0974748f23b5b","subject":"Merge \"Switch federation check jobs to opensuse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e6e3857ffcfbd1b82f114e9bf1f0974748f23b5b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-03-05 18:42:09.000000000","tz":0},"subject":"Allow domain users to access the limit API","message":"Allow domain users to access the limit API\n\nThis commit adds domain-scope to the scope_types for limit policies,\nallowing domain users to access those APIs when enforce_scope is\nenabled. This commit also introduces some tests that explicitly show\nhow domain users are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nThe policy that protects the GET /v3/limits/{limit_id} API was broken\ninto three separate policies. This was done because writing a single\ncheck_str for project-scoped, domain-scoped, and system-scoped tokens\nwhen project_id or domain_id might not be in the limit body exposed a\nFalse positive in the policy engine logic. For example, if a user\ncalls the GET /v3/limits/{limit_id} API with a domain-scoped token and\nthe policy contains `project_id:limit.project.id` to enforce project\nscope, the check will be true with domain limits with domain scoped\ntokens. This is because project_id is None and limit.project.id is\nalso None, resulting in a match.\n\nThis commit also modifies the GET /v3/limit policy to allow project\nusers to filter responses by project_id, which isn\u0027t entirely useful\noutside of just calling the API with a project-scoped token.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fdcb723129865573d646fcf991715e1b650168ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fdcb723129865573d646fcf991715e1b650168ce"}]},"branch":"refs/heads/master"},"b9510d489819612c60fb713bf3356e64fda7ffac":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-03-06 20:28:05.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/11"}}},"commit":{"parents":[{"commit":"e6e3857ffcfbd1b82f114e9bf1f0974748f23b5b","subject":"Merge \"Switch federation check jobs to opensuse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e6e3857ffcfbd1b82f114e9bf1f0974748f23b5b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-03-06 20:19:18.000000000","tz":0},"subject":"Allow domain users to access the limit API","message":"Allow domain users to access the limit API\n\nThis commit adds domain-scope to the scope_types for limit policies,\nallowing domain users to access those APIs when enforce_scope is\nenabled. This commit also introduces some tests that explicitly show\nhow domain users are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nThe policy that protects the GET /v3/limits/{limit_id} API was broken\ninto three separate policies. This was done because writing a single\ncheck_str for project-scoped, domain-scoped, and system-scoped tokens\nwhen project_id or domain_id might not be in the limit body exposed a\nFalse positive in the policy engine logic. For example, if a user\ncalls the GET /v3/limits/{limit_id} API with a domain-scoped token and\nthe policy contains `project_id:limit.project.id` to enforce project\nscope, the check will be true with domain limits with domain scoped\ntokens. This is because project_id is None and limit.project.id is\nalso None, resulting in a match.\n\nThis commit also modifies the GET /v3/limit policy to allow project\nusers to filter responses by project_id, which isn\u0027t entirely useful\noutside of just calling the API with a project-scoped token.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\nPartial-Bug: 1818736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b9510d489819612c60fb713bf3356e64fda7ffac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b9510d489819612c60fb713bf3356e64fda7ffac"}]},"branch":"refs/heads/master"},"9ba688fc3a49356f3150d6aa2c55edf1c859e05c":{"kind":"REWORK","_number":12,"created":"2019-07-02 15:03:01.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/23/621023/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/12"}}},"commit":{"parents":[{"commit":"41f812fd5936d5c52886541226de72e0d205d890","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/41f812fd5936d5c52886541226de72e0d205d890"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-02 14:54:28.000000000","tz":0},"subject":"Allow domain users to access the limit API","message":"Allow domain users to access the limit API\n\nThis commit adds domain-scope to the scope_types for limit policies,\nallowing domain users to access those APIs when enforce_scope is\nenabled. This commit also introduces some tests that explicitly show\nhow domain users are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nThe policy that protects the GET /v3/limits/{limit_id} API was broken\ninto three separate policies. This was done because writing a single\ncheck_str for project-scoped, domain-scoped, and system-scoped tokens\nwhen project_id or domain_id might not be in the limit body exposed a\nFalse positive in the policy engine logic. For example, if a user\ncalls the GET /v3/limits/{limit_id} API with a domain-scoped token and\nthe policy contains `project_id:limit.project.id` to enforce project\nscope, the check will be true with domain limits with domain scoped\ntokens. This is because project_id is None and limit.project.id is\nalso None, resulting in a match.\n\nThis commit also modifies the GET /v3/limit policy to allow project\nusers to filter responses by project_id, which isn\u0027t entirely useful\noutside of just calling the API with a project-scoped token.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\nPartial-Bug: 1818736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ba688fc3a49356f3150d6aa2c55edf1c859e05c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ba688fc3a49356f3150d6aa2c55edf1c859e05c"}]},"branch":"refs/heads/master"},"fcc72d7b2514e0e2791fbe20dca00276e2e5f42a":{"kind":"REWORK","_number":13,"created":"2019-09-24 17:30:57.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/23/621023/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/13"}}},"commit":{"parents":[{"commit":"e992b79fa3892bfd12ca9d45a67dc2b74069ecd7","subject":"Merge \"Allow system/domain scope for assignment tree list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e992b79fa3892bfd12ca9d45a67dc2b74069ecd7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-24 17:29:15.000000000","tz":-420},"subject":"Allow domain users to access the limit API","message":"Allow domain users to access the limit API\n\nThis commit adds domain-scope to the scope_types for limit policies,\nallowing domain users to access those APIs when enforce_scope is\nenabled. This commit also introduces some tests that explicitly show\nhow domain users are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nThe policy that protects the GET /v3/limits/{limit_id} API was broken\ninto three separate policies. This was done because writing a single\ncheck_str for project-scoped, domain-scoped, and system-scoped tokens\nwhen project_id or domain_id might not be in the limit body exposed a\nFalse positive in the policy engine logic. For example, if a user\ncalls the GET /v3/limits/{limit_id} API with a domain-scoped token and\nthe policy contains `project_id:limit.project.id` to enforce project\nscope, the check will be true with domain limits with domain scoped\ntokens. This is because project_id is None and limit.project.id is\nalso None, resulting in a match.\n\nThis commit also modifies the GET /v3/limit policy to allow project\nusers to filter responses by project_id, which isn\u0027t entirely useful\noutside of just calling the API with a project-scoped token.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\nPartial-Bug: 1818736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fcc72d7b2514e0e2791fbe20dca00276e2e5f42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fcc72d7b2514e0e2791fbe20dca00276e2e5f42a"}]},"branch":"refs/heads/master"},"91edc773b11d45358eb36d0f006b60d400dd0b54":{"kind":"REWORK","_number":14,"created":"2019-09-24 22:36:15.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/23/621023/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/14"}}},"commit":{"parents":[{"commit":"e992b79fa3892bfd12ca9d45a67dc2b74069ecd7","subject":"Merge \"Allow system/domain scope for assignment tree list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e992b79fa3892bfd12ca9d45a67dc2b74069ecd7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-24 22:30:47.000000000","tz":-420},"subject":"Allow domain users to access the limit API","message":"Allow domain users to access the limit API\n\nThis commit adds domain-scope to the scope_types for limit policies,\nallowing domain users to access those APIs when enforce_scope is\nenabled. This commit also introduces some tests that explicitly show\nhow domain users are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nThe policy that protects the GET /v3/limits/{limit_id} API was broken\ninto three separate policies. This was done because writing a single\ncheck_str for project-scoped, domain-scoped, and system-scoped tokens\nwhen project_id or domain_id might not be in the limit body exposed a\nFalse positive in the policy engine logic. For example, if a user\ncalls the GET /v3/limits/{limit_id} API with a domain-scoped token and\nthe policy contains `project_id:limit.project.id` to enforce project\nscope, the check will be true with domain limits with domain scoped\ntokens. This is because project_id is None and limit.project.id is\nalso None, resulting in a match.\n\nThis commit also modifies the GET /v3/limit policy to allow project\nusers to filter responses by project_id, which isn\u0027t entirely useful\noutside of just calling the API with a project-scoped token.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\nPartial-Bug: 1818736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/91edc773b11d45358eb36d0f006b60d400dd0b54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/91edc773b11d45358eb36d0f006b60d400dd0b54"}]},"branch":"refs/heads/master"},"f249c9e2b0f39b688ba356feaca7818adfc9f739":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2019-09-25 02:25:55.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/23/621023/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/23/621023/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/23/621023/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/23/621023/15"}}},"commit":{"parents":[{"commit":"e992b79fa3892bfd12ca9d45a67dc2b74069ecd7","subject":"Merge \"Allow system/domain scope for assignment tree list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e992b79fa3892bfd12ca9d45a67dc2b74069ecd7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:06:09.000000000","tz":0},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-09-25 02:14:17.000000000","tz":-420},"subject":"Allow domain users to access the limit API","message":"Allow domain users to access the limit API\n\nThis commit adds domain-scope to the scope_types for limit policies,\nallowing domain users to access those APIs when enforce_scope is\nenabled. This commit also introduces some tests that explicitly show\nhow domain users are expected to behave with the limits API. A\nsubsequent patch will do the same for project users.\n\nThis commit also modifies the GET /v3/limit policy to allow project\nusers to filter responses by project_id, which isn\u0027t entirely useful\noutside of just calling the API with a project-scoped token.\n\nChange-Id: I9b38f3fd2f83efd508b2d9a6c323bbaa7169d4cd\nRelated-Bug: 1805880\nPartial-Bug: 1818736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f249c9e2b0f39b688ba356feaca7818adfc9f739"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f249c9e2b0f39b688ba356feaca7818adfc9f739"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
