)]}'
{"id":"openstack%2Fkeystone~621025","triplet_id":"openstack%2Fkeystone~master~Ie0f333a9e8b60154711a24ba7d9ade531217eb71","project":"openstack/keystone","branch":"master","topic":"train-rc1-policy-fixes","hashtags":[],"change_id":"Ie0f333a9e8b60154711a24ba7d9ade531217eb71","subject":"Remove limit policies from policy.v3cloudsample.json","status":"MERGED","created":"2018-11-29 21:34:28.000000000","updated":"2019-09-27 00:09:18.000000000","submitted":"2019-09-27 00:07:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"621025-1569542861687-b42f53c0","meta_rev_id":"a971f5a8d96763286ffcf84aaf542f615d5df9d8","_number":621025,"virtual_id_number":621025,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 00:07:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-09-25 02:26:08.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-09-25 05:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-25 04:11:58.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-25 04:11:58.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-20 07:18:55.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-02-15 20:41:44.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-09-25 02:26:08.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-09-25 04:11:58.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-09-25 05:40:26.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-09-27 00:07:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7aea45f44d3894a9e78452855c462471b2eece32","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-11-29 21:34:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fde04122ea731e1da23b39b6427a59008330aac2","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-11-29 21:52:35.000000000","message":"Topic changed from bug/1805880 to clean-up-v3-cloud-sample","accounts_in_message":[],"_revision_number":1},{"id":"08762627d9c6dd6722e52c2424ee79b21d642b29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-30 01:31:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/1/check/openstack-tox-cover/af9f48c/cover/ : SUCCESS in 12m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/1/check/openstack-tox-lower-constraints/b85812c/ : SUCCESS in 38m 34s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/1/check/openstack-tox-pep8/8a98551/ : FAILURE in 7m 10s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/1/check/openstack-tox-py27/7277b92/ : SUCCESS in 12m 00s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/1/check/openstack-tox-py35/4928cc8/ : SUCCESS in 12m 39s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/1/check/openstack-tox-py36/8d596e3/ : SUCCESS in 12m 24s\n- openstack-tox-docs http://logs.openstack.org/25/621025/1/check/openstack-tox-docs/ed2b1e5/html/ : SUCCESS in 8m 09s\n- grenade-py3 http://logs.openstack.org/25/621025/1/check/grenade-py3/79ee044/ : SUCCESS in 1h 08m 29s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/1/check/build-openstack-releasenotes/cdabdb5/html/ : SUCCESS in 9m 34s","accounts_in_message":[],"_revision_number":1},{"id":"907785c799c8dcf653ed9d783e9b09866a0b1e57","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-11-30 23:37:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e4ddfa707033b9933477ff7935628660f76174b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-01 04:30:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/2/check/openstack-tox-cover/9540312/cover/ : SUCCESS in 14m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/2/check/openstack-tox-lower-constraints/d9c9043/ : SUCCESS in 14m 37s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/2/check/openstack-tox-pep8/d4ddb8a/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/2/check/openstack-tox-py27/b24745b/ : SUCCESS in 11m 14s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/2/check/openstack-tox-py35/fce00fe/ : SUCCESS in 11m 47s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/2/check/openstack-tox-py36/516c984/ : SUCCESS in 10m 51s\n- openstack-tox-docs http://logs.openstack.org/25/621025/2/check/openstack-tox-docs/000695a/html/ : SUCCESS in 6m 03s\n- grenade-py3 http://logs.openstack.org/25/621025/2/check/grenade-py3/48d163c/ : SUCCESS in 56m 53s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/2/check/build-openstack-releasenotes/bead07f/html/ : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":2},{"id":"631d2633cee4a37e66f3b8a5e4f6044008841d8d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-03 15:13:52.000000000","message":"Topic changed from bug/1805880 to clean-up-v3-cloud-sample","accounts_in_message":[],"_revision_number":2},{"id":"66f1227c319a9055a232f4abddeae078b009d424","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-07 17:54:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e5f469b7b91b31e452143a5853c12d12ec66e5ba","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-07 18:00:55.000000000","message":"Topic changed from bug/1805880 to clean-up-v3-cloud-sample","accounts_in_message":[],"_revision_number":3},{"id":"ddcf5842537fe5f046982c3e4df20e0eaa360212","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-07 23:39:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/3/check/openstack-tox-cover/97285f7/cover/ : SUCCESS in 13m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/3/check/openstack-tox-lower-constraints/52f98bf/ : SUCCESS in 15m 25s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/3/check/openstack-tox-pep8/968f678/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/3/check/openstack-tox-py27/102899e/ : SUCCESS in 9m 39s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/3/check/openstack-tox-py35/f87eb18/ : SUCCESS in 10m 38s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/3/check/openstack-tox-py36/f7c8773/ : SUCCESS in 9m 34s\n- openstack-tox-docs http://logs.openstack.org/25/621025/3/check/openstack-tox-docs/d759c4e/html/ : SUCCESS in 6m 45s\n- grenade-py3 http://logs.openstack.org/25/621025/3/check/grenade-py3/bb3e7dd/ : SUCCESS in 1h 01m 38s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/3/check/build-openstack-releasenotes/387e1c3/html/ : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":3},{"id":"c1ed02c07cc67e8bf2de6aedac162589e6314214","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-11 09:15:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"d8b395f854ab607450e58b3c3fa87d85238086d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 11:32:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/4/check/openstack-tox-cover/8b528d2/cover/ : SUCCESS in 13m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/4/check/openstack-tox-lower-constraints/b15dc5f/ : SUCCESS in 17m 55s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/4/check/openstack-tox-pep8/0808c70/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/4/check/openstack-tox-py27/2f9999c/ : SUCCESS in 12m 15s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/4/check/openstack-tox-py35/aaa4828/ : SUCCESS in 10m 43s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/4/check/openstack-tox-py36/571b842/ : SUCCESS in 9m 47s\n- openstack-tox-docs http://logs.openstack.org/25/621025/4/check/openstack-tox-docs/eeb445b/html/ : SUCCESS in 7m 39s\n- grenade-py3 http://logs.openstack.org/25/621025/4/check/grenade-py3/1e6c391/ : FAILURE in 59m 15s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/4/check/build-openstack-releasenotes/c0248ef/html/ : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":4},{"id":"db65ba28c372793bcb2d232834e7d1a6c469c303","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-11 13:10:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"84ccd7e828bf45beb283035c20dd1e2da0c8c9eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-11 14:19:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/4/check/openstack-tox-cover/627168b/cover/ : SUCCESS in 12m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/4/check/openstack-tox-lower-constraints/791c6bf/ : SUCCESS in 15m 33s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/4/check/openstack-tox-pep8/52d3ddb/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/4/check/openstack-tox-py27/3327f4a/ : SUCCESS in 10m 18s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/4/check/openstack-tox-py35/3ecb547/ : SUCCESS in 11m 44s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/4/check/openstack-tox-py36/3d9b7da/ : SUCCESS in 11m 00s\n- openstack-tox-docs http://logs.openstack.org/25/621025/4/check/openstack-tox-docs/1f41714/html/ : SUCCESS in 6m 47s\n- grenade-py3 http://logs.openstack.org/25/621025/4/check/grenade-py3/86f46f8/ : SUCCESS in 59m 52s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/4/check/build-openstack-releasenotes/47e1e1d/html/ : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":4},{"id":"ded884384b80b3786e37a7834dabd3efc46bb916","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-18 22:28:55.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"54084aa3a76e2d029ce4c937153198bb47a39075","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-18 23:51:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/5/check/openstack-tox-cover/7ee332e/ : FAILURE in 12m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/5/check/openstack-tox-lower-constraints/0d04292/ : FAILURE in 17m 11s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/5/check/openstack-tox-pep8/f6f7264/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/5/check/openstack-tox-py27/382bf2b/ : FAILURE in 11m 42s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/5/check/openstack-tox-py35/74ae17c/ : FAILURE in 11m 11s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/5/check/openstack-tox-py36/11b796b/ : FAILURE in 10m 02s\n- openstack-tox-docs http://logs.openstack.org/25/621025/5/check/openstack-tox-docs/1023a54/html/ : SUCCESS in 7m 30s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/5/check/build-openstack-releasenotes/8e85db3/html/ : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":5},{"id":"e19a4897b330b5a71d39aa4ce7c0efdd97073146","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-19 23:14:26.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"85dcd94992ff7f317f1e7f80a6ce112338b56be5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-19 23:40:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/6/check/openstack-tox-cover/c8ba3cf/cover/ : SUCCESS in 10m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/6/check/openstack-tox-lower-constraints/bbe9519/ : SUCCESS in 18m 24s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/6/check/openstack-tox-pep8/cedf036/ : SUCCESS in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/6/check/openstack-tox-py27/313785d/ : SUCCESS in 12m 16s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/6/check/openstack-tox-py35/f7d91b1/ : SUCCESS in 9m 13s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/6/check/openstack-tox-py36/8a5adca/ : SUCCESS in 10m 13s\n- openstack-tox-docs http://logs.openstack.org/25/621025/6/check/openstack-tox-docs/142a555/html/ : SUCCESS in 6m 05s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/6/check/build-openstack-releasenotes/02f5a00/html/ : SUCCESS in 10m 38s","accounts_in_message":[],"_revision_number":6},{"id":"3cb9f20e7df1bf09e181fd4e1630abf6735c0993","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2018-12-20 07:18:55.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f6e328f14be7c386b7234603fbacb9ec95b0ea89","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-12-20 16:16:27.000000000","message":"Topic changed from implement-default-roles to clean-up-v3-cloud-sample","accounts_in_message":[],"_revision_number":6},{"id":"cdc133e978c59a3ebdc1009ffc80abecb31554e0","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2018-12-27 09:15:09.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"8dcf39595a9f0cbd8004ef60bb6d75a32426a9af","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-01-08 18:24:13.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"eaa5a73208565a7a248ede84265d1f37eb4a1580","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-08 18:55:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/7/check/openstack-tox-cover/ef9a530/cover/ : SUCCESS in 13m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/7/check/openstack-tox-lower-constraints/bbfe5e3/ : SUCCESS in 14m 08s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/7/check/openstack-tox-pep8/cf80a5d/ : SUCCESS in 6m 28s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/7/check/openstack-tox-py27/7e1f078/ : SUCCESS in 14m 07s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/7/check/openstack-tox-py35/0cee699/ : SUCCESS in 11m 13s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/7/check/openstack-tox-py36/24c36ad/ : SUCCESS in 11m 19s\n- openstack-tox-docs http://logs.openstack.org/25/621025/7/check/openstack-tox-docs/7607a4f/html/ : SUCCESS in 7m 33s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/7/check/build-openstack-releasenotes/5db878d/html/ : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":7},{"id":"5c24f579779b39efe0434d8379d404d9d9ee20e1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-01-08 19:23:16.000000000","message":"Topic changed from bug/1805880 to clean-up-v3-cloud-sample","accounts_in_message":[],"_revision_number":7},{"id":"f627de80e0f73f6caef61a0fcd9458d1a2c82409","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-01-09 14:59:44.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"390789d83e60192b862d1e8c1965ae4d091d2ea2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 15:40:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/8/check/openstack-tox-cover/6d7b56f/cover/ : SUCCESS in 14m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/8/check/openstack-tox-lower-constraints/a371700/ : SUCCESS in 14m 39s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/8/check/openstack-tox-pep8/bc25569/ : SUCCESS in 6m 04s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/8/check/openstack-tox-py27/0cc6b3b/ : SUCCESS in 12m 34s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/8/check/openstack-tox-py35/503345e/ : SUCCESS in 12m 57s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/8/check/openstack-tox-py36/469ee24/ : SUCCESS in 13m 10s\n- openstack-tox-docs http://logs.openstack.org/25/621025/8/check/openstack-tox-docs/033327b/html/ : SUCCESS in 8m 27s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/8/check/build-openstack-releasenotes/7ff52e7/html/ : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":8},{"id":"c1d53a3a7c545dc024773dc07b3ee6d5c79513e2","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-02-15 20:41:44.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"d77db31906cc2433a5d3308ed2c388bd45983c3e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-04 15:42:36.000000000","message":"Topic changed from implement-default-roles to clean-up-v3-cloud-sample","accounts_in_message":[],"_revision_number":8},{"id":"e6903dbbfdc1e888679656f8062840a14d208b3e","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-05 15:40:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"649f1a239d2597712c76d334bc46f7ce7a85c0ad","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-03-05 15:55:33.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"f4eb65a68e1973b469dab127e90d679b249f20d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 17:04:27.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/9/check/openstack-tox-cover/2fc575f/cover/ : SUCCESS in 26m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/9/check/openstack-tox-lower-constraints/0352477/ : SUCCESS in 24m 34s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/9/check/openstack-tox-pep8/721092e/ : FAILURE in 6m 06s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/9/check/openstack-tox-py27/cabd96b/ : SUCCESS in 15m 59s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/9/check/openstack-tox-py35/2c8a82a/ : SUCCESS in 18m 32s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/9/check/openstack-tox-py36/7450aab/ : SUCCESS in 20m 30s\n- openstack-tox-py37 http://logs.openstack.org/25/621025/9/check/openstack-tox-py37/03e167b/ : SUCCESS in 18m 11s\n- openstack-tox-docs http://logs.openstack.org/25/621025/9/check/openstack-tox-docs/c877db5/html/ : SUCCESS in 8m 11s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/9/check/build-openstack-releasenotes/3c08054/html/ : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":9},{"id":"2f5863241c11a29a967b76fd407124d108ad2fbc","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-05 19:14:17.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"23978817c902612669d1096921dfdfa37cba8575","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-05 19:15:07.000000000","message":"Topic changed from bug/1805880 to clean-up-v3-cloud-sample","accounts_in_message":[],"_revision_number":10},{"id":"84e5e4d5ceec51f156f489b5c0062524d0fbc98a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-05 20:44:46.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/10/check/openstack-tox-cover/a2c2baa/cover/ : SUCCESS in 23m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/10/check/openstack-tox-lower-constraints/f2f945f/ : SUCCESS in 25m 44s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/10/check/openstack-tox-pep8/124f510/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/10/check/openstack-tox-py27/7530a61/ : SUCCESS in 14m 48s\n- openstack-tox-py35 http://logs.openstack.org/25/621025/10/check/openstack-tox-py35/bc009a3/ : SUCCESS in 19m 14s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/10/check/openstack-tox-py36/600cca4/ : SUCCESS in 21m 40s\n- openstack-tox-py37 http://logs.openstack.org/25/621025/10/check/openstack-tox-py37/5c7dbfa/ : SUCCESS in 17m 41s\n- openstack-tox-docs http://logs.openstack.org/25/621025/10/check/openstack-tox-docs/310f76d/html/ : SUCCESS in 8m 03s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/10/check/build-openstack-releasenotes/b150e24/html/ : SUCCESS in 10m 58s","accounts_in_message":[],"_revision_number":10},{"id":"092480821a3cb1bb8ef500268cd7d3e7b2db348c","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-03-06 10:29:25.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"be4bd1484807b3f435474df935cc4271f2ec8743","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-06 20:28:05.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"fce5f666186e9f3615aabbd690cf15ad5a1d22e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-06 23:09:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/11/check/openstack-tox-cover/8735aaf/cover/ : SUCCESS in 23m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/11/check/openstack-tox-lower-constraints/1cbaa17/ : SUCCESS in 24m 42s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/11/check/openstack-tox-pep8/1adb9bf/ : SUCCESS in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/11/check/openstack-tox-py27/08c5c7c/ : SUCCESS in 17m 19s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/11/check/openstack-tox-py36/78540e5/ : SUCCESS in 18m 30s\n- openstack-tox-py37 http://logs.openstack.org/25/621025/11/check/openstack-tox-py37/b0c8370/ : SUCCESS in 20m 07s\n- openstack-tox-docs http://logs.openstack.org/25/621025/11/check/openstack-tox-docs/5ba9250/html/ : SUCCESS in 8m 57s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/11/check/build-openstack-releasenotes/e9c0769/html/ : SUCCESS in 10m 13s","accounts_in_message":[],"_revision_number":11},{"id":"81e8261c5cb1be1bdf5a298ad02bcc4aa310419c","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-03-25 19:40:18.000000000","message":"Patch Set 11: Code-Review-2 Workflow-1\n\nWaiting on this until Train opens for development.","accounts_in_message":[],"_revision_number":11},{"id":"c196d5d47a588b3b03ef5cbb6ed2a9aeca3a58e7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-27 21:12:55.000000000","message":"Patch Set 11: -Code-Review -Workflow\n\nThis is ready to pick back up now that Train is open for development.","accounts_in_message":[],"_revision_number":11},{"id":"7115192be481b41c43f1ec9165a9c21bb82d4528","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-07-02 15:03:01.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"fcc724398cae8edb457ba1527f09f88a665c3fac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 20:13:04.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/25/621025/12/check/openstack-tox-cover/3d786b7/ : RETRY_LIMIT in 1m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/25/621025/12/check/openstack-tox-lower-constraints/999825b/ : FAILURE in 31m 57s\n- openstack-tox-pep8 http://logs.openstack.org/25/621025/12/check/openstack-tox-pep8/31cae40/ : RETRY_LIMIT in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/25/621025/12/check/openstack-tox-py27/7e962cc/ : RETRY_LIMIT in 1m 56s\n- openstack-tox-py36 http://logs.openstack.org/25/621025/12/check/openstack-tox-py36/9eabb0d/ : FAILURE in 35m 43s\n- openstack-tox-py37 http://logs.openstack.org/25/621025/12/check/openstack-tox-py37/2382396/ : RETRY_LIMIT in 2m 58s\n- openstack-tox-docs http://logs.openstack.org/25/621025/12/check/openstack-tox-docs/b81f0ce/html/ : SUCCESS in 7m 14s\n- build-openstack-releasenotes http://logs.openstack.org/25/621025/12/check/build-openstack-releasenotes/57424cc/html/ : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":12},{"id":"04ed3dac03e8aa83d5157b2d5594b4b47fe7fd64","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-09-18 14:53:23.000000000","message":"Topic changed from implement-default-roles to train-rc1-policy-fixes","accounts_in_message":[],"_revision_number":12},{"id":"6f86b29350a40b552ae0a3216e773b00511db020","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-24 17:30:57.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"baa3a48ddfda3d251551b99fecc44e56e5eb92ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 22:13:36.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/591425942e044099938bb11a48229a0f : FAILURE in 16m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/242327d3c9294646bde9c15d3aeab6de : FAILURE in 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27336be0959e403db1dff1c25d92fc2d : SUCCESS in 8m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/47971c224f604c668e7bfdd8ebd25fab : FAILURE in 25m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa7bcc806e154f9b9593da4c67cef506 : FAILURE in 17m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec73de3a879c4e60bb016642f3174c88 : FAILURE in 15m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c46931c7aa264f24b5ca890e25819d8f : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6235156394c4318bceeb625e1ec67db : SUCCESS in 12m 55s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/cfc0e5841d034bf088eb2cffeb533085 : SUCCESS in 46m 33s","accounts_in_message":[],"_revision_number":13},{"id":"aa5adada0c71bbd01a449a56daa58ae06c7fa2d3","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-24 22:36:15.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"bbf16a5aa37236f4ce6ebec1c49e2ecce9c2686c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 00:23:28.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25788b31cf7f4c8da9cef8e308e649c2 : SUCCESS in 17m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee8628b1576a40da83c2799cca6527a5 : SUCCESS in 18m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/105a35435e264ee8aabdac45acd2a85e : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9269a270a2af41229b039b3fd33a0c18 : SUCCESS in 14m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bd96f1f6180445494fe6abea4f4be26 : SUCCESS in 13m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2660f5385f1d4032b1ef7a7eeb4aae89 : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520b83c72c364ecab782c01b9b7068c0 : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a02cf970bc9e403aba7e02a38097a01b : SUCCESS in 10m 52s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/fb7b8bfce63f4ec8b48f44087e0e9357 : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":14},{"id":"bc67268faa514e37d2f78ebae969ee70ec89b9b5","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 02:11:40.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"853c4490d54d7cc5d0cebf6dd5c83fe93822642f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 02:25:55.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"790252d70eac99428180389ea13908cccf739dcb","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 02:26:08.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"ca0a11e7ad1bcf907290bc4d15144a9aaa8387f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 03:14:35.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a39484f72ed84e139818b05369c13a53 : SUCCESS in 16m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b788505efa2941ba96e2d75e9167719f : SUCCESS in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24b4b25c164c48b0b201d5e62af46a99 : SUCCESS in 5m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4fb64dcf28254b63a4f5a68a5932cdc1 : SUCCESS in 16m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b85b26b08a14116bc392fe92cf091a8 : SUCCESS in 13m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2965ee6c36a642d08faefd5b7621f5d7 : SUCCESS in 12m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c55c3d1b6c4d13b78761396cb8a6e4 : SUCCESS in 14m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea0d3facc8374a2ab3f7d9da2e7abf29 : SUCCESS in 13m 10s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/65096ad9f0fd4062a1512fdcdc042af1 : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":15},{"id":"3c289136a7a0f1d094b4718bb2f1a49bd88f6b22","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-09-25 04:11:58.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"2838b8d1f19e5b4aa592cf302c4ea1df72a566ea","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-25 05:40:26.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"e23d1f8339412cf72de42dc2c0cc82bea06b6d21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 17:28:24.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"3194aefa3d4a6d62ce781b213c03ab94e876c384","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 20:30:03.000000000","message":"Patch Set 15: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"c74acad190668febaeddbd09b3ec4fbb2dac4c69","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 20:34:52.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"4bcdb683415721af002fa6e1deb1607a7f498613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 21:46:34.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe977582ce5049bb9d66cda1459734f0 : SUCCESS in 35m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c33ea8a565b4fb99a2fd3bef4fbb77c : TIMED_OUT in 41m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5d7b4d7331f41629042e1d46fd03dd8 : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d7de82c28d44acea3364074664762f2 : SUCCESS in 15m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1017db3921b4243bbfff51339c899e9 : SUCCESS in 16m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2b6746a9bef5481ea3b337f66b31b61d : SUCCESS in 13m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da69086d0c742858b2b30c69344d128 : SUCCESS in 15m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/496b1a29637940928d6e791843c0eccb : SUCCESS in 11m 08s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/5b5715c3dce04e04b233f4acb0187388 : SUCCESS in 43m 01s","accounts_in_message":[],"_revision_number":15},{"id":"e95f827af2c1453d93d89bf5b5d40f55df21ff52","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-25 21:58:03.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"c5551b734c52fd5668247c81e7aefa3009568fea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 22:41:21.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d18a725db38a4632be5c5260e1cec7be : SUCCESS in 19m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0d5f83f37f44cc9b26e9ec192eba943 : SUCCESS in 14m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1332d40337049c5bc05e23ffd3c60f2 : SUCCESS in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1fcb2b8c242c450781204ed10476945f : SUCCESS in 14m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76b59bfdfc2840d78567d3c5bd9a361f : SUCCESS in 13m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8870657364134fb499dda6140719f961 : SUCCESS in 16m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ca7bae47d64043979c244c094dd5d9 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e6dba5d0de0437a8f77b0e5e6cec91c : SUCCESS in 11m 59s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/0ef07eced30a46efb45e3281c18a0c3a : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":15},{"id":"be2b13dd84793019e03f3e646d9f8114a33ed18b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 01:25:02.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"a84aab359d8d6cd2aa73d3b95fbf3f0d9dbaabaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 03:12:52.000000000","message":"Patch Set 15: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"dced66ffc6af1391534a93a2621e35cbb1313dbd","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-26 03:46:28.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"01b65bedd35f987a3de1f29ce4a4fafc04b9bd6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 04:32:04.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4c1b0fd47e4446e916c6befd632abff : SUCCESS in 25m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c111331c4c34a01858c64108f9df9c9 : SUCCESS in 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9709af86eb704452be805b694c61aa7f : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a2e1547c7fd4669a26df4182f45c478 : SUCCESS in 27m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad85264c4a2341759cb59da257d8dcd6 : TIMED_OUT in 40m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f578d56fa538455ca7fd64d841d67e55 : SUCCESS in 22m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb46db9b1974eca95f9ecbf4aabdd5b : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/787e9db802354f1e843d681b89bf56b1 : SUCCESS in 11m 58s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/ebf88ffe2bad458ba1ff769676497da7 : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":15},{"id":"566b1d39a5b13174ffb029c32dbe82fb1f1454da","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-26 06:09:17.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"7c1319c5223a792c1f48f2990af12e7febe1b048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:36:34.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd19e3aacfd74262a53f495eacd272eb : SUCCESS in 14m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/542f12fc2e464d529ba1fe6b13b75285 : SUCCESS in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91206626ed6a4454b26069bed5d67f18 : SUCCESS in 6m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7987bc404af24b6ea9d6094564c9d60e : SUCCESS in 19m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfbacf053e06482babde5cb8e899b92f : SUCCESS in 18m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa0ab5d7fa384138bf09a335a7288d94 : SUCCESS in 20m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c07c124d61e4e8b949e2d593fe0a53e : SUCCESS in 20m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/644c071468204038924ad4f4ad358af9 : SUCCESS in 12m 55s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/b4b17cbf0bb3429ab3f8f3d90ef55cc1 : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":15},{"id":"19823240774fd3129f11c4a3a48d51760e1743b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 12:04:18.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"a02f3110a63bc9183eca658c318caadc22956b56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 13:36:53.000000000","message":"Patch Set 15: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"f6467c6330b645b0b346465803647439b7eafc4e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-26 15:26:37.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"745c6444453c667efefcb40a7bcb5bd9beb0b90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 17:34:28.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54f6ed0b448940b3a11d4257b1269945 : SUCCESS in 16m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57a69fbb19464819b34834c81543e052 : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7295732d526f44938aaacb314dc88b48 : SUCCESS in 6m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ca778485aa74da1a36fc961ed9b0873 : SUCCESS in 24m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd2a421e1860495eb6bb76333ad7ec62 : SUCCESS in 14m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/afea12b4cd334738b2f4cfb0f6481c23 : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff1b683496ed475ea5bebdc98d5f420d : SUCCESS in 13m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/592311c98dc242669afb0692bd5fb940 : SUCCESS in 12m 30s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/cab0acfb8d954546b5688856a9529ab2 : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":15},{"id":"ebb7f3786c3d857cb2894ebc699292e2626b26ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 18:33:38.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"062adc16559364c0ce3698a8340d5dd74b35b58a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 22:08:52.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad13088bb2d94ea48e66269416263a05 : TIMED_OUT in 40m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b3c0efaf0b48eb9537e04fe498eed5 : SUCCESS in 7m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0f45bccc6484b52a86e64748a4726ba : SUCCESS in 20m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4777fc74f59647dfb60115cceba615fa : SUCCESS in 16m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4e232a71cfb41c28e3ac003013a5e86 : SUCCESS in 16m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8064fc63e134c2b95b9f3e34092da52 : SUCCESS in 13m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ce34163d8184fdcbb7ee400a7212eb3 : SUCCESS in 12m 23s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/a7af9919e0dc4325bf500d3e1545af27 : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":15},{"id":"9a1f2507946cf879584b1dd33b560a7e0b04898e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-26 22:16:38.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"8d491d335439a40b3c3bbe85b81d01edb658bef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 23:02:45.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c85edf28a3c4caeb343e71f2170239f : SUCCESS in 15m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/276d8bceb5a3440480443e3fb1ebceca : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dd648803b864a66b2486f63f8d4ae77 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d67d4cbb3c304dc194ef714aacc0d0f2 : SUCCESS in 12m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d994624a7a74caeb3ba48c88e1d66b1 : SUCCESS in 13m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c083cae9a6e045a985853ddfec18a850 : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd1eefad452e4555acd425df067beddf : SUCCESS in 15m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2bd6c54156e4a8bbc2cc460a5833ede : SUCCESS in 14m 03s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/49d3cea6aee54dd2a041202c54fb0237 : SUCCESS in 40m 52s","accounts_in_message":[],"_revision_number":15},{"id":"afe3c80ba9a2038792450237e765a7d98cd12a6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 23:03:08.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"7c1b9f533b7022436ae758f59c615d61ded0531a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 00:07:41.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1d6e95f44824ad8888b39893b26623e : SUCCESS in 39m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ffa03846a341c7b27a3af9f049e8b5 : SUCCESS in 7m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e41cbb9755f7435ea862cee8b4389de7 : SUCCESS in 18m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3020662617f74510b6b2dd83f214e704 : SUCCESS in 40m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8fa625af193431e8f9cc0bb0ead9c2d : SUCCESS in 16m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b794a82218024d4784677be41751d869 : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c71bd425541741cd86ad2fe2a0b4ba7b : SUCCESS in 15m 37s\n- keystone-tox-protection https://zuul.opendev.org/t/openstack/build/fd1a8ddefc234e9bb1e5952bb4855ea6 : SUCCESS in 39m 46s","accounts_in_message":[],"_revision_number":15},{"id":"ecd311c7e7b542e7a311f83262b77202a85aeff3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 00:07:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"a971f5a8d96763286ffcf84aaf542f615d5df9d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 00:09:18.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d3875fe345d40a98b3b05efa1369522 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fa03da8ef7448f7a2e0a49038e3f203 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"5b995cc8fbf0bb654ed0f6a88091c48548f53f6e","revisions":{"d3cac7044e947209582b6f94a24cb8d8ad7b71d2":{"kind":"REWORK","_number":1,"created":"2018-11-29 21:34:28.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/1"}}},"commit":{"parents":[{"commit":"b7b2495d7ca280da4ff67a81b0bab2e71cc1b13a","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b7b2495d7ca280da4ff67a81b0bab2e71cc1b13a"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:33:16.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d3cac7044e947209582b6f94a24cb8d8ad7b71d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d3cac7044e947209582b6f94a24cb8d8ad7b71d2"}]},"branch":"refs/heads/master"},"db46024de4273aaacb896ab1f65e6c35375bf915":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-11-30 23:37:19.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/2"}}},"commit":{"parents":[{"commit":"4828ea4a17df36c00b124f4951e7e4dc8275ee06","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4828ea4a17df36c00b124f4951e7e4dc8275ee06"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-30 23:36:48.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/db46024de4273aaacb896ab1f65e6c35375bf915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/db46024de4273aaacb896ab1f65e6c35375bf915"}]},"branch":"refs/heads/master"},"a3b3c4763aa96a122ff9a2c4e70b9ad14e63c0b4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2018-12-07 17:54:59.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/3"}}},"commit":{"parents":[{"commit":"148df7089f3fde2084ad5d5283b4c08535fff060","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/148df7089f3fde2084ad5d5283b4c08535fff060"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-12-07 17:54:51.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a3b3c4763aa96a122ff9a2c4e70b9ad14e63c0b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a3b3c4763aa96a122ff9a2c4e70b9ad14e63c0b4"}]},"branch":"refs/heads/master"},"ea8b580bf79183ce6d3fd0d711b2cec6383f9351":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-12-11 09:15:47.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/4"}}},"commit":{"parents":[{"commit":"4f631e7c42753b10d3feb5d4c068798ab7a73693","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f631e7c42753b10d3feb5d4c068798ab7a73693"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-12-11 09:15:13.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea8b580bf79183ce6d3fd0d711b2cec6383f9351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ea8b580bf79183ce6d3fd0d711b2cec6383f9351"}]},"branch":"refs/heads/master"},"286e86ddaadb07ee0923f7aa02a804ecaddec233":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2018-12-18 22:28:55.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/5"}}},"commit":{"parents":[{"commit":"a49fb7fe49d54c26acc9f392d9245f25e9a691e8","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a49fb7fe49d54c26acc9f392d9245f25e9a691e8"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-12-18 22:27:40.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/286e86ddaadb07ee0923f7aa02a804ecaddec233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/286e86ddaadb07ee0923f7aa02a804ecaddec233"}]},"branch":"refs/heads/master"},"010dd8e6c37c7b7f92719e5eb630272f9e2e3475":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2018-12-19 23:14:26.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/6"}}},"commit":{"parents":[{"commit":"0778ee9769fb1ecd1e2e53e2c8306190bd02b3e7","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0778ee9769fb1ecd1e2e53e2c8306190bd02b3e7"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-12-19 23:09:22.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/010dd8e6c37c7b7f92719e5eb630272f9e2e3475"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/010dd8e6c37c7b7f92719e5eb630272f9e2e3475"}]},"branch":"refs/heads/master"},"4bac86ad24bbd254482cb8161a3bd4328b59322b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-01-08 18:24:13.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/7"}}},"commit":{"parents":[{"commit":"bb9e447cba3a2e7b61f372c018c4f7d3e3a46a22","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bb9e447cba3a2e7b61f372c018c4f7d3e3a46a22"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-01-08 18:18:29.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4bac86ad24bbd254482cb8161a3bd4328b59322b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4bac86ad24bbd254482cb8161a3bd4328b59322b"}]},"branch":"refs/heads/master"},"401232d3e0c8fcbf09d7b7b1504faff812d21394":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-01-09 14:59:44.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/8"}}},"commit":{"parents":[{"commit":"c5143a065bd39292d052aabe5f90f3e2d17364e1","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c5143a065bd39292d052aabe5f90f3e2d17364e1"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-01-09 14:59:30.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/401232d3e0c8fcbf09d7b7b1504faff812d21394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/401232d3e0c8fcbf09d7b7b1504faff812d21394"}]},"branch":"refs/heads/master"},"20af61beb09103e8b1872758cfada1d9c981d93c":{"kind":"REWORK","_number":9,"created":"2019-03-05 15:40:35.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/9"}}},"commit":{"parents":[{"commit":"a42a9f8ed3742381b1d89322a065f261ae8b0244","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a42a9f8ed3742381b1d89322a065f261ae8b0244"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-03-05 15:40:15.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/20af61beb09103e8b1872758cfada1d9c981d93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/20af61beb09103e8b1872758cfada1d9c981d93c"}]},"branch":"refs/heads/master"},"27b22c2c1a3111859001a02807e0fc21004f7e49":{"kind":"REWORK","_number":10,"created":"2019-03-05 19:14:17.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/10"}}},"commit":{"parents":[{"commit":"f7564ab68361b0d72e4b6c809946576b06a13b6e","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f7564ab68361b0d72e4b6c809946576b06a13b6e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-03-05 19:14:08.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/27b22c2c1a3111859001a02807e0fc21004f7e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/27b22c2c1a3111859001a02807e0fc21004f7e49"}]},"branch":"refs/heads/master"},"d163ffd19115d6de0382bda7783e6541a4aa0cb3":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-03-06 20:28:05.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/11"}}},"commit":{"parents":[{"commit":"0f16fe31684f02e8ce8c56de11179ab08e071353","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0f16fe31684f02e8ce8c56de11179ab08e071353"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-03-06 20:27:53.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d163ffd19115d6de0382bda7783e6541a4aa0cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d163ffd19115d6de0382bda7783e6541a4aa0cb3"}]},"branch":"refs/heads/master"},"ec669efbec96f2c46c29d79267dd00dcd1702b6e":{"kind":"REWORK","_number":12,"created":"2019-07-02 15:03:01.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/25/621025/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/12"}}},"commit":{"parents":[{"commit":"79ffefc5f9196404d2d5055d9cd82692323b8aa2","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/79ffefc5f9196404d2d5055d9cd82692323b8aa2"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2019-07-02 15:00:03.000000000","tz":0},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ec669efbec96f2c46c29d79267dd00dcd1702b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ec669efbec96f2c46c29d79267dd00dcd1702b6e"}]},"branch":"refs/heads/master"},"c487d1e2ce5781fe51c8ecc963e29c1d35174607":{"kind":"REWORK","_number":13,"created":"2019-09-24 17:30:57.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/25/621025/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/13"}}},"commit":{"parents":[{"commit":"d69724f330fb5414c2741539d6df32b3d816fe86","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d69724f330fb5414c2741539d6df32b3d816fe86"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-24 17:29:21.000000000","tz":-420},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c487d1e2ce5781fe51c8ecc963e29c1d35174607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c487d1e2ce5781fe51c8ecc963e29c1d35174607"}]},"branch":"refs/heads/master"},"762731c6a96720d6f15dbf13c217bbe007995317":{"kind":"REWORK","_number":14,"created":"2019-09-24 22:36:15.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/25/621025/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/14"}}},"commit":{"parents":[{"commit":"5a5ee4e7f2bc819e19e6fced88be7b540aac59d8","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a5ee4e7f2bc819e19e6fced88be7b540aac59d8"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Colleen Murphy","email":"colleen.murphy@suse.de","date":"2019-09-24 22:34:27.000000000","tz":-420},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/762731c6a96720d6f15dbf13c217bbe007995317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/762731c6a96720d6f15dbf13c217bbe007995317"}]},"branch":"refs/heads/master"},"5b995cc8fbf0bb654ed0f6a88091c48548f53f6e":{"kind":"REWORK","_number":15,"created":"2019-09-25 02:25:55.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/25/621025/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/25/621025/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/25/621025/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/25/621025/15"}}},"commit":{"parents":[{"commit":"e938c496281daa6d1dab66d66bdb2d34abd5ddc3","subject":"Add tests for project users interacting with limits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e938c496281daa6d1dab66d66bdb2d34abd5ddc3"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2018-11-29 21:30:06.000000000","tz":0},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-09-25 02:25:45.000000000","tz":-420},"subject":"Remove limit policies from policy.v3cloudsample.json","message":"Remove limit policies from policy.v3cloudsample.json\n\nBy incorporating system-scope and default roles, we\u0027ve effectively\nmade these policies obsolete. We can simplify what we maintain and\nprovide a more consistent, unified view of default limit\nbehavior by removing them.\n\nChange-Id: Ie0f333a9e8b60154711a24ba7d9ade531217eb71\nCloses-Bug: 1805880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5b995cc8fbf0bb654ed0f6a88091c48548f53f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5b995cc8fbf0bb654ed0f6a88091c48548f53f6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
