)]}'
{"id":"openstack%2Fkeystone~630301","triplet_id":"openstack%2Fkeystone~master~I2f1a2de9d0974d43dcadb58f6623d15e4a0740c9","project":"openstack/keystone","branch":"master","topic":"application-credentials","hashtags":[],"change_id":"I2f1a2de9d0974d43dcadb58f6623d15e4a0740c9","subject":"Converting the API tests to use flask\u0027s test_client","status":"MERGED","created":"2019-01-11 16:10:16.000000000","updated":"2019-02-05 18:09:41.000000000","submitted":"2019-02-05 18:09:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":76,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"630301-1549390181826-77bea6f0","meta_rev_id":"bcc32ed91813020278f8f7ba00b105bcbbb871cf","_number":630301,"virtual_id_number":630301,"owner":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":2,"date":"2019-02-05 18:09:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-02-05 10:01:38.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-02-05 16:13:48.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"date":"2019-02-05 16:08:09.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2019-02-05 16:13:48.000000000","_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-02-01 09:24:27.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-02-05 10:01:38.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-02-05 16:08:09.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-02-05 16:13:48.000000000","updated_by":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"reviewer":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"state":"REVIEWER"},{"updated":"2019-02-05 18:09:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"adc807a655b846c5bad1d2af94cd3ed78034aa8f","author":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"date":"2019-01-11 16:10:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0006ac2db855e409f07930fa83f35face6bfd40e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-11 16:34:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/01/630301/1/check/openstack-tox-cover/48df077/ : FAILURE in 13m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/630301/1/check/openstack-tox-lower-constraints/559575b/ : FAILURE in 16m 42s\n- openstack-tox-pep8 http://logs.openstack.org/01/630301/1/check/openstack-tox-pep8/d67edd9/ : FAILURE in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/01/630301/1/check/openstack-tox-py27/ba0db1f/ : FAILURE in 11m 21s\n- openstack-tox-py35 http://logs.openstack.org/01/630301/1/check/openstack-tox-py35/f866f10/ : FAILURE in 12m 46s\n- openstack-tox-py36 http://logs.openstack.org/01/630301/1/check/openstack-tox-py36/8da82ae/ : FAILURE in 9m 57s\n- openstack-tox-docs http://logs.openstack.org/01/630301/1/check/openstack-tox-docs/0e621b1/html/ : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a65c441343bae08d61ae1adff9ed03db39ce0677","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-01-12 14:46:40.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nGreat start! I\u0027ve left some feedback inline.","accounts_in_message":[],"_revision_number":1},{"id":"7914930325b3517ea4799e00bb3c190c25333cc8","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-01-17 09:56:30.000000000","message":"Patch Set 1: Code-Review-1\n\nHi Islam. The changes you was trying to do in [1], you should update the comments in this patch by using git commit --amend. You can take help from the openstack contributors guide[2]\n\n[1] https://review.openstack.org/#/c/631255\n[2]https://docs.openstack.org/contributors/code-and-documentation/using-gerrit.html","accounts_in_message":[],"_revision_number":1},{"id":"4d44c30755a09b2519a5a81d693d0cba3644808e","author":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"date":"2019-01-17 16:16:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"651181621f81836cc5f879e5840f747012f88eb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-17 17:00:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/01/630301/2/check/openstack-tox-cover/9e1c537/cover/ : SUCCESS in 12m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/630301/2/check/openstack-tox-lower-constraints/5dae552/ : SUCCESS in 14m 48s\n- openstack-tox-pep8 http://logs.openstack.org/01/630301/2/check/openstack-tox-pep8/5c4abd2/ : FAILURE in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/01/630301/2/check/openstack-tox-py27/42ead0a/ : SUCCESS in 12m 22s\n- openstack-tox-py35 http://logs.openstack.org/01/630301/2/check/openstack-tox-py35/de4a16b/ : SUCCESS in 13m 37s\n- openstack-tox-py36 http://logs.openstack.org/01/630301/2/check/openstack-tox-py36/fb2e5de/ : SUCCESS in 11m 29s\n- openstack-tox-docs http://logs.openstack.org/01/630301/2/check/openstack-tox-docs/8888b10/html/ : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":2},{"id":"df863b7a9800eba28464391a4321e3cd3aebe72e","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-01-17 17:00:47.000000000","message":"Patch Set 2:\n\n(7 comments)\n\nHi Islam. Nice work done. Some more inline comments. Thanks","accounts_in_message":[],"_revision_number":2},{"id":"f5fbc6192d010cb0496aea5149c9b78b6c7bd975","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-01-17 17:01:51.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"c76c8ac44a4492ba3ced05389bb3fe90688a55c4","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-01-18 09:43:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(18 comments)\n\nGood progress. Take a look at all the places where gerrit has added bright red highlights, those are accidental whitespace additions that need to be removed.","accounts_in_message":[],"_revision_number":2},{"id":"8b9ee1617df4e58a70dfa1319278ee2449b17e16","author":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"date":"2019-01-21 19:22:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2aa87bd8777ead041b41fb4d3484b238da715852","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-21 20:09:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/01/630301/3/check/openstack-tox-cover/e298d6a/ : FAILURE in 14m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/630301/3/check/openstack-tox-lower-constraints/02bcdc3/ : FAILURE in 14m 55s\n- openstack-tox-pep8 http://logs.openstack.org/01/630301/3/check/openstack-tox-pep8/f13fe09/ : FAILURE in 6m 27s\n- openstack-tox-py27 http://logs.openstack.org/01/630301/3/check/openstack-tox-py27/0e75b49/ : FAILURE in 11m 41s\n- openstack-tox-py35 http://logs.openstack.org/01/630301/3/check/openstack-tox-py35/f9c71ec/ : FAILURE in 11m 17s\n- openstack-tox-py36 http://logs.openstack.org/01/630301/3/check/openstack-tox-py36/36f580b/ : FAILURE in 11m 39s\n- openstack-tox-docs http://logs.openstack.org/01/630301/3/check/openstack-tox-docs/00d4d20/html/ : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":3},{"id":"2d389e604b7771a62741dfd8e61ed7df49cc8adc","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-01-22 11:13:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(12 comments)\n\nHi Islam. Thanks for the patch. Some more comments inline. Kindly run tox -epy27 after all changes to confirm unit test cases not failing.","accounts_in_message":[],"_revision_number":3},{"id":"c249781408b2144b9fe251caa9591b546e278f17","author":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"date":"2019-01-30 22:03:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bc236fcfdfb6cbfde7d5d2cee3d62b5b3fb66bbe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-30 23:02:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/01/630301/4/check/openstack-tox-cover/ee48f0e/cover/ : SUCCESS in 14m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/630301/4/check/openstack-tox-lower-constraints/11a6a46/ : SUCCESS in 19m 50s\n- openstack-tox-pep8 http://logs.openstack.org/01/630301/4/check/openstack-tox-pep8/89a5752/ : FAILURE in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/01/630301/4/check/openstack-tox-py27/8a2cdf7/ : SUCCESS in 11m 30s\n- openstack-tox-py35 http://logs.openstack.org/01/630301/4/check/openstack-tox-py35/e23f3f0/ : SUCCESS in 12m 17s\n- openstack-tox-py36 http://logs.openstack.org/01/630301/4/check/openstack-tox-py36/74ad059/ : SUCCESS in 14m 54s\n- openstack-tox-docs http://logs.openstack.org/01/630301/4/check/openstack-tox-docs/09252dc/html/ : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":4},{"id":"7c8593d3a572757237bc92f9e899a48ad5365bb3","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-01-31 08:58:57.000000000","message":"Patch Set 4:\n\n(22 comments)\n\nIt\u0027s looking really good, there are just a few more pep8 errors to clean up, I\u0027ve pointed them out inline.","accounts_in_message":[],"_revision_number":4},{"id":"59c489750badf753b296a68d966c33d38aab2deb","author":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"date":"2019-01-31 21:40:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"247a89b7efd6483c7802b5b246aa3a22b83ffa2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 21:58:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/01/630301/5/check/openstack-tox-cover/4daf80b/ : FAILURE in 12m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/630301/5/check/openstack-tox-lower-constraints/1988b89/ : FAILURE in 16m 03s\n- openstack-tox-pep8 http://logs.openstack.org/01/630301/5/check/openstack-tox-pep8/e22df73/ : FAILURE in 6m 11s\n- openstack-tox-py27 http://logs.openstack.org/01/630301/5/check/openstack-tox-py27/b11777b/ : FAILURE in 11m 22s\n- openstack-tox-py35 http://logs.openstack.org/01/630301/5/check/openstack-tox-py35/016f3fa/ : FAILURE in 12m 03s\n- openstack-tox-py36 http://logs.openstack.org/01/630301/5/check/openstack-tox-py36/ca000cd/ : FAILURE in 11m 09s\n- openstack-tox-docs http://logs.openstack.org/01/630301/5/check/openstack-tox-docs/df80e5e/html/ : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":5},{"id":"7e231557046fa989ff16728a82230d12aea31923","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-02-01 09:24:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)\n\nHi Islam. Great going. Few errors left. It will be good to go. Find the inline comments.","accounts_in_message":[],"_revision_number":5},{"id":"de937d8378b1997433b6a1c5ed68c39d0547121b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-02-01 11:59:19.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d20365081dea43592005c6d8c115a83b5a338078","author":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"date":"2019-02-01 22:26:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dbad532a832b50d8ec47066d585820ea324e772c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-01 22:47:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/01/630301/6/check/openstack-tox-cover/7e1f1d6/ : FAILURE in 18m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/630301/6/check/openstack-tox-lower-constraints/2652730/ : FAILURE in 20m 31s\n- openstack-tox-pep8 http://logs.openstack.org/01/630301/6/check/openstack-tox-pep8/06d4ab1/ : FAILURE in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/01/630301/6/check/openstack-tox-py27/cbfdae7/ : FAILURE in 12m 19s\n- openstack-tox-py35 http://logs.openstack.org/01/630301/6/check/openstack-tox-py35/82e406e/ : FAILURE in 16m 15s\n- openstack-tox-py36 http://logs.openstack.org/01/630301/6/check/openstack-tox-py36/c57058a/ : FAILURE in 14m 47s\n- openstack-tox-docs http://logs.openstack.org/01/630301/6/check/openstack-tox-docs/59ddd22/html/ : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":6},{"id":"b5926fe3059a8933843b41dcb20d0b611d68736b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-02-02 18:07:49.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"671bf5bd832840fe3aca3a0e318de22975d06067","author":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"date":"2019-02-04 21:34:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a3a536768a3f6aff7e9b1b57b3a3867ab9e57c73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-04 22:01:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/01/630301/7/check/openstack-tox-cover/696f34f/cover/ : SUCCESS in 14m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/630301/7/check/openstack-tox-lower-constraints/e74bd36/ : SUCCESS in 25m 12s\n- openstack-tox-pep8 http://logs.openstack.org/01/630301/7/check/openstack-tox-pep8/19449f5/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/01/630301/7/check/openstack-tox-py27/befa1a6/ : SUCCESS in 13m 31s\n- openstack-tox-py35 http://logs.openstack.org/01/630301/7/check/openstack-tox-py35/ab26619/ : SUCCESS in 10m 59s\n- openstack-tox-py36 http://logs.openstack.org/01/630301/7/check/openstack-tox-py36/8366fae/ : SUCCESS in 12m 23s\n- openstack-tox-docs http://logs.openstack.org/01/630301/7/check/openstack-tox-docs/dda4a61/html/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":7},{"id":"c010919338928502adae59b64e3aae36d24a8afa","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-02-05 10:01:38.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)\n\nLooks great! I made one minor comment inline it\u0027s not necessary to change this.","accounts_in_message":[],"_revision_number":7},{"id":"735992f411413a5eedc1875886f7f8a1ad233a24","author":{"_account_id":2903,"name":"Morgan Fainberg","email":"morgan.fainberg@gmail.com","username":"mdrnstm"},"date":"2019-02-05 16:13:48.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b647d1d460b6185a17f0e70221948cf28f977553","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 16:14:24.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"6ee895aa9386b075a3f9588877f8b89ea0d988a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 18:09:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"2f249f89f688c90020801d13c8982f09ebdb487d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-05 18:09:41.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/01/630301/7/gate/openstack-tox-lower-constraints/30b181e/ : SUCCESS in 21m 31s\n- openstack-tox-pep8 http://logs.openstack.org/01/630301/7/gate/openstack-tox-pep8/63a055c/ : SUCCESS in 6m 54s\n- openstack-tox-py27 http://logs.openstack.org/01/630301/7/gate/openstack-tox-py27/2916347/ : SUCCESS in 12m 46s\n- openstack-tox-py35 http://logs.openstack.org/01/630301/7/gate/openstack-tox-py35/2a0804b/ : SUCCESS in 13m 51s\n- openstack-tox-py36 http://logs.openstack.org/01/630301/7/gate/openstack-tox-py36/af43ea4/ : SUCCESS in 13m 16s\n- openstack-tox-docs http://logs.openstack.org/01/630301/7/gate/openstack-tox-docs/ab62720/html/ : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"96eb890a086a62c2efd92e5b9dc585d167fb3361","revisions":{"a480aa8eb6cc47c0f9a182604ada5fa94ef75ed9":{"kind":"REWORK","_number":1,"created":"2019-01-11 16:10:16.000000000","uploader":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"ref":"refs/changes/01/630301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/630301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/630301/1"}}},"commit":{"parents":[{"commit":"e287f58fbbfb81f16270800581059be2083e547c","subject":"Merge \"Add role tests for system member role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e287f58fbbfb81f16270800581059be2083e547c"}]}],"author":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-11 15:41:50.000000000","tz":-480},"committer":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-11 15:41:50.000000000","tz":-480},"subject":"Converting the application credentials API tests to use flask\u0027s test_client For each API unit test class, the methods are converted to use the Flask app test_client context manager.","message":"Converting the application credentials API tests to use flask\u0027s\ntest_client\nFor each API unit test class, the methods are converted to use the\nFlask app test_client context manager.\n\nChange-Id: I2f1a2de9d0974d43dcadb58f6623d15e4a0740c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a480aa8eb6cc47c0f9a182604ada5fa94ef75ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a480aa8eb6cc47c0f9a182604ada5fa94ef75ed9"}]},"branch":"refs/heads/master"},"69211ec1e3ba028adb524bf87c519f916e9dd912":{"kind":"REWORK","_number":2,"created":"2019-01-17 16:16:56.000000000","uploader":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"ref":"refs/changes/01/630301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/630301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/630301/2"}}},"commit":{"parents":[{"commit":"e287f58fbbfb81f16270800581059be2083e547c","subject":"Merge \"Add role tests for system member role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e287f58fbbfb81f16270800581059be2083e547c"}]}],"author":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-11 15:41:50.000000000","tz":-480},"committer":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-17 16:14:03.000000000","tz":-480},"subject":"Converting the API tests to use flask\u0027s test_client","message":"Converting the API tests to use flask\u0027s test_client\n\nFor each API unit test class, the methods are converted to use the\nFlask app test_client context manager.\n\nChange-Id: I2f1a2de9d0974d43dcadb58f6623d15e4a0740c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/69211ec1e3ba028adb524bf87c519f916e9dd912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/69211ec1e3ba028adb524bf87c519f916e9dd912"}]},"branch":"refs/heads/master"},"4507b4a24989d24e7b9f9414cdf5586d97bb0fce":{"kind":"REWORK","_number":3,"created":"2019-01-21 19:22:22.000000000","uploader":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"ref":"refs/changes/01/630301/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/630301/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/630301/3"}}},"commit":{"parents":[{"commit":"e287f58fbbfb81f16270800581059be2083e547c","subject":"Merge \"Add role tests for system member role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e287f58fbbfb81f16270800581059be2083e547c"}]}],"author":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-11 15:41:50.000000000","tz":-480},"committer":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-21 19:18:48.000000000","tz":-480},"subject":"Converting the API tests to use flask\u0027s test_client","message":"Converting the API tests to use flask\u0027s test_client\n\nFor each API unit test class, the methods are converted to use the\nFlask app test_client context manager.\n\nChange-Id: I2f1a2de9d0974d43dcadb58f6623d15e4a0740c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4507b4a24989d24e7b9f9414cdf5586d97bb0fce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4507b4a24989d24e7b9f9414cdf5586d97bb0fce"}]},"branch":"refs/heads/master"},"313ef90b6ab99be8fe21c67bcf7733aac4980446":{"kind":"REWORK","_number":4,"created":"2019-01-30 22:03:12.000000000","uploader":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"ref":"refs/changes/01/630301/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/630301/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/630301/4"}}},"commit":{"parents":[{"commit":"e287f58fbbfb81f16270800581059be2083e547c","subject":"Merge \"Add role tests for system member role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e287f58fbbfb81f16270800581059be2083e547c"}]}],"author":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-11 15:41:50.000000000","tz":-480},"committer":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-30 22:02:52.000000000","tz":-480},"subject":"Converting the API tests to use flask\u0027s test_client","message":"Converting the API tests to use flask\u0027s test_client\n\nFor each API unit test class, the methods are converted to use the\nFlask app test_client context manager.\n\nChange-Id: I2f1a2de9d0974d43dcadb58f6623d15e4a0740c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/313ef90b6ab99be8fe21c67bcf7733aac4980446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/313ef90b6ab99be8fe21c67bcf7733aac4980446"}]},"branch":"refs/heads/master"},"b2c35f0e88db4fd5a96875605d5cc72a4c17f421":{"kind":"REWORK","_number":5,"created":"2019-01-31 21:40:43.000000000","uploader":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"ref":"refs/changes/01/630301/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/630301/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/630301/5"}}},"commit":{"parents":[{"commit":"e287f58fbbfb81f16270800581059be2083e547c","subject":"Merge \"Add role tests for system member role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e287f58fbbfb81f16270800581059be2083e547c"}]}],"author":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-11 15:41:50.000000000","tz":-480},"committer":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-31 21:40:25.000000000","tz":-480},"subject":"Converting the API tests to use flask\u0027s test_client","message":"Converting the API tests to use flask\u0027s test_client\n\nFor each API unit test class, the methods are converted to use the\nFlask app test_client context manager.\n\nChange-Id: I2f1a2de9d0974d43dcadb58f6623d15e4a0740c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b2c35f0e88db4fd5a96875605d5cc72a4c17f421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b2c35f0e88db4fd5a96875605d5cc72a4c17f421"}]},"branch":"refs/heads/master"},"0cd8e52a7302cf4f4365e5f631cc718d9eeb930b":{"kind":"REWORK","_number":6,"created":"2019-02-01 22:26:18.000000000","uploader":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"ref":"refs/changes/01/630301/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/630301/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/630301/6"}}},"commit":{"parents":[{"commit":"e287f58fbbfb81f16270800581059be2083e547c","subject":"Merge \"Add role tests for system member role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e287f58fbbfb81f16270800581059be2083e547c"}]}],"author":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-11 15:41:50.000000000","tz":-480},"committer":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-02-01 22:26:05.000000000","tz":-480},"subject":"Converting the API tests to use flask\u0027s test_client","message":"Converting the API tests to use flask\u0027s test_client\n\nFor each API unit test class, the methods are converted to use the\nFlask app test_client context manager.\n\nChange-Id: I2f1a2de9d0974d43dcadb58f6623d15e4a0740c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0cd8e52a7302cf4f4365e5f631cc718d9eeb930b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0cd8e52a7302cf4f4365e5f631cc718d9eeb930b"}]},"branch":"refs/heads/master"},"96eb890a086a62c2efd92e5b9dc585d167fb3361":{"kind":"REWORK","_number":7,"created":"2019-02-04 21:34:05.000000000","uploader":{"_account_id":29355,"name":"Islam Musleh","email":"islam.musleh@gmail.com","username":"imus"},"ref":"refs/changes/01/630301/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/630301/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/630301/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/630301/7"}}},"commit":{"parents":[{"commit":"e287f58fbbfb81f16270800581059be2083e547c","subject":"Merge \"Add role tests for system member role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e287f58fbbfb81f16270800581059be2083e547c"}]}],"author":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-01-11 15:41:50.000000000","tz":-480},"committer":{"name":"Islam Musleh","email":"islam.musleh@gmail.com","date":"2019-02-04 21:33:51.000000000","tz":-480},"subject":"Converting the API tests to use flask\u0027s test_client","message":"Converting the API tests to use flask\u0027s test_client\n\nFor each API unit test class, the methods are converted to use the\nFlask app test_client context manager.\n\nChange-Id: I2f1a2de9d0974d43dcadb58f6623d15e4a0740c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/96eb890a086a62c2efd92e5b9dc585d167fb3361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/96eb890a086a62c2efd92e5b9dc585d167fb3361"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
