)]}'
{"id":"openstack%2Fkeystone~659118","triplet_id":"openstack%2Fkeystone~master~I52c0290bbba0e8c5e9196a87d8123816b3a2de04","project":"openstack/keystone","branch":"master","topic":"docfix","hashtags":[],"change_id":"I52c0290bbba0e8c5e9196a87d8123816b3a2de04","subject":"Fix documentation typo","status":"MERGED","created":"2019-05-14 16:33:13.000000000","updated":"2019-05-14 23:43:58.000000000","submitted":"2019-05-14 23:43:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659118-1557877438890-cf8f5779","meta_rev_id":"26e73b0d2b76f7740dba2337bd62cc59132af389","_number":659118,"virtual_id_number":659118,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-05-14 23:43:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-05-14 16:59:32.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-05-14 17:30:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-14 19:54:32.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-14 19:54:32.000000000","_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-14 16:59:32.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-05-14 17:30:25.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-05-14 19:54:32.000000000","updated_by":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"reviewer":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"state":"REVIEWER"},{"updated":"2019-05-14 23:43:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"652bb7163af27a89b542d915f17161a441ea3403","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-05-14 16:33:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"651a9e598a5c155a3c6a3610f2c514e560c2fe5d","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-14 16:59:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a9cd60d29aa96dc0860f44a205016f8187be08de","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-05-14 17:30:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a7e5728a7097f5bf3eb37f78e3b599bd220d72dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 19:28:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/18/659118/1/check/openstack-tox-cover/5e76724/cover/ : SUCCESS in 29m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/659118/1/check/openstack-tox-lower-constraints/d6726a3/ : SUCCESS in 30m 02s\n- openstack-tox-pep8 http://logs.openstack.org/18/659118/1/check/openstack-tox-pep8/09c0b7b/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/18/659118/1/check/openstack-tox-py27/4e35cc9/ : SUCCESS in 16m 25s\n- openstack-tox-py36 http://logs.openstack.org/18/659118/1/check/openstack-tox-py36/6fe957c/ : SUCCESS in 26m 01s\n- openstack-tox-py37 http://logs.openstack.org/18/659118/1/check/openstack-tox-py37/38111d4/ : SUCCESS in 22m 22s\n- openstack-tox-docs http://logs.openstack.org/18/659118/1/check/openstack-tox-docs/4fd4461/html/ : SUCCESS in 7m 40s\n- tempest-full http://logs.openstack.org/18/659118/1/check/tempest-full/2138a3c/ : SUCCESS in 1h 44m 44s\n- neutron-grenade http://logs.openstack.org/18/659118/1/check/neutron-grenade/1d3184f/ : SUCCESS in 53m 54s\n- grenade-py3 http://logs.openstack.org/18/659118/1/check/grenade-py3/92e0208/ : FAILURE in 33m 30s\n- tempest-full-py3 http://logs.openstack.org/18/659118/1/check/tempest-full-py3/2fc9f0b/ : SUCCESS in 1h 31m 34s\n- keystone-dsvm-functional http://logs.openstack.org/18/659118/1/check/keystone-dsvm-functional/9670a29/ : SUCCESS in 32m 20s\n- keystone-dsvm-py3-functional http://logs.openstack.org/18/659118/1/check/keystone-dsvm-py3-functional/34e80f2/ : SUCCESS in 28m 57s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/18/659118/1/check/keystone-dsvm-functional-federation-opensuse15/1bdea45/ : SUCCESS in 34m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/18/659118/1/check/keystone-dsvm-py3-functional-federation-opensuse15/41be0f8/ : SUCCESS in 48m 43s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/18/659118/1/check/keystoneclient-devstack-functional/b1c0d86/ : SUCCESS in 15m 26s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/18/659118/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/cb27c47/ : SUCCESS in 1h 27m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf55acefe33215fff060b45d2b452dbc398b7cf5","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-14 19:51:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9af694673c58c0ed0ab546d4d6bea406dc83aabb","author":{"_account_id":11022,"name":"Rodrigo Duarte Sousa","email":"rodrigodsousa@gmail.com","username":"rodrigods"},"date":"2019-05-14 19:54:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc455eb1fe039d75cad3b825eba22dc882b70d4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 21:49:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/18/659118/1/check/openstack-tox-cover/bb16038/cover/ : SUCCESS in 32m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/18/659118/1/check/openstack-tox-lower-constraints/922719c/ : SUCCESS in 28m 29s\n- openstack-tox-pep8 http://logs.openstack.org/18/659118/1/check/openstack-tox-pep8/a7aa7b7/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/18/659118/1/check/openstack-tox-py27/b77ef75/ : SUCCESS in 22m 30s\n- openstack-tox-py36 http://logs.openstack.org/18/659118/1/check/openstack-tox-py36/61e835b/ : SUCCESS in 20m 51s\n- openstack-tox-py37 http://logs.openstack.org/18/659118/1/check/openstack-tox-py37/2c649b6/ : SUCCESS in 24m 03s\n- openstack-tox-docs http://logs.openstack.org/18/659118/1/check/openstack-tox-docs/72dd081/html/ : SUCCESS in 7m 23s\n- tempest-full http://logs.openstack.org/18/659118/1/check/tempest-full/a48aba5/ : SUCCESS in 1h 34m 54s\n- neutron-grenade http://logs.openstack.org/18/659118/1/check/neutron-grenade/f5ad7a2/ : SUCCESS in 56m 48s\n- grenade-py3 http://logs.openstack.org/18/659118/1/check/grenade-py3/5e479af/ : SUCCESS in 1h 00m 58s\n- tempest-full-py3 http://logs.openstack.org/18/659118/1/check/tempest-full-py3/22e6b2e/ : SUCCESS in 1h 28m 23s\n- keystone-dsvm-functional http://logs.openstack.org/18/659118/1/check/keystone-dsvm-functional/5d8f17c/ : SUCCESS in 34m 16s\n- keystone-dsvm-py3-functional http://logs.openstack.org/18/659118/1/check/keystone-dsvm-py3-functional/225f5a4/ : SUCCESS in 30m 40s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/18/659118/1/check/keystone-dsvm-functional-federation-opensuse15/3ed045c/ : SUCCESS in 37m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/18/659118/1/check/keystone-dsvm-py3-functional-federation-opensuse15/c67799e/ : SUCCESS in 39m 56s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/18/659118/1/check/keystoneclient-devstack-functional/03158aa/ : SUCCESS in 19m 35s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/18/659118/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/1674c19/ : SUCCESS in 1h 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4e95f5c3e3f9102d5b71dfee02a0bd912bb7dba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 21:49:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"717cc825ac7cb95d7e4a6846a4ebef6e343d060e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 23:43:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6e5c2bf99ee1b5a743930ee61a7230f9e47b0f23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-14 23:43:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/18/659118/1/gate/openstack-tox-lower-constraints/f39b18e/ : SUCCESS in 30m 49s\n- openstack-tox-pep8 http://logs.openstack.org/18/659118/1/gate/openstack-tox-pep8/b23733e/ : SUCCESS in 5m 30s\n- openstack-tox-py27 http://logs.openstack.org/18/659118/1/gate/openstack-tox-py27/947080f/ : SUCCESS in 24m 03s\n- openstack-tox-py36 http://logs.openstack.org/18/659118/1/gate/openstack-tox-py36/395f4c8/ : SUCCESS in 23m 39s\n- openstack-tox-py37 http://logs.openstack.org/18/659118/1/gate/openstack-tox-py37/c279963/ : SUCCESS in 27m 28s\n- openstack-tox-docs http://logs.openstack.org/18/659118/1/gate/openstack-tox-docs/028197d/html/ : SUCCESS in 7m 26s\n- tempest-full http://logs.openstack.org/18/659118/1/gate/tempest-full/7747a92/ : SUCCESS in 1h 48m 43s\n- neutron-grenade http://logs.openstack.org/18/659118/1/gate/neutron-grenade/e755598/ : SUCCESS in 46m 51s\n- grenade-py3 http://logs.openstack.org/18/659118/1/gate/grenade-py3/9bd9ab2/ : SUCCESS in 51m 39s\n- tempest-full-py3 http://logs.openstack.org/18/659118/1/gate/tempest-full-py3/128ce5d/ : SUCCESS in 1h 32m 28s\n- keystone-dsvm-functional http://logs.openstack.org/18/659118/1/gate/keystone-dsvm-functional/ec2a9ee/ : SUCCESS in 32m 21s\n- keystone-dsvm-py3-functional http://logs.openstack.org/18/659118/1/gate/keystone-dsvm-py3-functional/137b2a6/ : SUCCESS in 30m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c5d71cce917b09ae6583be90c6441e538ee8e44","revisions":{"5c5d71cce917b09ae6583be90c6441e538ee8e44":{"kind":"REWORK","_number":1,"created":"2019-05-14 16:33:13.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/18/659118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/18/659118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/18/659118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/18/659118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/18/659118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/18/659118/1"}}},"commit":{"parents":[{"commit":"d775a0448847322213662e5d4b341c6694b19479","subject":"Merge \"Update Python 3 test runtimes for Train\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d775a0448847322213662e5d4b341c6694b19479"}]}],"author":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2019-05-14 16:32:11.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"douglas@redrobot.io","date":"2019-05-14 16:33:12.000000000","tz":-300},"subject":"Fix documentation typo","message":"Fix documentation typo\n\nThe correct section for configuring token provider is\n[token] (singluar) not [tokens].\n\nChange-Id: I52c0290bbba0e8c5e9196a87d8123816b3a2de04\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5c5d71cce917b09ae6583be90c6441e538ee8e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5c5d71cce917b09ae6583be90c6441e538ee8e44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
