)]}'
{"id":"openstack%2Fkeystone~661790","triplet_id":"openstack%2Fkeystone~master~Ibb48983f7edef1c2df50ba998f6a95741f933168","project":"openstack/keystone","branch":"master","topic":"bp/whitelist-extension-for-app-creds","hashtags":[],"change_id":"Ibb48983f7edef1c2df50ba998f6a95741f933168","subject":"Revert \"Add API for /v3/access_rules_config\"","status":"MERGED","created":"2019-05-28 15:39:02.000000000","updated":"2019-06-26 17:45:08.000000000","submitted":"2019-06-26 17:45:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661790-1561571108046-de56cdb0","meta_rev_id":"ba4cafa10a755e5bec7772ddc9e7c32578e5f86d","_number":661790,"virtual_id_number":661790,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-28 19:30:59.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2019-06-26 17:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-05-29 05:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-06-26 15:11:55.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-05-29 15:17:45.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-03 09:20:14.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2019-06-26 15:11:55.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 05:28:01.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-05-29 15:17:45.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-06-03 09:20:14.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-06-26 15:11:55.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-06-26 17:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"648075a27fd9dec0722a2ab52cd2b71b5c33d04e","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-28 15:39:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be72679c36c7d6a610f52bfeb6662e006b4d3a7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 18:47:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/90/661790/1/check/openstack-tox-cover/45cccbe/cover/ : SUCCESS in 28m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/661790/1/check/openstack-tox-lower-constraints/592be4c/ : SUCCESS in 29m 01s\n- openstack-tox-pep8 http://logs.openstack.org/90/661790/1/check/openstack-tox-pep8/19ace45/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/90/661790/1/check/openstack-tox-py27/ecfea8a/ : SUCCESS in 22m 07s\n- openstack-tox-py36 http://logs.openstack.org/90/661790/1/check/openstack-tox-py36/16ce257/ : SUCCESS in 27m 20s\n- openstack-tox-py37 http://logs.openstack.org/90/661790/1/check/openstack-tox-py37/8b928d8/ : SUCCESS in 25m 43s\n- openstack-tox-docs http://logs.openstack.org/90/661790/1/check/openstack-tox-docs/8f3143b/html/ : SUCCESS in 7m 03s\n- tempest-full http://logs.openstack.org/90/661790/1/check/tempest-full/2d0f760/ : TIMED_OUT in 2h 10m 20s\n- neutron-grenade http://logs.openstack.org/90/661790/1/check/neutron-grenade/bdbd85d/ : SUCCESS in 55m 57s\n- grenade-py3 http://logs.openstack.org/90/661790/1/check/grenade-py3/505cdb7/ : SUCCESS in 1h 04m 52s\n- tempest-full-py3 http://logs.openstack.org/90/661790/1/check/tempest-full-py3/a25f206/ : SUCCESS in 1h 38m 24s\n- keystone-dsvm-functional http://logs.openstack.org/90/661790/1/check/keystone-dsvm-functional/cb110e5/ : SUCCESS in 34m 45s\n- keystone-dsvm-py3-functional http://logs.openstack.org/90/661790/1/check/keystone-dsvm-py3-functional/129cfbc/ : SUCCESS in 35m 19s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/90/661790/1/check/keystone-dsvm-functional-federation-opensuse15/b43ec3f/ : SUCCESS in 34m 34s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/90/661790/1/check/keystone-dsvm-py3-functional-federation-opensuse15/137a3ed/ : SUCCESS in 32m 05s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/90/661790/1/check/keystoneclient-devstack-functional/ee8f889/ : SUCCESS in 15m 33s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/90/661790/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/8defa02/ : SUCCESS in 1h 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0e047a779c95ce159f49ed8eb61b44bbe421ecb","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-28 19:30:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0ad0afe63693aa1a1a337b9a0ddcd28edf206d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 21:20:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/90/661790/1/check/openstack-tox-cover/588a6f3/cover/ : SUCCESS in 28m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/661790/1/check/openstack-tox-lower-constraints/a3274eb/ : SUCCESS in 38m 49s\n- openstack-tox-pep8 http://logs.openstack.org/90/661790/1/check/openstack-tox-pep8/cb49871/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/90/661790/1/check/openstack-tox-py27/8d71280/ : SUCCESS in 20m 24s\n- openstack-tox-py36 http://logs.openstack.org/90/661790/1/check/openstack-tox-py36/d1bacb0/ : SUCCESS in 26m 23s\n- openstack-tox-py37 http://logs.openstack.org/90/661790/1/check/openstack-tox-py37/12c2c49/ : SUCCESS in 21m 00s\n- openstack-tox-docs http://logs.openstack.org/90/661790/1/check/openstack-tox-docs/41fb145/html/ : SUCCESS in 7m 31s\n- tempest-full http://logs.openstack.org/90/661790/1/check/tempest-full/1280215/ : SUCCESS in 1h 41m 42s\n- neutron-grenade http://logs.openstack.org/90/661790/1/check/neutron-grenade/66bf583/ : SUCCESS in 59m 01s\n- grenade-py3 http://logs.openstack.org/90/661790/1/check/grenade-py3/a6e8184/ : SUCCESS in 1h 01m 40s\n- tempest-full-py3 http://logs.openstack.org/90/661790/1/check/tempest-full-py3/d675458/ : SUCCESS in 1h 35m 44s\n- keystone-dsvm-functional http://logs.openstack.org/90/661790/1/check/keystone-dsvm-functional/2c0be91/ : SUCCESS in 59m 21s\n- keystone-dsvm-py3-functional http://logs.openstack.org/90/661790/1/check/keystone-dsvm-py3-functional/25b8c8f/ : SUCCESS in 33m 20s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/90/661790/1/check/keystone-dsvm-functional-federation-opensuse15/1525267/ : SUCCESS in 37m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/90/661790/1/check/keystone-dsvm-py3-functional-federation-opensuse15/0b6372f/ : SUCCESS in 45m 05s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/90/661790/1/check/keystoneclient-devstack-functional/ca21695/ : SUCCESS in 18m 15s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/90/661790/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/f905574/ : FAILURE in 1h 25m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5813e53a036a52a9f6143be4869bdebce08045f6","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-05-29 05:28:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"10f62a95c8db8cbc4e284886c58671ef270bbeae","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-05-29 15:17:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c273f9d4ceb545e7f8b7390e84f83e0ebbf47993","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2019-06-03 09:20:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1435f0225e543130e49637f8bdc785b90f8d06b","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-25 16:14:26.000000000","message":"Topic changed from undo-access-rules-config to bp/whitelist-extension-for-app-creds","accounts_in_message":[],"_revision_number":1},{"id":"71d7cce78bb89ae70bee9c72e14c59d5eddc95b1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 15:11:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a05e9ff1f1a87ac300abc8a6e00fcff96e15d46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 15:12:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1960b9ec98b8bdc640d4da4630257a1d5ee8bfae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:45:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/90/661790/1/gate/openstack-tox-lower-constraints/d9a9b82/ : SUCCESS in 32m 17s\n- openstack-tox-pep8 http://logs.openstack.org/90/661790/1/gate/openstack-tox-pep8/6473144/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/90/661790/1/gate/openstack-tox-py27/816f6e0/ : SUCCESS in 19m 39s\n- openstack-tox-py36 http://logs.openstack.org/90/661790/1/gate/openstack-tox-py36/e622aa0/ : SUCCESS in 26m 58s\n- openstack-tox-py37 http://logs.openstack.org/90/661790/1/gate/openstack-tox-py37/493573c/ : SUCCESS in 23m 54s\n- openstack-tox-docs http://logs.openstack.org/90/661790/1/gate/openstack-tox-docs/26a62f6/html/ : SUCCESS in 7m 45s\n- tempest-full http://logs.openstack.org/90/661790/1/gate/tempest-full/6ae9eba/ : SUCCESS in 1h 42m 04s\n- neutron-grenade http://logs.openstack.org/90/661790/1/gate/neutron-grenade/f20a386/ : SUCCESS in 58m 37s\n- grenade-py3 http://logs.openstack.org/90/661790/1/gate/grenade-py3/2b82377/ : SUCCESS in 59m 43s\n- tempest-full-py3 http://logs.openstack.org/90/661790/1/gate/tempest-full-py3/3f23ac7/ : SUCCESS in 1h 28m 41s\n- keystone-dsvm-functional http://logs.openstack.org/90/661790/1/gate/keystone-dsvm-functional/9f749df/ : SUCCESS in 33m 30s\n- keystone-dsvm-py3-functional http://logs.openstack.org/90/661790/1/gate/keystone-dsvm-py3-functional/f458d81/ : SUCCESS in 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ba4cafa10a755e5bec7772ddc9e7c32578e5f86d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:45:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7a2dd48e4636462d686023c9ee8cedfbbec9838","revisions":{"d7a2dd48e4636462d686023c9ee8cedfbbec9838":{"kind":"REWORK","_number":1,"created":"2019-05-28 15:39:02.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/90/661790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/90/661790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/661790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/661790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/90/661790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/90/661790/1"}}},"commit":{"parents":[{"commit":"d2b5550214cecd8a753253066d61e3419e76e952","subject":"Merge \"Remove [token]/ infer_roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d2b5550214cecd8a753253066d61e3419e76e952"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:28:54.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:38:39.000000000","tz":-420},"subject":"Revert \"Add API for /v3/access_rules_config\"","message":"Revert \"Add API for /v3/access_rules_config\"\n\nThis reverts commit 8d31705806721242db6dab34b2c293e317753ea6.\n\nIn the Train PTG[1] we agreed to defer this feature until we had some\nkind of traceability or discoverability for APIs and that this wasn\u0027t\nfeasible or useful until then.\n\nThis change was merged to master but never released, so I submit that\nit is safe to revert.\n\n[1] https://etherpad.openstack.org/p/keystone-train-ptg-application-credentials\n\nChange-Id: Ibb48983f7edef1c2df50ba998f6a95741f933168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d7a2dd48e4636462d686023c9ee8cedfbbec9838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d7a2dd48e4636462d686023c9ee8cedfbbec9838"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
