)]}'
{"id":"openstack%2Fkeystone~661791","triplet_id":"openstack%2Fkeystone~master~I65ac52b8730221562391adc8b0dbccd22ea79b16","project":"openstack/keystone","branch":"master","topic":"bp/whitelist-extension-for-app-creds","hashtags":[],"change_id":"I65ac52b8730221562391adc8b0dbccd22ea79b16","subject":"Revert \"Add manager support for app cred access rules\"","status":"MERGED","created":"2019-05-28 15:39:02.000000000","updated":"2019-06-26 17:45:10.000000000","submitted":"2019-06-26 17:45:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661791-1561571110233-d253eb7f","meta_rev_id":"af30795332198c972ca9b8153c3982502f079ba8","_number":661791,"virtual_id_number":661791,"owner":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2019-06-26 17:45:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":1,"date":"2019-05-28 16:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-06-26 14:58:16.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-05-29 18:04:50.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-03 09:20:20.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":1,"date":"2019-06-26 15:11:58.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-28 16:58:21.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-05-29 18:04:50.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-06-03 09:20:20.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-06-26 15:11:58.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-06-26 17:45:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b3d81e564e52f45d17c2bdc5091fd95cde7bf45","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-05-28 15:39:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"285d74b712665796babd21db311c20357d38d039","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-05-28 16:58:21.000000000","message":"Patch Set 1: Code-Review+1\n\nAgreed","accounts_in_message":[],"_revision_number":1},{"id":"f35eaccb389e78468ca763e5e9228507b0a964a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 18:48:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/91/661791/1/check/openstack-tox-cover/a3a9f3e/cover/ : SUCCESS in 35m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/661791/1/check/openstack-tox-lower-constraints/b06e650/ : SUCCESS in 25m 48s\n- openstack-tox-pep8 http://logs.openstack.org/91/661791/1/check/openstack-tox-pep8/9f3e907/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/91/661791/1/check/openstack-tox-py27/3a58d7e/ : SUCCESS in 22m 07s\n- openstack-tox-py36 http://logs.openstack.org/91/661791/1/check/openstack-tox-py36/c14c34b/ : SUCCESS in 27m 58s\n- openstack-tox-py37 http://logs.openstack.org/91/661791/1/check/openstack-tox-py37/214c1eb/ : SUCCESS in 22m 35s\n- openstack-tox-docs http://logs.openstack.org/91/661791/1/check/openstack-tox-docs/c2df4c8/html/ : SUCCESS in 9m 25s\n- tempest-full http://logs.openstack.org/91/661791/1/check/tempest-full/97ffaa4/ : SUCCESS in 1h 44m 00s\n- neutron-grenade http://logs.openstack.org/91/661791/1/check/neutron-grenade/0000e8e/ : SUCCESS in 1h 11m 36s\n- grenade-py3 http://logs.openstack.org/91/661791/1/check/grenade-py3/6dd8ac0/ : SUCCESS in 1h 13m 42s\n- tempest-full-py3 http://logs.openstack.org/91/661791/1/check/tempest-full-py3/ce33c73/ : SUCCESS in 1h 44m 42s\n- keystone-dsvm-functional http://logs.openstack.org/91/661791/1/check/keystone-dsvm-functional/c764627/ : SUCCESS in 34m 19s\n- keystone-dsvm-py3-functional http://logs.openstack.org/91/661791/1/check/keystone-dsvm-py3-functional/d3f8b38/ : SUCCESS in 35m 08s\n- keystone-dsvm-functional-federation-opensuse15 http://logs.openstack.org/91/661791/1/check/keystone-dsvm-functional-federation-opensuse15/4e13a09/ : SUCCESS in 37m 18s (non-voting)\n- keystone-dsvm-py3-functional-federation-opensuse15 http://logs.openstack.org/91/661791/1/check/keystone-dsvm-py3-functional-federation-opensuse15/cbae44b/ : SUCCESS in 34m 49s (non-voting)\n- keystoneclient-devstack-functional http://logs.openstack.org/91/661791/1/check/keystoneclient-devstack-functional/c768089/ : SUCCESS in 16m 51s (non-voting)\n- legacy-tempest-dsvm-ldap-domain-specific-driver http://logs.openstack.org/91/661791/1/check/legacy-tempest-dsvm-ldap-domain-specific-driver/4e4fbdd/ : SUCCESS in 1h 46m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08fab67b61ef457c059436d25e387de024405d58","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-05-29 18:04:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"adc7d2f2c250b23ef2b9b555b6085c361b4179d0","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2019-06-03 09:20:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef08738e9ea3dfb91a4cf63ce0b433b64d79db78","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-25 16:14:16.000000000","message":"Topic changed from undo-access-rules-config to bp/whitelist-extension-for-app-creds","accounts_in_message":[],"_revision_number":1},{"id":"e6a1a08dddae5b827d65325b94ff74b7355e63d0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 14:58:16.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good, just waiting on one question in IRC before approving.","accounts_in_message":[],"_revision_number":1},{"id":"dcade0793158d005e036cb29ae9831edadc52b85","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 15:11:48.000000000","message":"Patch Set 1:\n\n\u003e Looks good, just waiting on one question in IRC before approving.\n\nColleen answered my questions in IRC [0].\n\n[0] http://eavesdrop.openstack.org/irclogs/%23openstack-keystone/%23openstack-keystone.2019-06-26.log.html#t2019-06-26T14:49:45","accounts_in_message":[],"_revision_number":1},{"id":"b43249c5cb758511bcfdf0584a3ca5d7b77531b0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 15:11:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e47f64de13b4a47ce1d05f32d7d17e40c780101","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 15:12:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"928afda00b81f9f837b83b32396d8b3903f74b10","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:45:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/91/661791/1/gate/openstack-tox-lower-constraints/79294da/ : SUCCESS in 30m 22s\n- openstack-tox-pep8 http://logs.openstack.org/91/661791/1/gate/openstack-tox-pep8/2bee9d7/ : SUCCESS in 6m 03s\n- openstack-tox-py27 http://logs.openstack.org/91/661791/1/gate/openstack-tox-py27/75dce06/ : SUCCESS in 31m 48s\n- openstack-tox-py36 http://logs.openstack.org/91/661791/1/gate/openstack-tox-py36/17a5618/ : SUCCESS in 26m 01s\n- openstack-tox-py37 http://logs.openstack.org/91/661791/1/gate/openstack-tox-py37/6d3cef6/ : SUCCESS in 24m 38s\n- openstack-tox-docs http://logs.openstack.org/91/661791/1/gate/openstack-tox-docs/a2933a8/html/ : SUCCESS in 9m 26s\n- tempest-full http://logs.openstack.org/91/661791/1/gate/tempest-full/95a9e15/ : SUCCESS in 1h 46m 34s\n- neutron-grenade http://logs.openstack.org/91/661791/1/gate/neutron-grenade/cc961f5/ : SUCCESS in 1h 04m 15s\n- grenade-py3 http://logs.openstack.org/91/661791/1/gate/grenade-py3/7894931/ : SUCCESS in 56m 00s\n- tempest-full-py3 http://logs.openstack.org/91/661791/1/gate/tempest-full-py3/1ac8d86/ : SUCCESS in 1h 26m 50s\n- keystone-dsvm-functional http://logs.openstack.org/91/661791/1/gate/keystone-dsvm-functional/fe820e0/ : SUCCESS in 36m 57s\n- keystone-dsvm-py3-functional http://logs.openstack.org/91/661791/1/gate/keystone-dsvm-py3-functional/b75d738/ : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":1},{"id":"af30795332198c972ca9b8153c3982502f079ba8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:45:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25b2f151a7d19d7725425ba76a8c8883257afb0e","revisions":{"25b2f151a7d19d7725425ba76a8c8883257afb0e":{"kind":"REWORK","_number":1,"created":"2019-05-28 15:39:02.000000000","uploader":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"ref":"refs/changes/91/661791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/91/661791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/661791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/661791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/661791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/91/661791/1"}}},"commit":{"parents":[{"commit":"d7a2dd48e4636462d686023c9ee8cedfbbec9838","subject":"Revert \"Add API for /v3/access_rules_config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d7a2dd48e4636462d686023c9ee8cedfbbec9838"}]}],"author":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:32:52.000000000","tz":-420},"committer":{"name":"Colleen Murphy","email":"colleen@gazlene.net","date":"2019-05-28 15:38:40.000000000","tz":-420},"subject":"Revert \"Add manager support for app cred access rules\"","message":"Revert \"Add manager support for app cred access rules\"\n\nThis reverts commit 37fc2b9120538e001daba7093d0295e5f65b811d.\n\nIn the Train PTG[1] we agreed to defer the access rules config part of\nthis feature until we had some kind of traceability or discoverability\nfor APIs. For simplicity of review, this patch reverts the access rules\naddition to the app cred manager so that we can reimplement it in a way\nthat doesn\u0027t require using the access_rules_config API provider.\n\n[1] https://etherpad.openstack.org/p/keystone-train-ptg-application-credentials\n\nChange-Id: I65ac52b8730221562391adc8b0dbccd22ea79b16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/25b2f151a7d19d7725425ba76a8c8883257afb0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/25b2f151a7d19d7725425ba76a8c8883257afb0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
